)]}'
{"id":"openstack%2Fkayobe~920126","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~Ibef1713d7de2e0a2c6fe046f607b2a8e60dae3da","project":"openstack/kayobe","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-05-30 11:45:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibef1713d7de2e0a2c6fe046f607b2a8e60dae3da","subject":"Fix issue removing docker volumes","status":"MERGED","created":"2024-05-22 12:08:21.000000000","updated":"2024-05-30 11:46:08.000000000","submitted":"2024-05-30 11:45:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920126","cherry_pick_of_change":909594,"cherry_pick_of_patch_set":7,"meta_rev_id":"16f50def6bff3fdeb570b54866decc614d92977f","_number":920126,"virtual_id_number":920126,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-30 11:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-05-30 09:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-05-24 10:29:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-24 07:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-05-30 09:13:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-22 12:08:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-05-22 14:34:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-24 10:29:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-05-30 09:13:31.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"e0abdd28a13bed476e1a1b139f66e7e5fc060ba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-22 12:08:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"89c523963cea7f987cefd53c636c3691b7d45f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 14:34:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9260e3f548e94584b26e6aa1bff8eeda\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1605eb116fe34e53b87a8417530bc945 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62ae92a081446999082ad17c6ea2a11 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccb2781d127145ddb5dc1435013e1d1c : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8007d9c1ae9844d18e726073c6d9a1ae : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1d890fef058f468a9398bf614192ad1c : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5828706a7d7346e0bb47a25f8545d065 : SUCCESS in 9m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/99a29bb7abda4d8f97a053b3f57b3c27 : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a49d9f95bb594838b4ff0498e08b6c86 : SUCCESS in 45m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fb6aa51cbed4f59a54eee736b2169a7 : SUCCESS in 57m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/68f24eb14c124cb887f0ba4be757f515 : SUCCESS in 1h 11m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/77fdce96309f4d0e8b659e78021308e0 : SUCCESS in 15m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8f2c3f4b52f64d0f900e1d19c87c6538 : SUCCESS in 17m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7fac5e467da41a08c27c6c44c7f551c : SUCCESS in 18m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3d06946eb5fc4b1e832e0781d3722cef : SUCCESS in 2h 23m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/579087afbd664e18b42ff9349d8048d8 : SUCCESS in 2h 19m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bf56496178fc4b1584b8edc57c0c2ff9 : SUCCESS in 17m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acc29b0f57884a9d88941fe433c95fce : SUCCESS in 14m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c419bf2e8d4746538060c0d6d8870be5 : SUCCESS in 41m 05s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/da233fe9742e443b9e3f18c48869a558 : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f5d248a40dc4267907915e7303b4c7a : SUCCESS in 28m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4ee51f11cace44c7abece72c099ce2ab : SUCCESS in 13m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e86e78e27316453c9fa7b697ba0cca62 : SUCCESS in 13m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4e4d80e45a244afa9e32382d2ffff99b : SUCCESS in 17m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d75aee49cd5458bb20394ccf16fe46c : SUCCESS in 14m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be27424cfe41433b8431d8435d3423b2 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0431329cfc9d40c6995e7d52cbc6199f : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ff20ef805e140b1a2df438bc6c34c77 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6bca7970e062833300e38be191b8c041db2f0a6e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-24 07:51:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0027765a441f2408f51dcfc5d0af463ce450f642","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-24 10:29:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"da2abe15029276fba17943e41c1add445bc14794","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-30 09:13:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35cae37fc05e6f2e1474bc88bb944bfb5519af1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 09:13:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92813597427ec6a33367cbb5c01da3e152dd1e8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 11:45:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb0157acc2944e0870038ce3b428ac5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef4eea067243421cbfdd5ed3245a735c : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6784844c0f74866a06b52229fb9fe29 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a61484a9ef3d474c8af7964245ea5f95 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/283cd529544342599a80ef6b8ff14fa6 : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c7a727cb7d1142ccb793002ac43efc2c : SUCCESS in 9m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/eecad9c654ee4f15b614e83e8398d6f7 : SUCCESS in 56m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ea25db2816d44b5ab57194c366cb7fa : SUCCESS in 42m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ccb4eac7178b4a81a2c515c544aba4b5 : SUCCESS in 1h 00m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/be5d2536a8154a6394a6f48b0a0a8343 : SUCCESS in 16m 27s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e0dcfea375e4d84be7862fd8340524d : SUCCESS in 12m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7588d08da59f4e80a5a70123e544decc : SUCCESS in 2h 26m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18a18125d03748968c6ff7681c2bd790 : SUCCESS in 2h 01m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/df5891d13a9448a8b956d460b140ca39 : SUCCESS in 16m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3ec76b575b44172a24320dd6227f863 : SUCCESS in 14m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/020cf35251b341baa0bed04130d23134 : SUCCESS in 28m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab89bf4617a8421f914a184e2efdf2ce : SUCCESS in 22m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/69a205cec6ea4cfeb071a65b729d3075 : SUCCESS in 15m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1f463fd23a446b3828864f5a8ac906e : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a8b317e3729e47e089cde2c842c741df : SUCCESS in 20m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc1c1976746148bea5b42ed8e67905cb : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0d1d87112d3ffb668ef2b064dab91f1b9058b6f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 11:45:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16f50def6bff3fdeb570b54866decc614d92977f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 11:46:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95b6cadc6ede4094919af76c6a3e25bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeabd65e26594e398a75c41d842144b4 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/792b1e8f264b46309b57b25004782bc8 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16147f173ffe253acf6392c8f674f716434cb07d","revisions":{"16147f173ffe253acf6392c8f674f716434cb07d":{"kind":"REWORK","_number":1,"created":"2024-05-22 12:08:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/920126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/920126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/920126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/920126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/920126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/920126/1"}}},"commit":{"parents":[{"commit":"d7f5f8ac3a9cb5857616385a5b0a7ae0b173c87a","subject":"Merge \"Add python3-devel to Kayobe dependencies\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7f5f8ac3a9cb5857616385a5b0a7ae0b173c87a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-20 15:30:56.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-05-22 12:08:21.000000000","tz":0},"subject":"Fix issue removing docker volumes","message":"Fix issue removing docker volumes\n\nThe error response from the docker daemon has changed to:\n\n    Error response from daemon: get opensm_logs: no such volume\n\nThis change should support both old and new responses.\n\nChange-Id: Ibef1713d7de2e0a2c6fe046f607b2a8e60dae3da\nCloses-Bug: #2050092\n(cherry picked from commit 27ece2622b934eb6f4d9712d1879ac9bab8d236e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16147f173ffe253acf6392c8f674f716434cb07d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16147f173ffe253acf6392c8f674f716434cb07d"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
