)]}'
{"id":"openstack%2Fkayobe~920375","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I8c9ce97badc3add0b39c5ef6948e7b8eebeeda25","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-05-24 12:03:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c9ce97badc3add0b39c5ef6948e7b8eebeeda25","subject":"Fix permissions on dev/rabbitmq-migrate-queues.sh","status":"MERGED","created":"2024-05-24 08:02:17.000000000","updated":"2024-05-24 12:04:25.000000000","submitted":"2024-05-24 12:03:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920375","meta_rev_id":"e881408d55a7cc9048fbc9e8a977ce5abaf779ff","_number":920375,"virtual_id_number":920375,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-24 12:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-05-24 08:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-24 08:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-24 08:24:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-24 08:04:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-05-24 08:24:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-05-24 09:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f526815958c9c6df35d56ddb0fdfd12888bc804d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-05-24 08:02:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b829889ebc58049c74affacc311bb903b37c7e0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-24 08:04:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5078d21e21d53208ce1ac19ab5fc1a20e284cd71","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-24 08:24:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e21635ce847c761d7f3a70095afd04c6a195ec20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 09:55:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a16870ee6d554dccaececef4e30aaf17\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8c65e9b7d7c4f7fae4ffa5acbbb74dc : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235a3f9ffc8944c39da732321ba9681f : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f30697b1772b45db938025ab3c834426 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2457525264454b91bfe441860975866d : SUCCESS in 6m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f387d1a289f41ce91f9667c28cfc871 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6aed84a052b4c91a0575129a010d1ba : SUCCESS in 14m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91c0cb8308a54352a3df217485692df4 : SUCCESS in 14m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ee80a492b19a47838ea667ea7da4c729 : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/dfa717c07bc04ff793eb9ec2ff3a149d : SUCCESS in 1h 04m 47s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7626a9c7f0ec4000b238390e5458dd25 : SUCCESS in 55m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/778ba9cf3f0342eca23a8033db68148b : SUCCESS in 1h 14m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3d39691733544f748c64e3f59bd834f9 : SUCCESS in 19m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a76bff14f2cc438f965b1ee14efd2014 : SUCCESS in 16m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91d0e0e1931b4484a889888e456a03bb : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b58e37fcc9ae499f90025d9b7e8787f9 : SUCCESS in 1h 46m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2d3ba20265c453282a31c34e33fe9e5 : SUCCESS in 1h 38m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5dad232f48d946a3b8bcca60609f6518 : SUCCESS in 19m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4620637dc2c4e3ab1b1f5213b1aadd3 : SUCCESS in 16m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fdc8bc329cfa4e279c7dd72ea044ccdb : SUCCESS in 35m 44s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0729e7e9d0f24f2d9866da1e99ab57f8 : SUCCESS in 33m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21b7e9b2c450413dafd13688907779dc : SUCCESS in 30m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9d8e2e8d3d01477c98fcfff2506b49e9 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37d18482ccb74921bad0696ec917e769 : SUCCESS in 9m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c97f98b83e4642cfafe27dcab181df18 : SUCCESS in 17m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/368002090eaa466d8815389844d74ec7 : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a1fc44cdefaf62e503f58f50526aabb5c46623c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 09:55:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1766b7a547a746dd4c5d8808c4320d1b8fe83a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:03:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa62e66847ed40aca8fd9a1a9a1efc21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56e2ca2d6204503ba706015aadc28d6 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79c7ccefeb6948748ed6e1c2375ba1f7 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca43533eeb04ea0b971295ae3e79a36 : SUCCESS in 7m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14e4baf4a6544fc684f9dfab4d3f8a2d : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f615344db7434eb3bb23bfb067242ac7 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6df9780708d4b88aeebc7c6cf86623d : SUCCESS in 14m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/791fc4c13c164a5485ea790922a24166 : SUCCESS in 1h 08m 46s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c71384d414b4fb4a3150a9a8aac6406 : SUCCESS in 53m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c1fc350cdfc04a22be842721255e74b6 : SUCCESS in 1h 11m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/45bbbefd98684190994dc04055e65ca6 : SUCCESS in 19m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/111e8886e600442b925fd22d4b00c13e : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/30f0ae5d486343b4b6382d1b5b0a1277 : SUCCESS in 2h 00m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e983a5e387034c828b5bc341f379b950 : SUCCESS in 1h 37m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ea6d6c3e29f7460f85ce09ae26745dd3 : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7409e0de7a4e472b8e19dc834667a475 : SUCCESS in 15m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1e545584981347c9b34409a530479e5b : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef32290f49cb4038b89a8fcd90ca53ff : SUCCESS in 21m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c799f63e1f2d4d77a5ad470f2a01eda4 : SUCCESS in 13m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2836af422bb94475bd1d1df7b4993928 : SUCCESS in 12m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5eca269c8a147b292a3bc5cd6c6f8a6 : SUCCESS in 17m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7c37a99088a4900a8d3e94671f9b920 : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":1},{"id":"00dd3372d1c00738d2d631b22bc146f019357679","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:03:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e881408d55a7cc9048fbc9e8a977ce5abaf779ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:04:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06cf77f2e02d495eadbee104788d33d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95c7fd3a44a4418a46624f69f01e2c9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c43b1152f62085bb0061752c6d8f4e4d21aaec6f","revisions":{"c43b1152f62085bb0061752c6d8f4e4d21aaec6f":{"kind":"REWORK","_number":1,"created":"2024-05-24 08:02:17.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/75/920375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/920375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/920375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/920375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/920375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/920375/1"}}},"commit":{"parents":[{"commit":"1edbcfd65d0e9634427f8a46d319305311473074","subject":"Merge \"Add script to migrate to RabbitMQ quorum queues\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1edbcfd65d0e9634427f8a46d319305311473074"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-24 08:01:13.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-24 08:01:13.000000000","tz":60},"subject":"Fix permissions on dev/rabbitmq-migrate-queues.sh","message":"Fix permissions on dev/rabbitmq-migrate-queues.sh\n\nChange-Id: I8c9ce97badc3add0b39c5ef6948e7b8eebeeda25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c43b1152f62085bb0061752c6d8f4e4d21aaec6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c43b1152f62085bb0061752c6d8f4e4d21aaec6f"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
