)]}'
{"id":"openstack%2Fkayobe~921005","triplet_id":"openstack%2Fkayobe~master~I84971f8daa4caf2227863810c7b12e59192c2515","project":"openstack/kayobe","branch":"master","topic":"remove-get-md5","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-06-03 16:24:00.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I84971f8daa4caf2227863810c7b12e59192c2515","subject":"Remove use of get_md5 when using stat module","status":"MERGED","created":"2024-06-03 10:21:32.000000000","updated":"2024-06-03 20:39:21.000000000","submitted":"2024-06-03 20:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921005-remove-get-md5","meta_rev_id":"6725bc3b3cdbcd489386b265429c87048459349c","_number":921005,"virtual_id_number":921005,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-03 20:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-06-03 10:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-06-03 10:31:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-03 10:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-03 10:31:50.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-06-03 12:11:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e0a0a8ddf1bea7e105e51c62a8abd5d7d122de2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 10:21:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a43ed0a07cd1c3f9cdebf546301776ca287d12f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 10:21:55.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"a6eb481fe1230aab8676112bfcac175938d13338","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 10:31:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f17066e2faf928668c88ef773808d7fc55de01d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 12:11:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c74032cf4e84e898408d041dfd4a84b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf7a01321e1043d48c66e8859d12a93c : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26235305c4144431a77c8d21292336a7 : SUCCESS in 7m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c3eaf0582be478db51d90d42eb723c5 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/793b1c47a0074884a34dadf29a5f68d7 : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c8b72d5c3f74a1695b7976a3478ba45 : FAILURE in 5m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4831423901da43e18375a7604577114f : SUCCESS in 59m 51s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8f4dcfbd1a2343849f9731d35892b962 : SUCCESS in 1h 07m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85cbf519e1d04e099d958e4b40d743ba : SUCCESS in 53m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a16e85639a3343d98f038578c8d9ab36 : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/77cbc0389d31402ab646a4eb9fe2f16f : SUCCESS in 20m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7152226397bc4593962018d50f7921e3 : SUCCESS in 16m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2659b0ae59947ddbe13238c6ec9f5e0 : SUCCESS in 12m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ff45e58c839f498a841aa00cc4f833ef : SUCCESS in 1h 44m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b77989ecc59474a9c83d553c076553b : SUCCESS in 1h 39m 49s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/917467b230a949ce831a2d88f5afef8d : FAILURE in 1h 20m 13s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c20c1ad1d0a042b0bf3d2c6b01c7d2a9 : FAILURE in 1h 01m 42s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7f2ab263711143948b355d024c48a38b : SUCCESS in 15m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abcbab27af664a94a16dae670c5f9027 : SUCCESS in 15m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1a348cbbff3546fe910b60a7d7f077f8 : SUCCESS in 30m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6281d3d038844b5a8c7633ff47cf1f5f : SUCCESS in 29m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/091df96186b84b768ca39fb232229002 : SUCCESS in 20m 06s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0eb5a610cb4642be9630894fe8556784 : SUCCESS in 31m 02s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5199d646b9c408d82698f0230930a32 : SUCCESS in 23m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/adb4731deedd4fd4b6f9fd764b751485 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05a6e66b8bdb4b6daf5ad9709514dd93 : SUCCESS in 10m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9bbd9f164f3d43e1816a601a81e834af : SUCCESS in 18m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b114478e4aef4908bade3fdbd1e27f35 : SUCCESS in 12m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c8d19d92a6b4f02aa228b67b0287dc4 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3014249f798e483d9b20798407f8a809 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e8e55689d3343d08f28d8cde17fe610 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f89236b3c471a98c581da71110aa9d9eb6cf9152","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 16:24:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87407f461e4cbe69f1c6bf31df280ea7bcb0b233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 18:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/771dd3f69e9b473484ae6023cda2a3ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0113f0059ded40ceaeb99d5447b33fdc : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9d1f1e044074edba856d4a46748bcdc : SUCCESS in 7m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6cc18633aecf41d089d5c3f8c999a1a4 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/24827655780144a6aa9b02b4fb4a0c36 : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be60ed6a52674a40b49da43ff12e2e5d : SUCCESS in 10m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/466903bb260748e089b162a10a5f6fd7 : SUCCESS in 56m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9c5b866199ac4c0886c745adffe2fff0 : SUCCESS in 53m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49daf3b4f33343708ef4aecd78c6ee8c : SUCCESS in 51m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/715b1326f15e4106909971a376bfe491 : SUCCESS in 1h 09m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f5d9b2378284442b9d5ad2610512d064 : SUCCESS in 16m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b9e09853b3b041caa6c2c9310e12d30e : SUCCESS in 15m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d48a5a96afc1422ca5c167482941740d : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5dc6f4a62f18479089056d6b1a57c814 : SUCCESS in 2h 01m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51a123fe7af944e1a273e1fcde5fbc27 : SUCCESS in 1h 41m 21s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a5aee4d9030545108aa71ab1db427886 : FAILURE in 1h 20m 17s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22e63523b1de4c68ba91df127f984cef : FAILURE in 1h 11m 30s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/594e093dd206445291f658caae272a3a : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be06f0d93f734d76af92b2c25316a88a : SUCCESS in 14m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/614ef3a8dfac4f02b888be6a5db51a42 : SUCCESS in 40m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/26012247dee74291850018bc2862a735 : SUCCESS in 30m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1c4f6d206c74717b7650a1a4f04cde5 : SUCCESS in 24m 27s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3744c24b87c24d9fbbc7eb4b1ea9dffe : SUCCESS in 35m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2daa6e3d4afd480880b7834b4369aee0 : SUCCESS in 27m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7bfeb5ef6c2a4672a63f65ef556c6948 : SUCCESS in 13m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7772dd288ff4fc1a90ac516f27dd2b1 : SUCCESS in 12m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dd962ac2a223462188fae909f2e876ab : SUCCESS in 17m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07f00f7c9a7a454a8cfcb4514991bbf4 : SUCCESS in 12m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bd5cf962b81466d9150fcbd1b4cd80a : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45e29d8bf8724df4b5b7b52e8fbbf0d2 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25f0119023124b0c9fbd7a1a84e5966a : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"909c54a9aceba7841ffc7f34ca738ac7f01a9c6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 18:37:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"455ecf43edef4985dc0f742e69373a48aab5f7d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 20:38:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9b6c864cfb483abd7637bb4b5fc2bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5b9b8f5ce144b40a44c72b5f6ed1b1b : SUCCESS in 7m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/306ad17cf7d940b086ccaa67bba461cb : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bb2102869dd149bb865f390b293ae401 : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02a0f0d811414afaaa0aafb985f1b1ee : SUCCESS in 9m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/804156e04a22457eb56f839d36bdd542 : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3ee1c5556dc4fdb9a0fba35be65abf7 : SUCCESS in 45m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/18cb6a98b5f24742ab0736f32546fb14 : SUCCESS in 1h 13m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3ef6ca985dc246cab878c532f64913d4 : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf92aa2c4c424ccaa2dca3c67203cfbe : SUCCESS in 11m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/10d03aa80e1c4d728b427b9d58b90f77 : SUCCESS in 1h 55m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/063b6aec03084f7a9ba633777741323d : SUCCESS in 1h 17m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f5da678bf6804292961c72f0d0aa3f1b : SUCCESS in 17m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31c975eb002c420fb5d5faa31f5b29d1 : SUCCESS in 13m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/92346159509c420db2e3fcdc1ee77a25 : SUCCESS in 28m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f05c51740d4143f3a5385a19a1a37f83 : SUCCESS in 23m 21s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c2118d8d2730444dbe8cd7a4c35ce3a4 : SUCCESS in 31m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4479ed123904f229353b8476df8ea17 : SUCCESS in 25m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/915e0e3c711e4a8da1ade69a366098bc : SUCCESS in 12m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a94296582fbe49a7b159336ee6971c69 : SUCCESS in 9m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2d67b6863fbc478cbf6a7e5538eb9ebd : SUCCESS in 15m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c0771ddde824793878dad0a3594eb22 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1ddf38f1f632caa75b7952bc92717d8a33a084d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 20:38:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6725bc3b3cdbcd489386b265429c87048459349c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 20:39:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b785d7ee10c5472180d4a3d0ce27ef12\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b89f88cec6aa43eb9e84c78fd06d57b2 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f39adcf328e609fbde80a7157035f8ead6db2b8","revisions":{"8f39adcf328e609fbde80a7157035f8ead6db2b8":{"kind":"REWORK","_number":1,"created":"2024-06-03 10:21:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/921005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/921005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/921005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/921005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/921005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/921005/1"}}},"commit":{"parents":[{"commit":"31728a180330e666bc4477262583f8ac9b757a15","subject":"Merge \"docs: fix link to global variable defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31728a180330e666bc4477262583f8ac9b757a15"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 10:18:16.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 10:18:16.000000000","tz":120},"subject":"Remove use of get_md5 when using stat module","message":"Remove use of get_md5 when using stat module\n\nThis was missed from I2ea9b2ec58913722c4defffbeee88cc420dcbdab.\n\nChange-Id: I84971f8daa4caf2227863810c7b12e59192c2515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f39adcf328e609fbde80a7157035f8ead6db2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f39adcf328e609fbde80a7157035f8ead6db2b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
