)]}'
{"id":"openstack%2Fkayobe~921045","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I8604c342e4dbc41dd5bee0f28a92c3c3cd37d631","project":"openstack/kayobe","branch":"stable/2023.2","topic":"molecule-c8s-fix-stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-06-04 10:20:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8604c342e4dbc41dd5bee0f28a92c3c3cd37d631","subject":"CI: Fix kayobe-tox-molecule job","status":"MERGED","created":"2024-06-03 16:17:48.000000000","updated":"2024-06-04 10:20:57.000000000","submitted":"2024-06-04 10:20:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921045-molecule-c8s-fix-stable/2023.2","cherry_pick_of_change":921013,"cherry_pick_of_patch_set":2,"meta_rev_id":"66f74b4be2306d1b7fa33c1578c8a894d7ebff62","_number":921045,"virtual_id_number":921045,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-04 10:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2024-06-04 07:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2024-06-04 07:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-03 17:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-03 19:24:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-04 07:47:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"91902ecc69a5b4687cd77f0fd449426e082f592a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 16:17:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"057baf8a9b1dea5affdd791770f55c1aedcd185d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 17:12:18.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"aff11d46ea7349bbb340e864a928debee5fd7cb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 19:24:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa5095821d24c239f111d69436a91f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59b0cd28db8246a2b5bb3811f9be9896 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e34eb4c2c7a4407daf61b1b77ee2d3b2 : SUCCESS in 9m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d769dec5b914ab4b4451431cdc3577e : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b10ab98ee144cbb991c5ce4bffc7672 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71f877fbe5334a648b214270763e43d9 : SUCCESS in 9m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3ac95633031b49b5b94080aeb8d0d265 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/11d6b0fdc2de4eb5aefde64c56240bb5 : SUCCESS in 59m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d0c7bcf0a2f485da071c8d7b1a2da1d : SUCCESS in 49m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c84087c8d54c404da058144bb4e2410a : SUCCESS in 1h 12m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ba75bbb9665344b49ce6a539b6e5114b : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/48cd69e5e11647178e2948551488cdd9 : SUCCESS in 15m 40s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9ad272c73ab41d98bc4883fb7923417 : SUCCESS in 16m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0de3c589a1a447348466ead16a28c18d : SUCCESS in 2h 57m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63822a7bd4d64b1b912b9963a7416fc2 : SUCCESS in 2h 05m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3f1120e4859146de925d000c76a30646 : SUCCESS in 17m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1b52cf714734e3293fa22a5645831ee : SUCCESS in 19m 00s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f0c5e7092cc74c6bac236a2babcecd20 : SUCCESS in 34m 19s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dd715b84882740d3b3ddd2e150240f65 : SUCCESS in 30m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95eaf77d20a54c488c61d678497bdafe : SUCCESS in 25m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dc6409875d09475b8ef46d696136eb10 : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b890422fbe74429c8e8ffc47294b897d : SUCCESS in 10m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3c909bce48444222a5d4f80acb01c21f : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c447f64576d45618570f5ac743ddc78 : SUCCESS in 13m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c2b3b356a3b449ca18d5e99b46a709d : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3907f64a35764ac3907136767ce835a1 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e924dd5235fa4633ab88daa3c612c11e : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b8ef73b74f677940e41f10951170b8aedf43ce2a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-04 07:47:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56dd1053d057b1f3cc91abe2b94d5ebc293953ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 07:47:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91795b84e74659aed26f518b98cb42e7d682384d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 10:20:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81f1502a40b0436baa38c7069a06fe54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f939347316a469e9f4f85d8e80a1d55 : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8051d6ba9495413da18d8de384b69075 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7809055dbd6842298f5248ea86563f26 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/165446c5c51c408c8c2f67304eac31fb : SUCCESS in 9m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8d9f12a87ef244c78f6251306d4abda1 : SUCCESS in 1h 15m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f3240faa47049c9a5c35ca3e98fada6 : SUCCESS in 49m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1f8b12a57d7746b781b0898485ca16ce : SUCCESS in 1h 18m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d6f0f0ad7b9a420789faa965fbe2f02d : SUCCESS in 17m 41s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2a17652ed2c429ca213d70445e54c2b : SUCCESS in 13m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4adeb53460b84586841720702760e716 : SUCCESS in 2h 19m 22s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/844a9b172e53466990347fc0208113f9 : SUCCESS in 2h 15m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/94af210beeac420f8acfc7a010269126 : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad8578b8dd584206ac72a239a2f17f31 : SUCCESS in 13m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a62913c3d2dd423b9530f5a4e42d31e7 : SUCCESS in 28m 50s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06556d6d6fed47adb7062bf400d14cda : SUCCESS in 26m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1accfa7a6a4249889f7eb6dad883ef14 : SUCCESS in 12m 37s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1137d2afca974be4af4d23d4a33558d8 : SUCCESS in 12m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/725b9bf014b94b73895c0a0f1f6c5abf : SUCCESS in 17m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbd80b770cb6485498d7b2b204a5114e : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f92f6f35961e474ab639cf0d0f2f890530d40851","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 10:20:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66f74b4be2306d1b7fa33c1578c8a894d7ebff62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 10:20:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c00e61ea3f4f4eb44eaa3a48534c83\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfa583a22ed4c738c8196b92cf0298b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30ecafdfac374709f1572c950423d1b9b9f6a9e2","revisions":{"30ecafdfac374709f1572c950423d1b9b9f6a9e2":{"kind":"REWORK","_number":1,"created":"2024-06-03 16:17:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/921045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/921045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/921045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/921045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/921045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/921045/1"}}},"commit":{"parents":[{"commit":"16147f173ffe253acf6392c8f674f716434cb07d","subject":"Fix issue removing docker volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16147f173ffe253acf6392c8f674f716434cb07d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 12:02:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 16:17:48.000000000","tz":0},"subject":"CI: Fix kayobe-tox-molecule job","message":"CI: Fix kayobe-tox-molecule job\n\nWe were still using stream8 images which cannot install packages\nanymore, due to the end of life of CentOS Stream 8.\n\nChange-Id: I8604c342e4dbc41dd5bee0f28a92c3c3cd37d631\n(cherry picked from commit 3b0a74dfc46392c1611b060c3048e3d990476548)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/30ecafdfac374709f1572c950423d1b9b9f6a9e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/30ecafdfac374709f1572c950423d1b9b9f6a9e2"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
