)]}'
{"id":"openstack%2Fkayobe~921065","triplet_id":"openstack%2Fkayobe~master~Idf705d9cf41766a897444898ac31c4635b70ec16","project":"openstack/kayobe","branch":"master","topic":"image-build-testing","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-06-05 15:57:43.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-06-05 15:57:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf705d9cf41766a897444898ac31c4635b70ec16","subject":"CI: Build default images in experimental jobs","status":"MERGED","created":"2024-06-03 19:25:05.000000000","updated":"2024-06-05 16:00:09.000000000","submitted":"2024-06-05 15:57:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921065-image-build-testing","meta_rev_id":"c3bb4dc96e49fd08381630de4f3c749dc4ef5e96","_number":921065,"virtual_id_number":921065,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-05 15:57:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-06-05 12:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-06-05 13:53:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-03 20:29:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-04 08:18:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-06-04 09:06:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-06-05 08:27:29.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-06-05 12:50:04.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"f25b59d79e07c944cf440937af2803920f3fdf1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 19:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4be5059236b4b6324bc0fea3adecd5b7c412c961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 20:29:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bf51e27c3c04846b0e8dc532f370699\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aaea5268ec34a0dac89fd8016ed5f33 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1389c5fb9949d8bb3315a0f1bb81db : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f5f64f9787c44e7a7c93ef81d1bc52a : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7134787fafb4fd9ae8cc721adea212c : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/862f0efb13fe493ab5aa7a54f5bf52d5 : SUCCESS in 8m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3d2e3c85e7824e048614ced5a9704576 : SUCCESS in 16m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89a4f713cb274dea9c9520ed37c6d1b1 : SUCCESS in 14m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/698ff4f987fd4ac0ae14ae794c92a6f6 : SUCCESS in 33m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/90874ffccdf2457987aa1d198011402f : SUCCESS in 27m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ba1ec343a294457bb8938f980b16239 : SUCCESS in 25m 17s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f3151044ce4c4a7c83bc0036d585d549 : SUCCESS in 28m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2ea673b2f7448fab72b863318252f41 : SUCCESS in 27m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/642f6cd6d16c42a6b4c5581d154964c6 : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/454074a7383f4adab23b959e7a249b97 : SUCCESS in 10m 31s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/7e098c63337b4d538ff105dd63676615 : FAILURE in 41m 35s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9f50b0646c844f0915633499e280972 : FAILURE in 29m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d442f9c9032b442696d9157e64a9d3a4 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28f07cbf902c4af4892130c897d1aff1 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e437bbd2dce4bd88b360fcd52864939 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ddaaba136515f1a6d3bf90f4a216127731898716","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 20:31:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c091b5d0deffcae6bb7d344f0b3b02e90a06168","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 20:32:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ceaf19bc2b7652c8bb1452771081c6a5ae0f111","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 20:59:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26c5e98d325f453974076fb0a1a70f34064dea3f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 21:51:23.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6125d4a1dc5541f4a025b022f01902d4\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/d109e867867a4bd587a2da1ad64c19ac : FAILURE in 45m 47s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1f3cf4fba694aa68fcc8491eec23d0d : FAILURE in 45m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5ec39c991e58385269369f6ddbf33fefc44b4ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 22:21:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05b7bcb0870a46f6aa20e736eb3047f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7120c7249f0a474fb2706664089cbbc7 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9dd8af51ab4f6ca971d0cc9e9f3dc7 : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7129bc85c5247488d274ab3466b7cc0 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5c2f5c6c81b417cb408049873a1fa19 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3a6b025e842418cab0272cedc77554e : SUCCESS in 7m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/349502cd405843a4b4b43a2ba9316b38 : SUCCESS in 58m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/97fea4904c3f41c0ac5b81d3705ebad8 : SUCCESS in 1h 04m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/366eb214f5814311b4bdb8d83160b9f6 : SUCCESS in 53m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ff26bc341d8d4d6493b3c437a90fd75d : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0e9624d1e0e447988e668607e04bdec2 : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/95ef8bfe11014fcc9d1abf958595299b : SUCCESS in 16m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbf3ff9add07493da2246b8a267645cd : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0a5c2925b10f44c7a18e9a1a9da5fb70 : SUCCESS in 1h 42m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d69b93e3077441eb6d6e2dedb5835f2 : SUCCESS in 1h 36m 16s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/503087f6a41a43aaa1a99035a35b783b : FAILURE in 1h 36m 22s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38cd5bd2a0674e1da72f0e31deb5d25a : FAILURE in 1h 11m 01s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3ff57652a07c4b01b026243bd4226da8 : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b58cc2ded5804dfc9da3b125cfa95cc0 : SUCCESS in 16m 49s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5927b9373da54013a6489f7fdb1e20cb : SUCCESS in 33m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2f42bd7acb57468aa6a21d16ffe9d049 : SUCCESS in 31m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c546891dc7e4b71a68602218fe8cd9c : SUCCESS in 27m 16s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ccdab13a7e1f4c069e4d5a36d5c90625 : SUCCESS in 35m 05s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5af6ada19f04ca5bea2c3caf1edc965 : SUCCESS in 29m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ebbff80d99604c4585aea5bf535b15c6 : SUCCESS in 12m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/484779a7cb474ad7a19d848164b3c1e7 : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/879bea008a774ba398271f6a637b8b9d : SUCCESS in 19m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/810029866a524ea7a7164b8c1582cb1b : SUCCESS in 12m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/073c3cbb27844e8c896ef7dd42bb5eb8 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a148673130748b38b8177df986a906d : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5ed73406b0e4a3caae5307930f5d4e9 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"9d0a00bce3cf6dea3a59de5549e7a37cb41f96f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-06-04 09:06:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5ff94b41728deb2516700d2c187a074f9f109c96","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-04 12:05:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4728dc2d3c5abe1abc2bdab2460cd57cd98e7a9e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 13:46:47.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/558c1c23e79242bf8a42e46909d028a6\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/cf727e560da14355a4ccb81e4e85c945 : SUCCESS in 47m 56s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e9574422bd741979fca3b2cf9a3a058 : TIMED_OUT in 1h 35m 19s","accounts_in_message":[],"_revision_number":2},{"id":"dd4a181568bf975fc1d2242eb11f017dfe29af05","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-04 14:34:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f5108105d512b52792f6138d7addc8d3e9186f2","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 16:18:14.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44d9c256a75a4fad8bc42b6bf4c9e4f1\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/2527b525624a4b6cb0d08f97852b4a69 : SUCCESS in 47m 26s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69de5a9f10d74d61b19d53fc0a7f4051 : TIMED_OUT in 1h 33m 57s","accounts_in_message":[],"_revision_number":2},{"id":"60f0198fb31b57790cb888f34e2fa821d177f10c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-04 20:04:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9cb242e1dcbbd59162b1709ba5774178d040e9ac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-04 20:04:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca9e8f03db2f206d67274cd86b8e8750097d9239","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 21:08:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/151b5cf85c2241b8aed5d3b1067d9980\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/2513749e51874d06815f227189a9f43a : SUCCESS in 52m 49s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81f1916ea4cb40ac94fb49bdd82388ec : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":3},{"id":"548201ea323e2ada303d027c268f456ad556dd5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 21:56:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9ab65ce9b8147199b87de5107935af7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f2dd4e8c6594607a502ba2776e9ebba : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121d39f1d70d45668de5aa85377587cc : SUCCESS in 7m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eeb1585b26cd4fdf8f9cea838c69bdde : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f34d5d762299459a92d002f0429b4572 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5280f9d765904c7e9a8c609f850227ca : SUCCESS in 9m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/736a2c49168345738930f0157a838700 : SUCCESS in 59m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ddb5ddc15c2745d4a7b809510d9de4d9 : SUCCESS in 1h 04m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d704de3a8b54d9aacfea64802704713 : SUCCESS in 59m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/57779b08049a4679aba78751d44160b9 : FAILURE in 56m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/135ef66ff2c04aa1b56b1252429b290e : SUCCESS in 18m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/88094fa9f81b44eebc692ae799ba50b1 : SUCCESS in 20m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46ff5057c4f145b181fcba8f56eaf57d : SUCCESS in 13m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9afc90da3b87408f948607a8f032e11d : SUCCESS in 1h 46m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65aefeccdd2f4f39bec83e8c41372aa5 : SUCCESS in 1h 39m 14s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5043ea421bbb42d08369cc19d725a7a4 : FAILURE in 1h 21m 14s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4bfe5a678f64c4785d0e4b6d4fcd822 : FAILURE in 1h 04m 43s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f7656d861d314f5ea7973bdfdd0fd040 : SUCCESS in 16m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5293a6ad74bc464d8575e60d32a58534 : SUCCESS in 12m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/810de353c66d4c5d872b98d548ee5152 : SUCCESS in 31m 07s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0d645a3d823f4ab78121244371900cef : SUCCESS in 27m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/603a9f7165144f319ee20b676540e184 : SUCCESS in 23m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0e13d74094064cc49e2c466de37bdac4 : SUCCESS in 31m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7c154c9ce4b4608b96e42910ee61a76 : SUCCESS in 25m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b14e39f36e4f45a8a03888a057c1ef96 : SUCCESS in 11m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cf0a61e377e42c7b167bdb6b838d3be : SUCCESS in 10m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ab3def4b31b4b559eeba7400741b7c6 : SUCCESS in 18m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f7c0caa596142d8990f409caa784162 : SUCCESS in 14m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4157324b942b4bddbf22b3b1e7c2be0f : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86b6e0cac71f4ee7a31965a1f3294517 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d152d82c41e34459bb57448021c00240 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"7142cb7412f7ccafd9b5e730b2d08eb93cad9fd6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-05 08:27:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee048dddbda1d139d761fa0120019a78709b4a37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-05 09:34:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"527382fe045aaa041eb23f9056574afaeb7b0c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 11:35:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d46767c228e473c816d76f36114aaac\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/570ecade0c1c4fa281291f4af87867a7 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de740e662344f0a9a9302d241873428 : SUCCESS in 6m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a04ed4de33224c1f915c5c64e4401d8d : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63e41a7e4ad949e98e388b224690a0c8 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5bec63186fe341eaa408fdd30ca0975c : SUCCESS in 9m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5f08d69bcfd8492ba467394fbeba2c14 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/83d83a04364b4bf1adcd636407a2ebb3 : SUCCESS in 1h 05m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad4680f26ba049e285b4a0e7af102e90 : SUCCESS in 53m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d26574e1271142e5850e67a448f9db94 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d2f1dd76aca047c78dec42842d961c7b : SUCCESS in 15m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c104fd4797e846f3a03910d20d8a67c7 : SUCCESS in 15m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/969e67de599b43119503b82c4b8f0ee1 : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/02fb538056874f14bd584e925112eed7 : SUCCESS in 1h 55m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/481f461fa4a643c598a237641970c872 : SUCCESS in 1h 37m 06s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4562be9e9dba4b44872efe2def4b827e : FAILURE in 1h 18m 04s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8572e2427f2f43039fe0f3e9ba4bb98d : FAILURE in 1h 17m 18s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/39af786cacb54bd4b932fd4b79d6bf30 : SUCCESS in 16m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77ac138d60804254a44b6ca64f0b5490 : SUCCESS in 11m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/45d042aa2590455c84cca4037fc64851 : SUCCESS in 31m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/723ece5a970040d3a814842665e1fdb7 : SUCCESS in 27m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fabc545cca1473da32eee93c1269f4e : SUCCESS in 23m 25s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e91af58ea58c47a184c70c3fe63578ec : SUCCESS in 33m 02s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84b00e96ff1d481ea40714cc03174400 : SUCCESS in 26m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82d497d48e8e4cd8a5be45f6e7325311 : SUCCESS in 13m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4d69232b27b4d49ad72abb6669344a7 : SUCCESS in 9m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/13ea4dd87e724007b5d23e417425857e : SUCCESS in 16m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7516c1fbc9a5462f834a0241457ed709 : SUCCESS in 13m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74ba24469ca7498e9c618a3bf29ef998 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdfd0d658fab43a6b3522c38fef8b8ce : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd32e7baaae4497ca1652762c9de511b : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":4},{"id":"8f4b6612db983653bbf5de3deac22157d1cf2a1c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-05 11:36:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e5cbc483fc3bba37369d60e7840874655c450ff2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-05 12:50:04.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f55fe47e4dcea1754ec13f53a3482598c5a876e4","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 13:18:47.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02cdd0c6151d4498a4bdddfed29be97b\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/f72df1561d4f4874b49d95146062810c : SUCCESS in 48m 22s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f83a0967c9964c72ae4d1bdd60c9ef52 : TIMED_OUT in 1h 34m 12s","accounts_in_message":[],"_revision_number":4},{"id":"17e9973ce168008b8248836567454d6f6c099435","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-05 13:53:34.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b5016aae5bbc42887212e1f54db031ce89b548a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 13:54:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0b61f04e923515aacdec5989f3438cd227e99254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 15:57:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5682e5f89bbe46d7af21d4d011e2a34e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b38bd84a2d4da7bc9eb4d68115a135 : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/43e496bdfa60480284190acac6aff3c8 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/88728076357c4efeb5161be3616a9cef : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1083c54604d41ed8281230c75d40021 : SUCCESS in 8m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/aa009bde28574ed38abf00c483f41a10 : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78ec3666a63f4c61922665254e802bd7 : SUCCESS in 51m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1284dcf66d654f789189bedd112c12a9 : SUCCESS in 1h 13m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0b0d72218e114eaea4a7ee54e97a7665 : SUCCESS in 15m 40s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f738a26d8e34f9e83a78d58de6a0584 : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0d59e54c2c324c0bb57d7501e9423c3e : SUCCESS in 2h 03m 31s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31d8648c46f5419d8509f4ef7e3cb72c : SUCCESS in 1h 33m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d70589a665a14b2d9891706c59a33fd5 : SUCCESS in 19m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78bae99c1d324b678380548bbd4c7216 : SUCCESS in 14m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8fd681979cd1421aad5b89b043c5f68c : SUCCESS in 34m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2019dfb8acf4c18b9022872dfa61825 : SUCCESS in 29m 35s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/329c893e7351460c916320e016da8ba4 : SUCCESS in 31m 13s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c08b6c149ea421ab959845e92d796cb : SUCCESS in 25m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5e13057503a14a85b46ca8ae879ce8db : SUCCESS in 13m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7480cea8ec794bab84d05004368dc0bc : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4820ee36ded84d959f50cfb10596a6f4 : SUCCESS in 18m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87f652dd5025454da4eb19ce18d778dc : SUCCESS in 13m 44s","accounts_in_message":[],"_revision_number":4},{"id":"35e599df75de007082739eaa447eeb9dbbcd30e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 15:57:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c3bb4dc96e49fd08381630de4f3c749dc4ef5e96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 16:00:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed5b37bc6736412d84be93250f317370\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18b24aacf2847998d2006be03ca489b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ee79fc5d4d4df567a882b1fb54b95f221f41f82c","revisions":{"4f4e7c1a5f50f5a394196964c727ee39a8901a63":{"kind":"REWORK","_number":1,"created":"2024-06-03 19:25:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/921065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/921065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/921065/1"}}},"commit":{"parents":[{"commit":"3b0a74dfc46392c1611b060c3048e3d990476548","subject":"CI: Fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b0a74dfc46392c1611b060c3048e3d990476548"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 19:24:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 19:24:56.000000000","tz":120},"subject":"CI: Build all default images in experimental jobs","message":"CI: Build all default images in experimental jobs\n\nChange-Id: Idf705d9cf41766a897444898ac31c4635b70ec16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f4e7c1a5f50f5a394196964c727ee39a8901a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f4e7c1a5f50f5a394196964c727ee39a8901a63"}]},"branch":"refs/heads/master"},"887427221a9bbfc30beefa804f72f674501d60e9":{"kind":"REWORK","_number":2,"created":"2024-06-03 20:32:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/921065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/921065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/921065/2"}}},"commit":{"parents":[{"commit":"3b0a74dfc46392c1611b060c3048e3d990476548","subject":"CI: Fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b0a74dfc46392c1611b060c3048e3d990476548"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 20:29:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 20:29:50.000000000","tz":120},"subject":"CI: Build default images in experimental jobs","message":"CI: Build default images in experimental jobs\n\nThis commit changes seed-images jobs to build seed and overcloud\ncontainer images, each with a customisable regex.\n\nThe rocky9 job continues to build only the base image, although for both\novercloud and seed instead of seed only.\n\nThe other seed-images jobs (centos9s and ubuntu-jammy) now build all\ndefault images. This takes longer and can fail for various reasons\nunrelated to Kayobe, but this is fine since the jobs are marked as\nexperimental.\n\nThis is to attempt to catch bugs in the image build code, such as the\nones fixed by [1].\n\nAlso update release documentation to remind developers to run\nexperimental jobs.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/921012\n\nChange-Id: Idf705d9cf41766a897444898ac31c4635b70ec16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/887427221a9bbfc30beefa804f72f674501d60e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/887427221a9bbfc30beefa804f72f674501d60e9"}]},"branch":"refs/heads/master"},"1ef277e6b54e31c745097ef39148fd72c4b183a5":{"kind":"REWORK","_number":3,"created":"2024-06-04 20:04:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/921065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/921065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/921065/3"}}},"commit":{"parents":[{"commit":"afa0b75cf543ad67bca88cf64a8de8cbfcf4952c","subject":"Merge \"Remove use of get_md5 when using stat module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afa0b75cf543ad67bca88cf64a8de8cbfcf4952c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-04 20:03:37.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-04 20:03:37.000000000","tz":120},"subject":"CI: Build default images in experimental jobs","message":"CI: Build default images in experimental jobs\n\nThis commit changes seed-images jobs to build seed and overcloud\ncontainer images, each with a customisable regex.\n\nThe rocky9 job continues to build only the base image, although for both\novercloud and seed instead of seed only.\n\nThe other seed-images jobs (centos9s and ubuntu-jammy) now build all\ndefault images. This takes longer and can fail for various reasons\nunrelated to Kayobe, but this is fine since the jobs are marked as\nexperimental.\n\nThis is to attempt to catch bugs in the image build code, such as the\nones fixed by [1].\n\nAlso update release documentation to remind developers to run\nexperimental jobs.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/921012\n\nChange-Id: Idf705d9cf41766a897444898ac31c4635b70ec16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ef277e6b54e31c745097ef39148fd72c4b183a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ef277e6b54e31c745097ef39148fd72c4b183a5"}]},"branch":"refs/heads/master"},"ee79fc5d4d4df567a882b1fb54b95f221f41f82c":{"kind":"REWORK","_number":4,"created":"2024-06-05 09:34:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/921065/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/921065/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/921065/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/921065/4"}}},"commit":{"parents":[{"commit":"ee0754b6f1ddc9f9076b5f06ac28a590e64d3534","subject":"Merge \"Introduce max fail percentage to playbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee0754b6f1ddc9f9076b5f06ac28a590e64d3534"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-05 09:33:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-05 09:33:48.000000000","tz":120},"subject":"CI: Build default images in experimental jobs","message":"CI: Build default images in experimental jobs\n\nThis commit changes seed-images jobs to build seed and overcloud\ncontainer images, each with a customisable regex.\n\nThe rocky9 job continues to build only the base image, although for both\novercloud and seed instead of seed only.\n\nThe other seed-images jobs (centos9s and ubuntu-jammy) now build all\ndefault overcloud images. This takes longer and can fail for various\nreasons unrelated to Kayobe, but this is fine since the jobs are marked\nas experimental.\n\nThis is to attempt to catch bugs in the image build code, such as the\nones fixed by [1].\n\nNote that the ubuntu-jammy job does not build the bifrost-deploy seed\nimage because the build is currently timing out. The centos9s job builds\nit successfully.\n\nAlso update release documentation to remind developers to run\nexperimental jobs.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/921012\n\nChange-Id: Idf705d9cf41766a897444898ac31c4635b70ec16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee79fc5d4d4df567a882b1fb54b95f221f41f82c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee79fc5d4d4df567a882b1fb54b95f221f41f82c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
