)]}'
{"id":"openstack%2Fkayobe~921278","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I234b7a0434ba3d533a66b766331fd489078fbcf8","project":"openstack/kayobe","branch":"stable/2023.2","topic":"kolla-enable-letsencrypt-stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-06-04 18:33:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I234b7a0434ba3d533a66b766331fd489078fbcf8","subject":"Fix container image build issues with letsencrypt","status":"MERGED","created":"2024-06-04 11:42:25.000000000","updated":"2024-06-04 18:34:40.000000000","submitted":"2024-06-04 18:33:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921278-kolla-enable-letsencrypt-stable/2023.2","cherry_pick_of_change":921012,"cherry_pick_of_patch_set":2,"meta_rev_id":"40b731be333cdff40a75f7ed76b6793f5f643da7","_number":921278,"virtual_id_number":921278,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-04 18:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-06-04 16:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-06-04 16:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-04 13:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-04 16:10:26.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"631979a39840146c8628e92de0bd196867bbc107","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-04 11:42:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d8736a23d402f7af87fd96b2f4f1716f66cf2b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 13:59:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e73918a09a145bdb2c36e6b536a3fdf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/336fadebaf1a46a5a709f3dbc5870a7d : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68a3dfccb6a4529b9b9fb7ca3a51a48 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a88ad3e680cc4da1baf9b077e0e756a6 : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92d8e154d6e24253b82e6c28a0e7dc3a : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/80b0c73c49f843e5b1a29b6b8a36d3f5 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0425b9d5fdd44c9b3214b386cf364c5 : SUCCESS in 10m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e6a250a8048a47e08f7c6ea494c61306 : SUCCESS in 1h 01m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7aff57cc7b3a41709f6fbd8e8f9aaa33 : SUCCESS in 1h 03m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20b712e904d047b1974b238189192641 : SUCCESS in 50m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a4ebabde9cd44494be51cc832fadafee : SUCCESS in 1h 08m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/66781dbb5c7f4af09f89f203fc9c9f3a : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e4eff5660b7a4b12b2827715d9d74aa8 : SUCCESS in 19m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24855ab237dc4860b4dc770fe24ffc54 : SUCCESS in 13m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/033e12db10194e07ac3f4004cea1b8eb : SUCCESS in 2h 15m 27s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edd3ed7d02804cd4b411564cc74fd07e : SUCCESS in 2h 09m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0303f098fdd1435fafd06fc22b3d0322 : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/435497ca5896452fb4994205700cf4ee : SUCCESS in 16m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ba212fac6c60443da1716441f54300c2 : SUCCESS in 34m 10s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/732713f4beac4d88b4ac393676d9937c : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e99063883a34a8196e8e6f92dc7ef69 : SUCCESS in 24m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ad86a5ad3134b69b93ef4f727340ed1 : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e5bdb72ee324f408a71119321a8d04e : SUCCESS in 9m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0c3adf66154c41e7a1cda269968c1abd : SUCCESS in 17m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/125fe1e4023c4819a810bb7de3ed46fa : SUCCESS in 13m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaed5fba50f648f2a4aef080a0a99261 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33c8c71166284f07bff17ddcd0b0bcfd : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ed634a1f7f545fca5ce8ccef7a374c6 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bfe78b8a55b6e4ce07b42add45c1e43464de2682","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-04 16:10:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7ec28758c01980f88636879c5a7e57866dc14fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 16:10:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77ba27a498701f6d8776c63b71d41a15eac269a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 18:33:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c45626754bfb4f90bfe51e18f1c03a43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a3a1bce6e845429515844d0ddd642e : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7096e2d767b044688b271086fdd5832f : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19cb4e172a6441f1adcd5f6a767207dd : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b7a150f34b94c13ade29eb25d6b6001 : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65cf79b9d7054c97b5843325c4e0cd3e : SUCCESS in 10m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/df2b1757d3ea4a86982a168b1843dac1 : SUCCESS in 1h 06m 45s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0f40ac8936740e8a5bd9dfaf10a8bd8 : SUCCESS in 56m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8fbdb62d8aef4f38aa0af9f948431538 : SUCCESS in 1h 05m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1dfa7857dc384125afdc3bcc46437860 : SUCCESS in 15m 35s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5ead9661f154710ac8eee53939f4e06 : SUCCESS in 14m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/eb12a7b8d8944473a55ad7cf11921695 : SUCCESS in 2h 15m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1dc4aba951d42e2a4557d0b3336fced : SUCCESS in 2h 04m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2e56635596b447178f5df686cbacb7c4 : SUCCESS in 15m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe0892cf3e1748d68504cf178883890e : SUCCESS in 17m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/adfc69bbf7da491b982ad0b943ccb3bb : SUCCESS in 31m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0e46d3ec124444698399cc6ad8a5a00 : SUCCESS in 26m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f06d86dfe1414877a6f0af2b0963dc28 : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afe011477d0f4ef9b6a0f5cf8b8d891d : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e07d4b978bb1481785aa71f55765ea46 : SUCCESS in 20m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4c7f14ed5434313b188635054914382 : SUCCESS in 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"292791fcba3a2c8327b642ab192626a67d70ade3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 18:33:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"40b731be333cdff40a75f7ed76b6793f5f643da7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 18:34:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a57a2c659aa640ba8d6d804793905aba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964f1d59ee144c8898cdb339891a96d6 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae90e4a6b3e448ebb20ada811a4c48c6 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e2a3aeb5e5c012e66194af1059043ab01752a05","revisions":{"8e2a3aeb5e5c012e66194af1059043ab01752a05":{"kind":"REWORK","_number":1,"created":"2024-06-04 11:42:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/921278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/921278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/921278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/921278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/921278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/921278/1"}}},"commit":{"parents":[{"commit":"30ecafdfac374709f1572c950423d1b9b9f6a9e2","subject":"CI: Fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/30ecafdfac374709f1572c950423d1b9b9f6a9e2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-03 11:04:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-04 11:42:25.000000000","tz":0},"subject":"Fix container image build issues with letsencrypt","message":"Fix container image build issues with letsencrypt\n\nThe kolla_enable_letsencrypt variable was undefined, which was causing\novercloud container image build to fail when the variable was not\ndefined in Kayobe configuration.\n\nThe bool filter was missing for kolla_enable_letsencrypt, which was\ncausing letsencrypt images to be built when kolla_enable_letsencrypt was\nset to \"false\" or \"no\".\n\nChange-Id: I234b7a0434ba3d533a66b766331fd489078fbcf8\n(cherry picked from commit 9669462208e1785c2d64485d6bdc10cc75210d39)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e2a3aeb5e5c012e66194af1059043ab01752a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e2a3aeb5e5c012e66194af1059043ab01752a05"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
