)]}'
{"id":"openstack%2Fkayobe~921375","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I57fb4c1697a31d4f592d79a31a04507cf91e97c7","project":"openstack/kayobe","branch":"stable/2023.2","topic":"deprovision_cleanup-stable/2023.2","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-06-06 13:42:43.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2024-06-10 11:06:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57fb4c1697a31d4f592d79a31a04507cf91e97c7","subject":"Add ironic config files clean up tasks to overcloud deprovision","status":"MERGED","created":"2024-06-05 11:33:40.000000000","updated":"2024-06-10 11:07:47.000000000","submitted":"2024-06-10 11:06:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921375-deprovision_cleanup-stable/2023.2","cherry_pick_of_change":920546,"cherry_pick_of_patch_set":5,"meta_rev_id":"ca79e8b48350722442732128cde383ee9fa2ae07","_number":921375,"virtual_id_number":921375,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-10 11:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-06-06 13:42:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-06-07 09:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-10 08:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-10 08:05:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-05 13:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-06 08:27:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-06-07 09:41:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-06-10 08:05:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"10c8aa44826b7ac7557334af5ac833c2aac3f343","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-05 11:33:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* ansible/overcloud-deprovision.yml","accounts_in_message":[],"_revision_number":1},{"id":"b98e002a4a2f93994855c78ca97136597be68c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 13:50:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32f9a6daf9f14e40b184223dc913c3b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9c6d045827d46e1a37489b56d149907 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ee429eca3a482d864dcd83c1a87178 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6966adb7e0f746e28fb191cd5ac5dba7 : SUCCESS in 2m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f2e3ce5ca8844b18978c1633ce8708b : FAILURE in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d770f3768fc4a268baa6b433331a2de : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8293108e384449c8adb022600a4cc2a4 : SUCCESS in 11m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e725630144184630a83030aa28eb02ec : SUCCESS in 59m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7ad36c3215a44304a1fde00245c44d17 : SUCCESS in 1h 02m 52s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c5e965babd64200ad7d80f92b2b1de4 : SUCCESS in 49m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/11ff9e2d8e614dc99c06732c059e8883 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/df9d5771302449ec915d81273527cc81 : SUCCESS in 16m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ce52be5d7165404b96c24a73a0901509 : SUCCESS in 19m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/680bbb561a33441a972ff65ac3c72181 : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/34b8631e6a164bb7becc16d2e57707e4 : SUCCESS in 2h 03m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31481206195c41b4ab719affdeac9280 : FAILURE in 2h 14m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a1f017c9d8ee45c5b9bdfa8dd2959318 : SUCCESS in 18m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88e8909a6a2a4df78c5a43cf8398f592 : SUCCESS in 13m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d3002de6d3d34cbebe1bcc80a8cf0cf1 : SUCCESS in 37m 20s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/99c582815745416687fd365c92270668 : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea3a5784991748ca9f30c2df60cc3645 : SUCCESS in 28m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a602011389a945f9bd1fcd53f34bb057 : SUCCESS in 14m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bbddcf03ed941988d5bb68c0e981ea0 : SUCCESS in 10m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/25241daa9af847818390787a7abad2f4 : SUCCESS in 16m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7adff98e6af04cdc8b33c9e80d9ad98d : SUCCESS in 14m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2581f22c37e94573bac5781a3d5237a7 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5fe97d84fd34216ac42ed4a8ac41672 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/220b780e9db74d0b810dbb23a7197505 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6541af6b213ebcfd04a1dbe90e71a7f0ca90a6e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-06 07:56:25.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"38fa6190e22a167b7f35d0f1c941b209c2d28a13","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-06 07:58:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5bcbefd4145d748de1ab71a3371872840c8f17de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-06 07:59:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7da657495cdbe3f187a8882d2f0d482f9958e749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 10:48:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b63fe48c614e29a467b5563a3b4bbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fb21284f81347a198f2022ba2c95fc7 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee82a7292ee48918ce5361f325c9799 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf8b0561f21140e6a332ab80216d5cad : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e59e4e5712284f4e9f52896a7fc6e143 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2fe5e14216e4ad7bd90fe86b11d0472 : SUCCESS in 12m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ef53e97627f4b31858faed9051e0c77 : SUCCESS in 8m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/259a1e8b832e4734bd9e8d934d10ed47 : SUCCESS in 58m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4c7f5e4951944349ad7cc8d87e2ac728 : SUCCESS in 1h 03m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/974827843ade49409d955f54f142719c : SUCCESS in 50m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9ae431404b7c4ca9a314e97c99bafcc9 : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5cddcaa346524fd289579ca04b9bf3fa : SUCCESS in 20m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cfdd7caf069d4699b904c41b7085072d : SUCCESS in 16m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/723b48081f0b442092118642b5788239 : SUCCESS in 14m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f7d7510371974eeeb6cb49f4c4150614 : SUCCESS in 2h 19m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cafe1c3149ea4b4f8d10fd0972cec7a6 : SUCCESS in 2h 42m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/da33444072ce4632a15c4933bbc51b33 : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2f84155763d4959931e9018ffec4e69 : SUCCESS in 14m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/25e4314013dc49f286c5b41171afba1c : SUCCESS in 35m 25s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ad927f1f99ac47fea25a18d016b30942 : SUCCESS in 31m 29s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0d672ee2e514cf5aa48e7cc38c45895 : SUCCESS in 32m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b1d50e976dd42f9b7647d053cde1ba7 : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90cd20035c5340b88fba149db3e51d14 : SUCCESS in 10m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/31243c7547af486fb9cce562f149b1df : SUCCESS in 17m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b77c293e5ccd46e08afa647b8b06d893 : SUCCESS in 17m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a483cbf05434a218287c47d0317c758 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d01348270ba4dd29c018153d6dd56b7 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e4f5b6dd75f4115b8426ee402efcd09 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":4},{"id":"ece63818ba185fd749fce804003b3e5084651e29","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-06 12:41:45.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"330a6068ccd4d5553831cc6a520f78199f43041b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-06 13:42:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d2a6773da7eabc17fe35704f217dff5c3847243d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-06-07 09:41:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3ef9e6dba5283b560dfd626142634169f97c4ba6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-10 08:05:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b4192dfb68907befaf483bbcd1b23d134902d961","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 08:06:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"99e9e0179c465456e74ec711f16b90a01e352e0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 11:06:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0c687984389459a9e3f0cc3e4a5abb6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e7b806b81f44f93999f26b10abbbf0e : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc7ebe94b4f449c28ed46213cbe02472 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d8d52a54bd7463b8b80232bebaf4392 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8bf4e5c1efdb4a78b7b0f9bfa2f1466e : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b25f4012ebbd46d6ae3cace7f072632b : SUCCESS in 10m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6f767041e1e242309ba1cd7cf7c92cab : SUCCESS in 1h 08m 33s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/814d43d3c4d049b1822f6bd19507f209 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/63cfec9414114c91b5ec1f9a87d6443d : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/baa1e24b724c44b6a1da390a8432c34c : SUCCESS in 20m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc70f48645604cd1a351a2108f1a0615 : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e0e55c3965ba470cbb6dcf0a8ec5c8b0 : SUCCESS in 2h 42m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ae93a83f75744d88b1140544bbc98c5 : SUCCESS in 2h 13m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b0223d8e6e8b41d1a3c211e5cee3c8bc : SUCCESS in 19m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f54936deb75e49a39c0b4b7401868140 : SUCCESS in 14m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/49f5aadf8d1646e1822ba3624ec053c1 : SUCCESS in 34m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63c7a0185dfd4300bba18eebe842a21c : SUCCESS in 24m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8e7719cfad79474ab1487f100f5db108 : SUCCESS in 14m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3aa819db0c54b87b50242a223934a9f : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d5e8d3e165c54f9ba0659c3042302f9a : SUCCESS in 19m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21c6b6b3a91a481092885997d6e4c935 : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":4},{"id":"25b22180b1b329cb779141077eb2e67991a3f95c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 11:06:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ca79e8b48350722442732128cde383ee9fa2ae07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 11:07:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f41914c379594d7bb4f5d06f95ea62e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d10c1125cce4ed893352e6370937e9d : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b178b0bf52824d768c57a321578c1942 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"67bee0c5e0e447f4e2c0833eaccf70bdf570e863","revisions":{"a126f0f197213aa7e3f3e3d766427e307292f354":{"kind":"REWORK","_number":1,"created":"2024-06-05 11:33:40.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/75/921375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/921375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/921375/1"}}},"commit":{"parents":[{"commit":"8e2a3aeb5e5c012e66194af1059043ab01752a05","subject":"Fix container image build issues with letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e2a3aeb5e5c012e66194af1059043ab01752a05"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-05-27 14:34:51.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-06-05 11:33:40.000000000","tz":0},"subject":"Add ironic config files clean up tasks to overcloud deprovision","message":"Add ironic config files clean up tasks to overcloud deprovision\n\nCloses-Bug: #2067288\nChange-Id: I57fb4c1697a31d4f592d79a31a04507cf91e97c7\n(cherry picked from commit 7837c9b7f8d3de78cebdfc37bcf38775b7f8c292)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a126f0f197213aa7e3f3e3d766427e307292f354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a126f0f197213aa7e3f3e3d766427e307292f354"}]},"branch":"refs/heads/stable/2023.2"},"71a6cc337d774126b054bf06b362cfa95ad1c986":{"kind":"NO_CHANGE","_number":2,"created":"2024-06-06 07:56:25.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/75/921375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/921375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/921375/2"}}},"commit":{"parents":[{"commit":"8e2a3aeb5e5c012e66194af1059043ab01752a05","subject":"Fix container image build issues with letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e2a3aeb5e5c012e66194af1059043ab01752a05"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-05-27 14:34:51.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-06-06 07:56:23.000000000","tz":120},"subject":"Add ironic config files clean up tasks to overcloud deprovision","message":"Add ironic config files clean up tasks to overcloud deprovision\n\nCloses-Bug: #2067288\nChange-Id: I57fb4c1697a31d4f592d79a31a04507cf91e97c7\n(cherry picked from commit 7837c9b7f8d3de78cebdfc37bcf38775b7f8c292)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71a6cc337d774126b054bf06b362cfa95ad1c986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71a6cc337d774126b054bf06b362cfa95ad1c986"}]},"branch":"refs/heads/stable/2023.2"},"77a12a4ca489b17f495013d97f9a4a5c42f29ab0":{"kind":"REWORK","_number":3,"created":"2024-06-06 07:58:27.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/75/921375/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/921375/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/921375/3"}}},"commit":{"parents":[{"commit":"8e2a3aeb5e5c012e66194af1059043ab01752a05","subject":"Fix container image build issues with letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e2a3aeb5e5c012e66194af1059043ab01752a05"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-05-27 14:34:51.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-06-06 07:58:10.000000000","tz":120},"subject":"Add ironic config files clean up tasks to overcloud deprovision","message":"Add ironic config files clean up tasks to overcloud deprovision\n\nCloses-Bug: #2067288\nChange-Id: I57fb4c1697a31d4f592d79a31a04507cf91e97c7\n(cherry picked from commit 7837c9b7f8d3de78cebdfc37bcf38775b7f8c292)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77a12a4ca489b17f495013d97f9a4a5c42f29ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77a12a4ca489b17f495013d97f9a4a5c42f29ab0"}]},"branch":"refs/heads/stable/2023.2"},"67bee0c5e0e447f4e2c0833eaccf70bdf570e863":{"kind":"REWORK","_number":4,"created":"2024-06-06 07:59:28.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/75/921375/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/921375/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/921375/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/921375/4"}}},"commit":{"parents":[{"commit":"8e2a3aeb5e5c012e66194af1059043ab01752a05","subject":"Fix container image build issues with letsencrypt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e2a3aeb5e5c012e66194af1059043ab01752a05"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-05-27 14:34:51.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-06-06 07:59:18.000000000","tz":120},"subject":"Add ironic config files clean up tasks to overcloud deprovision","message":"Add ironic config files clean up tasks to overcloud deprovision\n\nCloses-Bug: #2067288\nChange-Id: I57fb4c1697a31d4f592d79a31a04507cf91e97c7\n(cherry picked from commit 7837c9b7f8d3de78cebdfc37bcf38775b7f8c292)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67bee0c5e0e447f4e2c0833eaccf70bdf570e863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67bee0c5e0e447f4e2c0833eaccf70bdf570e863"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
