)]}'
{"id":"openstack%2Fkayobe~921423","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I57fb4c1697a31d4f592d79a31a04507cf91e97c7","project":"openstack/kayobe","branch":"stable/2023.1","topic":"deprovision_cleanup-stable/2023.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-06-06 13:43:30.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2024-06-11 11:14:44.000000000","reason":"\u003cGERRIT_ACCOUNT_34940\u003e replied on the change","reason_account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"}}},"hashtags":[],"change_id":"I57fb4c1697a31d4f592d79a31a04507cf91e97c7","subject":"Add ironic config files clean up tasks to overcloud deprovision","status":"MERGED","created":"2024-06-06 08:08:18.000000000","updated":"2024-06-11 15:00:28.000000000","submitted":"2024-06-11 14:59:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921423-deprovision_cleanup-stable/2023.1","cherry_pick_of_change":920546,"cherry_pick_of_patch_set":5,"meta_rev_id":"f526f11581d7b5733905271afadc3b1b951f7727","_number":921423,"virtual_id_number":921423,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-11 14:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-06-06 13:43:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-06-07 09:41:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2024-06-10 08:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2024-06-10 08:05:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-06 08:25:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-06-06 10:01:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-07 09:41:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-06-10 08:05:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"7f99063fc6c7996e1dc53774c731f1899a2351e7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-06 08:08:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* ansible/overcloud-deprovision.yml","accounts_in_message":[],"_revision_number":1},{"id":"bf7bd66f19a9079c4bda7c2c5bda87a01dd01838","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-06 08:10:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1f8c03b8c6e5408365b5410b5a2886b34ce3ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 10:01:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/364d74b8050142d1b28e567f918b702e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d2a1126d460412ebd59fb38eb2ba66f : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73db363632e841b08c27461ab9dd67aa : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18d9d5935d174ece93d308a9974ee9b1 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf15300e3fa74ae89c3b9312b3798f99 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1d39730a6084e99ae5e7eabcbfc38f5 : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5e603229389a4cfe9f1e60d2ea82c94c : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/232f749b561a4a189de94dc76fdf918a : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f31771c1f3884d0183ddaa9dc32477b3 : SUCCESS in 12m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6a37b31d1d6744bf82ca55f1504d2f6e : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d445a644f76a4773b44e5d3a3dcecc38 : SUCCESS in 1h 01m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52e47f412b004f73ad1b465cb1002dd0 : SUCCESS in 54m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bcd2f499226e401c9e205244adabe3ae : SUCCESS in 1h 01m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/11350cc31e384cc1957388e327489cd8 : SUCCESS in 17m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3aba0324c9c2477c88db3060a7a2836d : SUCCESS in 15m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/138e572ccb10481799b355be52d6cc5c : SUCCESS in 15m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b219f20d5ddb4df9a4702e4ee6407d46 : SUCCESS in 1h 45m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6332cb84c3014de6b3f164fa13d5b399 : SUCCESS in 1h 40m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/62b7a613aacc4b32b2b884a50b7dbf94 : SUCCESS in 17m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf9fce93258d44e1aebee594dd892760 : SUCCESS in 15m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a28fd435c2744c8bae131b4d5afc9670 : SUCCESS in 33m 24s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/14ad7204711b43559963762205a409ec : SUCCESS in 27m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0db1252a495a45a78f9aa2c153a958ac : SUCCESS in 25m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/564b4a24996d4248b2c0dc99b1ef23ad : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02963a93aa3b4855b93e797ff3a16a7f : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9751651392c7469e973e51a0b1c01bfe : SUCCESS in 20m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a86d16ef83384337944f0c44b64a7548 : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f5b427f75bbf80742c9af4eadb3ada901570b789","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-06 12:42:02.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"4efa51af8c012bb64e35ec2e118d55500d0fb417","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-06 13:43:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b24baf33c7ac37aa4610d99e81ef54a232ecd271","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-06-07 09:41:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1894be0e46352902df72371e5e31f263ab45a006","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-10 08:05:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ecf58f6417067526b0a4ae607a0945d4f5c2fa3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 08:05:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4c8d099afcf19d5b1469906fd554e519dc23076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 10:05:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f751201727f94b368356637b0c34f1fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71164570502432d8ca91de50bbc34d1 : POST_FAILURE in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd462a06d034942bfbf9ceb68be0f4c : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba46bfabeb94a95b0e8d34fede3fe80 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64ca37e11389446d94fea86e6c058a88 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54f93735afdd41069b083017dc14ccdd : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6bfa60f2297d4afd9d7cd0689bcf7956 : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1eaab384b9d6424693b344a81636cb96 : SUCCESS in 9m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bbac4db998564a07adc552c1d00bcd9d : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09aaf5bb97364274b9992235925ca270 : SUCCESS in 52m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a6f9f6282e5d4c0a8c5a29b0e8d6b0c1 : SUCCESS in 1h 14m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/937c2f76970c4353b407e5c4ecb3f369 : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/640838863a31457b995f514f7a1ce2d3 : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a0858286e343464ab2b9c05d8eb9a909 : SUCCESS in 1h 54m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0af4909c8a2e409aa05c8a029e96888e : SUCCESS in 1h 36m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4edca4f954e94b7cbd9bebf757b80b43 : SUCCESS in 18m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2d9afad2c2a415ea11c464f07e5dd58 : SUCCESS in 14m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d35bd871cd6e4620bb2530c1b5862c69 : SUCCESS in 33m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da79832e99c44b73a7bd4650040ae596 : SUCCESS in 22m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5b06c3aae3841baa528d68ba8e680e4 : SUCCESS in 17m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1def8b9db9b44f18bcede0a6f85dfb3f : SUCCESS in 11m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8c740bb89feb4c81b100c4bc54017110 : SUCCESS in 21m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70ba2887da70456a8a4b691be84f149a : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":2},{"id":"15664e60081f29b2851c1f76f2f04d6e63b21b56","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-10 10:07:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"132225d0a96d4c46dad191ab6f39002868a32b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 12:03:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ac4aaae593d41dcafbb6863a9548aa7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbc624ec989941668fb5862e98b74d02 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40d8329036e493e9b28f700ac149683 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6edabf9541c45a3b52148b5637f232f : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef23654c25a46bd90b598fe6feddcc9 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb669b70551543ce84169d44657d7b44 : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d06b4a3765414d4d8253e46dfe500d4a : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/afdaafa9ef3743efaba2ca4c8c86d9b9 : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1ad14edcb1a4d0786c0268b2d9eff06 : SUCCESS in 12m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e6d73e2e8a7b4f7fa19f58cdb5d60f06 : SUCCESS in 1h 08m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/34f0e341a5ad40d0a50598a1ba10c243 : SUCCESS in 1h 10m 50s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92e75fceed224de59ad9238172370dd8 : SUCCESS in 57m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b14fdd39310b40e79a4c8400b208d917 : SUCCESS in 1h 15m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9d79d3081e6c433d8dca134b6b7a6423 : SUCCESS in 19m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bdfb616cc1d54305a34af3ff2e195a1c : SUCCESS in 16m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ed0989d7bee40809acac7731063938f : SUCCESS in 12m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e713447bef314eefaf02a2545d83b05a : SUCCESS in 1h 53m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43a97a160eaf4448b916aa5ac42dc0d5 : SUCCESS in 1h 25m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0424902e75d14b71a4400982fdcf7485 : SUCCESS in 20m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a7cc733842d47a1be04fd5732f58cd4 : SUCCESS in 14m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5fd866df303245e888d948d362e896ed : SUCCESS in 42m 18s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f27c42ce16ab4246a712388503c56b89 : SUCCESS in 34m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b54b83ea10f04cc4ac92a93d9389caf6 : SUCCESS in 25m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c07f763c2484a1e8a33f666f070c256 : SUCCESS in 16m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2550909da9344707a2400e7560dec4dd : SUCCESS in 12m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8fff27d091834a7d94f72f814a6b9d16 : SUCCESS in 18m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3a60eb3b84b4986abfea7a84061a072 : SUCCESS in 14m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3af7796217ee8b8613e4534683aecac554e30e0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 12:04:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2dd7cdee2e1875821d153f6f564f68878ebe623b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 14:00:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6eedf5a1a9ab4f8b8f2f0d82f04b924d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c0565deb2eb44fcbf2f124c5d8861d2 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dfad696cc7247b794f00f2cb4e48f82 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e6adb99504478abda3c7710d0f9433 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a47c0917feff47d19f043d983144d2f2 : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6dff652f02ed48cf8737509f26f26056 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39ead694eb0b460a9b5cd009d63bafcc : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/504aaaa166814c27bd04d43e3c25cd3e : SUCCESS in 12m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bf28739250534f18bb28099087cc44a5 : SUCCESS in 1h 15m 21s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f98f2a96e52d4c4787014c183ed5cf00 : SUCCESS in 54m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cce7efb11db64c329bdf617a58d0811e : SUCCESS in 1h 13m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e19952e3fe0a4802a150b14c04eebed8 : SUCCESS in 23m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d24a1d8ba5e49c3ba6b03098fee1e2b : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/281d39cd0c744d6ca93a34fff63bc839 : SUCCESS in 1h 51m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/717ff7b27a3e43aa810ad4ea155f891a : SUCCESS in 1h 31m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fc9e3c4240354def8e94f18bfe92f4dd : SUCCESS in 19m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01251633286c4e7ab6eae65b4633ea11 : SUCCESS in 14m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b4e5ed3fc96a48988ee511f650cc9181 : FAILURE in 26m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c69e96fc88b486bb1dd4dcc209e9b3c : SUCCESS in 24m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b576d6836cb14461966259f355a80044 : SUCCESS in 17m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/785a88d8c87b4822892ce29f9d207f99 : SUCCESS in 10m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ce99a52ed4b3409a95d3a178f52b3018 : SUCCESS in 21m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88df6d5776464254aa9570c6275a1abf : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"513fa28e202fc162091ecd6a8d84d66250d23d8f","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-06-11 11:14:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8765317133b99ea94a97d0b84141cb662f60a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 13:09:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c55c1e611dd644e086b2a094790a25da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef8422779eb84dd6844253edb0220c7d : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/893ea56df5c645c7a45fee88bb853ea8 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87241c0d4fd840fdaa3e6876c26a3940 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b62db49512d408b864e9b2e49d9da2a : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a45dd35a40841ca826af4e61c828068 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/123b5508c4824154892cb834376b0ffd : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28c6de2436424bf1a3b0593afbac6dc5 : SUCCESS in 10m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75522dfdc8bd48d085c387eef0ea7830 : SUCCESS in 12m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c0b09c7483cb48a8a7b77dd4fdf78b7c : SUCCESS in 1h 11m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/240bc8d8d3a34486a3e6eca0a2a9753f : SUCCESS in 1h 03m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ef0530686d642108a4e5417e0342a41 : SUCCESS in 54m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/08762af5d75949b8a0e5b5442796a52a : SUCCESS in 1h 12m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/88aef8f9961a40b88acc29a71593052d : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9e6481d782864614a4f560af33d3082a : SUCCESS in 15m 40s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/639cb0edb4f14aa3a51e27ba807b980a : SUCCESS in 12m 28s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5388b259b7dc4e1ebbf6a708c2f31b5f : SUCCESS in 1h 49m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8756e19dc22f4728a2d3e749f4912f78 : SUCCESS in 1h 35m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7efde796b82e4e0e9666538188a43d94 : SUCCESS in 19m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2e2ebe9379447f4b1ceef0af7aa9d04 : SUCCESS in 13m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e4057ad86236495b9368a822bb3081d0 : SUCCESS in 39m 20s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8255c4da784141d9acba31458e092f3b : SUCCESS in 34m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6857a44020f44e769e37e7e1a7a87b9a : SUCCESS in 26m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/34f249bffe51412da72ba4d04daf5917 : SUCCESS in 17m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0f336db3c0843348f53227c6c89a921 : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/91e5918288c7414597bcc8d8a6b74a88 : SUCCESS in 20m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b4b650e99f34d5f9f1cdb19737ba25c : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":2},{"id":"60c8fd88cb13d7d8c2f011c6fe095455de7e5bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 13:09:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"44ec0d5bac8a4ae2a118d85180a787da7e246c1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:59:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb38192ded4f4db9823448970e0c86aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5717ede94e6049a38dbbe74a8ed867e9 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6128a08a84874c228e9f7e72c78c59ae : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c4a3284d574c90a7e418608a5eb434 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aae2c1940f94ff8b573f2fbb07e75db : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/040bdc1a2e7749e79ba2075cd70d38eb : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/620cad9f244541d0ad2961af3db4a243 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6efc530612294e55a302557d5de7ea6e : SUCCESS in 13m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/308fa141f86244cf911e0b7143839418 : SUCCESS in 1h 03m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47aca3747c324959a0ef989958e41294 : SUCCESS in 50m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/319aad53eeb745abaa75b9952a50f054 : SUCCESS in 1h 11m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/21ddcace048541fe991a6aaae9b80b9f : SUCCESS in 18m 44s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96449bbc6e714e9c8781fe8183311348 : SUCCESS in 13m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e92fce952a0241e2a1422762f52c30b0 : SUCCESS in 1h 45m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df7fdedb6a764e67a381be906bd619a3 : SUCCESS in 1h 29m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/12cad04b6ef541f49bf627d50d69c2ff : SUCCESS in 20m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ae413d087394082af3fcc9b2d9c8b50 : SUCCESS in 13m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a5f8e886b35641269cdf8a0f5061b745 : SUCCESS in 32m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00596fb07eb348d7893a833aaf98812f : SUCCESS in 27m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3fefd2812ad947faa833340601f6d7a1 : SUCCESS in 15m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efd2a0097c3f4caf82382ff5b64ed7b7 : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/705834648def4c6c994709c6402bebaf : SUCCESS in 19m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9812b513d7cb4a849f14a5352c43e7ae : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":2},{"id":"b535f6b6b03fb6fe4ab6c63a28f33ad13dd94fff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:59:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f526f11581d7b5733905271afadc3b1b951f7727","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 15:00:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bece87e46c734fdc94103efd1a21a082\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e1b4b0719b4aa39d9d3e20cdf4daef : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9296789b1a634162b0802e0b5d72e386 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c733c7a7143f6da6eff2823f289fdd8f6663efc","revisions":{"48d5f487b3c4ed12c69b2f2add743f7aef36cca7":{"kind":"REWORK","_number":1,"created":"2024-06-06 08:08:18.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/23/921423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/921423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/921423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/921423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/921423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/921423/1"}}},"commit":{"parents":[{"commit":"20582e40e469b605ed0340d68da080c0c03fe76f","subject":"CI: Fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20582e40e469b605ed0340d68da080c0c03fe76f"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-05-27 14:34:51.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-06-06 08:08:18.000000000","tz":0},"subject":"Add ironic config files clean up tasks to overcloud deprovision","message":"Add ironic config files clean up tasks to overcloud deprovision\n\nCloses-Bug: #2067288\nChange-Id: I57fb4c1697a31d4f592d79a31a04507cf91e97c7\n(cherry picked from commit 7837c9b7f8d3de78cebdfc37bcf38775b7f8c292)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48d5f487b3c4ed12c69b2f2add743f7aef36cca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48d5f487b3c4ed12c69b2f2add743f7aef36cca7"}]},"branch":"refs/heads/stable/2023.1"},"8c733c7a7143f6da6eff2823f289fdd8f6663efc":{"kind":"REWORK","_number":2,"created":"2024-06-06 08:10:56.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/23/921423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/921423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/921423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/921423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/921423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/921423/2"}}},"commit":{"parents":[{"commit":"20582e40e469b605ed0340d68da080c0c03fe76f","subject":"CI: Fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20582e40e469b605ed0340d68da080c0c03fe76f"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-05-27 14:34:51.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-06-06 08:10:54.000000000","tz":120},"subject":"Add ironic config files clean up tasks to overcloud deprovision","message":"Add ironic config files clean up tasks to overcloud deprovision\n\nCloses-Bug: #2067288\nChange-Id: I57fb4c1697a31d4f592d79a31a04507cf91e97c7\n(cherry picked from commit 7837c9b7f8d3de78cebdfc37bcf38775b7f8c292)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c733c7a7143f6da6eff2823f289fdd8f6663efc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c733c7a7143f6da6eff2823f289fdd8f6663efc"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
