)]}'
{"id":"openstack%2Fkayobe~921660","triplet_id":"openstack%2Fkayobe~master~I88f20112ab4740fe114a6484f88f4fbfb66863e5","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-08-27 11:27:14.000000000","reason":"Change was submitted"},"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2024-07-10 07:53:33.000000000","reason":"Reviewer/Cc was removed"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-08-27 11:27:14.000000000","reason":"Change was submitted"},"35511":{"account":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"last_update":"2024-08-27 11:27:14.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-08-27 09:22:35.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-08-27 11:27:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88f20112ab4740fe114a6484f88f4fbfb66863e5","subject":"seed-vm/infra-vms: Add support for EFI and q35","status":"MERGED","created":"2024-06-10 12:17:11.000000000","updated":"2024-08-27 11:29:41.000000000","submitted":"2024-08-27 11:27:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921660","meta_rev_id":"5f3196dec1d161967d2238c244d8239b2e2a5a20","_number":921660,"virtual_id_number":921660,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 11:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":2,"date":"2024-08-27 09:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-27 09:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-27 09:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-27 09:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-08-23 15:10:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-10 12:18:29.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"},{"updated":"2024-06-10 14:30:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-19 10:36:35.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-07-05 15:39:02.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-07-08 13:15:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-07-09 18:18:00.000000000","updated_by":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"reviewer":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"state":"REVIEWER"},{"updated":"2024-07-09 19:23:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-07-10 07:53:33.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REMOVED"},{"updated":"2024-08-27 09:07:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8a0fb52333aa3da90fef9f0723d3bf570351611f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-10 12:17:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9627c95247b7934f5359493eccf6c401433dc0d","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2024-06-10 12:18:29.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f08beacc056168e55ca9f539ffb74de70a0e5dd6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-10 12:28:10.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fdae80e618d9507ed77ef9899286a3147a26bd88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 14:30:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab2c648d785443a09b304389e1fce084\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a944ef5309e749b399d083fa116745a4 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c504ccae6df9467ca8ec7276a10937ed : SUCCESS in 7m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/50884c0016974fb59f5a5ecfba2bab67 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b70849de34141d4bc760bf4ffcabb58 : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02c0aa8dffaa46b28e078f1189e96add : SUCCESS in 9m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c8cd991759d24e428f08ec613e94c52f : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3e7d629e4a904ceb8cc75a035bb18e10 : SUCCESS in 1h 01m 10s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f60946193073421ea3fde658210519f0 : SUCCESS in 51m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/56812115ee4240aa8d52d147aa252d91 : SUCCESS in 1h 10m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/aecd21b0cda3402483d62c0f7cf615c8 : SUCCESS in 19m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c270c50c3ae348878313749aafabc002 : SUCCESS in 18m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55ec22a793224cdead6e24c0c4235a08 : SUCCESS in 13m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c94eafdcbb3040a78b3fa02ce3424562 : SUCCESS in 1h 56m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ab4a745acae46ee85642b40397011b1 : SUCCESS in 1h 40m 33s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c865113fc5a143e8bedc69e59c6c787b : FAILURE in 1h 27m 17s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0160e2bfc6c54f98a36b3928f3de4b34 : FAILURE in 1h 06m 40s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3aba4ed4c6df4d6a878eec2d887dcc2e : SUCCESS in 19m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5743df0512f442328cf9c370cf0e9872 : SUCCESS in 13m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/65c217f35c9f4990a7e7df0934318958 : SUCCESS in 40m 23s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e643ea74d1e4125a5a84f86baa19304 : SUCCESS in 32m 42s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cea29cb4b6e147dbada69599b8c68d97 : SUCCESS in 24m 35s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/59bbb2f55800480da9f34d6c8bcd51e8 : SUCCESS in 35m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6559146babc41eb82ff9cb6056628fe : SUCCESS in 24m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad1011e84b044353aca811210fff42fb : SUCCESS in 16m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f973509cce8d4ecb9b4ea238b3f95c79 : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab413d3598ed4118b23ac08a6a401d1b : SUCCESS in 20m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/538e8a9a19c84b34a4c7a1d81db041a1 : SUCCESS in 14m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/487c11b5ee224c3d8aee97ec061087a1 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2758ebcd125f4b2884f67abcc69d1bd1 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6461d5a99a27457bbefd84b9d3fbb055 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ff1e62e56d87626acbc517680009495a9ed4d8f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-11 06:22:15.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"13df380e7bdd792f440c9091319947299055efd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 08:34:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd6f2a3bdfed4ac49d82b22497f19bf2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ee9d56ad0a64c859af3e65cab19e3b4 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe272ef651c14a9a9a34c672fb244c65 : SUCCESS in 6m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/35e2069660534a5484cd255adca624f5 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/53025819da2f4119955b4da674e7e905 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fff5508c2577462e827343552bbee38a : SUCCESS in 8m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f4f62f01c00d4c5293c6ffcf682cbdb5 : SUCCESS in 1h 10m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/15d481913e434cee987949e189c8eb31 : SUCCESS in 57m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9aff0f85d65b4a1eb9ae70e0e2b5416c : SUCCESS in 56m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e55deccd7beb4f6995057310de5e96e2 : SUCCESS in 1h 08m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3e1e5dea42f345fab8a085639ee0a173 : SUCCESS in 20m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/21700ad7758548d2a3787ce3fc70fd2b : SUCCESS in 17m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fbef596a79041d581fb0a6191734589 : SUCCESS in 14m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8ab448c82c8d42d8ba394c599ec82a52 : SUCCESS in 2h 06m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98fcc45f612e4f30a97dc504f7bcd7ea : SUCCESS in 1h 42m 29s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b259bedd0ab344349d1041265702da4c : FAILURE in 1h 31m 15s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cc51362d9824ee3bbd4af5a791c4b12 : FAILURE in 1h 09m 58s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d0fd18ff785641239fc8cc44fac8b279 : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ad8899d8df440d09310e4cd3c543ab6 : SUCCESS in 14m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1d5722fe71924135a60e0d3422bea388 : SUCCESS in 43m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a32923988b0444b9362a6bb1c0c4621 : SUCCESS in 31m 50s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/067b32cf31f54e60974e1f537954a959 : SUCCESS in 24m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ee70946a95ab470a9af4477c98231630 : SUCCESS in 35m 04s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eddca71c991e46aeab9cc3588699ff9b : SUCCESS in 23m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4857af1abd234f2e9c232fd089db6873 : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78e1ff417c1f4b62b8b92678f45d1200 : SUCCESS in 12m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cbb9622b794a4712b32ba23d6a130168 : SUCCESS in 20m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b73fec67229745ac8873d89bcd755818 : SUCCESS in 13m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4dd74a50c8148bab414c859b63d5482 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/375a5e13d3d941adaf4935989b680b1d : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/087d879842e549abb2a4f992d8d0498e : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d3acb1e4a090c6e08f983a762a5878dd79204918","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 13:41:38.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2fa96a0521cc432c22e8494272f3d43f8eda6986","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 13:41:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"01c41502cb29669b019a0f1e0d1c09d2df67d272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 13:41:58.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"6ccfc91bdca47676692a7198dead5a8c3620e972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 15:44:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43a064bbcf774717818c478e02d532b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76739321498142f38655b2d73a69e3c2 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db56e43ba4d43cdaa254ff74de34d53 : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64eda927b4f14c23a1225af34bee8d21 : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/95b6ceae93284468a2c1024e2340b4e2 : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81c7d118672b408cb1d7ec42405aa5e4 : SUCCESS in 8m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7eb6304f86664129a86786e71ec5ef43 : SUCCESS in 1h 10m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8ce74385a6da4d728a588fbcbadc080a : SUCCESS in 56m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bcda6cdc1ec42abb2fd9bd06ec6200a : SUCCESS in 56m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6fa95994481f4c6fb8a1f7234995c2f8 : SUCCESS in 1h 16m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d26c13cd00864054b030c6e4115bccae : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/882c1dd46a8c47b9ba0b662426b76bfe : SUCCESS in 17m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95040dd922c44c9bab3b9666d47c5520 : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d7eeb23f22224d46a154f03e4625112d : SUCCESS in 1h 56m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fed46bbd21c465b982cedb236145f38 : SUCCESS in 1h 30m 08s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ec39d53b5ada46c082e370d0c79791a5 : FAILURE in 1h 24m 43s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8a451abeeaa43c58dbf97723bff45b3 : FAILURE in 1h 01m 53s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/76e0c40e5c1248e290f28222704676bc : SUCCESS in 17m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbd5ec0aa9f64ede8e46868589fb1d02 : SUCCESS in 16m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3d84764305974e8790d0a4cce0024384 : SUCCESS in 38m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4bf0d94305784e9b8be81c975a1fe190 : SUCCESS in 30m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/514758db87854dd9b36c0f06bf8d624f : SUCCESS in 22m 21s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d707c673234c4d34ba1697f7fc0122f0 : SUCCESS in 34m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e0cee2b255a4dcc9f6aad0604042f3d : SUCCESS in 27m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/65024dff73c64d448b5ed5e01a57d1ad : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/423bea24d6f0403582884e2d882474cb : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/86946b7ec7144f1a9eb77879aa19dd27 : SUCCESS in 25m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db6d0fb8b4214e5bb0db2c5bc15ae69d : SUCCESS in 14m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79d9c8300af7448aa749bac545bfcb40 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee4bf894ad954cd49a88394ca9f203ee : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbad2115efbc464489895ba6bf5a9a50 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":5},{"id":"7e88ba651f26548eecdaff440b14306069050d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 15:46:59.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fb638eda0f764353ace8e790b160a78021b7deca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 15:47:40.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"5fab52d2b15403101099b83fc8719128140b8dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 15:51:44.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":8},{"id":"40d2ca41adee1c212d6ad5344ba9e69560899636","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 15:59:56.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c35b524d9d686248f4cd68b88640c249eccf9bc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 16:09:17.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":10},{"id":"661d3377a1a9096aca6addde9325a4ef8776eced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 18:07:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62ad0eaa35464401b9a0f602758a37fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd280bf5958f46848cd59c58e805c4e2 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab1d452f93ad4914a486d84d2e215b1d : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/955e9f594f6a4f578f9cf27d64581939 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72f2245a2707436eb4b6f241748f1190 : SUCCESS in 15m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/feea2d68fc7d421ab5b1a52013558267 : SUCCESS in 9m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cd77cfabbe3945d08282072eaf8fd256 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7b8d44015cd042f3ad7252a173fd9156 : SUCCESS in 53m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86a29fafd13e46ee96911f1e5c267c31 : SUCCESS in 55m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0fe7bf82402948d880f2b2136c8dae81 : SUCCESS in 1h 10m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9c0604458896408c8f6ed64f0a97aa18 : SUCCESS in 23m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d0ccb733ea0c415889a11102bd27fca7 : SUCCESS in 18m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c61012975fb045e48816e975f9ada4d7 : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/41d2d7e6f95c45b8bee7b786d060535e : SUCCESS in 1h 56m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b669f358fc748e19fc0263ee8fc617b : SUCCESS in 1h 44m 15s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6d7812db23134c60997a504f66c00c57 : FAILURE in 1h 27m 49s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bafc0def4bf44053b783f2d2b265d155 : FAILURE in 1h 11m 56s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/291526a4116f482baa8317728ca9abc3 : SUCCESS in 20m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/500c8422089042b39fd172cd599e7955 : SUCCESS in 12m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3643165617ce4a5eb3240be0519ce510 : SUCCESS in 37m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/64a3d82409544bccad795f44f2df2388 : SUCCESS in 30m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16ca22f8e2c74b4db2e717f5e7d4cf4e : SUCCESS in 23m 35s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e331464659f943eb8181e15facb71dfe : SUCCESS in 34m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75159be499534d9abfe1ba37886a4200 : SUCCESS in 25m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d073a31ec2104e13871589668aa90e80 : SUCCESS in 15m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/960b4c3d02c7423798b9e4fe1aa1086d : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/90c4654dc8114d9c992188feba6d05a2 : SUCCESS in 20m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be1205cd9c124acdae41fb51158f941d : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4ffcfc77b4149539365b3cb37cb4868 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f77693716f034ffe889bcaae84f8e080 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f08226cc6ba4e719005afa5ce656f74 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":10},{"id":"f5d53b68670f9b51a1c008fe63bf57cc121a13af","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-19 10:36:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6ba8768b50a2cdc412fe3199d56853e7f5dd8b78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 10:57:20.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"a5c1982aafde45b03178de552ae4820cfc6de96e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 11:17:39.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0cf31e1ce10cb30ae755cfbb202c32f4fdddbd5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 11:36:36.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":12},{"id":"1a7e50fb49c4aca95e7480bc5a724262a82f7196","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 11:36:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":12},{"id":"1d9c6ab9de2bf7eb389fbc99c1425e60e506adcf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 11:54:34.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":13},{"id":"a1075b20bc7b294002f47ede0dba763d54f985d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 12:21:50.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":14},{"id":"8e936383b29c8e4f4ee6258afe6a8a428849dfbd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 12:44:04.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":15},{"id":"8c07c1170073d0282ecbdfed3e480d444a640af5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 13:03:07.000000000","message":"Patch Set 16: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":16},{"id":"fe13e1f0946209f303951245d505635fcd224034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 15:12:05.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ede3388814cf4d5bbeba38ea571675c8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73cc45ade29f458991bfc783add42c10 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936b8c84eae9458281854d8655c4c3f8 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e825c91745c44a1874326c2643dd23c : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ef34d5a25ba4d3da12ef2a574598dd8 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad3ebca252094f55a730b7eb35a97035 : SUCCESS in 8m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/81fec56b652e4acba37f31d717b48213 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/98c3c22f6b0b4b0ba7a21ea365045308 : SUCCESS in 1h 01m 09s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d25066b88c3b449485522416aebc185b : SUCCESS in 55m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f2389406a8a64d28a12832c9401f9436 : SUCCESS in 1h 19m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/84767b811171420ab0883d3739d8d5d1 : SUCCESS in 23m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d64fb6da99734dffad99d14ed727e4cd : SUCCESS in 18m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8008269f87de4b67b3a94860dc372edc : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e63ffcb77209476f8597bfa95a9100b0 : SUCCESS in 2h 06m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab4100ac2a304146b3800763e4d7c3ec : SUCCESS in 1h 50m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/20135c11ea9646db8f4d8138ef3233d3 : SUCCESS in 20m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3a05c3a10d5425c9e57c16574e5eca3 : SUCCESS in 16m 19s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/de6e226156474abe9828415e1e81b63b : SUCCESS in 45m 15s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ce8471864dd542368a617bfad5864aaa : SUCCESS in 33m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19fa542f0afe479a9d52c856ab26351a : SUCCESS in 24m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9e11109f06b548319f12b274320fb139 : RETRY_LIMIT in 6m 53s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f83a0c88792e4ed89f45aeaad8d4bf01 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cda4c8972ff446018b104830aad3cf9a : RETRY_LIMIT in 3m 36s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/7eeb7863e3b448e497a4539f23f442a5 : SUCCESS in 10m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4559e1ac71b14c7d8edce2f2c2fbe16f : SUCCESS in 21m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa43fbe639d04abb832a1663e2ae4e08 : SUCCESS in 13m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/050a818b67674288b972fc817cd3f500 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a3b823200fc4557a9ecebd9ecde40cc : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b421998235e64e37bc5aa18ba4ccfba3 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":16},{"id":"5368f41b3f055b5eea42169229daa0e90eea93b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 19:14:01.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"768082b4aa45119145d4a5744ae654389a451e1c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 19:14:31.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":18},{"id":"2719a1af9a32c63a148cbcc3c0cbbe83847ccade","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 19:15:31.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":19},{"id":"6c6e31914c50986dc299ebc51663b958348ec8d9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 19:17:02.000000000","message":"Patch Set 20: Published edit on patch set 19.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":20},{"id":"531f940de3e7e5e348e3b96460b1441c6515b6b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 21:19:36.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aee6798f4e6a44a187ade73f61799929\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e6318fcf0fa4f45b552ab61b31eec95 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656d338234d0470e96685f3c6934c042 : SUCCESS in 7m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ecc879d3ef5b46b69b5ec653d554a8c2 : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93b2d9c1979143588fd5ea93d34f2a0b : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad52842aff4649aabedb9f0624827fed : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/360c314399ab4b2ea7ee8312fbf74270 : SUCCESS in 1h 06m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/164f83a79a4c4fd9986618fb6045e581 : SUCCESS in 1h 08m 41s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33f24896eea741558ca82f3c01b6f0f7 : SUCCESS in 51m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5f3417acafb64b989ec1b76d8599fa59 : SUCCESS in 1h 14m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b480bffa12204452b30f7a1ad4011951 : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b5877b169900416c8dd5ed10d20860f9 : SUCCESS in 17m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8de41cd41e7f4c0791ce8fd5dff178c6 : SUCCESS in 14m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5f540cd2863040e7ac27b9f68791571e : SUCCESS in 1h 56m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aa7720fc6414c05a35917c962b40f51 : SUCCESS in 1h 36m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ed2e14964283477f8f210d258c328c56 : SUCCESS in 21m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f2a63a2176941a79c063e68d9429ba7 : SUCCESS in 13m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d64450b863034d2f94034332097067c5 : SUCCESS in 39m 50s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1b71df69f60149408cd69ea9dd1211c8 : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f64df4362814e4c87f173e24adf265d : SUCCESS in 26m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f27928641a814a22a68125839db13add : SUCCESS in 16m 30s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/036eed9e2c214c17a33156c8b54a6dff : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05747faf5eb34bceb77272f2f90fa58e : SUCCESS in 10m 36s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/cee885de98ab453eadc8a698ac84fddb : SUCCESS in 10m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d2877246583847af80a0b9651bde10f8 : SUCCESS in 19m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2d5398911eb42468bd03535fd15f017 : SUCCESS in 13m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a047d4fe9d743c0a90016b5fa41b0a2 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc504d3a01944613b48931b1c5ef4fd1 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1043d413b4fb44a39ded96dd48084d05 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":20},{"id":"66ce48c0c424779ffe7c057dcc2dcdc5492f451e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-20 08:24:52.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"eb9c300a25701146a8803018955acf2bed2fce35","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-20 10:47:30.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"3cf14f2a5b415ce5a16fce75ca179587a0dec54f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-20 10:50:31.000000000","message":"Uploaded patch set 22.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":22},{"id":"9fea67de6c4c9cf13723ef541463f0467e65bf47","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-20 11:30:18.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":22},{"id":"d9f36d87c37d42f26877d649d02bfd5af7219971","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-20 11:31:44.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"e93c900d6a09e6eed5d236b1f1fac5eec6029adf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 13:11:37.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55c21b4415404b798a16ba361d0bc258\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e4c657f510f401291c10434a9ca8a1a : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311dc30b14a942daacb485e08c87a25d : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c560d3f328c94868b28bb39ddc4ae8d2 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00dd52fdb6f6434e824d9df88c75afb2 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/52acd65c05054b25a1ecadc7572ce82a : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/514bebb1f9b44e48baf709add6c8b2fe : SUCCESS in 9m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/32f317a98d414aa6b196540449bbdedc : SUCCESS in 1h 03m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/938f8588346e4d089da43e8999105609 : SUCCESS in 1h 02m 56s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95c780a8ad5849088e51cbfb02b58d3f : SUCCESS in 52m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8d389f976f4c474ea008c40e21d7e12d : SUCCESS in 1h 10m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/20ef034a9b0b49babde157f0a17b0856 : SUCCESS in 23m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6ce086136fd24c6ea118c8ce4c330d69 : SUCCESS in 17m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c148dd5bf4204c58bef9da2ce24d10d3 : SUCCESS in 13m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/885858c4535c4aa797fd5eeaea4efaab : SUCCESS in 2h 14m 28s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ae77c8446974b39ac1de7552cfadc4b : SUCCESS in 1h 36m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/41008538c0c34b73ae087aca02008fc3 : SUCCESS in 21m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9c1ee4cffa040d9981153b522bdd46e : SUCCESS in 13m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ea050957b44749a7b832ca31ec0f16ae : SUCCESS in 43m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bba3cd80370947adaa9a8b078b36ace2 : SUCCESS in 32m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcc93342b06f43519f9d484dd3b3e4ea : SUCCESS in 25m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f7bb5f5ecff47d1b942839b2a1f9553 : SUCCESS in 16m 56s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/51c2553b5a2f4e3b9f2309480ee27d02 : SUCCESS in 17m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fde4409110cb41fba7527b780a084266 : SUCCESS in 10m 49s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/aa436ba6f4844b59a9f9904ad7476902 : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c4e73ee161a748df822f5a77bc4386c1 : SUCCESS in 21m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cf6090b7cb64f6994e76440943b8231 : SUCCESS in 17m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df9d5d381582457ebe03a5ec7fa7619e : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d33a2f18b8f4b1eb79a69e777d350d0 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa7548eca7e041b4a9be1096f960b306 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":22},{"id":"21eb4e641f9856cdabc920eaabdcee095d9ea314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-20 15:33:33.000000000","message":"Patch Set 23: Published edit on patch set 22.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"4f9286810f6627004a828fd2166242394db1825a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 17:44:12.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8165d13743e345fba30250ba6fb35477\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c597f24a0d3c4c4283585bb5199a120c : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92dc3871bc6470bb3cd64c8aadef0fd : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4e288c8143940a4b6f8f248824d435a : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/143968772aae40d886b88ac86e980bb2 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3b6a7cba391f491fab7414d11f618627 : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a00af34ec284fd6867f31ea6a39688a : SUCCESS in 9m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ee12b4deef934993938b9af3be0ea22f : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e0652b62784448109308df2715628cfe : SUCCESS in 1h 06m 06s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f31094116ed4bd0a33b3532aba8315f : SUCCESS in 58m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0fc6dd93e1bf42c39e9d592bdacdfeee : SUCCESS in 1h 13m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c27a0a30f3294b4cb68b1df3f2f335e3 : SUCCESS in 19m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fa33b4adfe1f47c7a2b8d1ce1374ff06 : SUCCESS in 16m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35c4b99e885b4dbbbfef18f5db4d1046 : SUCCESS in 12m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b2c85bb6986b435d95e99cd2eef50a44 : SUCCESS in 2h 05m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfb60c8eddc148cb8a96a660192c7f8d : SUCCESS in 1h 44m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ee6f4bff32314f7f9cfef9e653ca2967 : SUCCESS in 19m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cd57eb1dde04e2f8d43f1f743accf03 : SUCCESS in 13m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5ba2e1dead28423ebcbd80dfcd95aae5 : SUCCESS in 43m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0ad8ebdbf65740cdb05a39fc10eddc62 : SUCCESS in 30m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1f914952b3947d5854c705abeb57a2a : SUCCESS in 27m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b75b7aade86d415da10a2780469d2cac : SUCCESS in 15m 15s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0f1d19508f224bd8b273b36e50cd3acc : SUCCESS in 16m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ecacf360ce64162809b8155e32ecbba : SUCCESS in 9m 42s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/f26004e487d740f181944bf02e32786c : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ce807f3735f7492ebaab6cbbe36a9125 : SUCCESS in 20m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e76908f616464bd987f81c4370d4e38b : SUCCESS in 14m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a619044877a64db7b972b7b698795923 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b4b3c7850974315ba53f24e68a2957b : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ecbd14bc6fa45868f2ee8d34c192a15 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":23},{"id":"c54bf777870fde626acfab36b1baf83350ff73cd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-05 15:39:02.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"10d91951c8d7581344aa6a2eeb5d77a1ac2ce520","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-05 15:39:11.000000000","message":"Patch Set 23: Review-Priority+1","accounts_in_message":[],"_revision_number":23},{"id":"759145325f77175f341a9da346a2d8267699590f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-07-08 13:15:25.000000000","message":"Patch Set 23: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"ea53866c1d4b7a4f5e0a229f0fb4d3319ba64d75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 13:16:11.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"ba3a8da2351e0b0dd2349494b76f69128b369aae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 14:40:20.000000000","message":"Patch Set 23: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fc25d33ede846f8aedfc16be81016a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0816a50dc4954fedbdb7f4b79cf877c0 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d85c298a1e142e69cc6f860594c3761 : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1816e8c17a444d719298d7b86c972c39 : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/523c52a6bb9147d9af65efa49ed9b7d5 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0a3203da1474ac9905f885ccd428e0c : SUCCESS in 8m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a1af1613533848a7a389928d25d4c3af : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e2b3c6acbba4df0967c4f556d9a5453 : SUCCESS in 53m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5c3b4e844d164059b3a1d2047dd908d3 : SUCCESS in 1h 06m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7dd46985eb1c42a4a661753307300fd5 : SUCCESS in 22m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae374eaffafd42f1ba3c0b86e4049447 : SUCCESS in 13m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a0a32bbbfe8947b1a64e57a0c44e7434 : FAILURE in 1h 16m 48s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ae4eb74ab0546888a6c2d79b7ba127c : FAILURE in 58m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0225fd36218e4a58b1b834778affc9c6 : SUCCESS in 19m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/654c170313d74a11a5e3f91da548c839 : SUCCESS in 15m 54s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8ebfdd8de61d4253b2601cf1002d7f2d : SUCCESS in 31m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b40c970a6db4f339c6bc62209856fcf : SUCCESS in 24m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15690e67035c40f7b20c70400eda925c : SUCCESS in 14m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c44d09e346de49c59cfc59d6597d0b1d : SUCCESS in 12m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4e8706a1ff7b49fe8760ff9d1ba83ef6 : SUCCESS in 17m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c90e7d9af4194da4ade76f379156815c : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":23},{"id":"cbb5ae49e0e555f3462503144d99dbe6789df4ee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-09 06:46:38.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"418c2c500131204dffb5095dfd4e009a1b5e897c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-09 07:04:26.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"102a68885ae4efa9d7016b465f38bdaf4f60d3aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 08:17:17.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8c44df88a804b1080c59ff39991417f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59f9d820cdfc41c39aed06d3efbb000a : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54dd481223844a5682dae481245726f3 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7792d609d92b47c493345c69df02355d : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d8c76a3ef7b4b30b240f19e02b14c0a : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b0eea2c4e005459bac8dde9dd7c6844f : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2cd89eb380fd49eeaa2a54406f3f6501 : SUCCESS in 9m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c64d160f4ec04611a989eece7e9b490a : FAILURE in 21m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b8f753c9ba544df1aaefbc0fab0ee90b : FAILURE in 18m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87a7a9ec4a9e4302bb263443572aa069 : SUCCESS in 55m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e46e65cbe34842d380df515f366c191b : FAILURE in 24m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8f1c7993eb9d439e9d51b96377779a26 : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/eae1b4dd68244631a2204c5addb22cf0 : SUCCESS in 18m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6633ffb7f46540ccb644579565322b93 : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/636cd8d3502d4e3eafbd14c1edb1afb2 : FAILURE in 1h 19m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29d36847699949dcb7e589723046021b : FAILURE in 1h 03m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1721d1d27a4d45a9bc8bce2911c53ffd : FAILURE in 15m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69d8a85d1f514ec6ab9428c4445d284f : SUCCESS in 14m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bd7cddfe6d16482e8a4a69434481b913 : FAILURE in 12m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2799463c5e694002809eb7ec5fa94798 : SUCCESS in 32m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a365b9d7b094be5b943a30a6e4259b4 : SUCCESS in 26m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/04d8a8998817440ba4550eae5085fb9e : SUCCESS in 17m 22s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/809970f0810c4119ad8a736499e2964d : SUCCESS in 17m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c806c684bfa4ec3b90be6d31221f7aa : SUCCESS in 10m 20s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/277d491809df43cba385206051b2ad19 : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c3365cd4aba44fc8877645cc5799cbc : SUCCESS in 23m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aa5d6baea34419e8809d161271fee65 : SUCCESS in 15m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8e7da2fcb6c42878e7fd8768f185dbe : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56bd183e26fa482eba2059226f79bcec : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2454d88074234d359c8d9424c32a68e6 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":23},{"id":"c2a76145e095ac50664933bdb8b3e29c214b3837","author":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"date":"2024-07-09 18:18:00.000000000","message":"Patch Set 23: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"163771237012d673364bfb646b756d926aaac61f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-09 19:23:55.000000000","message":"Patch Set 23: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":23},{"id":"0ef2148a1455023e3248131af76066469c1a8c94","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-09 20:27:58.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"2be78ea3b9fb201ef161487b85a4af027b53c794","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2024-07-10 07:53:33.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_35264\u003e with the following votes:\n\n* Backport-Candidate+1 by \u003cGERRIT_ACCOUNT_35264\u003e\n","accounts_in_message":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}],"_revision_number":23},{"id":"e03d0b20640613613179e4a8837f88e48a704966","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-22 13:05:52.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"0dd7138c7fe6378735565062eeb96f5e9d18dcb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-22 15:36:31.000000000","message":"Uploaded patch set 24.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":24},{"id":"ae9b38d8991c1934874675b4da29077afcb380e1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-22 15:50:38.000000000","message":"Patch Set 24:\n\n(7 comments)","accounts_in_message":[],"_revision_number":24},{"id":"74cc2a9b0b60e0e74aa696c7e7ab057e4c26090e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 17:24:36.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fcb4d5837f84ac18db2184672450ae2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e80f4acb7acf4e53914829e69ffb85a8 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5975fb1bb44a61a6dce4ced26987e7 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0088cd979484f15ad34b4a165a6aded : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8576869c7d274ca6abec4040e54c6171 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/89fb5a8aad0c445f842366548dd9d9a1 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3b7bdb4fbbf4b4099cec9b963ca7b00 : SUCCESS in 9m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/83002713858b482385c0e7dff6e83a26 : SUCCESS in 56m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/377b0c9dfbc943b3a4919f21990cc097 : FAILURE in 54m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b5ced0609b3450284ce108fecfe51d7 : SUCCESS in 54m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/863e9eb7cee0441789e5fb26359454ba : SUCCESS in 1h 02m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8ac25d3138024c9cac6df574a6d167a0 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4f231181e4c24397b7f2113fcfda3351 : SUCCESS in 17m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7048a12a55b4cb698a7b3a20e549932 : SUCCESS in 14m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/77f8f692661b4d23a1ff186d54f994ba : SUCCESS in 1h 42m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c3a1c5a91a74334b67092c633a2ccb8 : SUCCESS in 1h 36m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a3bf907b8e87400cafc274de7a431155 : SUCCESS in 19m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffc7ba1128d64dff8dcfca84665706b6 : SUCCESS in 17m 19s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/38c425d11eb24b8ba985c4a4cd5d98ac : SUCCESS in 46m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6d734f6cd5d94a1dbdd41b420ac9d4ba : SUCCESS in 29m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51ed5a7ac19f4de8aa2bcb2bfec965cd : SUCCESS in 22m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df9c1995ce6841d28c5c3734b0cd3a85 : SUCCESS in 19m 01s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/52786a396f724808b344fe8c3202b6f8 : SUCCESS in 17m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7984d633e3a54763a558dbb758994581 : SUCCESS in 12m 14s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/15736131af624b71bf72c51e0fe9c37b : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/83e2e2fb24b24f7cbf159dcbfe302143 : SUCCESS in 21m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afb08f8d479e45b09bf75faac83511d6 : SUCCESS in 14m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56cad4deff9845deacdf66b04ff43df3 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc2bbcba689a4e5492bfa86b4f251221 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4091107cd0c04a0ab1bffbf7441b4557 : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66b6008a7f5644388195c5fc5c497c4a : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"f72b9ad4b55da810d42bb2a46340d1ae2d35f823","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 10:45:00.000000000","message":"Patch Set 24: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"878a5ecccbda6c7d5bcdf630048fd621306ebdcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 12:53:16.000000000","message":"Uploaded patch set 25.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"2a3227528441c39e76dd63e1ce2f9ffe97a3fbd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 14:53:41.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02e1602435204f43973e98618be15c22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcb9f00e65474201b2ff3ac613f33ff1 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c18d49f1c31496eb8f80ee24b5a0aac : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6332ead1ab54f19bee56a0de1dccd00 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0e8bae90a83491e8cee7ba9c67c4957 : SUCCESS in 5m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63487e17f94945c99f30700ec85d7f1e : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc80a1781c146c78deb96f0113d930e : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/019456c829da4e8990713d4a982b5f80 : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e63a34ef9ecb4f7f8c0901afb054f768 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/656c3922ef0d476d8da7aa2d73d8272a : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac4983ada3a64e308cd12efc79757ad9 : SUCCESS in 8m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e8cca367892444fa8876da830cab3944 : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/328c90ee738b48dbab8f0359bb2e40d0 : SUCCESS in 52m 39s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1248871478b246a0bf2ebfb6a62d9e7e : SUCCESS in 46m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fc35a3c1118747649e916fc10d902f65 : SUCCESS in 1h 11m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6edb656f57514def8cc0fc947d02c0cd : SUCCESS in 19m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f4c39f97ad2b4363afd0292d708ed329 : SUCCESS in 17m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abcbe11749fd4fcca1b56e352b850495 : SUCCESS in 14m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a13ed568525646c7a7496915dfcd28b7 : SUCCESS in 1h 57m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fa0d3d85c1748b49dde2b5a476d39e8 : SUCCESS in 1h 27m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b155319d7491462db3875b45a34bf874 : SUCCESS in 20m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7289a4b1e9842c18ef2ea34cd1aefb0 : SUCCESS in 16m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/33224c9724784b76add0732ec80f9db4 : SUCCESS in 40m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/303e9c0a692c48d5a0d906ac3aca14da : SUCCESS in 32m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f588fcec3a64d37933aad634129ef54 : SUCCESS in 25m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ba3268332514da18489392c0946e808 : SUCCESS in 15m 51s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b5fe94485e574626ad604e9dd176285f : SUCCESS in 16m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1b36f8dc8804c2cbb2459f2ac206da9 : SUCCESS in 11m 23s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/713356dc48bf4c0bb88ce6544f7f45f9 : SUCCESS in 9m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/49735cc2cbd9466fb9be6c43e7a9d38f : SUCCESS in 21m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5519983e1fa495c8a0a4947ce29e0ab : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":25},{"id":"7e2d9b36f20c5fac2cf081f5064c6e9af97a693e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 15:09:29.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"3f61c0a647129cdc23f0f7322acb37c671c049e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 15:10:36.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"93f464eea6d52f3989f57a4090d5daaa349ff52f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:25:02.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af62db5222284bae972400650f2e3124\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10a8b3f4862e4939bf0ef342b263a207 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e45fe6459645fb8bd83c99a9b02d35 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91e30895c31744389b6e7a8e6c84ba93 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c346ffeb10d648d480dee92da4bfe71c : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff7bcac0ec4f4a65982e7674d1905cc6 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/110919da5097445e81bddab2e39da692 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/578d60a8101340d1948f863e3bbcdd3a : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ef695d4b42440679c4c27a0de7d62c9 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9456e25ff49a4c73b556d93414082f18 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6be14042fe141c19d0d49e0a525d55f : SUCCESS in 11m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a13267c437c846bfa225b954fc2a4c60 : SUCCESS in 1h 06m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d943f4dfd91d4d65bb24a98a9eebefa5 : SUCCESS in 56m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67c5d654cb834ca5872a98b1dd8d5bd6 : SUCCESS in 50m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d27c8e1b50e04175ad2393de57f7b153 : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a92b8a68c5194af9a1b834659f7be686 : SUCCESS in 18m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/65f9896f3c474914935b38fa9ed60e71 : SUCCESS in 16m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58ed4d0b5e7643f1a9422e828408a6b4 : SUCCESS in 14m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/35d1141136db4cb58818be5fbaffda00 : SUCCESS in 2h 01m 44s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5696532d8ad4b8c959fb0356f20049f : SUCCESS in 1h 50m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ba41f29d9839460cabe1bea3934f7b2c : SUCCESS in 22m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42bc4bbba688488ea81cc3aa722f97ef : SUCCESS in 16m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/61a387b7e2a5447d92b135df3188d5eb : SUCCESS in 45m 52s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8695f0e867344ca2b3540ddb82c173b4 : SUCCESS in 37m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ffc14b6eef84fadb0ef76baff752392 : SUCCESS in 29m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/69c680dbc47c47d18f25a6d3a4803ed2 : SUCCESS in 19m 04s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f3d383857c154bd9848ce901ceee497b : SUCCESS in 15m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a81db1c73b334075918e2a4bf3404d72 : SUCCESS in 11m 15s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/745ab649af934ade9e5304e14ef00cb1 : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d5e2601d83a0489a8ed8146ff5968dd5 : SUCCESS in 20m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63661b8670654be0812f46d127b7af5e : SUCCESS in 15m 01s","accounts_in_message":[],"_revision_number":26},{"id":"03b3275f5f96e942176a1ecb4faf446d6d649143","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-27 09:07:24.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"318c421a6cf619079ba4445b2cbde8601c814c46","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-27 09:22:35.000000000","message":"Patch Set 26: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":26},{"id":"9e7c18973f3ff4de8d722b5fbf65f2325959b12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 09:23:09.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"6c79c7a8c58f56093ff52769b747129267de3e30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 11:27:14.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8246ed630dd744f58decb65e6a4631fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad14b6cb3518421b81247269f2651951 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ccb3d7553214b5d818b0a13e2810dc1 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d741556c1ce9466589765060e3da8b84 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7b5f3fa27048f1999d075edc386fc0 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f966e47f430475ca2cdf5ae75a72560 : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9de7195874b94e298b6d9332603c01b5 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ff9fd237c0e48a4bdcd0d0be57777d9 : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f414506281b415db4371124cec2a303 : SUCCESS in 10m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/458392699c84426895bdbda69a94245e : SUCCESS in 1h 06m 45s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a3eadeb5339471dba72f64aeae25117 : SUCCESS in 1h 01m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/92aa581ccb83483b8a89dd87f3a4d4f8 : SUCCESS in 1h 12m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/01c3cf6f6204469c8317ccebea92bc51 : SUCCESS in 22m 38s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7585f37110904056bcaa37b621d89e48 : SUCCESS in 16m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/315fa618c4f243739433c9f4b05adf98 : SUCCESS in 1h 56m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd2666d4c31e42029afa86d8e8fcfedc : SUCCESS in 1h 35m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/71c4891eeaee40deb25ab09dfac99549 : SUCCESS in 24m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/111b1760bf574822988d27ff883abc5c : SUCCESS in 15m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8e3e1382a0e14d239a1f4f7c2352ae4b : SUCCESS in 36m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42912d2bf8c949a9a4c49f04d65deae4 : SUCCESS in 26m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a6d53908001e45d4a8e7bfc3309d2756 : SUCCESS in 15m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/859d39706da4437cbae47f308f78cb96 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/13df6d54e1f941e8a574ec28782925b9 : SUCCESS in 23m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3aea47c027140a783e23d4641a84766 : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":26},{"id":"acb70c8540e5a42c9e13ef27a47d8231fe2f2053","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 11:27:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26},{"id":"5f3196dec1d161967d2238c244d8239b2e2a5a20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 11:29:41.000000000","message":"Patch Set 26:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85b065002aea4ac58f4774267f1c202a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87243982dce54ada98b6791c0b718e36 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6365913785c48628bf2fe4cd9715ef0 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"9702736b3ec3c0dc44657b34c90594ccc3c75493","revisions":{"5897e65f975ac8e5139346d5db17c295bd27b555":{"kind":"REWORK","_number":1,"created":"2024-06-10 12:17:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/1"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:17:08.000000000","tz":120},"subject":"Bump stackhpc.libvirt-vm to v1.16.0","message":"Bump stackhpc.libvirt-vm to v1.16.0\n\nAdds support for EFI\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5897e65f975ac8e5139346d5db17c295bd27b555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5897e65f975ac8e5139346d5db17c295bd27b555"}]},"branch":"refs/heads/master"},"990854bc5bd506512b8951b8ec72eb1964ee225c":{"kind":"REWORK","_number":2,"created":"2024-06-10 12:28:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/2"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:27:45.000000000","tz":120},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/990854bc5bd506512b8951b8ec72eb1964ee225c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/990854bc5bd506512b8951b8ec72eb1964ee225c"}]},"branch":"refs/heads/master"},"e0e3990109709a891a582c85ba9dab699f0a054a":{"kind":"REWORK","_number":3,"created":"2024-06-11 06:22:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/3"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-11 06:22:15.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0e3990109709a891a582c85ba9dab699f0a054a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0e3990109709a891a582c85ba9dab699f0a054a"}]},"branch":"refs/heads/master"},"f89418ac33324a2a778961e3300db59dba1c545d":{"kind":"REWORK","_number":4,"created":"2024-06-18 13:41:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/4"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 13:41:37.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f89418ac33324a2a778961e3300db59dba1c545d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f89418ac33324a2a778961e3300db59dba1c545d"}]},"branch":"refs/heads/master"},"870c8a891d91dc91a11f7122ff9603e6d6ea2b61":{"kind":"REWORK","_number":5,"created":"2024-06-18 13:41:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/5"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 13:41:58.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/870c8a891d91dc91a11f7122ff9603e6d6ea2b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/870c8a891d91dc91a11f7122ff9603e6d6ea2b61"}]},"branch":"refs/heads/master"},"d51455ba63e6067e8a32e855ad29cd43d5490159":{"kind":"REWORK","_number":6,"created":"2024-06-18 15:46:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/6"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 15:46:58.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d51455ba63e6067e8a32e855ad29cd43d5490159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d51455ba63e6067e8a32e855ad29cd43d5490159"}]},"branch":"refs/heads/master"},"e639ba4bd45e252a13a8829491b7b56fa500e6b7":{"kind":"REWORK","_number":7,"created":"2024-06-18 15:47:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/7"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 15:47:40.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e639ba4bd45e252a13a8829491b7b56fa500e6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e639ba4bd45e252a13a8829491b7b56fa500e6b7"}]},"branch":"refs/heads/master"},"bfb6e97498a984776278db5d39ebb23f8a032c4f":{"kind":"REWORK","_number":8,"created":"2024-06-18 15:51:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/8"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 15:51:44.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bfb6e97498a984776278db5d39ebb23f8a032c4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bfb6e97498a984776278db5d39ebb23f8a032c4f"}]},"branch":"refs/heads/master"},"5e8c93f6600af187d72da1f625dbc5ecb5ab70d9":{"kind":"REWORK","_number":9,"created":"2024-06-18 15:59:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/9"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 15:59:56.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e8c93f6600af187d72da1f625dbc5ecb5ab70d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e8c93f6600af187d72da1f625dbc5ecb5ab70d9"}]},"branch":"refs/heads/master"},"d38efac002579be331f188e96b5d7bf5f1208771":{"kind":"REWORK","_number":10,"created":"2024-06-18 16:09:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/10"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 16:09:17.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d38efac002579be331f188e96b5d7bf5f1208771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d38efac002579be331f188e96b5d7bf5f1208771"}]},"branch":"refs/heads/master"},"a58cb177c4ed3fd9924a50b170b5dd989cb2d01a":{"kind":"REWORK","_number":11,"created":"2024-06-19 10:57:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/11"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 10:57:19.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a58cb177c4ed3fd9924a50b170b5dd989cb2d01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a58cb177c4ed3fd9924a50b170b5dd989cb2d01a"}]},"branch":"refs/heads/master"},"e6430d3105651f8ecede2f7c25fdc412d792243e":{"kind":"REWORK","_number":12,"created":"2024-06-19 11:36:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/12"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 11:36:36.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6430d3105651f8ecede2f7c25fdc412d792243e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6430d3105651f8ecede2f7c25fdc412d792243e"}]},"branch":"refs/heads/master"},"fefb0de5d1d2b630434ce709d18296974bbf2ced":{"kind":"REWORK","_number":13,"created":"2024-06-19 11:54:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/13"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 11:54:34.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fefb0de5d1d2b630434ce709d18296974bbf2ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fefb0de5d1d2b630434ce709d18296974bbf2ced"}]},"branch":"refs/heads/master"},"4f14530f6758aabdc28fb95c47a10fd6284269ea":{"kind":"REWORK","_number":14,"created":"2024-06-19 12:21:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/14"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 12:21:49.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f14530f6758aabdc28fb95c47a10fd6284269ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f14530f6758aabdc28fb95c47a10fd6284269ea"}]},"branch":"refs/heads/master"},"223d06306f9e89df38d9d4212282b84bbc573ee2":{"kind":"REWORK","_number":15,"created":"2024-06-19 12:44:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/15"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 12:44:04.000000000","tz":0},"subject":"Add support for EFI booted seed/infra vms","message":"Add support for EFI booted seed/infra vms\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/223d06306f9e89df38d9d4212282b84bbc573ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/223d06306f9e89df38d9d4212282b84bbc573ee2"}]},"branch":"refs/heads/master"},"d2e2c55de378173371f4b94839602e1d6b6c437c":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2024-06-19 13:03:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/16"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 13:03:07.000000000","tz":0},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2e2c55de378173371f4b94839602e1d6b6c437c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2e2c55de378173371f4b94839602e1d6b6c437c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5ba43bb4d3061236022ca1f6b0477c3546dc14b0":{"kind":"REWORK","_number":17,"created":"2024-06-19 19:14:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/17"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 19:14:01.000000000","tz":0},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ba43bb4d3061236022ca1f6b0477c3546dc14b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ba43bb4d3061236022ca1f6b0477c3546dc14b0"}]},"branch":"refs/heads/master"},"e9e1f5ba7760488605aabc64280f303f44963034":{"kind":"REWORK","_number":18,"created":"2024-06-19 19:14:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/18"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 19:14:31.000000000","tz":0},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9e1f5ba7760488605aabc64280f303f44963034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9e1f5ba7760488605aabc64280f303f44963034"}]},"branch":"refs/heads/master"},"d559b613706430c020a86d2539c6b60b1ec1cb33":{"kind":"REWORK","_number":19,"created":"2024-06-19 19:15:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/19"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 19:15:31.000000000","tz":0},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d559b613706430c020a86d2539c6b60b1ec1cb33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d559b613706430c020a86d2539c6b60b1ec1cb33"}]},"branch":"refs/heads/master"},"2931f664ed06d9bf7824e65d310990da64c5466d":{"kind":"REWORK","_number":20,"created":"2024-06-19 19:17:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/20"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-19 19:17:01.000000000","tz":0},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2931f664ed06d9bf7824e65d310990da64c5466d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2931f664ed06d9bf7824e65d310990da64c5466d"}]},"branch":"refs/heads/master"},"0956ea9b3b13bf9957a4cb866ba54f112687ca97":{"kind":"REWORK","_number":21,"created":"2024-06-20 10:47:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/21"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-20 10:47:30.000000000","tz":0},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0956ea9b3b13bf9957a4cb866ba54f112687ca97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0956ea9b3b13bf9957a4cb866ba54f112687ca97"}]},"branch":"refs/heads/master"},"4ba4a60bae3782f7160062f805e49864aadea3e0":{"kind":"REWORK","_number":22,"created":"2024-06-20 10:50:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/22"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-20 10:50:24.000000000","tz":120},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ba4a60bae3782f7160062f805e49864aadea3e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ba4a60bae3782f7160062f805e49864aadea3e0"}]},"branch":"refs/heads/master"},"28b604ff2cac16d2bbb2eac2593fe299c289c291":{"kind":"REWORK","_number":23,"created":"2024-06-20 15:33:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/921660/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/23"}}},"commit":{"parents":[{"commit":"f8f8cabc4ffc14db1441b1858a34c48d4fe06f36","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8f8cabc4ffc14db1441b1858a34c48d4fe06f36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-20 15:33:32.000000000","tz":0},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.0\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28b604ff2cac16d2bbb2eac2593fe299c289c291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28b604ff2cac16d2bbb2eac2593fe299c289c291"}]},"branch":"refs/heads/master"},"bd3070e131189a973ecc9536b50118aad1e85507":{"kind":"REWORK","_number":24,"created":"2024-08-22 15:36:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/921660/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/24"}}},"commit":{"parents":[{"commit":"45518bbf36e35e30b0d8a0d50403a8ee980a4e26","subject":"Merge \"Remove unnecessary dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45518bbf36e35e30b0d8a0d50403a8ee980a4e26"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-22 15:34:33.000000000","tz":120},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.1.\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd3070e131189a973ecc9536b50118aad1e85507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd3070e131189a973ecc9536b50118aad1e85507"}]},"branch":"refs/heads/master"},"045c04f97c4e63048f4b1f8559c8fbda5dc7e18a":{"kind":"REWORK","_number":25,"created":"2024-08-23 12:53:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/921660/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/25"}}},"commit":{"parents":[{"commit":"45518bbf36e35e30b0d8a0d50403a8ee980a4e26","subject":"Merge \"Remove unnecessary dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45518bbf36e35e30b0d8a0d50403a8ee980a4e26"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-23 12:50:48.000000000","tz":120},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.1.\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/045c04f97c4e63048f4b1f8559c8fbda5dc7e18a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/045c04f97c4e63048f4b1f8559c8fbda5dc7e18a"}]},"branch":"refs/heads/master"},"9702736b3ec3c0dc44657b34c90594ccc3c75493":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2024-08-23 15:10:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/921660/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/921660/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/921660/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/921660/26"}}},"commit":{"parents":[{"commit":"882f1f767918f979f18a5e446ae2a122e2f495c5","subject":"Merge \"CI: Return to openstack-python3-jobs, fix pep8 failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/882f1f767918f979f18a5e446ae2a122e2f495c5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-23 15:10:03.000000000","tz":120},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.1.\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9702736b3ec3c0dc44657b34c90594ccc3c75493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9702736b3ec3c0dc44657b34c90594ccc3c75493"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
