)]}'
{"id":"openstack%2Fkayobe~922401","triplet_id":"openstack%2Fkayobe~master~Idde1272da3f17c4a09e77c98c3570a7bbc8d326b","project":"openstack/kayobe","branch":"master","topic":"deployMoreContainers","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-02-20 09:27:43.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"38288":{"account":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"last_update":"2026-02-18 11:26:39.000000000","reason":"\u003cGERRIT_ACCOUNT_38288\u003e replied on the change","reason_account":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-02-20 12:21:26.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-02-20 12:21:26.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2026-02-20 12:21:26.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-20 11:03:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Idde1272da3f17c4a09e77c98c3570a7bbc8d326b","subject":"Support Ansible control host configuration","status":"MERGED","created":"2024-06-20 14:06:47.000000000","updated":"2026-02-20 12:22:17.000000000","submitted":"2026-02-20 12:21:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922401-deployMoreContainers","meta_rev_id":"efebebc3870ad45d655d74344e667f267a8754fe","_number":922401,"virtual_id_number":922401,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-20 12:21:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-02-20 09:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-20 09:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-02-20 09:27:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}],"CC":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-20 16:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-13 09:59:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-01-08 17:34:59.000000000","updated_by":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"REVIEWER"},{"updated":"2026-01-09 18:21:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-29 10:27:57.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"CC"},{"updated":"2026-02-19 12:02:21.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-02-19 14:54:10.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"}],"messages":[{"id":"2ca6c4f95b1395a6726591505ccd1b5f02bf7b53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-06-20 14:06:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc0658bd779562abd68cb43fe44d0eb87a856ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 16:16:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1265c7f521a48cd957abb6a26e6e46b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/57fe421c2cdd49c8994a4c92b64af62d : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b07d02cf834d42e59d37ffbf7ae50959 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b018485118a4dff9dc734c37f0cfc9d : FAILURE in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3396be2d703942e48c0e995d6ecfdddd : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6bd78eb7c9e1491d92fc1fb661feca10 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b84c20a57434272a631665863e7fa4c : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84d7995088224509a330add4ebf9dd31 : SUCCESS in 9m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/406a1234780a4fbabcb01b2ac5f06e81 : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6dccc98dc23c408192663b3c32aa77f4 : SUCCESS in 1h 00m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ab7a16d321b45769f38970f827f621f : SUCCESS in 55m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e2bca1364fb8437aabdbe8f8489abdc4 : SUCCESS in 1h 06m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6e1d5bff60044ce7815a384076303ce0 : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4553683e1cf14c469b5f9104a3e24773 : SUCCESS in 15m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8156d631ff854bfa8d623223b5a027ed : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/83fe209ab85b4ae1b8a020af516ecbf2 : SUCCESS in 2h 02m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8c1aea1cac3418489ffafe35a90b6c1 : SUCCESS in 1h 29m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/570627595fca4a6e907a0b9747c4611c : SUCCESS in 20m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fb7037ed7f4406cab2ff7b99eb550f9 : SUCCESS in 13m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1e99597024724e0ba51ab87ff800823e : SUCCESS in 40m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8a2111c289ce496c82da1b2986035731 : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f6e44b5652741d5852821b050b369dd : SUCCESS in 28m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7579647c1b0c46238a1cae51b4011ff6 : SUCCESS in 15m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5318ad3234a43cf97bdc0965c4c0c12 : SUCCESS in 10m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d07abffd4ba94cdd820ddf23fcb9b864 : SUCCESS in 21m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bb704933f1a42f980409feb18433f92 : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bd2262fc501485985f5ee3e429dbdf4 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4b279a2b8b3443a94f161e7d3018692 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe1e619f41044754aab4d6b5a1131b00 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"93b9609502fd27135aa30688387af3fb660c8fbc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-28 13:30:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_14826\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"_revision_number":2},{"id":"4acdcad68ad6bfff93cf0406955583d4c0eef19b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 15:35:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2d93ea0636046d28c43adcb83c6e516\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/00bba65ff58e46cbbf9c8185f3802f73 : SUCCESS in 1m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a5e850ba46f4c1aa7e1184e72c4091b : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/935d11dfeb2148e7a81a337c70ae43fc : FAILURE in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5eb121e94409426584fe140062677fa0 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9972542f7ccc46c0847ca69f2c1ffc4e : FAILURE in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb46d61f90ad49c08459f11afae8ac6b : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4df9ec0e4d94399ac7860f7e662fbc8 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d84ea64be9a44b0bb1c2210505c562f9 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e05bd5c0ad6c4994868a03f1a3d21d8f : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9fcd2fa0a0ee45aab45696fbf0db5a7c : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/dc01ea8f56af45779bd643ae3664c043 : SUCCESS in 1h 01m 28s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d131ea9f75824893a5342860e4e3bdbe : SUCCESS in 56m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/034ee3bf2dec4819862e68e5154b5c88 : SUCCESS in 1h 15m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fdd4f54505d145c88054d22d9065064a : SUCCESS in 20m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2ac93da5ab49463ea7ed7b839d462c3b : SUCCESS in 21m 05s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d9d977759a9460f8b04f1e06c036763 : SUCCESS in 8m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7ee24540535c4c7386e78ea7287e017f : SUCCESS in 2h 02m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce9790b4768447ef8e3e801b0c65283d : SUCCESS in 1h 55m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3a7cd7c634d24e31ba0e71b42ba94e10 : SUCCESS in 18m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01c2c28f378b4d3a87373327d19ced1f : SUCCESS in 15m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b5f3e89e70a541f6b9c420337c02e2cd : SUCCESS in 42m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fada0482f8f24ba08ed1f69eabca8ed4 : SUCCESS in 38m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df81b7442d7b401684b0b5ad47970e8b : SUCCESS in 27m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4060315327904500ae00aa499ac198ad : SUCCESS in 16m 55s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/9254f32712754ed086c7741d5ef5275c : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/026ce7a9f38442b4a90081a0ecc6462a : SUCCESS in 11m 54s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/1533a5394f3548e8b544a8f0217c0fb6 : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f583f44a5fa84a568a8cc0ae195834e9 : SUCCESS in 20m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0206ff63418d4c788665e74b1da413c9 : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":2},{"id":"38e06da47a40537862704077558530ab2a5a97b8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-13 09:59:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"84dd25506513de750cc64353d75df6a325e72273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 11:47:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0eb496f4dd1a445fb7e4e16870761f5c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1b390ee773234e8a8ccc2610a6ca61d7 : SUCCESS in 1m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9725580ca4df4ab38fb56e525d8ba057 : FAILURE in 6m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/679f0eed32b148ac81db7c94d8b1411b : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a725e8c0eac440a9bd389a5e949fb939 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa30707d418742548b8e04affedd2301 : FAILURE in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f800b4ee5d447d0814c08e3b9e8061b : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/847d04596c8d40b596404d87b0880b89 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc01f5d02cf48e482db53d0edc07bbb : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23288e40da94484da5105e913de571c8 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e292e0d899b54aff9355d60879e6432d : SUCCESS in 6m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e974e4407a4f4202af9840ebdf50c05a : SUCCESS in 9m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bf6115e4787e44f393a1cd90598c779c : SUCCESS in 9m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59c1687d8aba4028904c4234d0c15dc8 : SUCCESS in 10m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9f020f8eb0d44e96840fd856c6eeee3a : SUCCESS in 11m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/696812a34c3e4a6192caeb0d7acb0f3e : SUCCESS in 13m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/05b5cb617b2542218a646c3f91513048 : SUCCESS in 28m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b798387d62ad48ff8680d07f13e76d72 : SUCCESS in 50m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/32397fc94ba1457abe52d78d91787397 : SUCCESS in 55m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5724d812799a4e7e9a1817b98bf8aae6 : SUCCESS in 58m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/055d41e5e6574bfa9f98f308b5fca59a : SUCCESS in 49m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5633eadce0ee4778af92dbeaa7bc452d : SUCCESS in 1h 35m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c6296a0f07d948309396c5b5c395b764 : SUCCESS in 13m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f3b7ec8308af4ec2a9480ace6d3e23ca : SUCCESS in 10m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f09607992b7b41d69c14f7351dccef69 : SUCCESS in 9m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e78ad2fa4ffc48da8731aa87d5f8ce2f : SUCCESS in 13m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5de66840a8aa4c198a025176b4aaab5a : SUCCESS in 28m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/69f52e08aace45fe851ca61d8a1dc8cc : SUCCESS in 11m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bde6ff33bb5a46eeb8eabe5ad0d19b5a : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":3},{"id":"aec70cc2d38250d55186079a3460858614becdc4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-04 09:42:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"82f1a8e92cfddc4875a5606f0aa596212a058571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 12:50:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77594cfce8bb4eebb037faad7ec9c8ae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e230d1adebb34a28844a6780e0553662 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa5eecf4b5d4f72818a672e2dbf9242 : FAILURE in 3m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59f4029ddbbb47158d2441dafd44658c : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afbad3d3dce34f978dcaceb3638e5d28 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcb2ede72148445e87c7a59c830dd93a : FAILURE in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a780b31d82894c569f862392711cc34b : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdd537fa8cf840f4a16ade49cefc9805 : SUCCESS in 2m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef99aed8bec841bd9ebc6ceb107a73c9 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6abf8e6bf25e402fa6710d241498ebc8 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/086d576b7b404f6badac9187651ed916 : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/779f7e2a0b524680b3482000c5cc1afd : SUCCESS in 8m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80d6f769ffaa4550871545afe6baf601 : SUCCESS in 7m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3453ed1096b24977a7a3b8cfe5308968 : SUCCESS in 11m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fb4a4a31d034ffcb5f5244d169a90f9 : SUCCESS in 7m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/516b899ab03f4338ac510b636f0e2fad : SUCCESS in 37m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7a3226a563f84592a0f3b147313ecf6d : SUCCESS in 39m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a58ca39e5b00427799e634ddea8ccd23 : SUCCESS in 57m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f9a37869511418ea0cdca29a8195b47 : SUCCESS in 33m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a9e96791504470ea17fdae4e50616f9 : TIMED_OUT in 2h 02m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a57a6da97c9404695944701d34b02de : TIMED_OUT in 3h 03m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/673188e04e3e4a11b4063ec6112b448f : SUCCESS in 9m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f2116ca46e7d44c5be64a8df88c32815 : SUCCESS in 46m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2e92abe8c9246ac84670c6d9438a684 : SUCCESS in 12m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/01f189bed4ad49b19bf1477d17f078c6 : SUCCESS in 28m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/550b5d958218492caa70d85027eac90d : SUCCESS in 1h 05m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6a1a5195627b421a91c9317827734da0 : SUCCESS in 12m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/662b8b8d886a4c27951bd5d727b1c626 : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":4},{"id":"eb81d0122c59f25170646734bba44258865513ae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-07 10:25:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cb5c0ab9557b323803d3974939e7acaacf8dada9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:39:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc6b28a63f7c47bfab521aa053ea4f4d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dbd7e104ebbe491ea615a46ea438f101 : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f2798ee1d24ab295f087b5d8f12b8a : FAILURE in 6m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f02a4b2b36c43ebac52a70599b4f599 : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e81174096f0f423d8fba6f2272c1dd06 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5504dce44214986992b9ede50bffdd2 : FAILURE in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adf78b2aba0e49109537cd4d4e5cbe50 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b29d310c2ef41a8944181c2e96c8cbf : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c44818ae5d154d8abd4f205de0f5aac4 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2e8ed321fa554ab2ae605493a4d2145b : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14924c95cf8644319c80ae72e6bd948d : SUCCESS in 11m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a773e37f252e4e00b6162e9cccc7a063 : SUCCESS in 7m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d81c91d48b9d4f509763d1700bfeb940 : SUCCESS in 12m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f89449b9faeb4798a918677f87e7a157 : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9afd5c2bcc6c4e129d175b11913e0364 : SUCCESS in 15m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/130b7414bbbc48a2913473b2df7e1d23 : SUCCESS in 37m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b8cf82efc4614a13ad81e3a6230d29c0 : SUCCESS in 1h 17m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b456bfd8f46a4cca8381e53f5e8d3e2c : SUCCESS in 39m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3dc4898935b4599b3b06df680afa250 : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b07f04331f0f4b4fa1e27caa9a5ec8c2 : SUCCESS in 1h 12m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cc4ad2ce85bd4dc08ae4f7f9add18e0a : SUCCESS in 2h 01m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53371882071c40ffb431c42f446e6276 : SUCCESS in 1h 51m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7ff4a34aaeb94a8a805d8f568d34ec62 : SUCCESS in 9m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e9e0b559187b492087bf24769a3e1aa9 : SUCCESS in 9m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8b38ec830ca46e087156054bad8f8f0 : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cc6d02a75451448aba6711794e9cebbf : SUCCESS in 14m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b73ce4314d75443598adb89d06aa037b : SUCCESS in 16m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba9d3883a81945bcbccdf56898dd3a3f : SUCCESS in 29m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b156ad6b7ca4424daa0e376588840b7b : SUCCESS in 13m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5373e1e85dbf4731ad1ee63c267697a4 : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":5},{"id":"c49be5aeb6a32d6d230fd610dcf1a332f8b905dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-01-08 17:34:59.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cb44beb0fd1dbc48619d74e7b1a483a32eca5531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 19:55:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056738c02aa3417a81564d10cb1e297a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e3401b8c1534fc09264f212a42b71ce : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf8067a7cf749679d82a3f177264963 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d6d92f1dcab641098916f6599f3d2a19 : SUCCESS in 7m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f82990f97764c8681c1b8e5cff55d6b : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6d55808e154bd38dd4601992e163a9 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ecaf45e26784e53bea9ac9267856fbf : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b89efb2bb7cf40769748529ee1a009e6 : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/739fef84cf714d23ab393309c872ad6d : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/535b32b2cf0242f5aee39e55e1eb6419 : SUCCESS in 12m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b86ca2a777e4827a39dd45efe00ea6f : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f9d53af84ee47abbbd3862f7abd8b29 : SUCCESS in 16m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0fb08eef0f545da92b2aabee026864d : SUCCESS in 11m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/33b4f215eaf94c6880b7ba1f2927239e : SUCCESS in 24m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d4e362c5d784613979a98ed59343fc6 : SUCCESS in 14m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f93a2dcf07b9478181fd25095377c25d : SUCCESS in 57m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bb1218dbd7f148498c7b212f0088ecf4 : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f0cfcb2b18c64dd3afbca617301aeab2 : SUCCESS in 1h 21m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85dacf68102f47cab5528d83d361aae2 : SUCCESS in 1h 20m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a3373a8a1124e88ac2c741d948b6b14 : SUCCESS in 1h 17m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/92749705edda4f0296202caa44c321a5 : SUCCESS in 1h 44m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b83b5c3f063f4d848d5a8812d03d9c05 : SUCCESS in 1h 31m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7f6baab71eb942cebe98c1a8350a1b11 : SUCCESS in 17m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/20719a6c641e49f08562bfe9f17f4a1c : SUCCESS in 21m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/428ac6e72cf44e168520dcf21ebe9682 : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c7045767be92475588491d95db878393 : SUCCESS in 18m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a1a9551467d04460bdc507cb01810483 : SUCCESS in 27m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ae01ea943f04b55ab1375e573ad8e5f : SUCCESS in 28m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/45602b6d8c704e46bf1b867e2a6d0f75 : SUCCESS in 19m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/148c8126226740c7b3aa4928529d627b : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":6},{"id":"cb56d1cb4c5d0ffe5dfd2f3412df64ccf4473726","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 10:38:50.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d9dfd920467fd328a6ddd4fcd19f0188c8af465a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 10:39:02.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"63e1dda4edd39587209bce3932cdd525311813be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 11:20:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9ec43009a6a4fef1c0d5f22e80c71255845d3865","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 11:53:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c1e613ef3b568eae2cb2fb7767043d012e52c04a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 11:54:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0e410bfb9d3d0a4a95431c3990e8231a9311df2a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 11:58:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"127351c68d097b12c369d4f92c129b14673721f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 12:01:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0bb8880e69af69ebc4af1ace620da695f2d9c8c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 13:43:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/224a614b655f4e8f80ff231c37712714\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a8500828c18645acb3bc0cc24da930f7 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c1f4fc43934c1998859c7dcfbbd166 : SUCCESS in 9m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db3f2c5a0f444042939113da9a79c196 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7796bc0d66df4019899822dd7fba1a17 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fce02c37e64d52ae81d747ed4727d1 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcd91b9670e1423cbe7edeb7225a4637 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cf83b79340d4b3ca55d604a97b35bf3 : SUCCESS in 2m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28b29c1d1dda4031a12aff8688a0bf72 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fac723d9cafb49128b4175527bcac3b7 : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/62cf19638d2b409bb29cc6dc60d635c1 : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/83ce3210645444a9811afd2531dd0fbc : SUCCESS in 19m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99baf892229947a98163e6b24906d06c : SUCCESS in 6m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c08c66d98f674613bc51b8269be4973d : SUCCESS in 13m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e536d2ace7b24867901e1c30593579ad : SUCCESS in 14m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/943d9f673e8d4b1e8965066ad3433b88 : SUCCESS in 38m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6e88abae74684504a38688d7fd1ff767 : SUCCESS in 36m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/47a5e6adbcd945cfa44d97a8f0594bde : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc59e1d2a485478aaaa3863d2f4b68f4 : FAILURE in 1h 25m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/306808d2a5b7474c9145e9d5d204c046 : SUCCESS in 1h 07m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6f0616f9b1264aabb61b813c70c237e7 : SUCCESS in 1h 24m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a813b51096374b6098d73bb3b02e5b95 : SUCCESS in 1h 36m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5332e67315e744389f9ce1fd1d854003 : SUCCESS in 14m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f0edab13b0d4f63aaec6a6836fedd04 : SUCCESS in 17m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eb514267c7c49e883eca7e3c1d8bc15 : SUCCESS in 15m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f134f8de42ba48759e55cf7b3983fcd9 : SUCCESS in 13m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2afaf8d06299434597c0bd6b32b86413 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee591f16372949af812076de1fc29b67 : SUCCESS in 15m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c9663a5bf04b4d79a007635ce7ad6cb4 : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73eec7c99bf34c9585058d32e36e93b9 : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":9},{"id":"ebd759fd27eb19c5edb73a2eef5ae4c6411ee364","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 10:36:50.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d536709e096caaab8e9c2061f1f47a0c36a12cc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 17:31:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"9112df57ef81c578ef4a3477062b62b0b84920f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 19:58:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f98b162291d4ea09ee3af29bfd3e0b7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c92b52707324464a824d5d6ce00f4fad : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96cbc49be99a4eab8398f20fba2a83dd : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a9525cb568f648d2ae279d0563f841f5 : FAILURE in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ee9337bff2e48a78697069471f19d82 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/557dbb551ee846c6aec38d9cfb378473 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d860915c6ca94c19ae8eeceb01a4d587 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1aa2fc67f9341088d31cb0d0acf8cb8 : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/400cf7074f4a431da1d545ce9c6d2b81 : FAILURE in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d8b1dc31f4640419c86a7f64958c549 : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4062af68236149b7b95803c91aa7886f : SUCCESS in 13m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/826e3189fe534811b5c16aba25bf49f1 : SUCCESS in 13m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fafec1add1446708218e1b4a011387d : SUCCESS in 14m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cb1e89bd08ec45e99bb9a85e31287b83 : SUCCESS in 11m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e0b785190564033bdd01a60bdd30989 : SUCCESS in 16m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9330b8cb4a684ae4bda54200e9eaeabb : SUCCESS in 40m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6d5cb3aa486247aba7ce79d0165a448c : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ce6fc8a365d3492bb0439c60715ad37c : SUCCESS in 40m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b8ad46ab6af46429d7123b3e1384603 : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/acb5010f787744c1b30488b6a0077943 : SUCCESS in 1h 08m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b11a9ddf3d7246b6aead6d67a861fea6 : SUCCESS in 1h 45m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ae53b49276d44a6bada5b3589b93938 : SUCCESS in 1h 46m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a2bb3061f4214ed1afd5ad1d3e14e891 : SUCCESS in 11m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a26ecfa18f5e45c78fc584749561b7b6 : SUCCESS in 22m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7eac983660b45f0ac3bd7699941cb81 : SUCCESS in 14m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/36c51b8358d44ca18b2aca6be473b94d : SUCCESS in 15m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c14b10d9d7194f4d90d41b15b5d0aab3 : SUCCESS in 26m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee119ea577494c2996cedaaaabebc5cc : SUCCESS in 30m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/79c04a411ab44b27a9a59b0a6bb81fc8 : SUCCESS in 10m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da598d3d4edd4693b0b8c4b70037e6de : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":10},{"id":"64235bc000a02b70bab5765226d2b0ba9c124148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 20:33:06.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"3476cb356ced493b171b8262022ec7e3f4e5a39d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 22:26:57.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/747d67a6cd0e41ccb30a33337744c2a0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4787f20208bc4aa5ba83669ba0ea4d12 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be9af8160144a4c80d28dbd4e910ccd : SUCCESS in 8m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/29360b9f23ce42eb97be8d2e2568c9cd : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/422ee95323304f10b0b5912a679fd811 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd854838476c48e9a46af5a6f65017df : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aad23c368c334289b1dc37b14c666df0 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9360f6ff75164b759b1a8524eca0e74d : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe24afcad894451e8a3c5f99a4d16e4a : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/731300e9a4fa4c088cd35a0d464a62ce : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b900ef94c69c401b873f07ecd7864b18 : SUCCESS in 6m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/879d443aa4494449b76c490e978a5521 : SUCCESS in 18m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75da0d9b09fe4a19bb00b346373637a3 : SUCCESS in 15m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8be416eb178f4f6fb920e22a593eb505 : FAILURE in 6m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba7f4ae129eb447fbd1e8b67d5955721 : SUCCESS in 8m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9fe8824032404c51be14cb80fd23a82e : SUCCESS in 36m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4a2609a770734c70a3366fb95f6c8300 : SUCCESS in 35m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ebd993b604544e06852dc37ba3e30cfc : SUCCESS in 37m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02b03239908b4c7d9b70ab8b365182f9 : SUCCESS in 1h 12m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/58618c98ae3f4070a407697d63f50e8b : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/63aab28006da43479cfec8ce844e8c75 : SUCCESS in 1h 37m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97f9c5f7b2bc4172b5636f91715565d8 : SUCCESS in 1h 48m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/18e7a63da8ee46ad83291eb422fd9fb8 : SUCCESS in 10m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/025d515cb3324d0daf2a7ea5858710d5 : SUCCESS in 9m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4d3c47f5f904414a588e97480a29f81 : SUCCESS in 17m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04ee532611274173a00ac060434d20ff : SUCCESS in 14m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d57f1b6265554f02b867ab07d890ef8a : SUCCESS in 23m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/457c00395dc74a83b32365d96458a694 : SUCCESS in 32m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b595918a8c4a45048a116bf01786abb5 : SUCCESS in 9m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0f18eaa712344e18d5e98779cac7013 : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":11},{"id":"c1cf4cb2da4750640b39775fe471ee70f39231bc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 22:31:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"db1605ccd8cc0ba9423dc96147b0a23513927352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 00:03:02.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6bcc881f75d4380aad6202640ad2cf7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/05927e15d6244c2eaa1ce5d912980d17 : SUCCESS in 1m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d75dd5b9d94c3aa6a60049323cafa1 : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/32e041e1f6de47609cbdfc94fa383c2b : SUCCESS in 7m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a6ed52ecf914336ac6e0f70f3d86546 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe74ab5377174a46a52e728c3259da39 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4698f7a7e8c4347b734c5ee0fe47d1d : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46ea1af793004eaf9e69622b64d07cd9 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ff01e550205435c84754b37c1b1f1d4 : SUCCESS in 2m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/734da49882ce480c848dff3bdc850d8c : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/843a29431a9d4c719b0b2914d20380b5 : SUCCESS in 9m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2ff93012cfc34981a65ad6fc4b54a014 : SUCCESS in 8m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/627f8c4e0179415dab6af1b77f389981 : SUCCESS in 12m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b0ac376d59a44555a61ec586eeda6acd : SUCCESS in 24m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1770ecea77e94fc2b6e607cfeb8919f0 : SUCCESS in 8m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1a2bbad5386544a0bda34ab5c37488a5 : SUCCESS in 38m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/94c98b72de2f479fbbfdc0efa8d354ed : SUCCESS in 37m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ab918713bcac4a5c86604d283a9a3075 : SUCCESS in 36m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/682957a297f84b9b966b99ca8a988992 : SUCCESS in 1h 14m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e76f1860a254952931192695a75d742 : SUCCESS in 1h 07m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9c97c3d0e17b4f6c90e1ee304ced3d6a : SUCCESS in 1h 28m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c28baf392cca40d999b53775e9cb2f83 : SUCCESS in 54m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/84371f5fdb554f06b2d939d08f0a1c59 : SUCCESS in 19m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0bec5448f8914671a6c92d7e5cc33716 : SUCCESS in 9m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50b452b63bee45a39013f6441d9f9305 : SUCCESS in 17m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c94a12bbf3de4a52a043c92a10020f7a : SUCCESS in 14m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/301292cfdc5b40c08ef2516a17d6a008 : SUCCESS in 16m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b864c120906b4cd0baff4c43e55c5906 : SUCCESS in 14m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9c47f090d87242ffaea732b5c600e4b0 : SUCCESS in 9m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2cd988a68724ebd9ce7e512eca1731b : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":11},{"id":"38f982eda3561c77f8ccff9a2be92211cae28e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-27 22:07:01.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"6a1e11d26aa4487e2d2851d2308387a7c17e4b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 00:19:20.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a034b9a28a5490593f55027d2e89b44\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7494130c15244706ab84cff23195a4fa : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a91b712f1749a89ced1f27347aee9d : SUCCESS in 9m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/50e24ee80fe24a04ae3c958ed6455b7f : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/953c0561ea9b4e4eb3fc38194339af3b : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b476dfbedf3345c099ebb553c4489ee8 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/961beed66b1e4f70b8405dc58cd66c1b : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e11bb70e294647e68af27379a36b7fc1 : SUCCESS in 2m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec8a5a39890a453ca234aca0c8a06e8d : SUCCESS in 3m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdba261e46814d75af3350ae2071a255 : FAILURE in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c8319a0f6ee542eb8c7797a2f2431cee : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/634424b165214ff7b5a7d1534b719953 : SUCCESS in 10m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aa111f51a49461c84edee6736d46ca7 : SUCCESS in 7m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cbb1877765d44ef1916e6d0b3d0900ed : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81957401701346bba053c0ab1431da1d : SUCCESS in 18m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e881d96ad04f471183deb1dcc909f311 : SUCCESS in 55m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/72a6303c02bc46d09e4c1b85ede1f838 : SUCCESS in 53m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fa7cd35ee5ef449fac935aad98fa6cc0 : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77a71011e7f14dbdb683942f602bfb17 : SUCCESS in 52m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d3875d1f403043a6a8062174744c2d48 : SUCCESS in 52m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b56950e1ee6451daca610fc91f61448 : SUCCESS in 2h 07m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/de537b4d8fe4470fa7603f63f9b45bb0 : SUCCESS in 1h 28m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10fe4b360b4d4c6896407fb2b5800fb5 : SUCCESS in 1h 31m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48f2504aa64a4d8ca17ea171e6f38f30 : SUCCESS in 1h 28m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5d90a5a067684de7b767025a0b111eb9 : SUCCESS in 17m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6aae8b51fa904e7cb82aed8114e1e3f9 : SUCCESS in 8m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa4bd41ff03247a488c0f726ee405688 : SUCCESS in 19m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f9c9bfd8b6e447ecb454ac23a590a480 : SUCCESS in 9m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/55bf55356028495d8ce05f54e50dcff2 : SUCCESS in 17m 20s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9546ce89f2ad43e0addc25681535acb9 : SUCCESS in 34m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6237264a79d486ba1aa5a0af17d08cf : SUCCESS in 39m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ca5244c080f47aab1c7521482091368 : SUCCESS in 25m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f66bdf137ccd46b3a6566a374f5a28ec : SUCCESS in 11m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f4673cef05a42b09cf26763fc5ce54f : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":12},{"id":"d38e897c5f41b34d0b5b48b5b3c53e2d54c5afaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-01-28 08:54:42.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"7a885a03948cfca48884ef93a341b3e57288c874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:30:15.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2380e8d72cf5453399385865fd80418e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0cb501ac96ab440db86b088a8d1e5ba0 : SUCCESS in 1m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bdf4cbf6c014a5182a836c8b4954a36 : SUCCESS in 5m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/01eacf3e4ddd49d4b16353dff09d2846 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a1ca28508ce452c8b645da3e95501ff : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2614c06a01548569b57a2c5311c5d1d : SUCCESS in 7m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b24a18e11a9e458da709c8d273ab56c7 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0397894211bf4ac18c895e09ea867039 : SUCCESS in 2m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a6d154abf7cc49c1b3c28ead1a5abdae : SUCCESS in 3m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ef7fbe3ce944b678991a29c5a61af3c : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4207032f4ba14b1f98d8ca46fc9db884 : SUCCESS in 15m 01s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/52142e6c9fc943dc81def30923e596ec : SUCCESS in 15m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9338f6ac1e1c47ecba6327be97a3ecbf : SUCCESS in 7m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a777f9dc562c4ddd83758af0b8d70558 : SUCCESS in 22m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c00b1d3e5fd8407898edc8a120e41ddf : SUCCESS in 18m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5006d6de942a457c93fc6ef873c89450 : SUCCESS in 1h 13m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a55959ae2712490fb085f79a1bd1838a : SUCCESS in 48m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b964014dc7d0456d892f174c53e5e3b0 : SUCCESS in 1h 21m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d6fbe5e67a74ae6a15b1df4e1a9c238 : SUCCESS in 49m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45affd5e6a7445e3a8b6e36fc7321167 : SUCCESS in 50m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3e0d775637154f0596cdf57b323f1ebd : SUCCESS in 1h 34m 17s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/92c5e25625034e5195a86dd906db16f7 : SUCCESS in 1h 30m 17s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b63b221d44004bfd988002deff70b7d6 : SUCCESS in 1h 26m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6db52ad5d8a349d2a0582414880a971a : SUCCESS in 1h 27m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ea3d23e595eb42ab8768b5619ad2f530 : SUCCESS in 13m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e750a3fd7914b878f95983bd4ffd8da : SUCCESS in 19m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/749461fd09414724a23065f0e9068fa9 : SUCCESS in 9m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4a113f83ba034a38a8315537eb70ed5d : SUCCESS in 18m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9b15bc98cb014a9d854009efe04e08e7 : SUCCESS in 30m 10s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6826f3a33f5947e785f31d016b0119e6 : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83ac196f55a84f4581f6f70faa8fbb7a : SUCCESS in 30m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff634fa5f779475f91df77514a10d755 : SUCCESS in 28m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c60739d57bd42b39be1df64757cf1da : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd4145ab9b014bf5b61ffd455aa00a56 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":13},{"id":"5155c62bea212af76932a8f34b9f2242d2209ad1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-02-18 08:51:22.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"7763da1cfd04ccce2b1627177170fd21ddc2dbb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-02-18 09:10:07.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.","accounts_in_message":[],"_revision_number":15},{"id":"2729eaeb8d621d824a19f7500617244777a2d007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-18 10:33:22.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e20f74ffb7fa52aa7a922e79dae24e7764c52c26","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-02-18 11:26:39.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"a102ea027898e037f15431caf03d9b12b7febf1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 12:09:45.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50e850770eab45c4a71bd7fe7c5d5a21\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0e6f6bfbaed54971997ad73ef76c05ef : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754d7b2cd39a412eba863958da82f64e : SUCCESS in 8m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e23142249cbc479bb023f709736c5140 : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96baa7c656c8437297f2ae799caa516f : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f0a0d467b4416e90900937fa453391 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54f656983f7f40ef83f2cedaef501869 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/898844a1b84d43c1a53fa9fe45a25493 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a55b7c219684f04bc39c808084e402e : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f7482b7ce4e24202a04e57229271bca3 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c0275af6d3a1454cae31743b685983cc : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7ad7487ff1bb4ce7a2c592e858c36b70 : SUCCESS in 19m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e8114dd6cf84a9dad367c0f880fd237 : SUCCESS in 14m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1335c3fd5d984c3ca86a72d09fbeae96 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e62f83aca8444ee69c3c2fa687fb83ad : SUCCESS in 15m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/00298a4c06d848bab93ef2222e4ed520 : SUCCESS in 57m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/39bbd6d159a94e55aa1c9135b658365f : SUCCESS in 58m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fbc8ea42db7a462093e1c24bfeceefe2 : SUCCESS in 59m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49d618c7d94e4591889548e4088dd543 : SUCCESS in 56m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1fe6748bf999404dad96c431b0130614 : SUCCESS in 57m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e06278b7b5be4b959ea37f3e3aeb854f : SUCCESS in 1h 32m 31s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d6ffde4022c846cc95903ed3cc25b176 : SUCCESS in 1h 26m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b18516d3faa9446b94cde0df9bafed27 : SUCCESS in 1h 35m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0551faf02e754a7b9abbfb97d7fa2ca5 : SUCCESS in 1h 34m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/81d05833a9664ea493b984d36fd08569 : SUCCESS in 17m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/881d9bef6ec948afaec8d4b607b09913 : FAILURE in 15m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c60aa0ecf1af42ccb02e336ad77274cd : SUCCESS in 14m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f98aa9dd419743e7b49d6d8e0c9b1fd1 : SUCCESS in 13m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3b4491d887ef4bc0819307bce644af5b : SUCCESS in 26m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/54051cf5c7394639ab06b33a3185053f : SUCCESS in 38m 01s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55bdc10b8a364d7cad7c174c4361b574 : SUCCESS in 29m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6be087ae905842a489f3e904fb6594d9 : SUCCESS in 26m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4c0936ebb6c4402e9a2a862ed12ae7d2 : SUCCESS in 14m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43b2dc8fcd8c42e28ef8b1dd569db8e1 : SUCCESS in 16m 08s","accounts_in_message":[],"_revision_number":16},{"id":"3c52e69dff39c2014648cd4c89b04b1bae21e8bf","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-02-18 12:10:41.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"5568ddc58f5269362d06b81a1c7574a9bf144049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:51:51.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c0a1d83807f4ea593d1a1d62d5caccc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fcc99ca928e8499fb6522d39e51bdad4 : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bcafeb73524eb19e1dfc323e646153 : SUCCESS in 10m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2a6dea43138f46989e6753dfb11319fa : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/987864926aac448c8f300d6c1fb3ae0d : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdead9382cb0442696b13a37c4dbce4a : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b834b1deca7a402b9c158b2e91a1da73 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94b785adf09840a08234c775f5916090 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/213fd4a07c30493fb3162d29017af5ce : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c8d1b1c8eb0414cbfd21b7062b88e67 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae8f44e04be442129b40390a20ed701d : SUCCESS in 10m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0101cbd25d324d2986cbcb3cf5d8876b : SUCCESS in 13m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fb50c0da0a7400cac33d736668d9c31 : SUCCESS in 17m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/48f1a2821fb44fce8a553dce15774162 : SUCCESS in 12m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09e322f3533740ea8638e76ba0c8e99d : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/247d66ac73e74c49b74cd9c57dababbb : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2d79f75220142bf8d58d153e1640fe3 : SUCCESS in 1h 08m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f80f1683e2f74f80a3d728b4d9e5a859 : SUCCESS in 1h 11m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/132738625f764dc2851f64bfdce05d97 : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb9fefc259a1475386c8ff03a8747417 : SUCCESS in 1h 03m 34s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2a8bf501139f459c8eb5631f18d2e055 : SUCCESS in 1h 40m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/edb8a139479d4310a7217cce174973d2 : SUCCESS in 1h 38m 11s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e44b39b29194e75a3d2e22a677788fe : SUCCESS in 1h 26m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d690d3aaf3b45f9b416dbdcc9c11c65 : SUCCESS in 1h 38m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f1c535a5cb9d40ab90eb30f22fb028d6 : SUCCESS in 17m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3bc6c64bde37495891e10b3cc9bc498a : SUCCESS in 11m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f9c8ba9c11f456fb3560d238a2f09a2 : SUCCESS in 20m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/316e9c2d0d004438afd8f039fc189a31 : SUCCESS in 14m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/621388ff7fcc4e7ca2d00aa0ae6cd0e6 : SUCCESS in 18m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/383d8179549444af84c3ed5657cd7da8 : SUCCESS in 35m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0619e697ebf842aa94473f959b82ab6f : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd39e54362ab4fc183113ede39d16956 : SUCCESS in 16m 10s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6543f86bd0fc4e81af723e21df5a81ce : SUCCESS in 17m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87ea4f15bb2a48e1b84fe7018e55bd3e : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":16},{"id":"0c00eca159d3192bffabc43132f025ee2a9d87f6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-19 12:02:21.000000000","message":"Patch Set 16: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"b605e9d29b094811fa5847006b8a791774046ec0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-19 12:18:10.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"2f7b6c45b6393c77fd94a4a30a2e2646a073d623","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-19 13:59:51.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"39e6273e68879399b4749b5ff7791b7dcb1f6cb6","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-02-19 14:54:10.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f2d9fec226285c05053314bb297704046f6bc9d6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 08:23:27.000000000","message":"Patch Set 16: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"9a46374a678b5c8a2fe8b04240785b2fd2b91d7e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 08:33:02.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"9f133f8d7fb8bff47c2d3fae64f281556f2be1fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 08:35:18.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"25772a393ebf5a8c43b2e3e77c2a1b2b83a4bd52","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 08:35:41.000000000","message":"Patch Set 16: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"b0848529f3241dadfe2aac0d79550421d64d2ef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 08:48:23.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"559676b33146b48a2d53a6211448ee1d98b338e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 09:13:18.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"04b861c1ad8ee45372eb05aa57b22597329d1f5e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 09:15:36.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"e0467099f98f9e9eebca4be5f92dbfb4ce14e74d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-20 09:27:43.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"9987e5d6512be0817d2201c3a22898257fb69ada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 10:46:54.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccdd2d89304c43d695e08081d8e4e6d1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/213a116924d64434952bad87f040230f : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/432e711a9e7b459fa23354cf2b79fa97 : SUCCESS in 5m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6e79c70e130f40ddbb72afdd77314078 : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f68ab35b5a164d59b0da0844234e588f : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132ddd386d1445aaa8d6e5610e5e79be : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ef1fe138e7f4ef096f4f676e7736054 : SUCCESS in 2m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13b1777032884c6db547fc1df1468bbf : SUCCESS in 6m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51272221f66d465697bebd1c1f5346ac : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7ce4b0513c94567a4f5c2bd4a5c68eb : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a40638acac0347208674fede9abf5271 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be42389e223444e8a36efb15c7df5710 : SUCCESS in 6m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a24b47ce0adb4a9f80686d9cc8acac91 : SUCCESS in 12m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5ab0946bdcb47e0bda94049d2255940 : SUCCESS in 10m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f0e9ecfb9f17469ca788bfd13084949a : SUCCESS in 16m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c843d7318fda4c3b803c6096581e8f83 : SUCCESS in 15m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/27c93c6767ce4ccbb9b7b8afee32b176 : SUCCESS in 55m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ca9baba5610f4807a653dfe37efd8ff7 : SUCCESS in 1h 00m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8bf0d38838064b4e9f99f05e9a876972 : SUCCESS in 57m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aff88dd70a544f7bbed74f363c6a6e4a : SUCCESS in 57m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/38c85b2fcd4c4362b5001198a4c633b5 : SUCCESS in 55m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ed2405c769f1448aa1f06c93c93cd0ee : SUCCESS in 1h 32m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4e41e1c29764477b82adcc97dc5c0ed6 : SUCCESS in 1h 23m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/754395936c364bfaa167d181788f86c3 : SUCCESS in 1h 21m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc69c1aa934c41818ffcaa61c6186017 : SUCCESS in 1h 28m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0e1b24db443c47f3b3d4d6e339ad0472 : SUCCESS in 9m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ce6f7deafa8c4b8f90dc10b98b507551 : SUCCESS in 9m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/584286e3bd8e4fd08b2f1a1a3c1f9758 : SUCCESS in 16m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c45b8f639ae4a3fb6f5f6bfa4dbab46 : SUCCESS in 7m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1a962496b1f94a638e2e5c40b5e467ae : SUCCESS in 15m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2b08d32630814b528965fde7b9cfdb82 : SUCCESS in 25m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b99b0ba9252e425fa7b4f196880163f2 : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3ea7310fda94b87aa59c018ab92c88a : SUCCESS in 14m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/15f7dc1c2a594e05acd7a74c55635754 : SUCCESS in 14m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71b6601e1670417aaf643523cc5b78de : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":19},{"id":"e1729bc45536ac936ac7b236e326fe676a08ff46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 10:47:10.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"3da24c14dffed7f3c76d8f295d21b0f55c1faf9f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-02-20 10:57:51.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"796f94c61e489389b91ddea4b5036b9599feec03","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 11:03:45.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"9b7787b5469d1a7362bfff154222fcd32f2d4bc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 12:21:26.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73ae217df768421aa36c98ccaab63367\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/81d0d8277e984ea499e0a2edd53f55c5 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ffe52fc28440c6a35116fec1763808 : SUCCESS in 4m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9696b497c9c2430b8307c19ff786d3db : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60bc666446724a94977ebc6488fd36b3 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa7f0e1d081c44fba5b464ced9c3819a : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a919aa741de2403c9128f564eaea776b : SUCCESS in 2m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1861052d7b6040d5a00c6dbc95a7ef65 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c7b2f0e9ff2a4595b1e501d464cde081 : SUCCESS in 8m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f6df22ba3b34fc8aa0adac7c02ddcbf : SUCCESS in 8m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f31a118cbc184e458997db283c8f2d5f : SUCCESS in 8m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e72bb37f47af4c809abb291dd631589b : SUCCESS in 7m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/aa4cd05180c24497af919513e5aeec40 : SUCCESS in 10m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1a12bf88e6649428509fadbb558929d : SUCCESS in 9m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/42efac453c5c44e4abdc83e04d109401 : SUCCESS in 55m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4304e542f2d84f81a3e3392c6336ae38 : SUCCESS in 55m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e98bb660ab0849af9914d7d9b06fe4ad : SUCCESS in 57m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01a95e2183f74a1fbc7d9040c843f9cd : SUCCESS in 54m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4340732eb6024748ab4ef590db4384b6 : SUCCESS in 56m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/499a763be46646da900d5cefd95cc719 : SUCCESS in 1h 33m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0517d51f60044ba2a9313ad062b9293f : SUCCESS in 1h 28m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b74ed93411a74bbc8d103631aeb3747e : SUCCESS in 1h 24m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3911fb92a8274f288f2c67b421ed305b : SUCCESS in 1h 33m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b6aab09e10834e6baec5d13825e69332 : SUCCESS in 9m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1bd205c29c954d678ffb75ca136f2565 : SUCCESS in 9m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe6152234ffe4b2d92db217d9e46e829 : SUCCESS in 18m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d64468e15d340ffb8a2ca6685761cc0 : SUCCESS in 15m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/97c752a8377b4af593c026b49c649f26 : SUCCESS in 24m 25s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/716dd4114b6e487198fb93d1001f0c62 : SUCCESS in 38m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c52030f1ec64632b80d05499c184ffb : SUCCESS in 28m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/390078e23fdf496580f88250e50d95f4 : SUCCESS in 15m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/486ee2782ab44d1d8eff339db867e033 : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9163b6f7627e467a896dd38b147059fa : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":19},{"id":"02a49bea42d482581a90387bb2ec82b07cbe7765","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 12:21:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"efebebc3870ad45d655d74344e667f267a8754fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 12:22:17.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c392cdfb18e94cf68e6ab68b241cce6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d754aaa299224fd3bfb3c339a3adc2e2 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d85cc902f0674773840f452c3b8d63f6 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"f5ca1646b571812b4a01f50087cee840c8567e6b","revisions":{"6f63557dcb656424e3fa786b4b498f60cff226d6":{"kind":"REWORK","_number":1,"created":"2024-06-20 14:06:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/922401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/1"}}},"commit":{"parents":[{"commit":"03053e62b8eea70c44c3d1143692b0d467eb7640","subject":"Add support for Cumulus NVUE switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03053e62b8eea70c44c3d1143692b0d467eb7640"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-20 13:43:10.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new \u0027kayobe control host configure\u0027 command, and\nuses the existing host configuration features in Kayobe.\n\nAlso provided is a \u0027kayobe control host command run\u0027 command for running\ncommands on the Ansible control host, and a \u0027kayobe control host package\nupdate\u0027 command for updating its OS packages.\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f63557dcb656424e3fa786b4b498f60cff226d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f63557dcb656424e3fa786b4b498f60cff226d6"}]},"branch":"refs/heads/master"},"eb03acd77373068025fbd93c5a21f11cbc310e95":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-28 13:30:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/2"}}},"commit":{"parents":[{"commit":"93ed146cdfd6a80b56509d2f1cce2b1e13bbf75c","subject":"Merge \"CI: use quay.io/podman/hello due too dockerhub rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93ed146cdfd6a80b56509d2f1cce2b1e13bbf75c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-11-28 13:30:07.000000000","tz":0},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new \u0027kayobe control host configure\u0027 command, and\nuses the existing host configuration features in Kayobe.\n\nAlso provided is a \u0027kayobe control host command run\u0027 command for running\ncommands on the Ansible control host, and a \u0027kayobe control host package\nupdate\u0027 command for updating its OS packages.\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb03acd77373068025fbd93c5a21f11cbc310e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb03acd77373068025fbd93c5a21f11cbc310e95"}]},"branch":"refs/heads/master","description":"Rebase"},"6c0dc778daa612f04837e06d2d3a7fe6b271edd4":{"kind":"REWORK","_number":3,"created":"2025-11-13 09:59:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/3"}}},"commit":{"parents":[{"commit":"6fc762886d55bb8d0c8e830f3a86e9bacf510d86","subject":"Merge \"Drop support for CentOS Stream 9 and Rocky Linux 9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fc762886d55bb8d0c8e830f3a86e9bacf510d86"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-13 09:58:13.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c0dc778daa612f04837e06d2d3a7fe6b271edd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c0dc778daa612f04837e06d2d3a7fe6b271edd4"}]},"branch":"refs/heads/master"},"80672e5fff5449e1b89a04ca9e8768c9b0e546e9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-04 09:42:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/4"}}},"commit":{"parents":[{"commit":"fa13678912cf57ff275768379a4af4e6d775f10d","subject":"Fix Bifrost host variable file generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa13678912cf57ff275768379a4af4e6d775f10d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 09:42:25.000000000","tz":0},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80672e5fff5449e1b89a04ca9e8768c9b0e546e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80672e5fff5449e1b89a04ca9e8768c9b0e546e9"}]},"branch":"refs/heads/master","description":"Rebase"},"1c603b4900c57757597dbd097ba36a2395bfdffd":{"kind":"REWORK","_number":5,"created":"2026-01-07 10:25:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/5"}}},"commit":{"parents":[{"commit":"afc25e4eae3877e66b6ca6c7a40b281c44d4d46b","subject":"Merge \"CI: Fix overcloud deploy with TLS enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afc25e4eae3877e66b6ca6c7a40b281c44d4d46b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-07 10:25:13.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c603b4900c57757597dbd097ba36a2395bfdffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c603b4900c57757597dbd097ba36a2395bfdffd"}]},"branch":"refs/heads/master"},"2a9bd57cd3cad7075552c9ba97edcd19ec660a64":{"kind":"REWORK","_number":6,"created":"2026-01-08 17:34:59.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/01/922401/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/6"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"lchamberlin-medd@outlook.com","date":"2026-01-08 17:34:27.000000000","tz":0},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a9bd57cd3cad7075552c9ba97edcd19ec660a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a9bd57cd3cad7075552c9ba97edcd19ec660a64"}]},"branch":"refs/heads/master"},"e8134e8e36c93b7964aa1788614b6c43651e2dcc":{"kind":"REWORK","_number":7,"created":"2026-01-09 10:38:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/7"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 10:38:23.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8134e8e36c93b7964aa1788614b6c43651e2dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8134e8e36c93b7964aa1788614b6c43651e2dcc"}]},"branch":"refs/heads/master"},"8927da563aa4746f45f04b4401a56fcc1323a721":{"kind":"REWORK","_number":8,"created":"2026-01-09 11:20:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/8"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 11:20:11.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8927da563aa4746f45f04b4401a56fcc1323a721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8927da563aa4746f45f04b4401a56fcc1323a721"}]},"branch":"refs/heads/master"},"b0ea93dad436eb17572e26f3564a30f7e8a02679":{"kind":"REWORK","_number":9,"created":"2026-01-09 12:01:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/9"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 11:58:25.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}]},"branch":"refs/heads/master"},"4883834b444f3056d0613c926d7521b58f7bb91c":{"kind":"REWORK","_number":10,"created":"2026-01-12 17:31:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/10"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 17:30:40.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4883834b444f3056d0613c926d7521b58f7bb91c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4883834b444f3056d0613c926d7521b58f7bb91c"}]},"branch":"refs/heads/master"},"ffd5d0279b3a446a96f13f8593f3c11b2e771421":{"kind":"REWORK","_number":11,"created":"2026-01-12 20:33:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/11"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 20:32:57.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]},"branch":"refs/heads/master"},"a2da6d69171ae2b5cbe2f7d2bbae1ed5943342aa":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-01-27 22:07:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/12"}}},"commit":{"parents":[{"commit":"b622c92b37b16148505821f8eeec8f9ac6d78422","subject":"Merge \"Add support for url sources in kolla_sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b622c92b37b16148505821f8eeec8f9ac6d78422"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:05:00.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2da6d69171ae2b5cbe2f7d2bbae1ed5943342aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2da6d69171ae2b5cbe2f7d2bbae1ed5943342aa"}]},"branch":"refs/heads/master"},"6e5bce58bc7b53e2b2059f3a653b26703eca749b":{"kind":"NO_CHANGE","_number":13,"created":"2026-01-28 08:54:42.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/01/922401/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/13"}}},"commit":{"parents":[{"commit":"b622c92b37b16148505821f8eeec8f9ac6d78422","subject":"Merge \"Add support for url sources in kolla_sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b622c92b37b16148505821f8eeec8f9ac6d78422"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-01-28 08:46:17.000000000","tz":0},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e5bce58bc7b53e2b2059f3a653b26703eca749b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e5bce58bc7b53e2b2059f3a653b26703eca749b"}]},"branch":"refs/heads/master"},"5df699ff74edfdad50fb22f327e41ff2daf2e6c8":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-02-18 08:51:22.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/01/922401/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/14"}}},"commit":{"parents":[{"commit":"a011c0268ea80590b9f3e6c3136dc9e49f86ffb8","subject":"Merge \"Split Python installation from user bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a011c0268ea80590b9f3e6c3136dc9e49f86ffb8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-02-18 08:43:18.000000000","tz":0},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5df699ff74edfdad50fb22f327e41ff2daf2e6c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5df699ff74edfdad50fb22f327e41ff2daf2e6c8"}]},"branch":"refs/heads/master"},"adaeb2c5b3acb72d82043139be3c264378d06b7b":{"kind":"NO_CHANGE","_number":15,"created":"2026-02-18 09:10:07.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/01/922401/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/15"}}},"commit":{"parents":[{"commit":"a011c0268ea80590b9f3e6c3136dc9e49f86ffb8","subject":"Merge \"Split Python installation from user bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a011c0268ea80590b9f3e6c3136dc9e49f86ffb8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-02-18 09:09:17.000000000","tz":0},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adaeb2c5b3acb72d82043139be3c264378d06b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adaeb2c5b3acb72d82043139be3c264378d06b7b"}]},"branch":"refs/heads/master"},"11ff1e829ae5665b63c01bd80dfe8fa83b2f6a94":{"kind":"REWORK","_number":16,"created":"2026-02-18 10:33:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/16"}}},"commit":{"parents":[{"commit":"a011c0268ea80590b9f3e6c3136dc9e49f86ffb8","subject":"Merge \"Split Python installation from user bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a011c0268ea80590b9f3e6c3136dc9e49f86ffb8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-18 10:32:08.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11ff1e829ae5665b63c01bd80dfe8fa83b2f6a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11ff1e829ae5665b63c01bd80dfe8fa83b2f6a94"}]},"branch":"refs/heads/master"},"c6c46daf2c9936b3e10312e3e80d3e5257ec0c2b":{"kind":"REWORK","_number":17,"created":"2026-02-20 08:35:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/17"}}},"commit":{"parents":[{"commit":"8b36acddab1fc16c35ef86852c6a816da3617546","subject":"Support Python installation through Apt proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b36acddab1fc16c35ef86852c6a816da3617546"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-20 08:35:08.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6c46daf2c9936b3e10312e3e80d3e5257ec0c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6c46daf2c9936b3e10312e3e80d3e5257ec0c2b"}]},"branch":"refs/heads/master"},"cf079c474ff1b179df89bfb2131a3aee809e52f6":{"kind":"REWORK","_number":18,"created":"2026-02-20 08:48:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/18"}}},"commit":{"parents":[{"commit":"8b36acddab1fc16c35ef86852c6a816da3617546","subject":"Support Python installation through Apt proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b36acddab1fc16c35ef86852c6a816da3617546"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-20 08:48:08.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf079c474ff1b179df89bfb2131a3aee809e52f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf079c474ff1b179df89bfb2131a3aee809e52f6"}]},"branch":"refs/heads/master"},"f5ca1646b571812b4a01f50087cee840c8567e6b":{"kind":"REWORK","_number":19,"created":"2026-02-20 09:13:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/922401/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/922401/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/922401/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/922401/19"}}},"commit":{"parents":[{"commit":"8b36acddab1fc16c35ef86852c6a816da3617546","subject":"Support Python installation through Apt proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b36acddab1fc16c35ef86852c6a816da3617546"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-14 14:12:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-20 09:13:07.000000000","tz":60},"subject":"Support Ansible control host configuration","message":"Support Ansible control host configuration\n\nAdds support for managing the Ansible control host configuration. This\nis provided by the new `kayobe control host configure` command, and uses\nthe existing host configuration features in Kayobe.\n\nAlso provided is a `kayobe control host command run` command for running\ncommands on the Ansible control host, and a `kayobe control host package\nupdate` command for updating its OS packages.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: Idde1272da3f17c4a09e77c98c3570a7bbc8d326b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5ca1646b571812b4a01f50087cee840c8567e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5ca1646b571812b4a01f50087cee840c8567e6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
