)]}'
{"id":"openstack%2Fkayobe~922402","triplet_id":"openstack%2Fkayobe~master~I19ba61e90c78e088b740feed3dac1f6d1ad262ab","project":"openstack/kayobe","branch":"master","topic":"deployMoreContainers","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-21 15:10:46.000000000","reason":"Change was submitted"},"38288":{"account":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"last_update":"2026-05-21 15:10:46.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-05-21 13:29:34.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-05-21 15:02:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2026-05-21 15:10:46.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-21 15:10:46.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-05-21 15:10:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19ba61e90c78e088b740feed3dac1f6d1ad262ab","subject":"Deploy more containers","status":"MERGED","created":"2024-06-20 14:06:47.000000000","updated":"2026-05-21 15:13:40.000000000","submitted":"2026-05-21 15:10:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922402-deployMoreContainers","meta_rev_id":"2851c77b7b849493847134ce2514395e2debe02a","_number":922402,"virtual_id_number":922402,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-21 15:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-05-21 13:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-05-21 15:02:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-21 13:27:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-21 13:29:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-21 11:51:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-20 16:21:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 14:06:45.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-10-27 21:36:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-01-21 16:00:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-01-21 16:47:46.000000000","updated_by":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"CC"},{"updated":"2026-01-28 17:07:32.000000000","updated_by":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"REVIEWER"},{"updated":"2026-04-16 15:17:13.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-05-20 16:46:59.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-20 17:15:16.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-05-20 18:02:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"af15960914ad7bd6e8845c1295b985208b782294","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-06-20 14:06:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5165d0f76ac395585d25d4c54e7ca06d0016fca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 16:21:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d6817ea3a47428d9fa4302b29dd7f6a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2beecd5ecbfd487c9fcc0dbb02b0eb40 : FAILURE in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/639a116e01af4f5f94f5ec7f047188ea : FAILURE in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16b1de3ecd70479f8918984db1a0248f : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06f556cb6aa745cbad1d22a1693eb232 : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/80932933bd554025bdffa3ba11ab289d : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37dd3d33524a43f7bbe917d6f6b1e68f : SUCCESS in 9m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4afbb75793d64b86b1a4f18ac03f668d : FAILURE in 21m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bc31200d70f44aa8b77fe5e9179581e2 : FAILURE in 20m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2852f5c7a0342cb95c6508055af63b6 : FAILURE in 13m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/52f81c608f8d4a43b9ad6bd3213d329e : FAILURE in 25m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/461f56a72d0e49758010f75e9ac738f5 : SUCCESS in 19m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9e9e500d7c9a4abd97e9edda8ed28847 : SUCCESS in 18m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c7931334eea46f1904f898b72dfc2f2 : SUCCESS in 12m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/53fb326f3fc6453b906460c5e5bc1aef : SUCCESS in 1h 47m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e62466e5b47a4adeb1cdc82a1ed0faa4 : SUCCESS in 1h 37m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f412ad2fbd9049f3b7c494cb64f1cc1c : FAILURE in 13m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a59a0715e1494e5a881d0dda34cb90ca : FAILURE in 8m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0334e18fd56d4077a6ad9bbc1e80afb5 : FAILURE in 13m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/97162f4fd93d46afba783e8b531587be : SUCCESS in 34m 12s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9c27433d1e5494282572d84c16430ed : SUCCESS in 24m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab52480259e142e0a0d8fbe2171b6e39 : SUCCESS in 16m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf8f9849cf56478486a41fe6334f32da : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/461b167425b8449f9100f6b800571fb7 : FAILURE in 19m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e17d34dab7a94130922963a9d6271ee5 : FAILURE in 14m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/696de4ad523b4de689f6ec71e543f281 : FAILURE in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/038b22071028434caa015e5607ae7bec : FAILURE in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec7c1c47024249c893e06b83d9a72743 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ec3bd4cd15d162bddb431a98e7d848b0bbb5e6a5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-06-21 09:55:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c94e8576538a4dd36aab223b575253e0b6a4c3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-21 16:00:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"23aaf36655c7f8fa25188434bcf51a2a247e0ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 17:28:13.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a6f975f1d294a86b1affa77470d25b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0291710aba43d7b061fc999ecc5ca3 : FAILURE in 7m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0ca0f5a4b301423bbff5c37d552b45af : FAILURE in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cf417edfc2746369d62a5224cf25cf0 : FAILURE in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e909a091ae3f44fcb99c5e4baf78980b : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b484cd0098324709970f82874fd96df6 : FAILURE in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13831120b6ba465eb09db8b2e2e8ed8b : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3fc6c0f539fe44bc8c93f260269be021 : FAILURE in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5ff152769e544f8ab130a3f33e83979 : SUCCESS in 6m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/005449c27c484d8683897f3c51ad460f : SUCCESS in 13m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/910b3f2e956545d5a546f4ba71903b1f : FAILURE in 8m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6c752c6d3824df9bf8b01c00bb7f246 : FAILURE in 12m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a3710b92278047028abde89862b16da3 : SUCCESS in 24m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4837b963a08e40efb77320635e384d44 : SUCCESS in 15m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dc35580effde4d059663355b8758896f : FAILURE in 14m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9fc99f1c3a9545a088388222df319ac0 : FAILURE in 15m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/42a47a1484bb40c987f03be5d0f1812c : FAILURE in 27m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00ba0944540f45299594ecd0cfc7dbd2 : FAILURE in 22m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d6b4807c9b37420fa710273573fdd477 : FAILURE in 17m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/20f3660960204fe7b3daab774438bb12 : FAILURE in 1h 00m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bee54488401494494970b8a2c1f15e6 : FAILURE in 1h 16m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0ea5773c54fc497e860ab88211520d60 : SUCCESS in 19m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/533f154feb5a4a96b2341c77657facf1 : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ed0dd0d92df5cbe6e4f4b280c974afab1d10e972","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-01-28 17:07:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b3da5ce979ed626b9daaadf47124ad25726a4cff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-01-28 17:13:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9fb6a21f79585ad3fbe24b21d08b2d8dd39a2955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 18:45:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97fd49a81de3455b824d6b13bfb9af3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca97d03056444daa8831646c632cc73b : SUCCESS in 11m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9ebe05acc6e5461389477d87d7eea509 : SUCCESS in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d94a997bdbcf4a208952d4364efb2f49 : FAILURE in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6ea5efc124420a96f10d5557174047 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0513981cc709430d8180ce37fd143c6c : FAILURE in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/765dc74415914c358f02428904a28829 : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b2f2d2144804da1b07320647eb1c82f : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd52840734fc401fb7b2c6620fed9657 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d3d45ae246ae4ea1976d24b624ba9cd1 : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2fe758c2e4a14657aa4dba7905a8ca2f : FAILURE in 10m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c99d5cd191a947309686e046b0661bff : FAILURE in 13m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/940951ffecb24610be6bc5b9f3c8aad5 : SUCCESS in 23m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdf5089a341e4a3e8661c51f4f57c286 : SUCCESS in 16m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c3e03e9520f94e4c8fbe35d8a65dea95 : FAILURE in 18m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5822b92350984262a28d907017c3d5f4 : FAILURE in 14m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1d785f51c16943a68a6c46b2e53a78ba : FAILURE in 21m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0007cc0745c5447dafb47112064fbfaa : FAILURE in 15m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b2e1959c78b74a858d5286ee82ac8b22 : FAILURE in 14m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5e6767ce7584a40ab5741a1d2726db5 : SUCCESS in 1h 29m 46s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2ce5ecc6429c40cdaaff54ceb6dbf395 : SUCCESS in 1h 25m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22858129cf3844b5bd494a2ab296d2e0 : SUCCESS in 1h 27m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f678426b2d5420ebed98039dca92e7e : SUCCESS in 1h 29m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8d568a95ca5742ca9681351251f686e2 : FAILURE in 7m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/253b55ca6c3940af8f4c5528c36b43d6 : FAILURE in 12m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f677a95cf9f45cebcf31114e34a8df6 : FAILURE in 12m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e473d5c1fdb041d4977e9c49f6bd7724 : FAILURE in 12m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9a4f26546d5944b29351d6b81a6de6c7 : SUCCESS in 17m 43s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3a3819c8b6474f1d88dfdaf5de79bf14 : SUCCESS in 20m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2c64a877a4a4554b42d0fe70f06cc52 : SUCCESS in 36m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f87e52ab46c45e6908412b47276c56e : SUCCESS in 28m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8d601c69025f4f3483d6b74d1a408a82 : SUCCESS in 10m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d809745bb95b46f68b961d26139c0613 : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":4},{"id":"9b100f533742e0656c8479a526f5c3d16483aa55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-02-18 09:10:07.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7c67cd579dcf074efb363f2951bc156d58e678e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 10:52:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af6c00b252444a2fa455aa076bdf165f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42f2cc065a90415c9d325a5b37477727 : SUCCESS in 8m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/581c0a1fcb704e718582f2cdbaf249bb : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e3bcc6e673b49b2b5d0d281c2592bde : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f39ed9b6a0494bc18b6fd0ade494b7f2 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7dd49c4df8b64ac093157b628615b4ab : FAILURE in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d5c6d7ed4c04d88a3712a343e0fc8a0 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/38811f68d3114d5d9e988fd26403f49d : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0f739f3da0e420c851bdf647d347d2e : SUCCESS in 7m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db2bee9fc3f6485299f1ea2c582c06a2 : SUCCESS in 11m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/143575186e5746df9e863a5c72354074 : FAILURE in 18m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc748ff978ea401393202bcfbc2de314 : FAILURE in 13m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4bb7466cb43f40ffabc6825996882093 : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d66b0ca8c064877b589334152ade1a0 : SUCCESS in 18m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/20112e331cf040efae9d7cc4ebd2bf5f : FAILURE in 18m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/aac296347ecc4f688e642a8682b08d66 : FAILURE in 16m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/49ff6ae0bf8f401d8233b5710a7d5912 : FAILURE in 20m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49562367696d40dd8cb2daf84e26bc32 : FAILURE in 14m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e509e6a897474981a796915676b4701b : FAILURE in 14m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dc8fee82697a4258beb348d6d8ecb6a0 : SUCCESS in 1h 37m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cba5f3cfb4764ce98bd6abf35fc4f3bc : SUCCESS in 1h 32m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33cb7e5824d0460ba4f36ec1587abd6f : SUCCESS in 1h 27m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30f8749596b347b781f9f58374a6aa21 : SUCCESS in 1h 41m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b60f03388cf4476593cd5f5dbe34bb06 : FAILURE in 9m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3733d25c266e405db55caccf4b36fe47 : FAILURE in 8m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/582c95f9ddbb4c8fb80ba44f7d1f9c33 : FAILURE in 7m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/70b685fb46f34988957cdfad717091a0 : FAILURE in 9m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0cce749e9d494785b209360a346cc9fa : SUCCESS in 34m 34s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fb9a91a3d94f47c287ae85328390c21e : SUCCESS in 25m 12s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e459341fc726443ea143af46263a5d7a : SUCCESS in 35m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e22f66817f424682982a22f182afaab1 : SUCCESS in 41m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a4da14ab10541c997f33a20c04c91df : SUCCESS in 20m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43605043fc31431aaa4adddd7833f30b : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":5},{"id":"319260f870e5680d11f0e548b921bfff76104139","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-02-23 09:50:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0a57d32aa858812fee7a94e93ca8efd934ef4354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:22:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b68f227142f94c37842dbce2a863ee49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a65be3485442729908fbe245256ee0 : SUCCESS in 5m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4994c1fc60a94c0baab664da4e9b154c : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/968a825342004bc4b5a11a9ec21932cf : FAILURE in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429cf167258144e6a50827b74ef9876f : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea30a3dcb04b44e78b7d1a893948a2fc : FAILURE in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/25c56ab8920a45528ef3e3c0c52cb42a : FAILURE in 6m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28222cca9bf24e4b9b8b89bcae129da6 : SUCCESS in 2m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/381699eb434f43a68c634c59057637ec : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f266ca25b771421eb7c5894975ff8c3c : SUCCESS in 5m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f51a4bcac434664a37c4ec40e8e9d29 : SUCCESS in 9m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/703d6ad68e604f91bf9fb7f2a148db4f : FAILURE in 15m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efeeae4df0614d7f97925332b60321b5 : FAILURE in 10m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a1e29695160d499ab081b33ae4a4b927 : SUCCESS in 11m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3727459bc1404ddaa209b1140f1fb94c : SUCCESS in 7m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/aff366f04961461eb41b704bd8ba1963 : FAILURE in 14m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f029e2dc6b3e437295f192d093d77fdd : FAILURE in 14m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/df8e5842029349dd9c60f82f9da5781a : FAILURE in 18m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b83e327183b42f383912c73423e60ac : FAILURE in 13m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4c5eeaf2391e48788fd60a4c24211121 : FAILURE in 12m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e7d33842a01a46bf8403b56e3795de6d : SUCCESS in 1h 29m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d8804bb6ec234decbdb31a01995e3d4f : SUCCESS in 1h 28m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d3e9654fab947e4a0d3841f7186b6e4 : SUCCESS in 1h 21m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/257bdb1b92494f8ebb59c2dc980f52be : SUCCESS in 1h 29m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b88447bede364f2ca945e0571b46bf2b : FAILURE in 7m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0508281e09d7493b8610d0b5f92ebb60 : FAILURE in 7m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f3d6d48b734413f9c144e580ec81f52 : FAILURE in 14m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/41dd688337ce4e4a8769f85b83b9bef9 : FAILURE in 11m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ebf8a2bf50fc4fb2b1b34e225db3af8d : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0c57ded8815f43f0a341238f2875795e : SUCCESS in 27m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9174e02f9b4c4d538667b745e977a149 : SUCCESS in 31m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34cbf6d112cd40de92ea007b5b95adf1 : SUCCESS in 30m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4bc1836f128c4623b76204180ab05c87 : SUCCESS in 8m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baf99f20b73d413890ec874543081755 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":6},{"id":"6f2a74c11320c63b54112fae1d90ec5617521997","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-03-18 17:33:41.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a088fec89855f76ca8b43a08d898e3d364df1f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 18:38:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01028bfb69424e1e85425d41d727323f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec796451efb46c6966edb09a5eb1742 : FAILURE in 7m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/18f76cdc8be945138951def1e378a6ff : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d7a0f85c6744ecd87a147e4defba5bc : FAILURE in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51a03295fe7410cbcba0acd2daac939 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0849b7525be4faf97c50be60746feff : FAILURE in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d1541a96a9044484823d873bc9db1d4c : FAILURE in 7m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48174eb757af41df8f49d05f6e5b03c4 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9723103d33d14bfdb603fa162629db35 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d1119af3bdfd4ce694a848771e18a75e : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f2a13bc4017e413e9ea1d3a677287cde : SUCCESS in 10m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d12b393e15724a0d87ea08bfc64c706b : FAILURE in 8m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe6a7f365b984439acb84a6d5411d803 : FAILURE in 12m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8a5fc0374ecb418594c12c4cea7e3733 : SUCCESS in 18m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46ef5952f46243ca9ae49fdce9a26c8f : SUCCESS in 9m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5fb734787e1e429ca4ce315f93ab9789 : FAILURE in 19m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2f3bc2e510e74ab3b621f5f7740c624e : FAILURE in 17m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/433a5b0d653a48188227f72b65cc6c31 : FAILURE in 20m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5b39e1c5d644cb5b4157f25ed9df858 : FAILURE in 15m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/11c5a7706853427b9312ce8f257d33fe : FAILURE in 14m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a7bb6492441445ae82eb7fbb87316f69 : FAILURE in 1h 00m 26s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/40bf84bf399d452c803dd3bb3e1ee7be : FAILURE in 1h 00m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4889a754532744e9b7c2d187ccc0eff3 : FAILURE in 59m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/469309cd3c7d480485a75835d30c7515 : FAILURE in 1h 02m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1a368d32da8f46dcacd2b3d6ef9cbbb8 : FAILURE in 19m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f0cc1ef440af4ae296eff3d170432d30 : FAILURE in 15m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c4dc7c39ab94f0ca77ddb793bc2adfa : FAILURE in 18m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/88fa57d7b44542b4855e303e28214991 : FAILURE in 16m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d5479ebfb6fb4e2c8417ae577162c6a6 : SUCCESS in 15m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6300e12d6e6f4c0eaf04098cf09eb370 : SUCCESS in 30m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aefdf190badc490faa1008919b9a0a7c : SUCCESS in 29m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ad76e5c20f54294a44fa3356b2e84c3 : SUCCESS in 27m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/71eb8f8cfe744694ae261aaa283b5deb : SUCCESS in 20m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/735baea111994166a445ee100325ff42 : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":7},{"id":"0b68446700161ea5466e8859beb1dde5618381eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-08 07:28:56.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1ef0eb236505cd15eaa97a8960aef28c96fa55ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-08 08:22:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b839cc90474df317d24190a0d42e5ebad4baf2f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:27:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/444779f66bf94fc7a7972d7b36b04edc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873b622d8d134d67995fd3b39179a01b : SUCCESS in 9m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/682b80f82d0348be89db22da484d9c6a : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbd7021b7b6542c39be4e42e34326ba0 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351201a2a3b2451a88604035f10c46a0 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21680cbac360477ebe3293ecbea742d4 : SUCCESS in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8869a890ed0147168cef8b94bfd14ed7 : SUCCESS in 7m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c11b2265836e4fb2afc435ce35eb2f02 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f0d166c61cb94c93a88ad438d76f9e7d : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b544757d899b49cabea49672797dc628 : SUCCESS in 6m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/21fcf1eadd7f4095be607a5b3f608d73 : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f56d465fd38b41a9b22bad9a14c2226f : FAILURE in 10m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88a3eb35dea94b919faff82025c2bb95 : FAILURE in 15m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9a99e6b24e6d4dcfa235ab7686bab13b : SUCCESS in 11m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51e4c759ed444d689beab2d51c291f77 : SUCCESS in 11m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3091f244c0bb42f6bd0ea09710f56751 : FAILURE in 16m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/165054f154634b298433456668562bee : FAILURE in 16m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/00683dc610354422bdb5a90fc4b9ced2 : FAILURE in 20m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3aca34b8dedd40d694477a0303a086b1 : FAILURE in 15m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/12402ba9271e435692e9a0fe40b7b070 : FAILURE in 16m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4cb609cfa2d742179a1578f790691f9b : FAILURE in 57m 21s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a8579134cc0146d79e2f1588b3bfd186 : FAILURE in 53m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d140bd5fc0f4de6a829d3e1ce462bbf : FAILURE in 52m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0811b2390414abea99d2d79ac174e8a : FAILURE in 59m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a5a5339f08ec40e5ac86f1aa5b32d6e6 : FAILURE in 16m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/97df7240e9ae419ea945315387f8f21b : FAILURE in 17m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ccf7a24292d44aea63aab294334c350 : FAILURE in 18m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fffe5578568e4fe39c7e6b924ac37086 : FAILURE in 18m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/923a87173a074469a9a1c466f85c4fbb : SUCCESS in 33m 40s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/45f9d86173d24c1db6e5deca4b561666 : SUCCESS in 19m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b9339d8b6ae4206a6cc63b82a44bfa0 : SUCCESS in 17m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74042f35ef3f485888c2c88566bab6ae : SUCCESS in 27m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4446dce042814b37a963cf89fce31bd3 : SUCCESS in 15m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8909283fec6c4b45b7b9245eef6b6458 : SUCCESS in 10m 00s","accounts_in_message":[],"_revision_number":9},{"id":"f74bce8e9268f2cf485dbef0972e406d15f9ef73","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-15 12:48:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"e3e9d035225e3305cbdce69bb91f82a1cd5b92b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-15 13:51:48.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"13fb6b527a47b9d72a1caa98d4c0ccf72925b3f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:48:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf66caa78e5346a1899fec66d72d945e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1939bd0ea8b94633898d6fc329773598 : SUCCESS in 4m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/437ff0301aaf4505bb272758e7b888d0 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fa752744cf04a68b5eb50fb6ef31cfb : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6340c58b432043c7b677f0087bcb4d1d : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8350b7e05f384ebeb9abd2d5f8f97a51 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a86e146c408c4493a40b0d6693b37460 : SUCCESS in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11baeac5b5034be0b9fbba7e506a57dd : SUCCESS in 2m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e70c1692ba9046afb349a317fec497c6 : SUCCESS in 3m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd04d9b20e354210a72b95553b38b79f : SUCCESS in 13m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/68b95643f9424972b155bc6d1e1ec04a : SUCCESS in 6m 01s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1d6a7bc31b044bbdbe232b34721bda21 : FAILURE in 13m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62b4bce588b64882a30d3e8eeb4e72b6 : FAILURE in 11m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/090ad1b5143042abb25c1bcb840408d1 : SUCCESS in 11m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ece2996f5cac4a5a980ea2b38b3b7f2f : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7ed11167aae341a4b2a4436d9d71ffd0 : FAILURE in 23m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/941934ea96304122b53d899534d92fcc : FAILURE in 22m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a3fb0975961d4fb8a41817f56125de49 : FAILURE in 26m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54c390bafba14f56a8458760b8311267 : FAILURE in 22m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ebf85335ea184804a01a4dee7f7274a5 : FAILURE in 21m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/616601bbe3d14cb29645d041c325995a : FAILURE in 42m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1108f4785e5d403a9e4e4f417f086d9e : FAILURE in 55m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b697e54bc78416ba7b441d32f385c41 : FAILURE in 54m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43b6c0703e054022aa0825a656bb3c41 : FAILURE in 55m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1ec84397dc044932a5409cbf0654a367 : FAILURE in 31m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1b26979e16704d49bb7de2ec7a90cd3a : FAILURE in 21m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17383daf555d4244a56d3d53724b43ea : FAILURE in 14m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/60fecaf0f1c34f409fa82a7b6677cad9 : FAILURE in 19m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8f3452e72bc04667814538f7060ca0c2 : SUCCESS in 31m 07s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7159e26aacdf49e7bdeb02901eb5e28a : SUCCESS in 40m 27s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eba0c8561ef4adfa79d851c6bdb93d5 : SUCCESS in 31m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e5882c8c0644729b62832acd0ea615e : SUCCESS in 16m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b9f076e9c24144cb8f539bbe52e11f9f : SUCCESS in 14m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce99b0fdf199463a8eba5e768298c110 : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":11},{"id":"808284b9b15eb5d73cda7823ca38a4578f837fb0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 15:17:13.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8d73cfef361c97582e08ab86713aad98fe10529b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:33:33.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28f27b4a69c3490ab3c365c00c65c3e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/850167957cc649c7a097e96262a6cfe2 : SUCCESS in 9m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/881f3ad88e4748599cd274f2fd0d162a : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d495c37ce8fe43ce8740bfd4fcce060a : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8a4204466e4f08a7d6c9f8766b1519 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/633700a482fe485a82618e6d1c9b4218 : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a4a3768d7b90428a87713082cb307d8d : SUCCESS in 7m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52aed0a5a208406094d5a2351ded1981 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5f861d2a5641426387bca969844fd3e6 : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2387a477d28c4b9983119588d3557699 : SUCCESS in 12m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/239022cb85a14a408c4fee60243e6f6f : SUCCESS in 20m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b5407d33abe14d7eb02036f46724f696 : FAILURE in 20m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c6bcf198c844d2781611b0446168592 : FAILURE in 19m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a510ebede30e4614b05861c387bfccd9 : SUCCESS in 14m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/346093a411014f50b2bf882ee35928d3 : SUCCESS in 18m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2c183545299b4ed3b3742600943c6883 : FAILURE in 23m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8ef414d6efa84474affabe00b4430c9a : FAILURE in 22m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/80940c4f86714898877e6888bacbd407 : FAILURE in 26m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76abe44754b1400880be5cd51c01f239 : FAILURE in 28m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e11d5cdb13fb43eaa851d01d8053be10 : FAILURE in 24m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1cabca50ad584435bc54f979bf0c5565 : FAILURE in 1h 09m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d1086d74c9314b5da1a746b3845f5033 : FAILURE in 58m 54s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f6adcd1347940d4bad0827960d41fe2 : FAILURE in 1h 03m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24232464065842eface00ce8d1575029 : FAILURE in 1h 12m 15s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ff7ab7e698e143c6ac3ca8a556347666 : FAILURE in 22m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/abe1e9105e34486c9bec63768ecdb5c5 : FAILURE in 18m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/130f82fcc8474f18a3bca3df48efbbe6 : FAILURE in 22m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2b70efcfcebb40389d3b7346e9722519 : FAILURE in 21m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/365617d8216247a68dcca9e161cb61cd : SUCCESS in 30m 19s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cbabba1f0490406997a811bdf8b19336 : SUCCESS in 19m 57s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f725f558806f4763aaf0bdf6bddc2674 : SUCCESS in 17m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/776c358e5a3a45d0ab1521e162422927 : SUCCESS in 29m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab76a86673a74e0ead63b24d03fbc636 : SUCCESS in 16m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f48119e5d69403a8d8546d52cac08b8 : SUCCESS in 20m 34s","accounts_in_message":[],"_revision_number":11},{"id":"c455b1f8fb9045d92537909d89355a882dc29876","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-22 15:30:25.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"df733d8ac04ac73c62255a3d79d6e06368dbdb2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 16:35:46.000000000","message":"Patch Set 12: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e742e114a33c47a599c87f146bba5166\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89d01a44ffd462bbc40ecc01a879e55 : SUCCESS in 10m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/34b09d196b8441c78f56cb7ba8d89870 : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee07feb7314f4336bfef1ddf3317710f : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6454d64fc0424985064256d7fade26 : FAILURE in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4df317f91a37441295c437f6aa9b0067 : SUCCESS in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c8b1243ae96e47da9f3e66968f631d28 : SUCCESS in 6m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c5dc5a596384629883d2e5a0378035e : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a766cd7872b7402791eae44719fa8045 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f65a96e2fe294dc2b117b07ff13e7d12 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/51c3b5633e8a4523883f03e800c09ea5 : SUCCESS in 13m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/82218266325c47c98252867c3abab5a1 : FAILURE in 13m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a14b74c71624511884be2c20b2e0258 : FAILURE in 18m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/becd46857dab41d489738f598b8982b7 : SUCCESS in 11m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6cab77cad0c4aed95b2248b8035cbee : SUCCESS in 21m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3661a53039c2441dbf2f1701b49cbc6d : FAILURE in 22m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/54d82f1f48f540e0b3fc1a76de349833 : FAILURE in 21m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5dbec411dfeb467084b0491f15d51432 : FAILURE in 24m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9dca4de596de42a8a8df3e201d4f3644 : FAILURE in 21m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7ca448e694f4fedae35a8a598301e2e : FAILURE in 22m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/deaf3a9ef2b44bf5b24c488b3153e4a7 : FAILURE in 1h 03m 39s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4572da964327421581c38e544d73bc89 : FAILURE in 59m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a91b08cbf6946598505afa3a48d34f4 : FAILURE in 1h 00m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4db8afb5a3e6468697bd299fc02b6ea3 : FAILURE in 1h 03m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e9059c02caaf41248f6181e9c077f4fd : FAILURE in 21m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/eb9eb02c6be849f3886d03257ca8941c : FAILURE in 22m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52a7e890fb16413f8e6a1129e4dce440 : FAILURE in 24m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f15362e8d78b47c2aa20490bbb3a3c72 : FAILURE in 20m 50s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/046fd4716c6f4e9787d15c232b86404a : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f76d93b51a5c409488e75e1d91bea81e : SUCCESS in 26m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18019348a61f4b4a897973a32ebb2132 : SUCCESS in 16m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23cf5b802c1f4d149fbc67ad909c1eaf : SUCCESS in 30m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d5fde70238b9421a9f83ce49f55ec743 : SUCCESS in 18m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03ff245b4cd340128f2a8c72053acc96 : SUCCESS in 20m 05s","accounts_in_message":[],"_revision_number":12},{"id":"d5f7b01fd72c84488e088b5bbf5d24939f9505e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 07:36:44.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"18beacf223900f17d440255b505589432c2d6552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 08:41:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf86b03120444313960820a6c4badd57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffaaf871076846928415801a9f30c966 : SUCCESS in 10m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d82792d5e50c41d58d607330d0717409 : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6be6ccb38a064b388e8220b928b4925c : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c526fd1cf0fc4d6291630b05219c92c4 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/046e77a549ef4f159610a00f60d092c0 : SUCCESS in 2m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f79dea0440bf4a6e9cc3adcf95ee694a : SUCCESS in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8abb84a09674457a99050b6d12e637ae : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36d7316225404e36b68b2e4aedba0f7c : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65670e424cfc41adb55d96deeae0a0c0 : SUCCESS in 15m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c48c650b39448c9ac0ec54dfdc52019 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/db16ddc3481a403a8cb1cc8285302773 : FAILURE in 21m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ca2753e8bf647b193e80d4dab3ec73c : FAILURE in 19m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b711306074a24def81b42bd3a0985ef7 : SUCCESS in 27m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63e670278d0440f18ecdb0a9c04e6153 : SUCCESS in 17m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2c1a1e9a8069401e9d5e0aba1c8e4d72 : FAILURE in 19m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/44f971976cae4f5489489b8e74c7618b : FAILURE in 19m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/940608753572487385e6fc28f7eb2246 : FAILURE in 25m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afaeb44b70204e25a9c6a23c36ab9d81 : FAILURE in 14m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/66f46b243ea0493697bdeff9f7ac56d0 : FAILURE in 18m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cea4ae00181249238118e6796497b042 : FAILURE in 58m 17s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e8e841ae131a49ee8f75143b18158993 : FAILURE in 56m 25s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91d0920c045e4954bf49ddd5fec75a9b : FAILURE in 54m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6c60e7b2f4b49d49f0351bebe88da0a : FAILURE in 1h 03m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bcb4b48a2e21475bb0ec56f3cdd544a8 : FAILURE in 28m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c6f668d9363f43ff9e394c08d7b6e92d : FAILURE in 26m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d268b014bcc412b9ee6989bac18a036 : FAILURE in 22m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ec7f4fffbc44b80a1c8486930624781 : FAILURE in 12m 50s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/766d1979df2a4d8884003aad5a43de59 : SUCCESS in 17m 03s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fc3451d44bdb4e3d8362c79478f5dace : SUCCESS in 38m 01s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87258aab1720446fb598c8961fb8a1d4 : SUCCESS in 32m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12e3aa5c7c65476ca259e3c1a4dd5240 : SUCCESS in 34m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a7ccac7e7cda495bb790357c108f3547 : SUCCESS in 19m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28e6da6c8b454f58b546ee3c0ccfb365 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":13},{"id":"2e8457b90412d62cc88f4ea572337f54cd96f82a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 09:11:55.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"8dc083e88f10f884f5f15335b78a075695077139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 09:13:11.000000000","message":"Patch Set 14: Verified-1\n\n(4 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027environment\u0027]\n\nThe problem appears in the \"kayobe-infra-vm-cloud-image\" job stanza:\n\n  job:\n      name: kayobe-infra-vm-cloud-image\n      parent: kayobe-infra-vm-base\n      environment:\n        KAYOBE_INFRA_VM_HOST_CONFIGURE: 1\n        KAYOBE_INFRA_VM_SERVICE_DEPLOY: 1\n      vars:\n        infra_vm_use_cirros: false\n\n  in \"openstack/kayobe/zuul.d/jobs.yaml@master\", line 490","accounts_in_message":[],"_revision_number":14},{"id":"e91b70ee341c0c20f10b508d4981ff04ae7d7815","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 10:30:25.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"b921bc4c61f6af80ecc70fbba82e1a3c45421ab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 10:30:51.000000000","message":"Patch Set 15: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Job kayobe-infra-vm-cloud-image not defined\n\nThe problem appears in the \"kayobe-infra-vm-centos10s-cloud-image\" job stanza:\n\n  job:\n      name: kayobe-infra-vm-centos10s-cloud-image\n      parent: kayobe-infra-vm-cloud-image\n      nodeset: kayobe-centos10s\n      voting: false\n\n  in \"openstack/kayobe/zuul.d/jobs.yaml@master\", line 478","accounts_in_message":[],"_revision_number":15},{"id":"dd70c2901632ab7531b94d095659466f6a380157","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 10:39:36.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"a49bcc2fc2a3a0c3872a91c81bc73e2bfae95d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 11:58:19.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8e11a1acc93481989d0d0105e7f4e87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbcfaf4f480a41818bb2c04a285a5bd4 : SUCCESS in 10m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/451f945d1dc94a1ea90314a37c345899 : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f01be8a38f84771b14dd82088b7bc09 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e770fa83d848f484dd8d00e1595360 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/124d0e8dc250451fa6523ac9b61cdb1c : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae15f239736e4c50a4bb7e708d24c8d2 : SUCCESS in 3m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b5cb3c73c924fdcad43a0bcfc250360 : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d6604f1acf3d4d36b73740b3a6852822 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ac480d53f0384d05abfa61191971c60c : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1444e107b6e44e7899ffa6e0d952aac5 : SUCCESS in 6m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b32544d65ccc4a349e307bb1ed8aed9e : FAILURE in 16m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bfb98d755fa4f79bb3126fbce66c273 : FAILURE in 12m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/46202f1781a342599e00e870da617460 : SUCCESS in 24m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/578d9c0a16944e31b34d286d71aee4e0 : SUCCESS in 18m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fa9b04a82b574e18a3c00c3261603e0c : FAILURE in 20m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1b01447b275347f081fcbb5f4317e81b : FAILURE in 18m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/88e69332c36140c58c63e606f412fdc1 : FAILURE in 29m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65b4595179754aba9f5b725b1bb04290 : FAILURE in 21m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6c5f5fffee540a69ed4ea56b0988dc0 : FAILURE in 17m 41s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9625b8f6b6f74a39aa523d53558e3493 : FAILURE in 57m 22s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6d6677edbfd043c7926b362b12d816c9 : FAILURE in 56m 29s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a16b0cc172b648beb72027d89938f15e : FAILURE in 1h 10m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7a08cabce6c408eb00a7c78f1761e2e : FAILURE in 1h 16m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4dac6294009c4dcd9d35d5c0e7ccb3b7 : FAILURE in 15m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/954df3437ba645d2a8829e8d47cc4664 : FAILURE in 3m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03dc058cf3e2459a84cae6b28717f9f0 : FAILURE in 22m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7b7d00bbd482469a85447ed63ec9fea2 : FAILURE in 16m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8729ad1c1732499a8ba6c03b1581b0e3 : SUCCESS in 35m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f64bb4f80dcd4483a9e37386b7562719 : SUCCESS in 23m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ad833512e664252975a6fde724d2016 : SUCCESS in 36m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b745081e6254e2fa39d43cd76658ca7 : SUCCESS in 27m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/de6ed2922d744fe58c2b5a53146f10b6 : SUCCESS in 27m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c957d01c03940e69457e94d8663a661 : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":16},{"id":"27006e54f8efcb6362d40d56f208bf4e206b05cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 12:13:26.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"7d4849c19e67cfddd82d155fab84512a92e5807e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 12:20:33.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"2984cfd68b65eb256025027cc9d33ac4b5712d37","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 12:25:20.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"316f226f8ca8a4100b7bd8849c630450004b6be4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 12:40:02.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"0f86646fd63f60522bce9a62eb5aef32f3be04ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 13:00:55.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"cb26d254250cc3b614e47798fee9684c4a2fb411","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 14:03:41.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"98b1676853223c32c2b6b3460a22cbd84a65d8b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 14:37:12.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4a75eef33fb4a3fb34edd046d652523\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0037d4f68f264bc4ac7ce2f36a7211af : SUCCESS in 9m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa2139b196da4a1daae65c4c0eee05e7 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6e65ed83a1841508babde6ad618de02 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/514c0fdc8f104f80be0d298de5b36fe7 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28109423e2c04ac39a07dd359b811afc : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c909ec82c024b88be4886902660c2f2 : SUCCESS in 6m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3787c34c393840ffa65d262e793c2cf1 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ed37d3b74e04275801c99e98e5a89a9 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b34feb7115ae44f783c6ed5e34c24265 : SUCCESS in 15m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4eff1e4d5c99477abca7e08961317243 : SUCCESS in 7m 25s\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/83f414be737f4c9c80e42c0095514b82 : TIMED_OUT in 1h 31m 59s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/4a9af0250e5249789c17717e5d7b6c98 : TIMED_OUT in 1h 32m 10s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/82685fa971074e63ab9a9bcd8843420d : FAILURE in 34m 07s (non-voting)\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/86db6c56ed3244a5b87d8e2cd07f1864 : SUCCESS in 25m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/821044eaa46d46119fef8bb20e66a013 : SUCCESS in 14m 28s","accounts_in_message":[],"_revision_number":20},{"id":"88676763061c147a3566cfe4ad48c2b1bfadecef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:18:49.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a27326bb53dc48a786db12f4d2f73143\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6df218103c4b71848c5099e714087e : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6fbf2d28d64a4cfd842fb4b7d72f0e72 : FAILURE in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0903c7d0cd442a3b4418d7a81bd4759 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343544f968b74652a10b8bd4dbd4e7aa : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/179e0d0bae114d11806ffa1d95f0f7df : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8bfc27fd5478419d916bae338416e41f : SUCCESS in 5m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a262d2056e84e48909f34728132e16f : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06d2d05f60b84567b3d4461dc2208ead : FAILURE in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e8d4d800fb14c1b88931292e7f98817 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/935216fe6ec44b638fc5d954b56b8aa1 : SUCCESS in 7m 20s\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/ade68bbe03834149b2c554b7cd48db63 : TIMED_OUT in 1h 32m 57s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/99183380439e48c39214ef3aa603c825 : TIMED_OUT in 1h 32m 54s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/5dd5012c00234d8f8449430c0d877393 : FAILURE in 19m 08s (non-voting)\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1d9a24010b9549b6a666ceda09593918 : SUCCESS in 10m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60b9aca9d42143f0812d05dbbcdfd19b : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":21},{"id":"956f45b883342c5ef7555e50feafeb15ce413ba5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 07:39:55.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"c70e73bc6bdb3817de8cfd7e373b65d7549a9667","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 07:50:55.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"69e2405940396baf6b0bc79d911c8c08c229554a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 08:04:56.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"ef8ace7c574d01e598e1f44f155c63e3da2c4bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:39:22.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc45657c2ae942819f52326c8547bd56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107299a3621d46ec86ef36b262312b91 : SUCCESS in 8m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3809b1679db246f28da2f398368f47e8 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f28d16e6b534c469d850713bf1224a5 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4c994ca4533473b8dd5e416787a2ef8 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3da9a5ff18a41e8a3035a1555a0c43f : SUCCESS in 5m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bee68eba9bf0468783043c31ea4d9260 : SUCCESS in 6m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03b200ccb258428eb2bec7cfbe0db12a : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f3acfec8a644cbba2b7cdf0ec0d3119 : SUCCESS in 3m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd9bd440edb74a1cb5a3bf117f766718 : SUCCESS in 15m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/43d8ae8a83244604a780bd0590c6582e : SUCCESS in 11m 22s\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/2c4e1ad2fff94f2fa78cba313098558b : TIMED_OUT in 1h 31m 26s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/f768d93d91b8443faa66ed2a4622419a : TIMED_OUT in 1h 32m 18s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/ad496aff7ae74a3993caac5d0f567c17 : FAILURE in 13m 12s (non-voting)\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d94ba6d0a3104307a62a6126fca5ea4b : SUCCESS in 14m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f722fbd10a4e4fb8b6492c4acaf01735 : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":24},{"id":"85ac4e2775602eb76a59c68e0069ecb09c79388f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 12:19:45.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"8fd3dff2cb30fc0ea8cbdabe86dd3300ee55b335","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 12:40:07.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"a059e77d07a68c3721b5bf6b3208e64ee7327881","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 13:07:10.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"4bf5254ce2036c5409db3146abe2d94332b0c723","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 13:32:39.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"a90d05dc6247e9feb6c5dd1b23dec3732bf40e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:03:42.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02423a7bde5940e294246c35652901c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88df590431664bbfad81d897fc19bc53 : SUCCESS in 4m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d88deadecfd441d8b219a73fd5af6c77 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efc546a7f0934a5d930ee539170439d5 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee9ff354be34f3fb35f2d7e102a93cb : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d60ed7865e24bdc8fcb0fad0f049a0b : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f70a9b232f774903b2c4a103ce01e8bb : SUCCESS in 8m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b09ae4c5e8504e14a222639941416d03 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/499d2c41a32847019f53fc8b9a432d21 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6b7542720c74626b877e7ca05001aa4 : SUCCESS in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe795c0185f842b4a4502151b384c60a : SUCCESS in 10m 35s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/97247d68c3904916916603b730932c03 : SUCCESS in 12m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09fde7f37d0a47f09dece1d28794a40f : SUCCESS in 12m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f66748e697d649828b5e1b8b6ccd47b1 : SUCCESS in 19m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95bb3a128085493ebaa44cbdf81d1091 : SUCCESS in 11m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c2589bf68815477c9521b4090bffcb58 : SUCCESS in 24m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4f462c5d75545a9af8e4e18b8f6ddd2 : SUCCESS in 18m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/59052df8d6824d18bad94c336a937eb1 : FAILURE in 23m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/927cefbe593f4b59baa0a1616126b725 : FAILURE in 18m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/803a1031443749f9a5477f075612fa51 : FAILURE in 27m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff468e8d312341f29a736e2aeb4d1c37 : FAILURE in 18m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5d22efee5db4b0793de3e80776e50f6 : FAILURE in 18m 04s","accounts_in_message":[],"_revision_number":28},{"id":"40524f46a18f20ca2fc6a0bf93c03bb261ac8347","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 14:09:09.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"50f6a3afa4a0e603ef959670d2437bd8563c4b00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 14:13:20.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"f5761bf600445bf9858f38b8cc76300b3fb66106","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-13 15:01:55.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"39bbeafa23fb723eb930a38dd0f00755df62e4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 16:20:36.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72f37c1c508f4f1db0c4077fe11212df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa1129586ad4f8296d75a3dd2c9cded : SUCCESS in 10m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/45bd6bd904e74975b47a2f6412474d75 : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/756bc309046a4cc3ab22d32df8f3773e : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/152caba7a6ed45c7bd61f9c96d05429b : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ce84de055a244a99d9f8638825ab513 : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20ff08007bbf432ba10a753bacc0f9d5 : SUCCESS in 5m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5293d3631fe48798e9efba31e492e64 : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75f23d0066244d91ad39aef385a68310 : SUCCESS in 3m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0207211608624303bfbc2b3d10b26d03 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a84b5461fae4edf84da394bf6213ebe : SUCCESS in 9m 58s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/86e4c49e28084b049bcf866f8dc142d0 : SUCCESS in 12m 06s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/1477047012f141e282b45aec88e29f0d : SUCCESS in 11m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/259f90baea5f46f5a57d45dcf1798b71 : SUCCESS in 10m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8662496cc7764752a13563b561ca19b2 : SUCCESS in 10m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/181350b9bedd4383ac0989780ab85750 : SUCCESS in 17m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99e7a060d890474090c5de55e2d0d00c : SUCCESS in 12m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/23fa974c765042dd9f144ca49d940780 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/174a8d56a7254a1697ef5e4ae87b3249 : SUCCESS in 17m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7de2d2fe36504a2aba2dde9b4df9dc67 : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/61cacf461272492ba30f09c359cefc97 : SUCCESS in 1h 14m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7d48a3de6f444bc488125be138ff9dd4 : SUCCESS in 1h 15m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c147de3a429247c6b01365ac9bb1f246 : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/901200d7dc70485bb0354968d0c36b6e : SUCCESS in 1h 05m 51s","accounts_in_message":[],"_revision_number":31},{"id":"8f47c4167041a3d20d4e0e90ab35952f153d1714","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-14 10:22:19.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"0e9a303767d287f1f6713b828506fc8871eb8fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 12:05:03.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaf2b414a24a45499bc1fcb72887569a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f798202db5354c0990164293112c7abf : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a3453cb0e7914610831fe5a9c2fd878c : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b530656406648d39f056b82b1cca183 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9332c88441446d6a9602e02df6a716e : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95ac3020094443c585394c1d311fe32d : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/105aeebbf4c14d1d9adf7a35a87e1c6a : SUCCESS in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d77e38a4ca2418d9c7fb807c8e33544 : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4723f72ed2946378453d6c338cb4307 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c431e17cc5824a44987d6e219568e301 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/42a4c16f78334abdbfc5af7e329d0fab : SUCCESS in 10m 43s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/910e2ad775824b6da7175a588b045078 : SUCCESS in 10m 51s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a0453358930c43c38c6d12dad897cae9 : SUCCESS in 23m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f58e429c3e804430b01c0744cea9ff1b : SUCCESS in 16m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0617611ebd154ce2b0a646193e97108b : SUCCESS in 16m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d68b7df2df94cd1af5e679a62bd6872 : SUCCESS in 20m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db1cb26e7dce474287e83d4ea0b2bf0d : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6779d6b23e164d52a07c4f1505eba296 : SUCCESS in 10m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dfb1109641141bdb8d482e2f3c3b342 : SUCCESS in 17m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/472b8bf0fa954768ae7c6c5f8050d440 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0bd4c5a10f9b4c0d8a31420986b77b9a : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6869a54e264642679873124413942565 : SUCCESS in 1h 05m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d129772ebdad47c1a9aab1d75ca22f52 : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a4389390be0e494cb1b8948f14c716e4 : SUCCESS in 50m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/de700f0c8ac84b498f1a3d0ec953d78e : FAILURE in 1h 28m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/65e6c16b0f584d72a06737923265ec45 : FAILURE in 1h 21m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8c1b375494346828356cdf7964a85cc : FAILURE in 1h 35m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c835cd1664e24fb4ab704fa01febf980 : FAILURE in 1h 39m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c077bb4c579e4f38b55d45ed468af176 : SUCCESS in 22m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0e79837805254c94b4c764140531d42b : SUCCESS in 14m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b547674a3dc413cb350fb56c395253f : SUCCESS in 22m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/240a5acba8ba4c95b49b758fa66d2873 : SUCCESS in 21m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/edf5bb24119d4b6ea882c053d5e3de6d : SUCCESS in 24m 46s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/05f6d7ca992e4f349c512a9d1a11e5c1 : SUCCESS in 20m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b442536e2584b7cabfca10b6a29073e : SUCCESS in 30m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3eee3bef41ed4248a42392ac5aa29531 : SUCCESS in 14m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/02d796cdcce6400b8326c8d5455711d9 : SUCCESS in 23m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00ad4d62b56c40ec89bb77e698e68865 : SUCCESS in 19m 31s","accounts_in_message":[],"_revision_number":32},{"id":"2196b900bd8ca8a17c190a635ecb56aa48713bb1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 07:39:00.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":33},{"id":"3f18e41bd89e9e3752ab5fcb306476e4b75c2034","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 08:11:18.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":33},{"id":"1244d2a886677ab490a10a87d4c466f162dc846a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 08:44:34.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"ced2d329bcadad06f43386be46d81a953cb41310","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 09:09:56.000000000","message":"Patch Set 34: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":34},{"id":"f1186c5beae6e2b37231dfaf6ccb68f8906e20b3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 09:17:12.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"7f12b9b076a43d0da978ac75d887fd8772b54e1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 09:59:37.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":35},{"id":"d262a3c59eb03111403f4c26d4329cf3cab02c6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 10:56:53.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"2631b17fc16d54bce74e90dd082c9cd5b8cdd583","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 10:57:07.000000000","message":"Patch Set 36:\n\n(4 comments)","accounts_in_message":[],"_revision_number":36},{"id":"fde6f1b6605058460d3ab45ef1ec5c07a563b049","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 11:24:39.000000000","message":"Patch Set 36: Review-Priority+1","accounts_in_message":[],"_revision_number":36},{"id":"3bb7fd50deabc96244d16706b2d073bbe8dc59fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 12:34:41.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e85fabd36a1f45908a3b43571385d2d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3a6f20e70340c589cb5578ef919cc5 : SUCCESS in 8m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3e5f2f369ebc445e9de98945266e2c83 : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/056ff07b1cd045c4b5b997fa6289918c : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105b7078d61246a5b2382f3982b6c3b1 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60a00873d0fb440c84011650a5bb9acc : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6516067d31a47eeb4220f63cd184f65 : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75567364dc7a41a38a9d1e6fe38f0c13 : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5af2de2d1aa644e09677b5a2384bf61c : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d3fd3cf0823449be8422604a08c984b6 : SUCCESS in 7m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0a07e4424ca84e489e6e5e93e92ebc36 : SUCCESS in 9m 30s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d6e494ae04a341f09ba392225629204b : SUCCESS in 15m 01s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/efd182c484c24f6eb1b139345bec322a : SUCCESS in 15m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84951bd8ba1b436988cca33680e1bf9d : SUCCESS in 16m 23s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/acacaf91dce64db59d3cd66528a1ba95 : SUCCESS in 12m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7721352fd8404113912127f546649256 : SUCCESS in 15m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8e7cacd54724a43b25b0cdc499756ca : SUCCESS in 20m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bee60cf368ee4c71aae2c6622e7bd2c7 : SUCCESS in 13m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/742bcbd99faf42d1896bb26b06acd624 : SUCCESS in 15m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cfdaaaf9f9ce4bfb955d01ca21377110 : SUCCESS in 1h 03m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1b648cdc8e8542bb90bd9f0c5c44874e : SUCCESS in 1h 15m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a9b7d723699a468fadd76d4d94a9c29b : SUCCESS in 1h 14m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/524df096b5714ec7b82da0d24b90ce8b : SUCCESS in 1h 08m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c23bede35424e8ba5f29459d594b398 : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/42ee209c4a444aaca4af551e97d04ba0 : SUCCESS in 1h 36m 02s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e45a291aa4c64af4822d23a35f6e1e0d : SUCCESS in 1h 34m 15s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a91089ac5326436f8fd8c6c583a9122f : SUCCESS in 1h 27m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea6c7f4322ae449b989ae3aab2002229 : SUCCESS in 1h 35m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5da625bec000472083424450961e9ba7 : SUCCESS in 22m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9c6428ae4e89482380333aee4751029f : SUCCESS in 16m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b07dbad34bb04f96b983da9720de2bab : SUCCESS in 25m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f1f83fc9542e4ba990081b4af2d1e96b : SUCCESS in 14m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3fa7ce181ded458297b808bd16c97e6d : SUCCESS in 27m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d6a55dc0c40a4261ab08fb9398c0d3e8 : SUCCESS in 18m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a32a1406ef4242b49856f9ac6eab7064 : SUCCESS in 34m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5412db8ab1e94fb1834feec4067da2c2 : SUCCESS in 34m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/467ae7a461fe49fcae4fbcfb1dc53074 : SUCCESS in 21m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cb685d547074a8cbde96b9c8284e1ab : SUCCESS in 20m 24s","accounts_in_message":[],"_revision_number":36},{"id":"8e7588c8521f1885d5800a2c4e53e7ce30390d4b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 13:02:14.000000000","message":"Patch Set 36: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"6899b68043deaf06e070f72e0f1f8b55f4037c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 13:05:40.000000000","message":"Uploaded patch set 37.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":37},{"id":"3a5237f2733aedf9546adfa14a45f469d6cc69f2","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 13:06:02.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"8bd7de7eb2f8f053b8b3be05045abde9c5eb838f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 13:06:44.000000000","message":"Patch Set 37: Code-Review-1 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"02740740a85389f4409c83d40b1b4538bd277673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 13:34:56.000000000","message":"Uploaded patch set 38.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":38},{"id":"cde4d3ce5fafec104759772a0239970f0cd5d9f8","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 13:35:45.000000000","message":"Patch Set 38:\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"bec00c95f444f58464a04f063d8944bd23cb4ad5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 13:48:12.000000000","message":"Uploaded patch set 39.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":39},{"id":"acc9d76616c3051359b3f51474f8b099973fa162","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 14:49:46.000000000","message":"Patch Set 39: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"6ca1b88ca0da34253c44a487fc2a7e8bde33dfe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 14:50:07.000000000","message":"Uploaded patch set 40.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":40},{"id":"396309998b0f304e2afe2c2f320edb0f43d4ec0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 14:58:25.000000000","message":"Uploaded patch set 41.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":41},{"id":"602cd1a7a689918b4c7e439146aabcd5fa6d3d84","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 15:01:43.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"efb9a5fc7fceaed5dd7ff251f3b9c523b3cc96f9","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 15:07:29.000000000","message":"Patch Set 40: Code-Review-1 Review-Priority+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":40},{"id":"60063ed60c36190742810220f7392e30b41d7afb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 15:16:12.000000000","message":"Uploaded patch set 42.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":42},{"id":"a33660b4f869a69119ea421584541c053440bc87","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-20 15:18:01.000000000","message":"Patch Set 42:\n\n(3 comments)","accounts_in_message":[],"_revision_number":42},{"id":"ac925c8f60e2677eee5c8ccacbd7298c84f3f907","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 15:28:04.000000000","message":"Patch Set 42: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"80e9494dc930e79a073739f7fc715979c8a702bd","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-05-20 16:46:59.000000000","message":"Patch Set 42: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"8331843ec080257e2aa5ef791bd5888dec731c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 16:52:51.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/617c4eaa43274ad09ae5dfd0cdf77bd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea29e9e3061248d2b7e1fd42b5ecdc22 : SUCCESS in 9m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8fa0db917a3f424abc12d038f025e3fa : SUCCESS in 8m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2857f8bd4114fbbb121a0d9b4a512e4 : FAILURE in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/553a4acd645940ea922bd14f115b1251 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3092b8923589476bbfa3044590b08519 : FAILURE in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/594a8ad05e4147b4baee92294d16bf1c : FAILURE in 4m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d3b3297dbbd4d448b580afe1c233295 : SUCCESS in 6m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c9db1bd739c94aceaf60c6c430dfbeef : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da4feb6ad1024ed282dcba6df55c3e21 : SUCCESS in 6m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef52735f132f4d4eb19905809ffc6f6a : SUCCESS in 7m 01s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e8ae9ae3fe864d25ad112bea44ec5ff8 : SUCCESS in 17m 15s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/73b34f1a70fc4cf8b34fe71ee221be1f : SUCCESS in 20m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/850be24586f448f9ba299a154d030fcb : SUCCESS in 16m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/203291b61c544524bbeefaa4d0592401 : SUCCESS in 12m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ae69183ecb394f739aaace65316acae5 : SUCCESS in 18m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1f039b047d24eb2a90eedb500fb97ba : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/22a9ccbba8b74eae91aef8aa3e0fdac6 : SUCCESS in 11m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f67749b48e7e4c0ba543281949dce1c9 : SUCCESS in 18m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/84db91826153428cb59d6e55b4a8832e : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/59555872e32d40f1ba7220b013b4095b : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9943f89836f64e4eafe45d5c96108adb : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/702be98efc6c4e11a2f7d5b406100eed : SUCCESS in 1h 02m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ac5452ad0e8e42e19d831a51c52ea539 : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2acb09e2016145329d141e29b514a623 : SUCCESS in 1h 34m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/abb4f4e440b54f238420ef9e47324b99 : SUCCESS in 1h 29m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd38d2b6988c448382a5495255125f48 : SUCCESS in 1h 29m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50fb6b150fac448bbb3eadd9c3d97aac : SUCCESS in 1h 32m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8d135a6bc7b341f6bda8ecc5fb538158 : SUCCESS in 16m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f26ddec534504911b79bb8cc2a15132c : SUCCESS in 24m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d45eab2ce6442868daaeadef7b35e5f : SUCCESS in 25m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba6e41250c8544a6b75fb3915d900333 : SUCCESS in 20m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1f5ff9079e1a410693419e39f3533415 : SUCCESS in 36m 20s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f173158e382641ba929b6a2cb856ddf8 : SUCCESS in 37m 04s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f44444c39df4ec1afad923bb7373cb1 : SUCCESS in 29m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/743b280a8779460da90dab3ee5b97d81 : SUCCESS in 28m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ad1be9108aba4cd29ba182cdeccb4171 : SUCCESS in 27m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/047c7dad5d1f4401bfa8d8d9e7db3b92 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":42},{"id":"a447fdbf7f197d5c6d03dcea18968c1fff437b7f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 17:14:34.000000000","message":"Patch Set 42:\n\n(2 comments)","accounts_in_message":[],"_revision_number":42},{"id":"a2fec458d843d145842ce7dec68e21a89abd0429","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 17:15:16.000000000","message":"Patch Set 42: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":42},{"id":"e8ed2aa1784633293d72bd777c109a5cdd8d35d3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 17:16:23.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"5e743f9f3e668eb5454c07e43891144967965619","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 17:50:03.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"0cbb0ee6067e1a962b0ee3b5a9b92d3e2c4b0d1d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 17:55:47.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"609e6a1d47904a77b22dba0998a9096d7d28bfae","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 17:56:11.000000000","message":"Patch Set 42: -Workflow","accounts_in_message":[],"_revision_number":42},{"id":"df0d1645aeaca2f966f6139308aa13dd66c7aace","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-20 18:02:56.000000000","message":"Patch Set 42: Code-Review+1","accounts_in_message":[],"_revision_number":42},{"id":"9b6b686fdefb3e050908632ecfaa25029f4412a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-21 08:36:18.000000000","message":"Uploaded patch set 43.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":43},{"id":"44fa328b8aa4eb40ad1f97d92b77379af473fd55","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-21 08:41:24.000000000","message":"Patch Set 43: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"da318ad60f1c50a6263a9083992883d00f625d55","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-21 08:41:30.000000000","message":"Patch Set 43:\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"a446c033e1095c55b250655feb402eab3cbe0131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-21 08:44:39.000000000","message":"Uploaded patch set 44.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":44},{"id":"3c12c2bf3b25f63ef443e5fdf0f8cd667bd84eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-21 09:12:15.000000000","message":"Uploaded patch set 45.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":45},{"id":"5902b7ea55512deea581dca31ddd65111b0b6e95","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-05-21 09:18:37.000000000","message":"Patch Set 45:\n\n(4 comments)","accounts_in_message":[],"_revision_number":45},{"id":"c41dcc4b3c81ddf24fdbb4f99b83af3697802c10","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-21 10:24:51.000000000","message":"Patch Set 45: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"0f443c29d54548f76935138c5e694c94632af5e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 10:54:22.000000000","message":"Patch Set 45: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3b837b662cf4f82a1468ddb601aff9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5edba2679334077b6bc7533aa7fce82 : SUCCESS in 10m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8f571a9a10564bf6b7c5668d0832722a : FAILURE in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3642969895464f2b8c3bc2e23648ba9f : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e948c3b4980048f9bb8b5aeaf3f583a3 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b02995d5ee2241c2bba42edf73dec4f2 : SUCCESS in 5m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5fc191ec9a0f4dacaa5918d25de1a623 : SUCCESS in 8m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb238ccad5834453bf21ef192540460d : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/abedb8100ef64c56b037c0c96c4e99c2 : FAILURE in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d8b7df6fcb964ffc8c4db1efccbc23d8 : SUCCESS in 14m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2648b835218b458c911d6c418a522d91 : SUCCESS in 11m 27s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f8a78465282d4a3f889b0252fd416150 : SUCCESS in 10m 48s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/955603c9d67640449546f01ea87c2454 : SUCCESS in 17m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9dd6053bf14d4191addf1f3e3693dbf8 : SUCCESS in 20m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e6308449dfb444385481eea4b183246 : SUCCESS in 15m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/135d5eb01daa41a988e915fe787c8edc : SUCCESS in 14m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f7522439b04404cb9bbf869db7ad87a : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b4c03e55ce4747fd88b24550627e2393 : SUCCESS in 25m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f889a80a008477d8e1a82141e99b427 : SUCCESS in 17m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9047f4ea509548d197209bcf998def69 : SUCCESS in 1h 08m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d1793900f65e4e2a85d0c7ea5760abe4 : SUCCESS in 1h 02m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0ccc7b9c4b38430f851a4581b86b2405 : SUCCESS in 1h 04m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1a0c0f2e63843e4a27baf7bbd960f6a : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d5631a0f60f144ea9e86a8eab12781e1 : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/db5ace8a7a9b4f2c9c8e72e96bdf6365 : SUCCESS in 1h 38m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3888f8fa1ab846a6a21f6c8040ef9097 : SUCCESS in 1h 36m 16s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edc17a68b3b845a7bbdd20472454ff73 : SUCCESS in 1h 35m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4a87dd74ef04f8eb8750d29457c8c97 : SUCCESS in 1h 40m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9ceb899033b74808a385539be4b9bf5e : SUCCESS in 15m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/abd8e5360e254a6b9595d9ec9b57396f : SUCCESS in 28m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee770d375f7147cc85270ab951b3b691 : SUCCESS in 14m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/564107fa280b4c12aa288b2e24e7cee3 : SUCCESS in 26m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/71d1b0ab8aa2450aaa111fcc68c5032f : SUCCESS in 29m 43s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ba9a055308a546d18bee5901a04e886d : SUCCESS in 37m 34s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3764771055164877b796a5b191bc2de0 : SUCCESS in 29m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ff8eeca9d5c42c7bc83d6e8ba6c4306 : SUCCESS in 36m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8402a886c8be47c382c13e1d8b9f0b5f : SUCCESS in 14m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd0014730fa94433baaaed13dea320ed : SUCCESS in 20m 09s","accounts_in_message":[],"_revision_number":45},{"id":"68e8ebd53eaaf3344008874d3e6b46d7e7086a8b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-21 10:56:40.000000000","message":"Patch Set 45: Code-Review-1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"e980167de3c0500e37dc6ff51e89e732074f6f7e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-21 11:01:38.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"6e2392c6cbd07b12330e9b2761bfb6fd0750565c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-21 11:51:43.000000000","message":"Uploaded patch set 46.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":46},{"id":"4f1c25761c384b59e93e6ee0e1f31e9bfe302b68","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-21 11:51:54.000000000","message":"Patch Set 46: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"0aca2197c48b1331b6d937766747869ec3287ef9","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-21 13:27:34.000000000","message":"Patch Set 46: Code-Review+2","accounts_in_message":[],"_revision_number":46},{"id":"2eb959af042db552be9e0e2b2e51fd2b38de22fc","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-05-21 13:29:34.000000000","message":"Patch Set 46: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":46},{"id":"c6c08a7c3a1d38ba1b36a87e6b509e02deb0c617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 13:31:28.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b7d7ad5b9449079a94a3c2f69e9889\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a92816427e4bb4b164a365f0c72b20 : SUCCESS in 10m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cdcbee101c8c40219993dff4a4dd103e : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8aa3611c67ee430f97473110a643d1ba : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82a194ee3cf4713963fd888fd60bb0d : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/407a02d8197b462893b1b94bc1edafb9 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb8bd137b28941008c0a01e41d387dcf : SUCCESS in 7m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/370fad5b08ba4355b0db22512182b9e7 : SUCCESS in 6m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba39cf0625bf4e05bd10df4ddea38e10 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b92c17710ef40e19908b6233caebf6f : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8546e3b3f5c04af1aa3021322cc018a7 : SUCCESS in 10m 07s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/faab3da68f514ca7bbe28182a8bdd097 : SUCCESS in 16m 23s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8dae99e7a2b74bb29a3bb93d8099fbf1 : SUCCESS in 11m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64c7420042fa4c3084150714f59985bf : SUCCESS in 15m 20s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8e40f7e73edf49008e24f9b0b913ff4e : SUCCESS in 16m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/514d47d136334741ad33321a4e5caaef : SUCCESS in 14m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca47c79982bd45e4ac1710d6895f367f : SUCCESS in 14m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d5b3fed474134471a20fe606fe6cec43 : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbc6c2be22a34ef7889f09a4e58deda3 : SUCCESS in 18m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/35566dc01621487cb510eb945338d65b : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8029ada231cb4232a627e90c9ec7cb35 : SUCCESS in 1h 06m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d8467962dbed4c5ea46d75ee4c2e73c9 : SUCCESS in 1h 05m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9be52ab0ccb0485b83789499b1dc733b : SUCCESS in 1h 01m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0844aadd56ee402985da262463707277 : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/da623977b2084b8b865fee0de0fd6f15 : SUCCESS in 1h 36m 02s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/95227c1a3eea46c990ea1054a9e1b7af : SUCCESS in 1h 31m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c7cc1571d5b4e9eb2fdfb8bb8c9ca46 : SUCCESS in 1h 28m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c6705c8424f4922a81b35e8bf6c5ae2 : SUCCESS in 1h 32m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7a0939cb49094da4908c5f18962f15e7 : SUCCESS in 9m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8ab7b04778a7472eb06ca48cfbbd40a4 : SUCCESS in 14m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/678e93e8d8564cbc857f803ac74177ac : SUCCESS in 22m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/32d595f5d8be46b9b0993c3e4de5e032 : SUCCESS in 25m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0d904058a10f4dc78e11e9546143e613 : SUCCESS in 24m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/86b7ce038355419ea52b2735a6d0e5e4 : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2c2d197670c4c06a0be9cea48d069f6 : SUCCESS in 32m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/566adfe7f9e441198cf8da74cf3cfaf1 : SUCCESS in 33m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e22c3e312d2849e38b75a800edb3fbd1 : SUCCESS in 21m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c56140888204ec485d96d3586b6dffd : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":46},{"id":"8205d5fad39af7bcd4d22fa300eb35126cb6ddbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 13:32:01.000000000","message":"Patch Set 46: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":46},{"id":"987166209a414f3df15164a99a08aa97b39e6a99","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-21 15:02:16.000000000","message":"Patch Set 46: Code-Review+1","accounts_in_message":[],"_revision_number":46},{"id":"178ee4bf717f831ff46b9b1b647cb88b1f06c85d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 15:10:46.000000000","message":"Patch Set 46: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c00dfb9988b4473e87e334644af5d859\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52150f29b43f438985beb3366ae630b5 : SUCCESS in 12m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e3896364256b4c54a08f6ebed62dd01b : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c602066a3e34748a3a27eaf860ec1b7 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56d8cbbb83b9486298f78f8e56ab5ace : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7071c8cf977c4fd59090a7c89b09e85e : SUCCESS in 4m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/18e252fa092841f3bd5919c0e9656fe6 : SUCCESS in 3m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/929a148672194740aa18b2b36199ff9e : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/31ce3f981f234b0a9093871cce9a1a22 : SUCCESS in 6m 33s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6de0912d401643428481972f213b46fb : SUCCESS in 22m 00s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ee86f749e08492db7f584b1eb845648 : SUCCESS in 8m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8941efb18ede4992a92f73e80d863e15 : SUCCESS in 13m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aacaf0e21efd43ed986368f139343ec0 : SUCCESS in 11m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9f2f52b5570b4aa591888a40f36ef5c1 : SUCCESS in 11m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/077685a322014a45853e4f16d1d42929 : SUCCESS in 22m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/59f6d089f1844323be0db057db9e7268 : SUCCESS in 1h 03m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/783e69b00d3c4fbea131c685a562c84a : SUCCESS in 1h 09m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2e69a4a5673648a6b3ef18c4c778892b : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/271a6814bc174ed29f78bf9289af163d : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/76a537b366064291acece0865385c6a0 : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/60851d4f0414496eb7314e9169021cc0 : SUCCESS in 1h 37m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3a4bd85b8a0e44e48460e2fe722f2705 : SUCCESS in 1h 32m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9cf40d86e2b486bb99c4bc08f564a5c : SUCCESS in 1h 28m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a16a0c6e4a524ddbb70b7679847374b3 : SUCCESS in 1h 37m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/648d28fa7a3240c6aecbf6e382b47945 : SUCCESS in 24m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e72c08fd1e064217a78ca7937c0e7c9f : SUCCESS in 14m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/734a4d686b024b11ac53ab3d718fe06c : SUCCESS in 24m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9aedaf63dbc24ad18065de96ac605d34 : SUCCESS in 24m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3ec48256b03f4a2284f4312ca125c12c : SUCCESS in 35m 51s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/11024cd4f4184c578b3c01154f3c39dd : SUCCESS in 16m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73ec27a4a58e473aba654b6dd6a6c7c5 : SUCCESS in 38m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6bac0df0ed54f1f92ef424c582853f9 : SUCCESS in 14m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/21c55031d79e42cf8e1f969f7e24d0ae : SUCCESS in 21m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cd14b23157b4693a8f7a8d6d1292a39 : SUCCESS in 25m 06s","accounts_in_message":[],"_revision_number":46},{"id":"6b702c6bb897385597a360a89e8feca3ac78c70d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 15:10:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":46},{"id":"2851c77b7b849493847134ce2514395e2debe02a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 15:13:40.000000000","message":"Patch Set 46:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e3c8956d62426e959f7c40a6e87623\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad21b4664b3448d68e7b50567d3e194f : SUCCESS in 1m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d51f875d73a14d4b873b95224e4aff86 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":46}],"current_revision_number":46,"current_revision":"18cc4b921f35aad396e0bd4e198ec45622cf1e5b","revisions":{"209a49475bf2567fe894cdea0696f0d043794ec6":{"kind":"REWORK","_number":1,"created":"2024-06-20 14:06:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/922402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/1"}}},"commit":{"parents":[{"commit":"6f63557dcb656424e3fa786b4b498f60cff226d6","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f63557dcb656424e3fa786b4b498f60cff226d6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-20 14:05:52.000000000","tz":60},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/209a49475bf2567fe894cdea0696f0d043794ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/209a49475bf2567fe894cdea0696f0d043794ec6"}]},"branch":"refs/heads/master"},"0edb0f152b9922efea326c3ae5a1f0d7d6638650":{"kind":"REWORK","_number":2,"created":"2026-01-21 16:00:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/922402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/2"}}},"commit":{"parents":[{"commit":"ffd5d0279b3a446a96f13f8593f3c11b2e771421","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 16:00:05.000000000","tz":60},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0edb0f152b9922efea326c3ae5a1f0d7d6638650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0edb0f152b9922efea326c3ae5a1f0d7d6638650"}]},"branch":"refs/heads/master"},"c5ee30de72d3a7179b89698cdb43d78276d2f2f8":{"kind":"REWORK","_number":3,"created":"2026-01-28 17:07:32.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/3"}}},"commit":{"parents":[{"commit":"ffd5d0279b3a446a96f13f8593f3c11b2e771421","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-01-28 17:07:21.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5ee30de72d3a7179b89698cdb43d78276d2f2f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5ee30de72d3a7179b89698cdb43d78276d2f2f8"}]},"branch":"refs/heads/master"},"5ea066a4b716c5b84765784c465f283366d4f638":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-28 17:13:09.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/4"}}},"commit":{"parents":[{"commit":"6e5bce58bc7b53e2b2059f3a653b26703eca749b","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e5bce58bc7b53e2b2059f3a653b26703eca749b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-01-28 17:13:09.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea066a4b716c5b84765784c465f283366d4f638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea066a4b716c5b84765784c465f283366d4f638"}]},"branch":"refs/heads/master","description":"Rebase"},"cb4cb55935bbe96b00d9ee61fc6179e7041e6939":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-02-18 09:10:07.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/5"}}},"commit":{"parents":[{"commit":"adaeb2c5b3acb72d82043139be3c264378d06b7b","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adaeb2c5b3acb72d82043139be3c264378d06b7b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-02-18 09:09:36.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb4cb55935bbe96b00d9ee61fc6179e7041e6939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb4cb55935bbe96b00d9ee61fc6179e7041e6939"}]},"branch":"refs/heads/master"},"b4b3f49e635798fb3a8adbca6584bb0fbe9b077f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-02-23 09:50:21.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/6"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-02-23 09:50:21.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4b3f49e635798fb3a8adbca6584bb0fbe9b077f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4b3f49e635798fb3a8adbca6584bb0fbe9b077f"}]},"branch":"refs/heads/master","description":"Rebase"},"8d16dcc2c59492cac0136af85b6d1b3ea05af9ec":{"kind":"REWORK","_number":7,"created":"2026-03-18 17:33:41.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/7"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-03-18 17:33:39.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d16dcc2c59492cac0136af85b6d1b3ea05af9ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d16dcc2c59492cac0136af85b6d1b3ea05af9ec"}]},"branch":"refs/heads/master"},"cc61753052f3a8fd12d89a55db1e4d405a3a549c":{"kind":"REWORK","_number":8,"created":"2026-04-08 07:28:56.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/8"}}},"commit":{"parents":[{"commit":"290694c0feb523a85b4812c7625a6ec2f6698a45","subject":"Correct reno for kolla tags/limit deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/290694c0feb523a85b4812c7625a6ec2f6698a45"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 07:28:31.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc61753052f3a8fd12d89a55db1e4d405a3a549c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc61753052f3a8fd12d89a55db1e4d405a3a549c"}]},"branch":"refs/heads/master"},"b292034cde47f932b0cdfbcc9f71992691310e1d":{"kind":"REWORK","_number":9,"created":"2026-04-08 08:22:16.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/9"}}},"commit":{"parents":[{"commit":"290694c0feb523a85b4812c7625a6ec2f6698a45","subject":"Correct reno for kolla tags/limit deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/290694c0feb523a85b4812c7625a6ec2f6698a45"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 08:22:05.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b292034cde47f932b0cdfbcc9f71992691310e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b292034cde47f932b0cdfbcc9f71992691310e1d"}]},"branch":"refs/heads/master"},"de21e1232d4da4b6dbdf9be1159a89f870aa7fed":{"kind":"REWORK","_number":10,"created":"2026-04-15 12:48:01.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/10"}}},"commit":{"parents":[{"commit":"a5c425c04331553f41f236ead2b195e5c9245361","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5c425c04331553f41f236ead2b195e5c9245361"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-15 09:16:25.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de21e1232d4da4b6dbdf9be1159a89f870aa7fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de21e1232d4da4b6dbdf9be1159a89f870aa7fed"}]},"branch":"refs/heads/master"},"5c783d8ed57e70408b388fc79e8792a085efd97e":{"kind":"REWORK","_number":11,"created":"2026-04-15 13:51:48.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/11"}}},"commit":{"parents":[{"commit":"a5c425c04331553f41f236ead2b195e5c9245361","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5c425c04331553f41f236ead2b195e5c9245361"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-15 13:50:24.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c783d8ed57e70408b388fc79e8792a085efd97e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c783d8ed57e70408b388fc79e8792a085efd97e"}]},"branch":"refs/heads/master"},"97479a5c2eaf30ab8c941e2cc7fe7822b293f3f3":{"kind":"REWORK","_number":12,"created":"2026-04-22 15:30:25.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/12"}}},"commit":{"parents":[{"commit":"a5c425c04331553f41f236ead2b195e5c9245361","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5c425c04331553f41f236ead2b195e5c9245361"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-22 15:28:31.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97479a5c2eaf30ab8c941e2cc7fe7822b293f3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97479a5c2eaf30ab8c941e2cc7fe7822b293f3f3"}]},"branch":"refs/heads/master"},"fe50334ae7e43a2fb38bb361e7705cc7ffcfd395":{"kind":"REWORK","_number":13,"created":"2026-04-29 07:36:44.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/13"}}},"commit":{"parents":[{"commit":"a5c425c04331553f41f236ead2b195e5c9245361","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5c425c04331553f41f236ead2b195e5c9245361"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 07:36:11.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe50334ae7e43a2fb38bb361e7705cc7ffcfd395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe50334ae7e43a2fb38bb361e7705cc7ffcfd395"}]},"branch":"refs/heads/master"},"9bef6a607c8c7d2b6063eb21c3857fdfd4de78c0":{"kind":"REWORK","_number":14,"created":"2026-04-29 09:11:55.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/14"}}},"commit":{"parents":[{"commit":"6ae6d313e709016023324932c8ec8b12a237ceb3","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ae6d313e709016023324932c8ec8b12a237ceb3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 09:11:17.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs and overcloud hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bef6a607c8c7d2b6063eb21c3857fdfd4de78c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bef6a607c8c7d2b6063eb21c3857fdfd4de78c0"}]},"branch":"refs/heads/master"},"e83e68aecf60e0be7b85ddf1e5d65a050bb9a678":{"kind":"REWORK","_number":15,"created":"2026-04-29 10:30:25.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/15"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 10:23:28.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e83e68aecf60e0be7b85ddf1e5d65a050bb9a678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e83e68aecf60e0be7b85ddf1e5d65a050bb9a678"}]},"branch":"refs/heads/master"},"b51a394ff73c71ac90877ae85d933b785d0c0a84":{"kind":"REWORK","_number":16,"created":"2026-04-29 10:39:36.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/16"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 10:39:01.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b51a394ff73c71ac90877ae85d933b785d0c0a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b51a394ff73c71ac90877ae85d933b785d0c0a84"}]},"branch":"refs/heads/master"},"e96639f32cef4bc39a2ee53dde273fecff9c9f4f":{"kind":"REWORK","_number":17,"created":"2026-04-29 12:13:26.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/17"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 12:12:48.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e96639f32cef4bc39a2ee53dde273fecff9c9f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e96639f32cef4bc39a2ee53dde273fecff9c9f4f"}]},"branch":"refs/heads/master"},"4541dc91a61f4f5c31ca8aa415c9df9e115fa6fe":{"kind":"REWORK","_number":18,"created":"2026-04-29 12:20:33.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/18"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 12:20:16.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4541dc91a61f4f5c31ca8aa415c9df9e115fa6fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4541dc91a61f4f5c31ca8aa415c9df9e115fa6fe"}]},"branch":"refs/heads/master"},"cbcc30879a23d7006d6b06edeebfa0c61e7c9036":{"kind":"REWORK","_number":19,"created":"2026-04-29 12:40:02.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/19"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 12:39:34.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbcc30879a23d7006d6b06edeebfa0c61e7c9036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbcc30879a23d7006d6b06edeebfa0c61e7c9036"}]},"branch":"refs/heads/master"},"b0f37d1cdf0fba4131dd6df0a563c3bc381544ab":{"kind":"REWORK","_number":20,"created":"2026-04-29 13:00:55.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/20"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 13:00:21.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0f37d1cdf0fba4131dd6df0a563c3bc381544ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0f37d1cdf0fba4131dd6df0a563c3bc381544ab"}]},"branch":"refs/heads/master"},"36567ecb6b4d5ce2d299788741b556a2ee8de336":{"kind":"REWORK","_number":21,"created":"2026-04-29 14:03:41.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/21"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 14:03:10.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36567ecb6b4d5ce2d299788741b556a2ee8de336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36567ecb6b4d5ce2d299788741b556a2ee8de336"}]},"branch":"refs/heads/master"},"180ab960c61cc624ca654812882fe32b87482b0d":{"kind":"REWORK","_number":22,"created":"2026-05-13 07:39:55.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/22"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 07:39:20.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/180ab960c61cc624ca654812882fe32b87482b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/180ab960c61cc624ca654812882fe32b87482b0d"}]},"branch":"refs/heads/master"},"ebbee15f4979b3b5bc41f0957718cb5f9f1b8d76":{"kind":"REWORK","_number":23,"created":"2026-05-13 07:50:55.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/23"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 07:50:24.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebbee15f4979b3b5bc41f0957718cb5f9f1b8d76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebbee15f4979b3b5bc41f0957718cb5f9f1b8d76"}]},"branch":"refs/heads/master"},"7c4b559860fe8954d8815c270a5feb80625e451f":{"kind":"REWORK","_number":24,"created":"2026-05-13 08:04:56.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/24"}}},"commit":{"parents":[{"commit":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","subject":"Add control host service deploy/destroy skeleton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 08:04:08.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c4b559860fe8954d8815c270a5feb80625e451f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c4b559860fe8954d8815c270a5feb80625e451f"}]},"branch":"refs/heads/master"},"cb687583b74e348bfaf05f376316405c25dbfea7":{"kind":"REWORK","_number":25,"created":"2026-05-13 12:19:45.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/25"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 12:19:19.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb687583b74e348bfaf05f376316405c25dbfea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb687583b74e348bfaf05f376316405c25dbfea7"}]},"branch":"refs/heads/master"},"6f23f25e385af68d89f7426ac6977241e6d0266e":{"kind":"REWORK","_number":26,"created":"2026-05-13 12:40:07.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/26"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 12:39:54.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f23f25e385af68d89f7426ac6977241e6d0266e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f23f25e385af68d89f7426ac6977241e6d0266e"}]},"branch":"refs/heads/master"},"d92aa5eda29a9583b843aee53bf832c677e756a8":{"kind":"REWORK","_number":27,"created":"2026-05-13 13:07:10.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/27"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 13:06:59.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d92aa5eda29a9583b843aee53bf832c677e756a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d92aa5eda29a9583b843aee53bf832c677e756a8"}]},"branch":"refs/heads/master"},"e2a82c7704671454f99012d5b0b5ec9a6e3c8b91":{"kind":"REWORK","_number":28,"created":"2026-05-13 13:32:39.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/28"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 13:32:08.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2a82c7704671454f99012d5b0b5ec9a6e3c8b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2a82c7704671454f99012d5b0b5ec9a6e3c8b91"}]},"branch":"refs/heads/master"},"f88c11efd2c544d23180ba1f5915f66e8dfa12c3":{"kind":"REWORK","_number":29,"created":"2026-05-13 14:09:09.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/29"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 14:08:49.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f88c11efd2c544d23180ba1f5915f66e8dfa12c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f88c11efd2c544d23180ba1f5915f66e8dfa12c3"}]},"branch":"refs/heads/master"},"8c91ddc168f32e37bf5c97ce07fc97ab223ad3d8":{"kind":"REWORK","_number":30,"created":"2026-05-13 14:13:20.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/30"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 14:13:04.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c91ddc168f32e37bf5c97ce07fc97ab223ad3d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c91ddc168f32e37bf5c97ce07fc97ab223ad3d8"}]},"branch":"refs/heads/master"},"db6384cb263ad05ab55317c2ec3c5c15cc8ce633":{"kind":"REWORK","_number":31,"created":"2026-05-13 15:01:55.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/31"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-13 15:01:25.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db6384cb263ad05ab55317c2ec3c5c15cc8ce633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db6384cb263ad05ab55317c2ec3c5c15cc8ce633"}]},"branch":"refs/heads/master"},"68f31fe0687208e9031077092dea6926ad31ce4b":{"kind":"REWORK","_number":32,"created":"2026-05-14 10:22:19.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/02/922402/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/32"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-05-14 10:21:51.000000000","tz":60},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68f31fe0687208e9031077092dea6926ad31ce4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68f31fe0687208e9031077092dea6926ad31ce4b"}]},"branch":"refs/heads/master"},"de439378fc242416e4cb561e1f7ae67b8c0bd8ef":{"kind":"REWORK","_number":33,"created":"2026-05-20 07:39:00.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/33"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 07:38:47.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de439378fc242416e4cb561e1f7ae67b8c0bd8ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de439378fc242416e4cb561e1f7ae67b8c0bd8ef"}]},"branch":"refs/heads/master"},"ad16cde276b0fba3704f4d2a3da9e32d750b1232":{"kind":"REWORK","_number":34,"created":"2026-05-20 08:44:34.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/34"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 08:44:15.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad16cde276b0fba3704f4d2a3da9e32d750b1232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad16cde276b0fba3704f4d2a3da9e32d750b1232"}]},"branch":"refs/heads/master"},"7dd4bee2c0d0b6cf8e57ab30e52293793a6bed0a":{"kind":"REWORK","_number":35,"created":"2026-05-20 09:59:37.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/35"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 09:59:27.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7dd4bee2c0d0b6cf8e57ab30e52293793a6bed0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7dd4bee2c0d0b6cf8e57ab30e52293793a6bed0a"}]},"branch":"refs/heads/master"},"c9ad6ab0fc1929dad778a3843ed38776bf862474":{"kind":"REWORK","_number":36,"created":"2026-05-20 10:56:53.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/36"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 10:56:42.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9ad6ab0fc1929dad778a3843ed38776bf862474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9ad6ab0fc1929dad778a3843ed38776bf862474"}]},"branch":"refs/heads/master"},"6f421f71a061ef77d72ddccc57e814739e64bc76":{"kind":"REWORK","_number":37,"created":"2026-05-20 13:05:40.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/37"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 13:05:25.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f421f71a061ef77d72ddccc57e814739e64bc76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f421f71a061ef77d72ddccc57e814739e64bc76"}]},"branch":"refs/heads/master"},"515561e7294f42d3371b7c3c16880060a19f7363":{"kind":"REWORK","_number":38,"created":"2026-05-20 13:34:56.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/38"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 13:34:45.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/515561e7294f42d3371b7c3c16880060a19f7363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/515561e7294f42d3371b7c3c16880060a19f7363"}]},"branch":"refs/heads/master"},"0608905fb51e027cb2df0affc1799210e7aa20f5":{"kind":"REWORK","_number":39,"created":"2026-05-20 13:48:12.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/39"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 13:47:51.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0608905fb51e027cb2df0affc1799210e7aa20f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0608905fb51e027cb2df0affc1799210e7aa20f5"}]},"branch":"refs/heads/master"},"16e33a2b6b81a7d0403e0b608d9a799bb3163af5":{"kind":"REWORK","_number":40,"created":"2026-05-20 14:50:07.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/40"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 14:49:56.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16e33a2b6b81a7d0403e0b608d9a799bb3163af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16e33a2b6b81a7d0403e0b608d9a799bb3163af5"}]},"branch":"refs/heads/master"},"923a32719658c3bdad056f0c886ec663871f8599":{"kind":"REWORK","_number":41,"created":"2026-05-20 14:58:25.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/41"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 14:58:15.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/923a32719658c3bdad056f0c886ec663871f8599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/923a32719658c3bdad056f0c886ec663871f8599"}]},"branch":"refs/heads/master"},"b16f34ea04e60b051692813913f8f463730cf167":{"kind":"REWORK","_number":42,"created":"2026-05-20 15:16:12.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/42"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-20 15:16:02.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b16f34ea04e60b051692813913f8f463730cf167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b16f34ea04e60b051692813913f8f463730cf167"}]},"branch":"refs/heads/master"},"66c70a8865b79fbbd0745229f63f31dbd6904b58":{"kind":"REWORK","_number":43,"created":"2026-05-21 08:36:18.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/43"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-21 08:35:46.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66c70a8865b79fbbd0745229f63f31dbd6904b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66c70a8865b79fbbd0745229f63f31dbd6904b58"}]},"branch":"refs/heads/master"},"7e7fb676228acbf540e8951e10c562fc35f98304":{"kind":"REWORK","_number":44,"created":"2026-05-21 08:44:39.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/44"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-21 08:44:25.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e7fb676228acbf540e8951e10c562fc35f98304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e7fb676228acbf540e8951e10c562fc35f98304"}]},"branch":"refs/heads/master"},"c6b028f6d726830b9556012c43bf12706817f38f":{"kind":"REWORK","_number":45,"created":"2026-05-21 09:12:15.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/922402/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/45"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-05-21 09:12:05.000000000","tz":0},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6b028f6d726830b9556012c43bf12706817f38f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6b028f6d726830b9556012c43bf12706817f38f"}]},"branch":"refs/heads/master"},"18cc4b921f35aad396e0bd4e198ec45622cf1e5b":{"kind":"REWORK","_number":46,"created":"2026-05-21 11:51:43.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/02/922402/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/922402/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/922402/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/922402/46"}}},"commit":{"parents":[{"commit":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","subject":"CI: Add kayobe-ansible-control-host-configure jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-06-17 12:53:28.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-05-21 11:51:35.000000000","tz":60},"subject":"Deploy more containers","message":"Deploy more containers\n\nExtends the seed\u0027s \"custom containers\" feature to support deploying\ncontainers on infrastructure VMs, overcloud and Ansible control hosts.\n\nCo-Authored-By: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n\nChange-Id: I19ba61e90c78e088b740feed3dac1f6d1ad262ab\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18cc4b921f35aad396e0bd4e198ec45622cf1e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18cc4b921f35aad396e0bd4e198ec45622cf1e5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
