)]}'
{"id":"openstack%2Fkayobe~922480","triplet_id":"openstack%2Fkayobe~master~Ia11ac49986e2bebba17e533232d8cfd116ba9e73","project":"openstack/kayobe","branch":"master","topic":"bug-2070051","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-06-28 08:39:33.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-07-31 12:07:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia11ac49986e2bebba17e533232d8cfd116ba9e73","subject":"Stop replacing dashes with underscores for Kolla interfaces","status":"MERGED","created":"2024-06-21 14:51:49.000000000","updated":"2024-07-31 12:08:55.000000000","submitted":"2024-07-31 12:07:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922480-bug-2070051","meta_rev_id":"bfaf17e9311c1ecfd363296ef33964e29cb9a32a","_number":922480,"virtual_id_number":922480,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 12:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-06-28 08:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-07-01 21:58:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2024-07-16 09:20:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-31 10:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-31 10:10:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-21 14:58:38.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-06-21 16:41:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-01 21:58:45.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2024-07-16 09:20:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-07-31 10:10:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"4c184eed66b76ec39282ef1ecacb84349f0fd2b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-21 14:51:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"443bb27acaf2885f53d378b02f0e2e28a1b85f06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-21 14:52:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"154fecd4ea77c4a5782fc29f7021d141a7973a41","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-21 14:53:39.000000000","message":"Topic set to bug-2070051","accounts_in_message":[],"_revision_number":2},{"id":"39f556d88aab37872d1a5f082a5386fde3604cf8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-21 14:58:38.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a293aa262f72eeb7b0aa081de631d6d77c09d8e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 16:41:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/497683f89fa64d82b50b7cbc585f80d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea792591e5e142e6b2f2d1ea5d19dafa : FAILURE in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f163a29c8c774df1b7494566ef02aabf : SUCCESS in 7m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/643838322e154afba02f7d0432065208 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dae3420c4104425f840e5a9aafc5a74c : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f91f0b508da4bf29b9db1b54130f0fe : SUCCESS in 9m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f07415e140d04e3f81db87002f8c868d : SUCCESS in 59m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/18eedb1a3eae4442933581362e951a69 : SUCCESS in 1h 00m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d07d7a41062543688118f3792bc71741 : SUCCESS in 53m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b22f8f38045f44cdaa8ae28f658a3942 : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d05eaed0f0944f38afca034af0c74647 : SUCCESS in 19m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/810c0e46f56b4b1a8189bf983b17a8c4 : SUCCESS in 15m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59ca353b95f746d7bd01b867eb9aff48 : SUCCESS in 14m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8103ae8193e14509bc85cbc883fac0f4 : SUCCESS in 1h 44m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb6a7237daa74ac680ab856e02c25ce3 : SUCCESS in 1h 33m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/40621114583e4ba1bbc578f8315034ae : POST_FAILURE in 9m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d37f893b068e4a57bac9f8dedf0ee5c5 : SUCCESS in 12m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9a13fe978c1f4898a0e63d03e982616d : SUCCESS in 43m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b1bba09b8a394540b52b2c8fbd2fa9d4 : SUCCESS in 31m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f44f5aae49d746ed8783805e7864e947 : SUCCESS in 27m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23ea8bcc685d4ed8a686b131f1a832ce : SUCCESS in 14m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3e072e508144c9c80378b839a9cad87 : SUCCESS in 10m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5e2e003da014bb88e5e1f79c9b0c669 : SUCCESS in 20m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff818318752b43ec91d978e3380424b0 : SUCCESS in 13m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c576d65d13b745348a7315a34e4ba657 : FAILURE in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d27c13cc02244c6cab330a1ac678eebe : FAILURE in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d709ccac6f234ea7a067372947bb5b28 : FAILURE in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"08730951c1ea8ae2034c700b28817f1f2dcbc2b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-24 07:00:39.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a6ab93f8be284c9f3c0ca7b223ecdf5c9da88ef5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-24 07:16:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"84aa145a8b2450b81a5d1bd45f9d9325895139e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-24 07:16:38.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d9d4f717555b95810f214849dbea363b1dc190d7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-24 07:16:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29cb63e74760c21c998efebccd2228e7dc765ce8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-24 07:17:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cac9f4a7e4c973016ddb365a38cda3edd65064c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-24 07:17:23.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"89d7689f7eab61399cab40a443b0caa32bfc4bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 09:01:21.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/922478 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"329506f0fee040de14f7e354097fa8d7465e4585","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-24 09:05:14.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"44e3230b2ceb609b3f9175cc08bd1e1a78d86df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 10:54:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f724ce2c6144248981d71f575dfdd63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84b46619975742758e8edf8702d37664 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5a6635bd8f4678a33bbe758319095a : SUCCESS in 7m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/55ef4e38f6df493f8f8219b1c58fc643 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/60558d059739450da4f82d34278a5974 : SUCCESS in 12m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f4ae3c66caf4426afba101bf541a8d1 : SUCCESS in 7m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5afb0d92494849b9bb6bbc97cfa6ebe7 : SUCCESS in 1h 11m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/be1743b7b2b34ad095a0a41a56aa8536 : SUCCESS in 54m 11s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15433b77a18440dc811186ad6db1b735 : SUCCESS in 50m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f2cd885e982344419b51fcebf17fc57a : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1c4b748f5bfb436c8e90c978c6914d22 : SUCCESS in 17m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/505f83b7e4c04f1584f00a7aa3e713a2 : SUCCESS in 15m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fea68eac03664b0883662b8ee4bf56fd : SUCCESS in 12m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ec773ad655e04b388fd0779f2e0ccdbc : SUCCESS in 1h 44m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a92b7d4248aa4809b745e31461247207 : SUCCESS in 1h 30m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3362abfe4e6444699f9325591cf8d4e2 : SUCCESS in 19m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c189259ae964f03a55f0c0da3697312 : SUCCESS in 13m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/90f2466509044f74a24dc7b459fcfd33 : SUCCESS in 39m 18s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/eb4ecff89940427d89bb4972153535f7 : SUCCESS in 30m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6024b5d41bdd493ea497f6af511f7583 : SUCCESS in 22m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/90d316b9e6de49008b7f88b8b17a761f : SUCCESS in 15m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83b1d765e7e547c5a22052d3fd8e07bc : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/18a2bdb7145348daac9cf57b06b1461f : SUCCESS in 20m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ebc47d7acdc4465a3e055de6846a71a : SUCCESS in 12m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6beb1ff013b04c68a31af711c0b5d53c : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c955322988de4cabb25e2dd26dd015c0 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2cde398a6af54b219527ea2c2a78a815 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":6},{"id":"068fb869478ebf7f6b5a44cf39cd58757c74e15a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-28 08:39:33.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"80774db8100150d392d5ced9d437c2940ccb126f","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2024-07-01 21:58:45.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"07a5677ca2c153d9fbf02438485e41da23ce82bf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-16 09:20:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"bc478a06c076bc06a1bf49f4b33ce1682d63aae1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-31 10:10:36.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7445894be8246f8143a21cdd9c50ab18e2263fc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 10:11:03.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3ed0186aad3fcc33ea26f24ad3aca598d0c6e531","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 12:07:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddb3c05e7a184775ab42d6967f880747\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b2cbc33daa4a1686e387a152a801d4 : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce9bd144e71e44829b5c16c1e42869d2 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d4594ee8dd747b2877e7e4a2df0d9fd : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/398f92dd2a4e43cca21ef5e7c66e2cf6 : SUCCESS in 12m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/162613dc8544420996b37c44c9dbb7b8 : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07c935dff8cb45cfa81cbda418d61abf : SUCCESS in 52m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6b6e892a8ff5473a8281f52cb7527109 : SUCCESS in 1h 10m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f627da083e1e4eaca10387a8c83e1a95 : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cac2d8d76434498c9f8ddb564a4c25c7 : SUCCESS in 15m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d394329681cf49518a291438781a3df9 : SUCCESS in 1h 49m 56s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea881a9f820c4d06a5dacd8fd0055edc : SUCCESS in 1h 54m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ba22f5abd4c740e38b1de168db9032ed : SUCCESS in 19m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8b3a401fbb4415cbe08b42ee32815f3 : SUCCESS in 12m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/42c470bb0551428882b26410934b5ce8 : SUCCESS in 35m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0276a1523b104e05b0fa86504be00f87 : SUCCESS in 22m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1456259037ea42719a3222bb1efdbf4f : SUCCESS in 15m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9d7b6347d6b4c9b91566de0f6300165 : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1dd2b48c108243c185c61464ede862c2 : SUCCESS in 19m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6923310c5b9c4e2cb2959afe327ddd1d : SUCCESS in 13m 16s","accounts_in_message":[],"_revision_number":6},{"id":"17aacbc117ba4650be8108afb34978c562009e9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 12:07:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"bfaf17e9311c1ecfd363296ef33964e29cb9a32a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 12:08:55.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013d0783ad6048db917ea482dbff386e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/179aff7249524feaa1672019005b5065 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f2b042dcec55860bd7c57ca6c8db9e9a9580bab7","revisions":{"ce3c28dfd6db19f1849fab3b7ef8dde1cb5369e4":{"kind":"REWORK","_number":1,"created":"2024-06-21 14:51:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/922480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/922480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/922480/1"}}},"commit":{"parents":[{"commit":"3a20094cc43b5e14e94805309ada89c26cc8f3af","subject":"Merge \"Docs: Link to kolla contributer guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a20094cc43b5e14e94805309ada89c26cc8f3af"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-21 14:50:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-21 14:51:47.000000000","tz":120},"subject":"Stop replacing dashes with underscores for Kolla interfaces","message":"Stop replacing dashes with underscores for Kolla interfaces\n\nCloses-Bug: #2070051\nChange-Id: Ia11ac49986e2bebba17e533232d8cfd116ba9e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce3c28dfd6db19f1849fab3b7ef8dde1cb5369e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce3c28dfd6db19f1849fab3b7ef8dde1cb5369e4"}]},"branch":"refs/heads/master"},"38bae993638c74aa1c5578a0edfd5e749a19cbe1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-06-21 14:52:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/922480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/922480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/922480/2"}}},"commit":{"parents":[{"commit":"3a20094cc43b5e14e94805309ada89c26cc8f3af","subject":"Merge \"Docs: Link to kolla contributer guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a20094cc43b5e14e94805309ada89c26cc8f3af"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-21 14:50:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-21 14:51:52.000000000","tz":120},"subject":"Stop replacing dashes with underscores for Kolla interfaces","message":"Stop replacing dashes with underscores for Kolla interfaces\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/922478\nCloses-Bug: #2070051\nChange-Id: Ia11ac49986e2bebba17e533232d8cfd116ba9e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38bae993638c74aa1c5578a0edfd5e749a19cbe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38bae993638c74aa1c5578a0edfd5e749a19cbe1"}]},"branch":"refs/heads/master"},"36402f82609ce849f1e9c19c986dd858c12a33be":{"kind":"REWORK","_number":3,"created":"2024-06-24 07:00:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/922480/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/922480/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/922480/3"}}},"commit":{"parents":[{"commit":"3a20094cc43b5e14e94805309ada89c26cc8f3af","subject":"Merge \"Docs: Link to kolla contributer guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a20094cc43b5e14e94805309ada89c26cc8f3af"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-21 14:50:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-24 07:00:39.000000000","tz":0},"subject":"Stop replacing dashes with underscores for Kolla interfaces","message":"Stop replacing dashes with underscores for Kolla interfaces\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/922478\nCloses-Bug: #2070051\nChange-Id: Ia11ac49986e2bebba17e533232d8cfd116ba9e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36402f82609ce849f1e9c19c986dd858c12a33be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36402f82609ce849f1e9c19c986dd858c12a33be"}]},"branch":"refs/heads/master"},"26b7202623242713c2aee234fe1224db74453d11":{"kind":"REWORK","_number":4,"created":"2024-06-24 07:16:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/922480/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/922480/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/922480/4"}}},"commit":{"parents":[{"commit":"3a20094cc43b5e14e94805309ada89c26cc8f3af","subject":"Merge \"Docs: Link to kolla contributer guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a20094cc43b5e14e94805309ada89c26cc8f3af"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-21 14:50:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-24 07:16:38.000000000","tz":0},"subject":"Stop replacing dashes with underscores for Kolla interfaces","message":"Stop replacing dashes with underscores for Kolla interfaces\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/922478\nCloses-Bug: #2070051\nChange-Id: Ia11ac49986e2bebba17e533232d8cfd116ba9e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26b7202623242713c2aee234fe1224db74453d11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26b7202623242713c2aee234fe1224db74453d11"}]},"branch":"refs/heads/master"},"34fff0aecbf29f32086828e5776a06f0455b56b3":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-06-24 07:17:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/922480/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/922480/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/922480/5"}}},"commit":{"parents":[{"commit":"3a20094cc43b5e14e94805309ada89c26cc8f3af","subject":"Merge \"Docs: Link to kolla contributer guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a20094cc43b5e14e94805309ada89c26cc8f3af"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-21 14:50:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-24 07:17:22.000000000","tz":0},"subject":"Stop replacing dashes with underscores for Kolla interfaces","message":"Stop replacing dashes with underscores for Kolla interfaces\n\nThis also removes test__run_one_with_dashes, because it\u0027s not needed\nanymore.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/922478\nCloses-Bug: #2070051\nChange-Id: Ia11ac49986e2bebba17e533232d8cfd116ba9e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34fff0aecbf29f32086828e5776a06f0455b56b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34fff0aecbf29f32086828e5776a06f0455b56b3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f2b042dcec55860bd7c57ca6c8db9e9a9580bab7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-06-24 09:05:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/922480/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/922480/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/922480/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/922480/6"}}},"commit":{"parents":[{"commit":"17bb1a7ad6ab862077dcb9cf4d167025a1430016","subject":"Fix eof newline on hosts.rst docs page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17bb1a7ad6ab862077dcb9cf4d167025a1430016"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-21 14:50:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-24 09:05:14.000000000","tz":0},"subject":"Stop replacing dashes with underscores for Kolla interfaces","message":"Stop replacing dashes with underscores for Kolla interfaces\n\nThis also removes test__run_one_with_dashes, because it\u0027s not needed\nanymore.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/922478\nCloses-Bug: #2070051\nChange-Id: Ia11ac49986e2bebba17e533232d8cfd116ba9e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2b042dcec55860bd7c57ca6c8db9e9a9580bab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2b042dcec55860bd7c57ca6c8db9e9a9580bab7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
