)]}'
{"id":"openstack%2Fkayobe~923871","triplet_id":"openstack%2Fkayobe~master~I2d592c5479530b2fe9a60c14ee59e817b8402490","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-07-23 11:46:02.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-08-30 18:04:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d592c5479530b2fe9a60c14ee59e817b8402490","subject":"Support setting whether an Apt repo is trusted","status":"MERGED","created":"2024-07-10 15:39:25.000000000","updated":"2024-08-30 18:09:48.000000000","submitted":"2024-08-30 18:04:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923871","meta_rev_id":"eddfc105d0fe5156472a40766d0d70bd6f63b967","_number":923871,"virtual_id_number":923871,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 18:04:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-08-30 15:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-07-23 11:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-08-30 15:30:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-10 16:03:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2024-07-10 16:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 11:46:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-08-30 15:30:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"baf56b7cf100d7c6b41a4c474930276e70ffc872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-10 15:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d94b1b480dd9d04c379cfae6ee309229bb1f895f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-10 16:03:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3d7d72ff6865164721913f964ad307bf4070096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 16:56:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6c22c7001d04f74b2e9ee5e408957e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52111426c4be48278d50fcc7a42505dc : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/502599154cfd4225a1ac3a8fcd084761 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a9d48bcb77a4991851fb538cff9dc7c : SUCCESS in 5m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e6e332e7e6964ebb8cec69f92c9ac198 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a68d501c4164f87841bdd89c987788e : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb34fdee3c744907b519576e4dd8f757 : SUCCESS in 10m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9e5385dc41604b5ba63ed1ab0479c38a : FAILURE in 19m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/de9c0994d1af4cff9f93c23bc7dcd8e5 : FAILURE in 17m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34259f22440547a181f1d941070929e4 : SUCCESS in 49m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b741aa7895ec45e5a18c0e0b9bfd0a6a : FAILURE in 21m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1dcf1c036ef744bdaa884506119cbdd3 : SUCCESS in 18m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1111165893824df49bc610b224cbfb9b : SUCCESS in 15m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86d5289e446a49019afccb57d8f17bcc : SUCCESS in 13m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8fe4638b8987463e9e406c0ab2dcf61b : FAILURE in 1h 14m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fe63b53e610492a8c8702abbbf4f5e1 : FAILURE in 1h 03m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/edb1e099614a4c3f9309d7c45aa1cb2b : FAILURE in 14m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7518896ff06b4935865d072ffe33f835 : SUCCESS in 12m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c17de13572644c3995e6b441c74d1cd0 : FAILURE in 13m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8a41e0095c8b4c9f96c95e77cb1650d8 : SUCCESS in 31m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7936c583467249c7a93065b1c49c4651 : SUCCESS in 28m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f199b52e5564c77bae3ce24b756ed8c : SUCCESS in 17m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ad389f84dc34d89ae801f891d2fbba7 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9479f0d2bae54564bd1d192356fc2971 : SUCCESS in 23m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab677965c0cf4609b32e90742f5b02a9 : SUCCESS in 15m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1735bfb5a644893b69ff7ec29775e64 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b1aab08fec2415c8c370fb5c393b3a2 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5ee4d3665824dae83edc07cf8962914 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"541a2da78d159c186aeb2bdd9c1257f5d8fcb98e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-11 08:54:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9815143c12472686f88e25065f5e4375be4db012","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-23 11:46:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"08ea9d24b08f274e7a701ebbd07eef720eeab1d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-05 14:43:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b564df9b59755b106acc6c207aaeafb538b4cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 16:38:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5df59725bfdd441aaf3684843ad58ed5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6640a0fb82b43b597e6a82e09261c75 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e985cf7e99be461e8298c60ef54824b1 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd0d7345ccca4c539c035a859ef0ab74 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc8f1b6982a449f3b95ea5969e71858e : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d1b3116d03443318e38175b1264b3ce : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/51c2f10bb47745eaa34ec685acb7e561 : SUCCESS in 10m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8445ca681c004b418b544511cab25d21 : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/37bc5683fa7143d2a1e5f2c633292773 : SUCCESS in 1h 05m 10s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2abd66e47df14f29837f7b5531aa8e44 : SUCCESS in 54m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/904875c67c6c45279e23a1a1eb9f36d1 : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/615470f59b094884a795adc266b35894 : SUCCESS in 23m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/39ec0a6770fc4491b89bb435fc26b936 : SUCCESS in 17m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e60ac43a2bed4edbbeec42bda5bb2d6e : SUCCESS in 14m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/732cb8f2d24b48919f4e25391c61ad82 : SUCCESS in 1h 50m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fd4cc087f3644a6b1fbe83ca0800470 : SUCCESS in 1h 34m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f7cf899271e746b0852524fae22ded47 : SUCCESS in 21m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59acca4901264425b83db9abb0d0b2e3 : SUCCESS in 12m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c9a8c5efa587486aa20e3db1382cf3ee : SUCCESS in 45m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1a287c70350140b0bb428351bcdd37d6 : SUCCESS in 33m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f55118f00b15467fb8cbdf9714206333 : SUCCESS in 23m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/400fd7600cfa42e99f413610108dd53f : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef2f28707b5f4c3a9173f5ac573b41f4 : SUCCESS in 10m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eca2d1d5a4164342b1b6dcd44b2a0343 : SUCCESS in 21m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1a78440458341ac96b43c5938062bd5 : SUCCESS in 12m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d71bf1bd50f94119af97f16f80d6a315 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/464d00d5daa3460989e10baecd5fd0b5 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e632b6f4826d457f9ce0d5220244473b : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69e044f3e87a499081927ab645de8ffa : SUCCESS in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bf69069e4924c3ce342f6ac8cba25c1f377dfc7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-30 15:30:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e94326e6c020dccbbf52c16d4af131a7e45cc60d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:30:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8da9d40860a3bc373af5e3474ebf91c70cbafa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:04:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77b07a09db6b46e4ba5fd99d02b6fd5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57ce9a134056403abc05113f70c48997 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efda952c8f8d4bf9a335c591cf3d3772 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a77f537896ea499e835f029311e2fbd0 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11af98bdc134441aaa4044ecea90f82 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af2976e9270a4adabdf5895346cc353d : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c0a8afa98a840489f4155126027c75c : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/072aece96605421c82b717248acadfae : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eed6d61f8e094a5795e9cf63b3e8199e : SUCCESS in 9m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/173b58cb453b43b9813ba61424f52546 : SUCCESS in 59m 03s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77c8b181a231417780771666d0c2145c : SUCCESS in 45m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8a2b95e9ee254314a80eaefdf73b3880 : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/43e9910a257945c0977489c3755dc6a0 : SUCCESS in 20m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b72868421e5a43e495ebe75cb271fd96 : SUCCESS in 12m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/25bf89848c404e46a0b9011a491b454c : SUCCESS in 1h 53m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c153cff82d2747e396ae9c72d7b131ae : SUCCESS in 1h 36m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/47c2ae39ff214f38904db9efde61451d : SUCCESS in 20m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/381a904840e24643bebdb9f6266ee48b : SUCCESS in 12m 54s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d1f44a9793fe4cca9d21430998b11927 : SUCCESS in 29m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79e6b084433746518e76c72db856a688 : SUCCESS in 21m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f65eb76cf02446e483c6523813907fa4 : SUCCESS in 19m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27c2c2f43afa4a7cb5718e855a9990bc : SUCCESS in 10m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/79d9ff37005d49c78097d1e57a2ac496 : SUCCESS in 19m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d23b630b5574006b5ee157023508d88 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5e4505037a7a80302b43736a13a17fb92302a3b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:04:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eddfc105d0fe5156472a40766d0d70bd6f63b967","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:09:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae6d537ec5584551a754c47db226a886\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9be81b7cd804778b7cf5681fb726ba7 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ca3c6bceaff4e748157f6d2d8a44876 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b06c6a39acfb4e07e2756667380d92e63500c5c","revisions":{"2b06c6a39acfb4e07e2756667380d92e63500c5c":{"kind":"REWORK","_number":1,"created":"2024-07-10 15:39:25.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/71/923871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/923871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/923871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/923871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/923871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/923871/1"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-10 15:26:32.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-10 15:39:02.000000000","tz":60},"subject":"Support setting whether an Apt repo is trusted","message":"Support setting whether an Apt repo is trusted\n\nUseful if you\u0027re configuring your own private repos, and don\u0027t/can\u0027t\nsign them.\n\nChange-Id: I2d592c5479530b2fe9a60c14ee59e817b8402490\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b06c6a39acfb4e07e2756667380d92e63500c5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b06c6a39acfb4e07e2756667380d92e63500c5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
