)]}'
{"id":"openstack%2Fkayobe~923978","triplet_id":"openstack%2Fkayobe~master~Id9b19f10bc8ac38927914fb9782892b7daf1be82","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-07-25 18:29:21.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-07-25 18:29:21.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-07-25 15:10:34.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-07-25 18:29:21.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-07-25 18:29:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9b19f10bc8ac38927914fb9782892b7daf1be82","subject":"Use explicit version of community.docker","status":"MERGED","created":"2024-07-11 16:57:03.000000000","updated":"2024-07-25 18:30:36.000000000","submitted":"2024-07-25 18:29:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923978","meta_rev_id":"bd9091e19e0077ab098530ac45a3c820b28363a4","_number":923978,"virtual_id_number":923978,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-25 18:29:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2024-07-25 15:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-25 14:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2024-07-25 15:10:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 19:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-19 11:17:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-07-23 13:52:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-07-23 13:52:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-07-23 13:52:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"16e17e7e2e766f9ac457e14273df196165ebdfd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-11 16:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3633b07c12882389158b3c1df7ae9db8246a6e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 19:56:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3156a6253f34bcba2ae504aff32ba5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fd3862379aa42f9856134dd9dceb7bc : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e07663fb8e442ba84310e1233c1e2f7 : SUCCESS in 6m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb46d885d66e46599f54aeea98865657 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44c8c18238ec4a1ebc7a119f84faad55 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dcd6a9b502bb45dab1a7aca15fcded74 : SUCCESS in 10m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a5af761280374c0d83fea579c4c7becb : FAILURE in 22m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a6258fa98ffb41f18e4c2279fda2b308 : FAILURE in 18m 03s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ef798d631324484aa62afde56b56578 : SUCCESS in 55m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/011b9ac123da475ebdb08e5fe67a0b42 : RETRY_LIMIT Host unreachable in 33m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/53a2e25de3cc4357b8ecb4f1d0384de7 : SUCCESS in 19m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/20b3856446bc44bf977831990b34ff95 : SUCCESS in 16m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ff97a81fb324298b14eb8c0e01e60fb : SUCCESS in 13m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8cc4ec1d378c4875b8265fda909d6d78 : FAILURE in 1h 37m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f44fc05e640348fca5e09ed5e3748463 : SUCCESS in 1h 26m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2bcf8d74dd854c1d9e4642585887e278 : SUCCESS in 20m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/487e0489c435423ca23caf1a988f23bb : SUCCESS in 13m 56s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/12bd2ca239554064842532a03e4192c5 : SUCCESS in 41m 27s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/39fbba3fdbef4b4eb2df0c2d40ad2ccc : SUCCESS in 35m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3510677d873c4b6d8e1d1997db601c46 : SUCCESS in 22m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f435149e266d4703a725f6f0a68233dc : SUCCESS in 15m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c403ddd61606472ca459471203290d72 : SUCCESS in 9m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/900244698fff4e97a740138503e84cb7 : SUCCESS in 21m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1831a49c22f4325ba83d74af8093d49 : SUCCESS in 13m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96b5ea73104b48558392e5a03af95d3b : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5064e3ec8afe4ac4bfbcc9995099a444 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b60a84003e64d25a4abc5aa1cbcaa71 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"324cdff41b60896e95a35373e9cd99d58166f108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-12 09:04:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"072bbdece840eeb012fb8a4dd2893829be477a4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-12 09:35:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"10b542246fdeba0f704952eb3a052730ddc511d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-12 12:09:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"47b66df6a27dfcb80718e468b50ea6554f04cd3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-12 12:50:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"045842bfa18c337196d0cffaabf60dc15c3cacf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 15:08:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92148717e47c40da97e67527e47ebc74\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/847dd6852ebb4679b2bc31a3d017de00 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254fa4eb252a46dc8d225dfeb1d9a9e2 : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c70f3c8ceb3468d881d0ca30c69126b : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9c55168c347f4a59a825e8ba6362e839 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6982820dbb7842bf9aced44d565498e0 : SUCCESS in 9m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1b00fdcf14d24a7496f94f85fabf079c : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/399e90e8e4c04653ac33eda729abdfe0 : SUCCESS in 1h 03m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0f6cc0e4ffb4ee7896671a8c6cb027e : SUCCESS in 56m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c35e1fe288cc441e9ab285720f4121e8 : SUCCESS in 1h 23m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/52f6a447d619442c99de4a2c1fa54c29 : SUCCESS in 22m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b98db72d0cb7452e811edc6c470c038f : SUCCESS in 17m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f7a6cbdfd3b410c929e7b1a680ef75e : SUCCESS in 18m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6f84f57e78e048a59f40dae40c776c79 : SUCCESS in 2h 07m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad031f9f1abd4e58b0298f0277c9634a : SUCCESS in 1h 53m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/595899d8e84846a4b40cf8da0fc2e78e : SUCCESS in 20m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4c1a3dc4bd940eca25bed18b9cacec1 : SUCCESS in 18m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/92dd0475065b4c4693590a43aa8c3fc8 : SUCCESS in 40m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/da056116cf2c44c4bcfdc2815481fee1 : SUCCESS in 35m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/258dca6104a74736955ffb39b96dec5f : SUCCESS in 27m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ff637ae5453c40af907cc6f8eae36288 : SUCCESS in 17m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f826b919c284b1cbe95199df51f3c56 : SUCCESS in 14m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/192d8c466cc941cc922954087f374973 : SUCCESS in 22m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24cc88145abb47efb39483f40abde28a : SUCCESS in 19m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f5107ac565f4b1380f3de2bd204ce1a : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e71dd0bdfc3f49daadfc79b0ac78407d : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/adb1b7e0c6394b31802c2e6946b70f01 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":5},{"id":"2ef5c6fabfbd3bb0fb3d25657f21d4a81484721e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-12 15:17:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ab495a088968ea7bf50c47f435531882b5b7ca23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-12 17:53:07.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"e2559a0206b8be774fbf9cca41229d1c9842bef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 19:51:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d3a511ca8b4a71a7fa423bd0605710\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/773bad3427424705a75b98177f0e3b31 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a43d4e30f27410098b931aef6cfd71d : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b28363f243a64cb79624e99d1c9d9ade : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/440fb57f76ce44339e5972783c79f85e : SUCCESS in 3m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36c5a69eb29d4aa78a09911f14ce308e : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c0647107db954deea15957d8e73134a9 : SUCCESS in 8m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f4a3687eeadf4d8ab7dc206a9e6e1573 : SUCCESS in 1h 09m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ff385f8800f144f48d3283217429ec30 : SUCCESS in 54m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcc2b96ee9c04b53a61a0fded573fce8 : SUCCESS in 54m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4d29721921d54d79ae6224e3f06aa606 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8d3b87608db24e958f3fb3144aaad51f : SUCCESS in 18m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/96efeda6611940c0a007ac124039ed11 : SUCCESS in 15m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/252ba065cb7c4ac5842f5305f96d42b8 : SUCCESS in 13m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/75a169159e1246fc921522c05c7bff81 : SUCCESS in 1h 53m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92bf7a7d85204fb390a4e0ce0713e503 : SUCCESS in 1h 26m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/92837470d78c45fa92f2c71119c4ae68 : SUCCESS in 19m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09cf9eb08d174fe181a3d3e15df3366e : SUCCESS in 12m 59s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/60aa1277878e442ca7c4972d92872e52 : SUCCESS in 39m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c88421ff28a245e5994e780a490aeaec : SUCCESS in 28m 50s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67555866431746f687a855e7bd8704a6 : SUCCESS in 23m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/018474b4c4f9415ab4bafc659b27cbe1 : SUCCESS in 16m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cf4be81aad64f069ccb788e8e204d37 : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7c6c00dabddd43f6a503d7e4bc3a824d : SUCCESS in 21m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f701d5c07d44f23b2d5fd0ba8d30f18 : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d260ed30cb484816bfde69c749f99c1e : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6dbb251377c145dfa040729d991fdb8c : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40a394f3e766434e817ff1dfec0fdea3 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":7},{"id":"387071b712681f60eb499b2d796a3e16d8ad4459","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-19 11:17:44.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d093ce1a293fa57685f239e10baac7747eae2f8e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-19 11:18:12.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0f7227141eb571245bb87bfe0f15be4f1f65236b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-19 16:12:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f3823e14e7de5df789fdb8630cd8fc10c151434b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-23 12:08:49.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1f13c2f6b0ec558315d520cb3d0a565f0bbc95ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-23 14:20:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e73d6de551c99430f641e593208728b31bc55aab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-23 14:20:03.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b0ea6a0c35d74207c4e7624dfb87a40078c9d47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 15:48:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/633b195ceb92488da28b61c851dff714\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb119f10ab7b477483559dd694b2ae32 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fade958387543729158ddb7c2d9f552 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05dff6c1390644b7b002b7822d70fa01 : SUCCESS in 4m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81079f73fa114477954195d871c10f65 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6235a9b9f21443f5b3ca3064ef202028 : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dccd25399cbf424c9b462aafd2266438 : FAILURE in 5m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8f8b2a92853845b1b6a53105238e497e : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/14f4c3a962de4d50ba1308d2201f6e29 : SUCCESS in 1h 06m 32s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36fd290d39984bc5a418b5fbfc9eb206 : SUCCESS in 54m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f75bbff2073445c9bea7c841297aa8d7 : SUCCESS in 1h 14m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e408c71fd80c47ef872013e2f7c1e272 : SUCCESS in 18m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/857342d0ded04127a9016fbe02552764 : SUCCESS in 16m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec6b54a3bb9b4fa4b578d5bcab476be3 : SUCCESS in 12m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/95b830e0cdbb44e8849bd04d3a25164e : FAILURE in 1h 27m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb7ebfe7b0564a48878812d3e087f414 : FAILURE in 1h 17m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b371391f472046b7bd8ad195569e9016 : SUCCESS in 22m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df3fc2da775d41a7b30c5a4dde0b7b24 : SUCCESS in 15m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5e3f9b3e37a749e4a9ccc8723f26e4e2 : SUCCESS in 42m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/25f520e63dfc4e4896448a8ef5837109 : SUCCESS in 36m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6916cf7c402b4ad7bc004b62b24b60bf : SUCCESS in 25m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76112751d4414ef5ba589470dac43e8c : SUCCESS in 15m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f59963329bf94b559a477b1dffcbc4d5 : SUCCESS in 9m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4d4a17655024414e8081b175a6bd41fa : SUCCESS in 20m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/637e6276b8ca4d0c9e401368a0d5cc35 : SUCCESS in 13m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad4420f7e3c548dd89bd4abfb2a98e4d : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fe1ca70fca4481c9ca813d634e56a05 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8eb296c0186c4446aecad60f674786fa : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8656d99ad6d042f19c309debff6c43f0 : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"feca26102610ed46c74ed3211ef57bff20c79611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-25 14:08:48.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"8e1eaf0af0ff0074127cb50be304431cb5f6bddc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-25 14:21:51.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a65c0df9108fc51d4881ca6e4a3ccd2d53fe2725","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-25 15:10:34.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"6780122c3edea0a7639b680e0a69d71845856123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 16:14:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbfb7f35a61b4890a18a16a1ab29f12a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41daae3d1564413ab28b2fc0fd69d9d2 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3af776194664e8c8e7dad60b71e1992 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50d654558fa7467cba74db9ef62182d4 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6fec66fc78a84830b55b4dbc2956d498 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa6b47e3539b4a3aa0f49784ccc3cd7f : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c356993a5cdd467f98750b99f927b9a0 : SUCCESS in 8m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/69b271b10f7442e1ae89bbcf61ded86f : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1276da13bea04f6fa4d809735ec873d0 : SUCCESS in 1h 03m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b65a7fe669a44c2baede0d2d0497e5c5 : SUCCESS in 56m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5244258fcc474060abb0dd0ad3ca3bbb : SUCCESS in 1h 20m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b8b93c93befd4536a9c38a123ded2ddf : SUCCESS in 23m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/01fec5f2f2924a859f7f7d7d06e51bc7 : SUCCESS in 15m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1800e74c618740448bfafbde886e2350 : SUCCESS in 12m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1a827ad8e5984c00b60e5aeb5f75764c : SUCCESS in 1h 58m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b04ada551a5145bea1a08cd21ce2e5e5 : SUCCESS in 1h 49m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e529010a83bf4e65950e5cb0ebb6f1f1 : SUCCESS in 18m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5ae3f6fa5c346a69dbff304d67a4a14 : SUCCESS in 12m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bd1019f6fd6d4501bbd212194a2bd643 : SUCCESS in 47m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1dfc0753b38b4c2a845a9af08391b5f6 : SUCCESS in 39m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fe07c337c5f4befb27940ebcc4774fa : SUCCESS in 34m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/02fd3c27397f46d7aa950611ba8dc053 : SUCCESS in 15m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7a9f6a697a043048e97a3b4595c6dbe : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a7b9002a936b43da8170e97a99a1721a : SUCCESS in 23m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e67f5e864c0492e8bfc0dd3e9555eed : SUCCESS in 14m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7874294848f4a2a8d67ff06cf2a1025 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2666921d3174c97b10ded6e06f2c55a : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1509e3e12ee341a8b6a9abbed5c5e3ab : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7a47f6f7709455ab4d4ea080ffc83c1 : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c9b29f6da118ea0a5d020662c46e5ba54eb25618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 16:15:04.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"cb14105fe16c61edb271e1d6d3cc2aa761be975b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 18:29:21.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b6256ee0dc14819aaf40664d9a5bda1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4671980dbef24a1b8583fc611830d644 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f41e55bcb2242f1a388376c4980ef96 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/53435365f20f4447a70e4099b2f29a2c : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3cc1e7e976384d829b4129160f5775eb : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bbedcd9bbbab40bb8776487a947938d8 : SUCCESS in 8m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dc53494206584a6f90bbc04c745d2b4e : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dff6b283f6b4da19fa28cb636fb6e7b : SUCCESS in 54m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/37a4d024a2264e278e02feaed4ae156c : SUCCESS in 1h 14m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6141253fed9b417499f22596c0c1b37a : SUCCESS in 20m 12s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/595a95ae0f6b45179737b75bfd236a62 : SUCCESS in 12m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4250d1201d584b13a365163a2e582f8c : SUCCESS in 2h 07m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ff8def18e824a6fbd3dcb89694baf4b : SUCCESS in 1h 29m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c80a1dd6249d49c781baf3540d08997e : SUCCESS in 22m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ef7cdc1540d44268070dd3637c5d3eb : SUCCESS in 13m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08162866585547eb9e5659acde2ec29d : SUCCESS in 31m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b64ee92e85234a0d81131ab46c6bf478 : SUCCESS in 23m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/254b5c7398fd481c81c0a8824570cc17 : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37e3ed50e8544f72bd7e4ecb798e8412 : SUCCESS in 10m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/09413a91d86942fab0d55d3b49722f7a : SUCCESS in 19m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/785172d3fef141a88651c60a072df296 : SUCCESS in 11m 38s","accounts_in_message":[],"_revision_number":9},{"id":"48a29799889a3a6de4a75d7118b020dff2f75f4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 18:29:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"bd9091e19e0077ab098530ac45a3c820b28363a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 18:30:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40358bfacf81435791d412853c7d3b6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4789e0706d4f3c96dcde7ffc77e0c8 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6792fa697b5542b88d68e06eefefcc68 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"a22bbd9363478d75560e25fa922a88170615867f","revisions":{"7c7e25e01f0073bbf15a3a3ee3b4347df8dab837":{"kind":"REWORK","_number":1,"created":"2024-07-11 16:57:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/78/923978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/1"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:57:00.000000000","tz":60},"subject":"WIP: Use community.docker","message":"WIP: Use community.docker\n\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c7e25e01f0073bbf15a3a3ee3b4347df8dab837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c7e25e01f0073bbf15a3a3ee3b4347df8dab837"}]},"branch":"refs/heads/master"},"87a39dc9f7a2598418d28cfe65def50c3067b373":{"kind":"REWORK","_number":2,"created":"2024-07-12 09:04:29.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/78/923978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/2"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-12 09:04:27.000000000","tz":60},"subject":"WIP: Use community.docker","message":"WIP: Use community.docker\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/924043\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87a39dc9f7a2598418d28cfe65def50c3067b373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87a39dc9f7a2598418d28cfe65def50c3067b373"}]},"branch":"refs/heads/master"},"6813bcbeadc0ce749db038e876f33cacc0022a1a":{"kind":"REWORK","_number":3,"created":"2024-07-12 09:35:18.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/78/923978/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/3"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-12 09:32:46.000000000","tz":60},"subject":"WIP: Use community.docker","message":"WIP: Use community.docker\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/924043\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6813bcbeadc0ce749db038e876f33cacc0022a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6813bcbeadc0ce749db038e876f33cacc0022a1a"}]},"branch":"refs/heads/master"},"5c3646388ada99674aef227d5ac0cbd3e97e89f1":{"kind":"REWORK","_number":4,"created":"2024-07-12 12:09:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/78/923978/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/4"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-12 12:09:48.000000000","tz":60},"subject":"WIP: Use community.docker","message":"WIP: Use community.docker\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/924043\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c3646388ada99674aef227d5ac0cbd3e97e89f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c3646388ada99674aef227d5ac0cbd3e97e89f1"}]},"branch":"refs/heads/master"},"d84298ffbdc933d1e9546bd77e7a605642bbb6a9":{"kind":"REWORK","_number":5,"created":"2024-07-12 12:50:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/78/923978/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/5"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-12 12:50:25.000000000","tz":60},"subject":"WIP: Use community.docker","message":"WIP: Use community.docker\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/924043\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d84298ffbdc933d1e9546bd77e7a605642bbb6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d84298ffbdc933d1e9546bd77e7a605642bbb6a9"}]},"branch":"refs/heads/master"},"267c58c3c074f71206e0d52359405fe40d599b68":{"kind":"REWORK","_number":6,"created":"2024-07-12 15:17:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/78/923978/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/6"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-12 15:11:47.000000000","tz":60},"subject":"Use explicit version of community.docker","message":"Use explicit version of community.docker\n\nThe main driver for this change was that there is a bug in the ansible\nshipped version when using requests\u003e\u003d2.32. But it is more flexible if\nwe control the versions. Also fixed an issue where kolla roles would\nnot be installed if the same role existed in the kayobe ansible\ncollections path.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/924043\nCloses-Bug: #2072980\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/267c58c3c074f71206e0d52359405fe40d599b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/267c58c3c074f71206e0d52359405fe40d599b68"}]},"branch":"refs/heads/master"},"bc92fe1c11df5460062d192a8d562d84113dbd5d":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-07-12 17:53:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/78/923978/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/7"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-12 17:47:36.000000000","tz":60},"subject":"Use explicit version of community.docker","message":"Use explicit version of community.docker\n\nThis works around a bug in the ansible packaged version of\ncommunity.docker when using requests\u003e\u003d2.32. It is also more flexible if\nwe can control the versions.\n\nAlso fixes an issue where kolla collections would not be installed if\nthe same role existed in the kayobe ansible collections path.\n\nThis relies on the support added in kolla-ansible to use ansible-core,\nsee: https://review.opendev.org/c/openstack/kolla-ansible/+/896406.\n\nCloses-Bug: #2072980\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc92fe1c11df5460062d192a8d562d84113dbd5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc92fe1c11df5460062d192a8d562d84113dbd5d"}]},"branch":"refs/heads/master"},"69f7474d6eeeff1bc0199cff9531abd027e0624f":{"kind":"REWORK","_number":8,"created":"2024-07-23 14:20:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/923978/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/8"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-07-23 14:19:40.000000000","tz":120},"subject":"Use explicit version of community.docker","message":"Use explicit version of community.docker\n\nThis works around a bug in the ansible packaged version of\ncommunity.docker when using requests\u003e\u003d2.32. It is also more flexible if\nwe can control the versions.\n\nAlso fixes an issue where kolla collections would not be installed if\nthe same role existed in the kayobe ansible collections path.\n\nThis relies on the support added in kolla-ansible to use ansible-core,\nsee: https://review.opendev.org/c/openstack/kolla-ansible/+/896406.\n\nCloses-Bug: #2072980\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69f7474d6eeeff1bc0199cff9531abd027e0624f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69f7474d6eeeff1bc0199cff9531abd027e0624f"}]},"branch":"refs/heads/master"},"a22bbd9363478d75560e25fa922a88170615867f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-07-25 14:08:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/923978/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/923978/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/923978/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/923978/9"}}},"commit":{"parents":[{"commit":"73e38dc2fbcdf318aa6a41be5a353bf8e5ec7f68","subject":"Merge \"CI: Add non-voting Python 3.12 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73e38dc2fbcdf318aa6a41be5a353bf8e5ec7f68"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-07-11 16:56:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-07-25 14:08:48.000000000","tz":0},"subject":"Use explicit version of community.docker","message":"Use explicit version of community.docker\n\nThis works around a bug in the ansible packaged version of\ncommunity.docker when using requests\u003e\u003d2.32. It is also more flexible if\nwe can control the versions.\n\nAlso fixes an issue where kolla collections would not be installed if\nthe same role existed in the kayobe ansible collections path.\n\nThis relies on the support added in kolla-ansible to use ansible-core,\nsee: https://review.opendev.org/c/openstack/kolla-ansible/+/896406.\n\nCloses-Bug: #2072980\nChange-Id: Id9b19f10bc8ac38927914fb9782892b7daf1be82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a22bbd9363478d75560e25fa922a88170615867f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a22bbd9363478d75560e25fa922a88170615867f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
