)]}'
{"id":"openstack%2Fkayobe~924480","triplet_id":"openstack%2Fkayobe~master~Ied06ee3380a149c4c8d082562c93f9b771457cb4","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-07-29 09:01:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-07-29 14:16:20.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-07-29 12:25:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-07-29 14:16:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied06ee3380a149c4c8d082562c93f9b771457cb4","subject":"Bump stackhpc.linux collection to v1.2.3","status":"MERGED","created":"2024-07-19 09:57:39.000000000","updated":"2024-07-29 14:17:38.000000000","submitted":"2024-07-29 14:16:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924480","meta_rev_id":"d77b8ebb4c97087540ea6a5daf50d531cb26d667","_number":924480,"virtual_id_number":924480,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-29 14:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-07-29 09:01:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-29 12:25:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-29 12:25:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2024-07-29 08:52:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-19 11:18:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 05:21:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-07-24 09:20:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-07-24 10:26:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"66d725609ae29dbaaf0f9a8836cc12fd6214531a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-19 09:57:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c2561b43c0127567c15347f4bb28b3fd6cba085","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-19 10:02:03.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b414639a7163ed439b85ee2d2a55c85e9f04202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 11:18:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3060e96abe224bcfa71c9b7fbe5bc02a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73035713267341208fe213b06202f577 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692360945a124643bde20ca283eeffa4 : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e82c619da09d4838ae32656d7ea49b28 : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65f5663df7ff4256a5f8ea2799045c81 : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/984abe40a1f842f7a5d41b7e87549375 : SUCCESS in 10m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4653e540db224adeb7d3ba073f30ebf8 : FAILURE in 19m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fbc7cae7bfd74277b056ef98cb33e9d2 : FAILURE in 15m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aeda8e17c66545499cbe94cb2fc49712 : SUCCESS in 48m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9b5368164e19409db2042260aa62ce38 : FAILURE in 24m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4a476db6f78747c69bb750303a20ec1c : SUCCESS in 21m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/75b1daed94d8493bbca5bad7b3a8dcc8 : SUCCESS in 14m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10e3e8ba00b8415fb8b0c2ffe68a42b4 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6f8efbb9ac3a451ea824f48deb2b234f : FAILURE in 1h 15m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e473e435e0634007a23185fef948a3aa : FAILURE in 1h 03m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/012da3e974dd42049e0f6a67e7a7869a : FAILURE in 14m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98a7e9f7105d42f19ba8907c9d073021 : SUCCESS in 13m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/feac68d9bcc849668b17b97fb72a299b : FAILURE in 14m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f82bdf7e530e4f6199cd2ed67a8990d6 : SUCCESS in 31m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83c363afd485478a9324563f701d6ed0 : SUCCESS in 24m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/238deb71449840388ac8b8e28a128ba7 : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07781ac33a0349acba51d5ffaa9fc6f3 : SUCCESS in 9m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59ee010b19ee4b018dfab671bbc06eb0 : SUCCESS in 18m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a831eb82e316486ea5aaf5b1187036fd : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/504bbf7dc2b444f1a589db0942a5ab6c : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5675d17a5e84a5795fbec18aa6a17b8 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bdc58b6c4f2488099a450d59f829bd0 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"716ceb450ca97bb7acef46670a6ff89b8b028786","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-24 09:20:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d663d291f882d652b2bc1c60311505bb4e895bb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-24 10:26:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"704f2ad6120074c4d7cc9b13d86d6717efc3e07b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 12:05:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59ec67bf15a642f7a61bdb51e40153d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2be9a67adbff47028529ece06149a186 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce7a903c605459485a80f7cdf54b4d1 : SUCCESS in 6m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4cf79169dfee4b97addd61a9c7df41a9 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92179645493245be84f9f9a02d3034f0 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a38bfb7fa6c42dbb6ac7c3bce179136 : SUCCESS in 8m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/092996894f544b7cb1e66866e1c0ca09 : FAILURE in 22m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fb9a35bb199341aeb1f142555833903f : FAILURE in 17m 03s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c83c6c327704a338846f4e3aa26ff69 : SUCCESS in 49m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ef79c0940e04486e94d437b50f683be3 : FAILURE in 22m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fa8442a0ae2440069f3edcb28231413f : SUCCESS in 19m 53s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4ca7a3be30744ab3be8a8564ccd6648e : SUCCESS in 15m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbf5ecb7b73b4ebc8599153276842fcc : SUCCESS in 15m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e9874193f7094f198971afaf5fed857d : FAILURE in 1h 21m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f11369400d3343d6b731e557f4d4f6e2 : FAILURE in 1h 03m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5ac77529360f479fbd8a009572afd284 : FAILURE in 12m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54bca7befed8454bab74a0c5efcea9b8 : SUCCESS in 15m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/aee4407b998f42c69e563b424763b84d : FAILURE in 13m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bde684f086ce4123ba6fab8592a3018e : SUCCESS in 31m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb1667c16f4546ef88e8939ee05075c9 : SUCCESS in 28m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c1e03f6d9dcd4e6fb3428fc26a02c64c : SUCCESS in 15m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89642d7b135d4fcb89c3c58ba21ea11b : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76401b88213c4c79af71e88b3bb80ade : SUCCESS in 23m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0ce9cff67e14476a34ec0b74a2dbb48 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/511799b1873046559840763e9c61dc2d : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad920bd1997346499b60718ea15a59fd : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df89fa4affe14a0690f2a2811c21a7d9 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2e4c227035b484b97f4210eba058923 : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b9a20fd1b49278f5fd0d0b498d70a989f124393","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-24 13:14:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"896b9c20812551a41d6cd0bf048870c60d0a9560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-25 20:10:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_35263\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"_revision_number":2},{"id":"6031ab1c4db52930dc9783e09f5b978ba1d13479","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-25 20:12:49.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a058f031c93c8c10ee91d7a0ac61fde62203efcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 22:17:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce7b8e06e94948cbb69996ba6da2274e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb43ea234e014a9ab15c9cf12e706b5c : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39c4cbbb4a614db99889badd467ce3d8 : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c0147d449584ffcb36ce638a8b795dd : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7c8959b7a03f4fe980a0ebaa8c4b5d7b : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e341035ffdb6473da906849e95a08cef : SUCCESS in 7m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/570206542d70441989153df16d502418 : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f1e0faae402b42d2a5d217d64851c28a : SUCCESS in 1h 00m 37s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91ec507bc0c54e36a48d496163fd848a : SUCCESS in 52m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3fc9e64a66e44bf69aac354724e4a98f : SUCCESS in 1h 16m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c1aaf3f487bc4465b3c14d1f60dac43e : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e67dfe3120094a15a41434e4c785cb2f : SUCCESS in 15m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e198a25c4c164dba86f9d4831165b22a : SUCCESS in 12m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3e0809902e2643b4858bbfd1b96cdf2d : SUCCESS in 2h 00m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/941648deda114525a678fb9b3d01459e : SUCCESS in 1h 16m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/595a20ee7b564485b52d74a6d691acb7 : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de1ea850de424d848f14c5a040f79867 : SUCCESS in 15m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9e5b753e5e544ed297d933293ec38ca9 : SUCCESS in 39m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1cd6addd8a0545a681005928738654d0 : SUCCESS in 33m 19s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79e96b03919241f78ff1349cc2b04c43 : SUCCESS in 21m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c17ef4f4bd524e0e921205a3ee525a3f : SUCCESS in 16m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1899ca6818824b07b86e4c962933cc75 : SUCCESS in 10m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/be083932c3584dcba40acd7a85d64912 : SUCCESS in 18m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abbd610205444614bf2d481322bfe303 : SUCCESS in 12m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55c17517ed55472bb30d835293be51dd : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32ec5358d57b40789ad9eb2aae7c279a : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/619ec74a93d04d079a0abd4bbca99bc5 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe7e301f2c2b4a70ac101c4347093687 : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e3f53313f04ea5bd31dd79e6ed9a14e581a388d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-26 12:05:30.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d3c7a2814dd5e15e5e8256737b1b7118fd5c4d3b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-26 12:05:48.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b812b66555b3c5baebae2cc7858d9d690b9d4aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 14:05:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec3a9880ad1489cbd8ebf6b2cefaa8f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e16ad96aa394fd9bf69ccebe6e1bec9 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3859908ee64fb1afba4e4ed8405d26 : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be2434656f6a4889b0a7b1c2e6974864 : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f46c8dd0fdf4a74898558cccc6d5fcb : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c6fd145cb404cc6ab25468172d2506c : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c738581506f45658fcf0456b9da0270 : SUCCESS in 8m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b7a0228ee39641ee9a5d5fca23cdb807 : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/321b54bb20aa40ebb5f962975f882d52 : SUCCESS in 53m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccb4534ef97541e08bec7c52216cc49d : SUCCESS in 51m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e9ac10d656984957ba974e5dd8d1dfb3 : SUCCESS in 1h 11m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7dc7789782354623b0c161ba5dcf3df4 : SUCCESS in 18m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/79637d38dd6049139dbb5524761c7966 : SUCCESS in 14m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78ef105b509945a190ea927a56c33910 : SUCCESS in 12m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ba2b7547ecf84fd88ecbd943917700c8 : SUCCESS in 1h 55m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c89ffc61caba4734b19ea2e92112e2fb : SUCCESS in 1h 22m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0a0c2664307b461abe9fc8c2a9f513b0 : SUCCESS in 22m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f14fe9bc17c464e9c5462d96d06ed7d : SUCCESS in 13m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/199791cb13174ee88be1c38525f78540 : SUCCESS in 38m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/76c785cab84748338ba3bac5ea168837 : SUCCESS in 34m 50s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36ad23fca42845dbaeed4afd2a9dec6e : SUCCESS in 29m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2734d176450e4bf6bd05877a8578a8de : SUCCESS in 16m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/168c7297a4324a5d932d1eeafdd585c3 : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7907a3e8992947109e94a02ce9a1ca16 : SUCCESS in 21m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ffa9a8548704e47a55d23710544b779 : SUCCESS in 14m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74a7e3cdfe8644aeb72ea6bbd88587fa : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6fd7e968e994b8795722010693ef829 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c8dff11f146451fbeb5b30bf0db45ab : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5056ca7f3684826998714350e4f109e : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0d2bb64a0f60981ee9b7eeabfe67fcddf241e126","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-26 20:11:21.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2602478e55460d66b336dc82a3c6653ded7d4cb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-29 08:52:05.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"56189b6f80b2de7ee7dedb6d243074ddfcd2b07d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-29 08:52:08.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d23e87867235f7e755c5d8d5912b79f6fd2825e7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-29 09:01:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5358a4bfe40a177cb82662c9d22cddf50b2d82e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 10:44:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01cee166765e498f93c24ceb613a0859\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c102204511944a0ace6cae8be0917ec : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98288606f97a4ac6896fc645e6dc99b1 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b78be4e0b9d94af89e0aed0d97bb9376 : SUCCESS in 2m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02e2daeabc334544af97351331396ef0 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a170a163c9e64528b36f4dacd6ea3861 : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bbc2a18abbc4c59bdaf82b55834c081 : SUCCESS in 10m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/def4c3797fb044b48cdfe20cd114ed03 : SUCCESS in 1h 06m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1508819310b348489b071c68aca49a0e : SUCCESS in 1h 02m 28s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60877e5d494c4f1cb92fa523e2070984 : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a83969fd024f4e25bec8c1ac4793f01f : SUCCESS in 1h 12m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/74c4ac1cf31a4b1397a6106b63cf0350 : SUCCESS in 18m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/86be0d16175b485aa89dafadf86ffc17 : SUCCESS in 19m 11s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16526a406b6f4264a97de111da415956 : SUCCESS in 12m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/47970156bfe043dd8c350120bf5469ae : SUCCESS in 1h 51m 09s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c45cb944ab84c4d81f501cc24d3c411 : SUCCESS in 1h 31m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e87b932993974b96ab4911f5ddd37db2 : SUCCESS in 18m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45717d059e4b422ab009f0dbe5df3c58 : SUCCESS in 11m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e12590227c40470cb6e2457bb5416508 : SUCCESS in 39m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f610ef860bd84f20a0f2480ea9b25cc4 : SUCCESS in 31m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f7f133e22aa4597b317838c198596c2 : SUCCESS in 19m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d09e10b138bb45d69bd1848ee691c6ec : SUCCESS in 15m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ee7eb1a3ce34d3d85f8451d95c1c3c2 : SUCCESS in 9m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/158b43c902c744d6907d18872321ba92 : SUCCESS in 22m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea06576604ee42bdb2b61281380a1abe : SUCCESS in 11m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9533e9a6106a477691899cfb9ee4f48d : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c48c31888624bda893b33fd0b571439 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e4eb752053149d18e1653520549c029 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/827eb26838d746c6be26c08954df4d8b : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b3bda18dbd8a0231289d227ed32506689f81a3bf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-29 12:25:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"30b59f891ed12e7dcf071052527ed9b09e528e25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 12:25:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e867a6ef515782498971a22009c11e22769ae115","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 14:16:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8410e94b786f459eb0135656624c6f62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6321c796907f4452a001f2e590af3917 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a951e808e06a44bc861e69cd881171b6 : SUCCESS in 2m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0af5e9b6e48c406a80284e93bb606a63 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d752fabef54a472ea79e01ab32616501 : SUCCESS in 7m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb1285aa3e174e1ebaa35b2a1447515f : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b8242efcf0714a25b73807e91fb1b940 : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64f6e7c9c88344c09b5160d9695d0160 : SUCCESS in 53m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6c8ad3935b0e4336912a2b76b06a47e2 : SUCCESS in 1h 12m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/db790f1c46914d71ae8d51002894e71e : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6427223616041f18331891958d389f6 : SUCCESS in 12m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/09ee1dfff3d546158c38939a6ed0ee97 : SUCCESS in 1h 36m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b469f9cd62c44c1ba46c020e46b494e : SUCCESS in 1h 49m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d836cd0f908d4143b4a065fb5e0a2196 : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/490c9d48ce6341598faabd64eafca0b1 : SUCCESS in 16m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/01fcb46ad6434bdb8688562c7aba6ada : SUCCESS in 31m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98d527f1f0364a179e459ab5c914c18b : SUCCESS in 27m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/12aeebcc22694e389a82a08411532f95 : SUCCESS in 18m 37s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4a50807fe994912816725c73f5543eb : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d863b3ee859d4c7abc93e9091392f08d : SUCCESS in 20m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/116e27581e2b46a8bc010e99ab195625 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0480f1acf4e62a4daf8337bf073bfb76e3ee51ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 14:16:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d77b8ebb4c97087540ea6a5daf50d531cb26d667","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 14:17:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19e86e84012b46aeace5cc34cb1f7618\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ba54d4fe1d4c29b01b198685c4f463 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/127e072d4c924541b863b2584381a84a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bcdc4e6c2c43f96ee5eab772adfd7e06366f7ee6","revisions":{"ecfa88319f2b080c6ad64ae778492343f7314e1b":{"kind":"REWORK","_number":1,"created":"2024-07-19 09:57:39.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/80/924480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/924480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/924480/1"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-19 09:57:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-19 09:57:18.000000000","tz":60},"subject":"Bump stackhpc.linux collection to v1.2.3","message":"Bump stackhpc.linux collection to v1.2.3\n\nCloses-Bug: #2073100\nChange-Id: Ied06ee3380a149c4c8d082562c93f9b771457cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecfa88319f2b080c6ad64ae778492343f7314e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecfa88319f2b080c6ad64ae778492343f7314e1b"}]},"branch":"refs/heads/master"},"5927e3b63b15a6b12479615981a7419e5fbd6ff7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-25 20:10:31.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/924480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/924480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/924480/2"}}},"commit":{"parents":[{"commit":"a22bbd9363478d75560e25fa922a88170615867f","subject":"Use explicit version of community.docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a22bbd9363478d75560e25fa922a88170615867f"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-19 09:57:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-25 20:10:31.000000000","tz":0},"subject":"Bump stackhpc.linux collection to v1.2.3","message":"Bump stackhpc.linux collection to v1.2.3\n\nCloses-Bug: #2073100\nChange-Id: Ied06ee3380a149c4c8d082562c93f9b771457cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5927e3b63b15a6b12479615981a7419e5fbd6ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5927e3b63b15a6b12479615981a7419e5fbd6ff7"}]},"branch":"refs/heads/master","description":"Rebase"},"1657827f987ad8db28707871ebf252a508f87b3a":{"kind":"REWORK","_number":3,"created":"2024-07-26 12:05:30.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/80/924480/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/924480/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/924480/3"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-19 09:57:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-26 12:05:20.000000000","tz":60},"subject":"Bump stackhpc.linux collection to v1.2.3","message":"Bump stackhpc.linux collection to v1.2.3\n\nCloses-Bug: #2073100\nChange-Id: Ied06ee3380a149c4c8d082562c93f9b771457cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1657827f987ad8db28707871ebf252a508f87b3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1657827f987ad8db28707871ebf252a508f87b3a"}]},"branch":"refs/heads/master"},"bcdc4e6c2c43f96ee5eab772adfd7e06366f7ee6":{"kind":"REWORK","_number":4,"created":"2024-07-29 08:52:05.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/80/924480/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/924480/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/924480/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/924480/4"}}},"commit":{"parents":[{"commit":"6e302108a3fa7e93b3de72e33a87ffd1004ac0db","subject":"Merge \"Set slightly more accurate name for the manage containers role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e302108a3fa7e93b3de72e33a87ffd1004ac0db"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-19 09:57:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-29 08:51:52.000000000","tz":60},"subject":"Bump stackhpc.linux collection to v1.2.3","message":"Bump stackhpc.linux collection to v1.2.3\n\nCloses-Bug: #2073100\nChange-Id: Ied06ee3380a149c4c8d082562c93f9b771457cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bcdc4e6c2c43f96ee5eab772adfd7e06366f7ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bcdc4e6c2c43f96ee5eab772adfd7e06366f7ee6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
