)]}'
{"id":"openstack%2Fkayobe~925104","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Ied06ee3380a149c4c8d082562c93f9b771457cb4","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-08-05 15:56:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied06ee3380a149c4c8d082562c93f9b771457cb4","subject":"Bump stackhpc.linux collection to v1.2.3","status":"MERGED","created":"2024-07-29 14:41:08.000000000","updated":"2024-08-05 15:58:03.000000000","submitted":"2024-08-05 15:56:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925104","cherry_pick_of_change":924480,"cherry_pick_of_patch_set":4,"meta_rev_id":"5ba39639a33631c277f2a313c06f182a8ec832f4","_number":925104,"virtual_id_number":925104,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-05 15:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-07-29 18:24:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-08-05 13:55:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-29 17:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2024-07-30 09:19:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-08-05 13:55:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-29 14:41:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2024-07-29 16:49:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-29 18:24:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-07-30 09:19:33.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-08-05 13:55:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"fa6d20d310f02caa9a75bab5a7ea98620b69a4be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-29 14:41:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bfd6db3059619626ee2914ed7261cf9e9c0b1871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 16:49:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e095ee03303749b3a9f56485a6c1b5fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8c346d4f9e64ea6b0741628e4266e39 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed0f633febd4ccf99b281d8ef02a2c9 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/003e3dcb55164e698fdbdd07fead08a2 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b827b6c4d1c84c1788ca1dcde812b7b4 : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f360cf62321466a8d08f51f984fd03d : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35e1e5c999714e0980318c3d77494ee0 : SUCCESS in 9m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0d7aa0fc95214d9790ea319d8b98dada : SUCCESS in 1h 09m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/08e66220b65b47ba8c8ef47e404c76a6 : SUCCESS in 58m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9aa2c59910d4358b72fa8162f080248 : SUCCESS in 50m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1344b077098d40cf92d787fcaed3467c : SUCCESS in 1h 11m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/80692e76134748549850df3404e1817a : SUCCESS in 23m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0896ef8336ee4dd8bedcccbf4233dd0d : SUCCESS in 19m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e882ce0338b4ee2bdf62ecf4ea77bcd : SUCCESS in 17m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0760565f1cdd4fa88749630bf5e80cfa : SUCCESS in 1h 58m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90bde5e1ee2944c6b37ac04e7ccfb860 : SUCCESS in 1h 33m 32s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/13f88d1b1ff94637a2879f4ac2129347 : FAILURE in 1h 31m 04s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee0dca9beae943e0b28cb111e75db6c6 : FAILURE in 1h 11m 52s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0420739956754d12b0df6b14c8dba9e5 : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f105661fb7ff4322a2528029c95a28d3 : SUCCESS in 13m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6a0ad71fc2a74296ba8070e964ae3eb5 : SUCCESS in 43m 07s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd2a2a2778d44e9a9f96100df7f842af : SUCCESS in 36m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d6de03578a8453eb8782f319faeb49b : SUCCESS in 30m 35s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8bb3a7ef589943cd88fd5512caf1f5d6 : SUCCESS in 33m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f7db91985ff444ebdc82968329b29bd : SUCCESS in 23m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/411a0bf1d5564e4f9bae1fea22081f88 : SUCCESS in 15m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcabe8b14c6b48ceab85631c7fe2877e : SUCCESS in 10m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/63343362b2c7467e82d747f810a518e1 : SUCCESS in 18m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e42d12093724206932ea63b65083499 : SUCCESS in 12m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8be391fcff654671bf7196d1f5c49862 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4863339b479a4f0a82fa34e8979ddc7d : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2014974aece0448487584510816eee0c : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"27470e4d164b57ee64e27a14a03ec0343cb68007","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-07-29 17:51:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60186baa584ef8be537c459206f8a559e435d962","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-29 18:24:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6067a7abf21024e1394311fcfeb756eeab770d4c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-30 09:19:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"58b8e51e888265227570b2eb079c4bbd778ec4f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-05 13:55:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd5684ebc5f428068d6c4b5f58f0cfaf6b9d6a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 13:56:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a33df73627652c3c675a1ce781f93fa7c55355e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 15:56:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6abb75ff7496440aaf7abf497c4df079\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b3743099f441a8ab401b1dd1e72e8a : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1c67bc2535245c8b3b6c42630f5994b : SUCCESS in 2m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/705b1ff428c247559bc8753f5a1d5475 : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9fea1ff0d2245a5988c8195c2106d3a : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ddb9f2284d3432893b8e38cca672ba4 : SUCCESS in 10m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/41446083976f4dab8ed67c4ca9d435ac : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76e3d4dddbbd4e89ba452f840401a760 : SUCCESS in 57m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/339ac9bacec64aa6a0d96ee1806823d6 : SUCCESS in 1h 11m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6ff35b7cb69d47bb9cf7660d71600bf2 : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6d86831bf53404097a6b914ba73ef21 : SUCCESS in 13m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ac158a3d88c848309ed5d802584905a9 : SUCCESS in 1h 51m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4330fce11314456c86c94b33cb2191d6 : SUCCESS in 1h 49m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9fc2546477f949b1a8c3ccb63a326d1d : SUCCESS in 22m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26c907a1692546388d6ef5b975a7b348 : SUCCESS in 14m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2d364c519dce481a8889a8d973157ace : SUCCESS in 30m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/684be44b79f14431888cbec9ab2b1337 : SUCCESS in 25m 04s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/68156d59cdd94338ac60486774e3b75f : SUCCESS in 35m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f96535a6e91446c89587cef77ded877 : SUCCESS in 29m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea17aeaab28a4804830b3ff80a17223e : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b82c2788fd7e4b6f9155bb338abe4eed : SUCCESS in 10m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/11879232d79d41bbb6a97abf0b2ed504 : SUCCESS in 19m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d8450d62c51441593f47064865545d3 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"da0126af0829a6be709b705f3d0b70311efbc769","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 15:56:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ba39639a33631c277f2a313c06f182a8ec832f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 15:58:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/863bc6a5191b45e497235d78be4c592e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a60a1232e9745728a5dd1c34c2117cf : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4649e4367d2a4eb997dc7bcc3d671dff : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"688f3775e06782834bf5759fe7dcf4b895e62b2c","revisions":{"688f3775e06782834bf5759fe7dcf4b895e62b2c":{"kind":"REWORK","_number":1,"created":"2024-07-29 14:41:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/925104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/925104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/925104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/925104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/925104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/925104/1"}}},"commit":{"parents":[{"commit":"a59f4ed2c3215b1d7a3cc15c0bb2c786b559e3de","subject":"Update .gitreview for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a59f4ed2c3215b1d7a3cc15c0bb2c786b559e3de"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-19 09:57:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-07-29 14:41:08.000000000","tz":0},"subject":"Bump stackhpc.linux collection to v1.2.3","message":"Bump stackhpc.linux collection to v1.2.3\n\nCloses-Bug: #2073100\nChange-Id: Ied06ee3380a149c4c8d082562c93f9b771457cb4\n(cherry picked from commit bcdc4e6c2c43f96ee5eab772adfd7e06366f7ee6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/688f3775e06782834bf5759fe7dcf4b895e62b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/688f3775e06782834bf5759fe7dcf4b895e62b2c"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
