)]}'
{"id":"openstack%2Fkayobe~926639","triplet_id":"openstack%2Fkayobe~master~Idfbf0eed0b2333d550f822e40a0d96ae227f6159","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-11-05 14:26:57.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-05 11:27:19.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-11-05 11:31:15.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-11-05 14:26:57.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-11-05 14:26:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idfbf0eed0b2333d550f822e40a0d96ae227f6159","subject":"Align Kayobe Ansible version with Kolla Ansible","status":"MERGED","created":"2024-08-20 13:10:14.000000000","updated":"2024-11-05 14:29:21.000000000","submitted":"2024-11-05 14:26:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926639","meta_rev_id":"c48f30cabbdce19605ab5f82a3dcf8016d98099a","_number":926639,"virtual_id_number":926639,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-05 14:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-11-05 10:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-11-05 11:27:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-11-05 11:31:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-05 11:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-11-05 11:27:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-11-05 10:47:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-20 16:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-21 12:58:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-06 08:31:34.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-09-06 13:41:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-09-07 10:42:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-09 10:02:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"3594d36927ce2df15538c4cc19d00f6f301b166b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-20 13:10:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0ee743b64e0c3fcbab8cab43b92dc7a6e9060f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-20 13:24:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"49a9abb755dbf74fa414b520e8886a6e56b6f30f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-20 13:31:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a754984382d9032a97b804aa28007a85f96dbd97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-20 14:44:18.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"08dab3daf018f2fd5af3cec78e4a02a112cee551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 16:35:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e88c2d31ee0c4488a9fd6deaf2355fbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/559028cafdbc48988687798102f17a5b : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c8790388ef4324a1ff81b3e6a5e548 : SUCCESS in 7m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ba8bbfb450848fd8e5e543dc337d08c : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe213d265b4d4f4abae687e48a222735 : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/31921934f7144f62a6c1a27b3c3193c5 : SUCCESS in 9m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/92dedbb2fe384f88aaaf9d805ec424d9 : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b62c40bcd34e49d795e6fdb959555eb9 : SUCCESS in 1h 05m 36s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea8e1aecbb884c6a9b1a3d521da87c7d : SUCCESS in 51m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/260e0ec9191f4d78bfd4eae26db71b1b : SUCCESS in 1h 13m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/643c40c1ade34f9398435860e667926d : RETRY_LIMIT in 8m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bf002f23002c48c7aceb898e31bd53e0 : RETRY_LIMIT in 6m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4326d6fa1c55478fbdca8718597d40be : RETRY_LIMIT in 4m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c8c445b207c840b88886f4b9ad460cdd : FAILURE in 1h 03m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a2f4c38d98e476a8c981c80d9c14912 : SUCCESS in 1h 41m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/16534cd3e0af412ca07e3e439e1edac7 : SUCCESS in 18m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1db3a2911c7d410191e4d3d090ec9d10 : SUCCESS in 12m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f45f1d6e0ae74038b4ad5536ce35e129 : SUCCESS in 41m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9426914c1c804c62904346c11607862b : FAILURE in 21m 42s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b387cdc99474654ae68aa8d356b79a4 : SUCCESS in 26m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e9232648a1e24147ae0017650dbd474d : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4c24420040a496891514f644bea33e4 : SUCCESS in 10m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4bf60e8399aa47f4be2aefae6708a3a7 : SUCCESS in 20m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a0b974f81d043b6aff86490d67bdf26 : SUCCESS in 13m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6a3c85c5f4d4e74ac4d8d823a2338be : FAILURE in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ddb95fe1c344d8f839cf30c4ee6fddf : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f08e79b335045f98c04729bf896b23d : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4343eadf89a7419aabf9077087d91ec4 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"683f5f599442d8f15661ec33865d6d4db2bac1dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-20 17:08:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"989aaa9851652db37c42cd9ba3cf1a2c26dd573e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-20 17:15:47.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e76788ea32c651db8f5551b0a7aeaea172c3d6f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-20 17:46:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"85203403ed39f0b95a1adae3fe7fa54ff75708b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 19:45:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a55a06f7130c4d9bb106bd4865d314bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f152466ee6c4eb18c6125c39e7265eb : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f255d82403da40d094c141220787f28b : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79dbe3c22b024f589f81fc1d82b20dfb : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e86f876ee9e640419306d0e81bd2e021 : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd27727b007843e991c8fa24c00358bb : SUCCESS in 7m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d2c4a9ff750441a7824470633e338e8e : SUCCESS in 1h 13m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a5726fb3d0f84b2b875c330ad685b55f : SUCCESS in 1h 01m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8e742718f2d414c9ba736033208cd47 : SUCCESS in 51m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4ec4be37166d44ae90f0bc6e21d9c2d0 : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/96fa58ef8f8b41afaa046cfd7e4558c2 : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8c0286de972e4504a8114da4b7a3c62a : SUCCESS in 13m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc36549f90884ee2a8fc42f9ad0fb93a : SUCCESS in 12m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b0b4af7222fa444587f64fd33d302e0a : FAILURE in 1h 03m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14bc83c001454501a5530f0c94643803 : SUCCESS in 1h 44m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b3851ca0a42745a4b740a0214ce07d0d : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c0721b5e65743b0b4ae8557060f75be : SUCCESS in 11m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/51a342dff43c4baaaa478589e3857569 : SUCCESS in 43m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c125ad92b9c54075a654857a554534a6 : FAILURE in 20m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0dae902b2864437a30f747c6d26fd17 : SUCCESS in 20m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c2b037eb832545389c6a35bd3a2e4299 : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1125995ccd4a407b8eba1643ad77919b : SUCCESS in 9m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6481e419fa644155b5976d812a162262 : SUCCESS in 20m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bebd20ba5b3a438f96ebb86fa0cc6dbe : SUCCESS in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cf4ba477e834a10bfa49828b52b8c26 : FAILURE in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c1deed43f90434a9d4331719fb3dc60 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58226f7e2492452a85ad04818a12f864 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d68722bce206412cbf5286ce967a4a07 : SUCCESS in 3m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f3a2d4dc334ae00270fbd6c3626ea9865620d285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-21 11:31:35.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8337c61fa629b00f533ca30c17445d999c6e3c91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 13:20:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e53afaaef9146f4bf1dff49cb42f540\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f910c2f50ce944c6a15954fddbd0b26d : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcda044a493945fc93087eb5b5998406 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a28077dd82bc4013b0f1e5a234e9a625 : SUCCESS in 2m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca528a7ab2924eb6a830c692f0ef6e9a : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b827127f7c0540cca97d2bdaed6e0c99 : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/526bec6e996b4d1192d94a2456924ed1 : SUCCESS in 8m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3c03c5b4b4224ba4acf0642de24e3dbc : SUCCESS in 1h 06m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a51db13b926e45fab87abcd496b7cb2f : SUCCESS in 1h 04m 42s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cc5330458bc4841ba1e0e033689ebd7 : SUCCESS in 55m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3fd2da59ae854f27839538879752ba1f : SUCCESS in 1h 17m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f0dbb1df01f5444880df2763fd267b98 : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9890954c75a34995abb68e463788de30 : SUCCESS in 17m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/831a657b95324415895bdd03335ee826 : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/db5e9adcc6264bca8116e66a5ac5e823 : SUCCESS in 1h 46m 57s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39313534b31c4aeba9b1da891cf36c87 : SUCCESS in 1h 38m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/92235f37e9dd409f91569ff11e5ba305 : SUCCESS in 19m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b823f95ffddd45c1a2f9c0e1ac3b9d58 : SUCCESS in 15m 19s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ef81cce6fb664e63ad47612d5c986302 : SUCCESS in 39m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d56b3a67c5064535828aced2a8655598 : FAILURE in 21m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a9ee292b0f642ad812e36e8349d8f38 : SUCCESS in 25m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8b86d272314f42af93ebccbc9c5400ad : SUCCESS in 14m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/222256b447a44616b0fc7394afde5ffc : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/644c145ce50046bd925386926fd6c018 : SUCCESS in 18m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f2fda8e4a5144fe9467847402693895 : SUCCESS in 14m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c6c3ffd0aa0401ca56807113fabe554 : FAILURE in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13126d36844541d9b954bf657da6a330 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0562c8a0c4b142cabab9cb86549c5d99 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d4c630fc14a432e8902480c364eebfa : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fe3b45537b754ece13427efc67df539011aa632b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-21 16:02:12.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"e43caa9fcc77f1609edafa151b292478045aaaa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 17:37:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/475e503f9d394924943f722ecac8b934\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/857db0716a7b4e2f8b78a152c33a71f3 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a902605af78b4052a50c22536cddb8d1 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c19dce775814fe1bff73d4d771583e8 : RETRY_LIMIT in 2m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd2829f494604397a5c28283e51db8c4 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f53ba0eed87d455d86333d07a49cd985 : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f49a97af6f4d437f9bcc39778fd30958 : SUCCESS in 9m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c74ae898b24442e193d796c966f4cc31 : SUCCESS in 59m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fb1b5fe4210549b69288a2a9dce32501 : SUCCESS in 48m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20534e486d7c410f892217402e3726f6 : SUCCESS in 46m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c622fa266b274effa20c781a3a1e8d6a : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a7a171ac2de14c4f9e124a9429cd4d04 : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e9b3207506c049e09aeca2cef76fb79b : SUCCESS in 13m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38f00aa9bcf74f899a477c51d96ea026 : SUCCESS in 12m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c3923fc699a74e3a9973e477c3735b7e : FAILURE in 13m 27s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/047eda0f887c4743b4cd5b4542cef55d : SUCCESS in 1h 33m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/abc901546d5a4fd9b32db686d04ad25c : SUCCESS in 20m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ac86a7c900c4228bb5b9ba290f2a9e7 : SUCCESS in 11m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5b06575b121d467bacb8a4528bf51eab : SUCCESS in 37m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1dfd69fd795b49d6b114be5b47d0e180 : SUCCESS in 31m 50s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/319232632f1d4796b6e62889338fb66e : SUCCESS in 23m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4fcbbe242e75444291c6b3e07d3b72e4 : SUCCESS in 17m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3636bfa045cf48609c5073455388bef5 : SUCCESS in 9m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d1ee018663b94bfb84746e53caccea59 : SUCCESS in 17m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12ba993311c04f978acb6e34140fbb65 : SUCCESS in 13m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4abc9b1743e44b67a4dddc74710412dd : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54c12c1afd344461ade2eb95ba5e1e5f : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd2404e067f4472bb8a86dc1fd7826fd : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":9},{"id":"5d5e57832c8cf34d89d840c6106792af046339c5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-06 08:32:58.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nThe following files contain Git conflicts:\n* zuul.d/project.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"0d96fb05be622118c7554e4b11c5fd2fdb4c84f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 08:33:20.000000000","message":"Patch Set 10: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kayobe/zuul.d/project.yaml@master\", line 34, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kayobe/zuul.d/project.yaml@master\", line 38, column 1","accounts_in_message":[],"_revision_number":10},{"id":"52d2d8e814153e5118251dcef226d71c3252b25a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-06 08:37:10.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"f0c3d1a4d69389fcc4ad55d87228d97d1cfa9e2a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-06 08:38:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"549933028af7b8d5c0387239f46db918085b403a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 10:33:09.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c75e73944d94847850900a99cd81dd3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa43892f40ca456ca0807f07d7eb6c80 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b48e0b8eb10f4dba8943acee7946d66d : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8aa9998e59742bdb7636fe3cf374473 : FAILURE in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dadde381babe4b459bdbfde315b744dd : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9324164e4364710b8fee8b565dce230 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c50bd6845db7410aba4d6e0d1bca79f0 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd05989e990b47cd82addde9cb3ae7ef : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c4b0c54f1a33479e8bdf6ce7897baa3a : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d493884ccc344415a86d41eac2fc835c : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b25ab6f493ef43d69731661e57c7e0f6 : SUCCESS in 9m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f83d1474c22443e3a91f9e70e17ecd5e : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1c6282758f4b42139befcac5c76ac326 : SUCCESS in 53m 25s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59bf7450e8b344a281af0f9220d7a848 : SUCCESS in 48m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/058aae26559544c9b76787c2407aab5d : SUCCESS in 1h 09m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0ed1ded87809419f889d057412c79c3f : SUCCESS in 19m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ed130f0725e74a12b573e063f59a37ef : SUCCESS in 16m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/455b237663ff4d788ef70e9daebdc78b : SUCCESS in 13m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3111519360254a4dbbc7655e260f0d8b : SUCCESS in 1h 52m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfb4457888854f4dbcc4b0d15a2ea7c7 : SUCCESS in 1h 43m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/97d7dba9303c4baf90da4ac661282118 : SUCCESS in 18m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73f9d0d14ace41b78d151821ca4327c8 : SUCCESS in 12m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a2cc7538db6246ac88e1c15728c74a50 : SUCCESS in 39m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/84a09f6629d641399d0dadcbe75d5063 : SUCCESS in 33m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95ea93f16656454fb20f260667b959c6 : SUCCESS in 26m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dc1c14dd89974219b38c0fed5cca6ccc : SUCCESS in 13m 55s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/5e1504f51c504c17836a33cc8f856f60 : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2479b775ff64e28a07c49a50dbd1958 : SUCCESS in 10m 52s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/023a23463fe046cea3742716c3233eb8 : SUCCESS in 9m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3fbba5ab97fc4736bd66a3281928b961 : SUCCESS in 20m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82a809db4f1d464b841677e63995e4a6 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":11},{"id":"d234a19843fecd28c3808d154d24dccdd6c527a2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-06 13:41:19.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1a362d6b8c926b6a282ee0df341cdf20d2c93655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 15:25:27.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"92f4267a9af18c4fb49b74282880538bd9804bf0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-06 15:36:50.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0afa83fea0aa268bc1945b4a683704cbda4423ec","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 15:45:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4f467bbc0665becc53c2410bf7ed18f2b44c3fb5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-06 15:49:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f1150e6264bbbb416266d51376078956de216a89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 16:01:50.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"6530844507b4c1c00fe6c5ccd959e85445c7243a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 16:01:56.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9e01f61ed6918dceb589c65e47cf8ba394de5cac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 16:50:28.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"1644bfe76a58a7c9f98f352ac989ba5e4fbaf7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 16:53:06.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"ea1e69b4f42b67d1acfb8f6b291881d80890304e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:53:36.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  expected a list for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027openstack-tox-py39\u0027][\u0027branches\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      queue: kayobe\n      templates:\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n  ...\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":15},{"id":"3e2ab34bd977d1d426dd7c9dba2ae4981d7d6aa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 16:54:44.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"5dbadfb5b38d3c97f156305dea1f7f3f62234239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:55:06.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  expected a list for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027openstack-tox-py39\u0027][\u0027branches\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      queue: kayobe\n      templates:\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n  ...\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":16},{"id":"4c2b7789a4ed545262abe947027e6e13eaada138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 16:59:42.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"52b335fc1d745b8a91b4fd4ed63b3701155dfa28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-06 17:03:04.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"bfb29ba2824a7d87e80af58ffd07d05e3c547532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 18:57:35.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b23735e7b0344e4b88ae20a19c17a1f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc7e5466efb3462589730dcc3a29f6b5 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5acac460cc0347329e8c0e409d81eced : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e353a5c0bde44d19bc10b5a1be3ae503 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7783f90a51824f578646ee9fdc8e50d6 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd1e18e3e2b4b7685afb9663c3d51b2 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9983b6afe39c4beca3d450c00e802ad3 : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/179e7918fed84cf1838c00745dd7bcdf : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a895ac9c1ec44586b79c9f88c288266d : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91dcaad0a18340ac9280ceafeaf6d1db : SUCCESS in 9m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/63f8459b161447e8ac6750b9c8ee3cfc : SUCCESS in 39m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5291859ba3e944ff89f18705bbc2c285 : SUCCESS in 35m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7be23344ae34afca0f08b7ba932b98e : SUCCESS in 31m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c71a6854f4fd4e40b5d48f615cdb1641 : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ca64cddc376e43d48cd69b361e3a5401 : SUCCESS in 14m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/80f491d0c9354266bfaf70c744d04724 : SUCCESS in 17m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17e29bce38e849cdb009d7d3955ad1ca : SUCCESS in 12m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b2f6304d7e544507bbdc9be0fedbe8fa : SUCCESS in 1h 53m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b6b84a10f7144928cb802e2c1fff680 : SUCCESS in 59m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2f79202c297c4af28820264e940d3fe5 : SUCCESS in 20m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc0782506df644ecbb1cb41b844bbbdc : SUCCESS in 11m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b68559a11f094bcb8dee255be2be511e : SUCCESS in 41m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ec19998134a44d72893c346aeed45332 : SUCCESS in 30m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bdd0c68771940b8a5d7e43e94b29d33 : SUCCESS in 19m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c45d2b2bc62b4d5ea9dc24eef669fb56 : SUCCESS in 10m 43s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e7f9098657724f5bb4e95f9d5ae808dd : SUCCESS in 16m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6506df5b14e0483eaba51193e5bcd325 : SUCCESS in 10m 19s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/89b4c8d2923c4ec8afde22e7724618fd : SUCCESS in 9m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/93f6b858af8b47758f41a04907cfbf60 : SUCCESS in 19m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/184c32ac91784d9fb46d8ced791985bf : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":18},{"id":"8fafe83e5d3c7d3ce6fbb92e6f2b0847b8d85cd6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-07 10:42:37.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"be947f6a5a85b02871801feea55affb2a3d448cf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-07 14:57:58.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"839193e4f40551ed747e3e80a2e90e958c50302a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-09 10:02:05.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"ce47272adbf87b42cb9840573e0fe4879eca13b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-13 14:33:05.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"dad8d905135667f2c0023f10ddb55444707e48c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 16:43:59.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991e3ac7218141b8b180e88a8e5fb66a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3c4ba00cedb04a86850552dcc30d0a3b : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d0f8123d9294be38c49d28c90001500 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dde1127177e4046b5f4c498f8ef08bb : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8de72bfc7748461e8efac280ccb495a3 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e117a9709cf48638c9c97bf43d39e1b : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889b86bf63c745ef8475651113e1fdf6 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/452c9a9a1c03416ba262f0d44180ee32 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0031c7a0cbf14964bbd42484018a11ad : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7450fc930ac4b56a1e057e4529b294f : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5cac80de3f14ecdbd854684e6eb61ab : SUCCESS in 8m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/69ba45fdb1ef492fb7741b8457851d2b : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d5a667896ee64a1087ee4a422e1e6f6e : SUCCESS in 1h 07m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6d39191317147d1b2f8a5b91edd88d0 : SUCCESS in 50m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ed33fae07b794b06b6323f0f20afb5a7 : SUCCESS in 1h 11m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/983678016ce544dc841fb0affe53be09 : SUCCESS in 21m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/131125d41fc94d8981205f2f4c1de0cc : SUCCESS in 14m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e3a3cbb86ab4a36891a318c19afdf41 : SUCCESS in 8m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ead729085f834a93aabcb3167c34d9d7 : SUCCESS in 2h 08m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fea529d9762543fb878b63be24d39ff2 : SUCCESS in 1h 45m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/82eb61e1127d4a07be59cfd0d96309f0 : SUCCESS in 19m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c2c94e37b7d4ebf8b2d5c76f0480331 : SUCCESS in 14m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/66065bac61ea447981cdc5ac3c65cbcc : SUCCESS in 38m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/43659ff4aea44e96bc647c0a23252292 : SUCCESS in 31m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b2ab8aa5ba8444287b70790c71abb02 : SUCCESS in 23m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c96923f14cdc41fbafa52630a023bb91 : SUCCESS in 18m 47s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a76637bcd0164462b0a20107e5743b37 : SUCCESS in 16m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15c495de906746f1b7102bdb197a9342 : SUCCESS in 10m 33s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/fc2f8aa33ba1425ca99118959219a0ed : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2ba0c61c8ea84fe7be47857f541f95f1 : SUCCESS in 18m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9506871a852435cba510f0671affb96 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":19},{"id":"a2baa47fad8a314769e7d845d760ace73927d1a7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-13 23:13:45.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"3a8fe42bd6584cda148578915ec5ad3b5f123779","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-15 20:47:11.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"bfedfcf636caed501bf012bef612fcd7dd8165f7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-15 20:48:52.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"4cc54e680d16bc88b7b53bb82e88649eb3bc5263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-16 08:31:41.000000000","message":"Patch Set 20: Published edit on patch set 19.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"3a79dee4ef4a301bff481ca240963a549087581a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 10:16:10.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/209a051870da4747a4a5f8d4d52bbce6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/897bd92665df44148bafa21d4d82f965 : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93ced6ae9ee64be88a426bb3365c318e : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188c3103c94d4e308fb5f1957efa2351 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff09b3cc42fc42189ace98742d0683bd : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cef6503546d4af899dd26884d85666e : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3428467fa44b0687219533d13e8182 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/976d1db41eab44ad8428440006d5cc77 : SUCCESS in 2m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e7feb6790114cd8a22dbb21581fe33b : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5556bcd18b6f4fc89c65b2fe654bee97 : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3877ed8e70914efcbb6b871a4169d4e7 : SUCCESS in 8m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ece5182421444fbb8c49d69b12f110f2 : SUCCESS in 39m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/90a5cf8fcda24136920371d77ddb6c00 : SUCCESS in 57m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/351cc004593d4022ad820c9f7cbc6654 : SUCCESS in 47m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/581631fd56064356b54dec39abf3e028 : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5c4a9479d4c64f4c89c61011f2c98184 : SUCCESS in 12m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4111979dc55d45a58b3ad9eecfe5309f : SUCCESS in 15m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3263b3a0b80d45e891dffc6cae7774e1 : SUCCESS in 10m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9c9cf2157e40485cb88a3b1f3fcc6904 : SUCCESS in 1h 09m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ca8ebee72504b41964e9304986e63b0 : SUCCESS in 1h 34m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d438dc1b7b894160986575d696bd4cda : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0e81e80c99b49ea98392d995898e0d5 : SUCCESS in 12m 29s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1485d2f93e4f4145890e986a0ffd4e31 : SUCCESS in 25m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f524933042eb41638d9d3ef6574e5e54 : SUCCESS in 33m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f0ba92819fc43cbbbbc919760498988 : SUCCESS in 24m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/51621afdf35340d4b741d6f8a283fe17 : SUCCESS in 12m 24s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d721bdd79b1649e88f01aabb0f19b263 : SUCCESS in 14m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaeaa8073c2745ec85e42ff143a05842 : SUCCESS in 10m 47s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/da53ec8a9da2464e948965b83de05084 : SUCCESS in 10m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/081d706b4e1c4ee4ae58f6c586c81281 : SUCCESS in 13m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de760cf718094844bf3a8b7de578fd2b : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":20},{"id":"94a28a5aef5e2e4581a1e315a38ea04af0585d98","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-26 09:45:34.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"629a3e24040d15fac5890890f7482e0d33641572","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-26 09:45:42.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"0788a1331df8d1af25cdf5e583a6f01720387cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-24 16:14:28.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"cb2da4e4dbf6b54f1a50ca7308baf1cd82507aea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-24 17:22:26.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"e5f1a7d711b091d21b4fd737fc37136b02b65a73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:13:39.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6081f0ef9c444f0fab2fcb2663ed1ce3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7541b256ba734ef48047aadca6c3823b : SUCCESS in 1m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/738804d1f3c547e59b45240831ab3883 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0e6ae2496e4c298e198b62b241e497 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/508be14dc18a471495db212e5abb37c2 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8b9649ae8b4351b3f224703a43f0cf : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a614faa915b747d18908c196bd15de99 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9678b79b6b7e41fbab2c22598b5094f4 : SUCCESS in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/62b3678883754864a02f0a7a3bea5b1a : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c618ad0b3d0847a3b4c2779b7f4f4e7c : SUCCESS in 7m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/efa9c4cd5ea642d0a3c959c12c3d27b4 : SUCCESS in 57m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3d89b03b8f834b92ad5e2e891f81b94b : SUCCESS in 50m 41s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a42b91716b2b40ffa0c7783f5f82c023 : SUCCESS in 50m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/847bb75ff5ae4b0892cc8a562460f89d : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ee6675db1abf48dfad8cba5b14947628 : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/020f13b10c734a0592e8be2596489f6a : SUCCESS in 18m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94ac9e895a884152b8443b1a51343376 : SUCCESS in 12m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5293109b71dd4c419d7cf7692b82bef4 : SUCCESS in 1h 44m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea64b225415248c7b9c6c30c0233479a : SUCCESS in 1h 26m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7420f4ac6fe14701a58410f849078603 : SUCCESS in 20m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c097c021146495b9b14af3dfd35552c : SUCCESS in 13m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ce4c2f3d5e28422d9599fe2dacf50e8e : SUCCESS in 36m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/200a68b690c54e909e8b7bbffe6787b6 : SUCCESS in 31m 50s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0df310cf2464eee890ce25edc7456b2 : SUCCESS in 27m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0e9bcb1e444443e083341ece640f609c : SUCCESS in 13m 48s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b6c5b4d82278481387b825026196edd0 : SUCCESS in 17m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cfbb50e03ef4e5f915774eec1187e3f : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f58115213e9d4b088afab3affdc368b8 : SUCCESS in 10m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e3f7ee81c5d47e382a170fea4664fdb : SUCCESS in 21m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad4f2bb5f22d4fd88c630c2b5ec5c39a : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":22},{"id":"499d066365f49b9e38d86062c513e999867b6480","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-24 21:49:38.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"7b34954f19a2bc985eccae8d23a64e7a4f795d7d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-28 15:46:58.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"3a82f8cbc14b33b5012d5e5d58d83555402c63b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-30 16:46:15.000000000","message":"Patch Set 23: Published edit on patch set 22.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"ca38cdd1c7b1afd6afcdfa805dab8e51958850f7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-30 16:47:14.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"55a5ef86f8b499cca3aee53c59d7901a9270c2e8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-30 16:48:28.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"c4628fc39f7e13891d15b7b591c19a3e310ebcff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 18:42:16.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe9371e3ee984b0da8b87465cab9ffa2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3fb7e21c0a8a42fabbfc51c5ed0d0063 : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2afa383d7795497faea0d7f821a3d31f : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640b4e616d99450b892f2e9a5c67bf6e : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c00f1f384864455ebaebe804af911136 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8afec3b73d406c9ff5608b5d7bb31c : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01cf3fc972524b5da224073d6b871533 : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01154cde8bd942faa99858e41e8a7f66 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9572568abb07428e98ea16f8c6a94d99 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd6a80cc681c4c848846203f0814d976 : SUCCESS in 10m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a086f48878224fafb0a3a0526918a342 : FAILURE in 14m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c9e6324ead504097bb04a1dda07d3f14 : FAILURE in 14m 54s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8520010d588642dd8e249ebfb9d61966 : FAILURE in 12m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/10bef294b768435999c31a734ee5bf54 : FAILURE in 11m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f0ae996f4d6c4f809e210fd56978d7a5 : SUCCESS in 18m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/da883132fd4a43a7a646fee412660031 : SUCCESS in 16m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbeaa399f0754f03ae52162359b9ae56 : SUCCESS in 17m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/09701526ee9241aa8020239abfbc6a6c : FAILURE in 48m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39642ab1a650413b85b58dc7e81fa30d : SUCCESS in 1h 50m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3c01546aa5924dee88ec84a800d07eb2 : SUCCESS in 19m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec224ea7d2b5494490da79195de66eb8 : SUCCESS in 14m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/26c4281e483b4f54ab325135da2f80bd : SUCCESS in 38m 05s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b59e9580f13d430db2be6a4b54de3b7e : SUCCESS in 34m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dd7635493b74d269b6cb05d0c209796 : SUCCESS in 24m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5d9b65a375a34d16b3806218cdc027d4 : SUCCESS in 18m 32s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b0ac01554dad4bacb1759e88d1de16dd : SUCCESS in 16m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19dbb958a40e4f818378bb43f4384da1 : SUCCESS in 12m 21s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0c02d312da3249539fdac5c7627a1c96 : SUCCESS in 10m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/afc649769dc4477ead8cef6a558f1caa : SUCCESS in 18m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f89a8b7b56124b7c9ab01eea6437fb60 : SUCCESS in 17m 09s","accounts_in_message":[],"_revision_number":23},{"id":"a0c4f00b4562873bd28fd521d59b1630c3df03e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 08:56:53.000000000","message":"Patch Set 24: Patch Set 23 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":24},{"id":"0b5a38a799a82e042c497f3bb19ab0b15e117558","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 09:48:31.000000000","message":"Patch Set 24: Review-Priority+1","accounts_in_message":[],"_revision_number":24},{"id":"869d5e503240643eb77aa29caf13f0e5603e6ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 10:19:35.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80ce4ff4a08f4e71a67e864d8bf256f6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e628c8c3e33e49f9bd4616d2227add77 : SUCCESS in 1m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f265f2b49569484ab8b729e90454b3a5 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b70e44c1e3479fac15939f9090bb22 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b132c7be9bfc49ac91936c9397405ef8 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34febf292b8f40038101d6f2a73d579b : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff13feafa1f94380a0eb3b03fb2a27f1 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0aa841fab0d43dbb7a970c0fadfff83 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31eae4f29aad488eb76610cabc6c5619 : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/708b196e48b140edac55c756d2a364ca : SUCCESS in 5m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8cf38ed174a14c38a52bea9f539a0c96 : FAILURE in 52m 31s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8a840ccdd19e4e049bb26716bb6a0670 : FAILURE in 49m 31s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ab06988e36b4f49864ff9ddb711dcd2 : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a9dfebc6f7bc4adf9b5013149bfbc29e : FAILURE in 56m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d2d0107b81e241709c2920a08a027061 : SUCCESS in 13m 53s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/050731d09cc0441c805499d9c7e5330b : SUCCESS in 15m 52s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/349d12869b114251b3538d8b1cd41890 : SUCCESS in 17m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a6b208058cbb41208263410226543364 : FAILURE in 50m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40436a0d78ff4505b2789cc98e0e4ed2 : SUCCESS in 59m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a200eea3a12e451888d5515dee27cc2b : SUCCESS in 23m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b85e63534d4498e835769c74c2f55ba : SUCCESS in 18m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c17517d89ee74fefa2e021a1f4e3e851 : SUCCESS in 38m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1c40e4ba81374d7ba0c9e2f1ae505d0f : SUCCESS in 35m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9b0eded96354814b4d98e5d6521eef7 : SUCCESS in 27m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eb890f1164ce47bfa67614fa15ab84d6 : SUCCESS in 10m 19s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4f46eb7e140f4b869474cd326e6fd9a2 : SUCCESS in 15m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/388e3504deba441c9387048d659a794c : SUCCESS in 11m 18s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/68da1225ddfb4f5abf54326f81f631d3 : SUCCESS in 7m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1ff7aa38eeff4e69b61aec2fe458d96f : SUCCESS in 19m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3294662697fc46a5a3bc910ec27fbc37 : SUCCESS in 15m 59s","accounts_in_message":[],"_revision_number":24},{"id":"b84fc997376730dc4d2c5c1630b7dc25903cad30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 12:43:53.000000000","message":"Patch Set 25: Published edit on patch set 24.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"a5f2cb2d64d969953b4ecf5cb9acf313a823cb8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 14:43:05.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37cf13a7094f471eac6dc7b0bd16e6dc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fdd3c8f2f8ea4d979ae8bf35b7f53c08 : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eef07714811f4a1c9e4de6dfa2e6443d : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c81a02041654b8ca9fded765034e282 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76c6e543a68d4bd88fd3ff975cad1ac2 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8957e02fcf04151ae842963d9c1a4b3 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f05a18dbac0344329976a1faf5adfeba : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8b621bcc3f44eab9ffaa4cf676d2bf4 : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a482cff7872b426f9b695daeab79eb04 : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75a48c32d6134316bf9247fba3de3ec5 : SUCCESS in 8m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b5da5d68bea34d08a3f660c0d9d980ba : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1b3076618e1c48efad0341fdb2a141b6 : SUCCESS in 1h 05m 42s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a95b69f24234bed8ee40e35da7f8bb0 : SUCCESS in 56m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2aea9620de8c425684254b160dbd86e3 : SUCCESS in 1h 12m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2cbf3bb145804c1c99c329f89336c367 : SUCCESS in 19m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/54cea817165e4a25afcca62a0dcae5cf : SUCCESS in 20m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d65772ac1f454fbbbdbefdf2ec17ff1f : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c65a5dd400fa40d0954a323d5849c52b : SUCCESS in 1h 18m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac1bef9fc3e04517a93e7f79b205c8d2 : SUCCESS in 1h 46m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b727a931e47e4611ba5c7a315e99365f : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/115f58225e1a4d88a4c27b19b8a3726b : SUCCESS in 16m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ffac069f8e3f4037b42f349340bc1282 : SUCCESS in 45m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cce9e3e666e64484aa52c1b1ef11d1a2 : SUCCESS in 32m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b7d10a14d874c4db502d53518fad9f3 : SUCCESS in 25m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d13b5a262661498594a07cb312d91375 : SUCCESS in 11m 11s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/6624dc768c9a49d490f3072792fcacec : SUCCESS in 16m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c7a40661e4b4086aa641041ad38ad37 : SUCCESS in 10m 36s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0edaca82ae48474e9305a0eebdd53af2 : SUCCESS in 13m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/83d6916ee31644c3908fe6ec5cf9d3a8 : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23e80e15825541d3a90db194ff30cc58 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":25},{"id":"9e4450af2b49ec9dfe4c85a452261993105cc5d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-05 09:05:47.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"49ba707d3fc394e9b10722c6dbfcd1717c2ee1bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-05 09:11:31.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":27},{"id":"5795460e31b5905e28e3412b2e43dbee8384e511","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-05 09:20:41.000000000","message":"Uploaded patch set 28.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":28},{"id":"c9f17796bd55f70ba098328aef6017850e9470cc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-05 09:56:13.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"a426b67f26cc12005fe9b56e0868cbca6e6f062d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-05 10:21:18.000000000","message":"Uploaded patch set 29.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":29},{"id":"aa85fb255c80c69c3ac522582c25febfd61c8f4a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-05 10:47:18.000000000","message":"Patch Set 29: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":29},{"id":"a46b3ef51bcc207f06f9427a766a04a62b8fc984","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-05 11:20:00.000000000","message":"Patch Set 29: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"625756ba98a63b136c006152dfff4a1815a56b23","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-05 11:27:19.000000000","message":"Patch Set 29: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"9652c46fbffe57aab669f6608473c25f9c387142","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-05 11:31:15.000000000","message":"Patch Set 29: Code-Review+1","accounts_in_message":[],"_revision_number":29},{"id":"0db99fa4013fbef17668d77c9b000b3626c1d8eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 12:22:38.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2fa3836585f406182b606983cb3fff9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/859fa99c40dd4f0591dd08eacd99baab : SUCCESS in 1m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cd9083104f040028e1dbc283ea4a63e : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7d4d83bb534dfab807d6cfabbf191b : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45c77822d3934fcbaf4a5b4f8b1d957b : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f034e9a506b14c769d02a9bb971522ec : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47fb3c7842604ad0adce0aab75e196e2 : SUCCESS in 2m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/38d908c3e3d845eab0616a9e483bd5bb : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da2df1d2772f45599db0e57cb9a96ace : SUCCESS in 7m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa79c8571ab14b18a92ba212c6701d5f : SUCCESS in 7m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2435129b329a48c7b26250b269c6fdfd : SUCCESS in 59m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1c7bc548ea96463fb0a9bd059ded6210 : SUCCESS in 59m 54s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3d7d533eb064080827ec144c4085fd4 : SUCCESS in 54m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c763b13765e84728b3d1c5b5155382d2 : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/21f497b50e7a4e76a73a31d2acab5253 : SUCCESS in 18m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f9bd3062c8b54a61a1e4f41c747849d7 : SUCCESS in 15m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49f54ba3c3814bc4b08f77d96b4de62a : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6b2c414dbb974c42be617905fd2bfc22 : SUCCESS in 1h 55m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/428e1be5f76b47768faae06ad41cb76e : SUCCESS in 1h 39m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a92a3cdbf72847d8aa722397237e2be0 : SUCCESS in 19m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40b2c00339834954ac72ca0f1ee33224 : SUCCESS in 15m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2193b25a98b842a69ac08c214d093b26 : SUCCESS in 39m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/993ef0451e8e4200ad39456c68dfe1b8 : SUCCESS in 20m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f304ca079a346ad8a5993bf090ed498 : SUCCESS in 27m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/42f90355b994491dbc0fddbf95afe33d : SUCCESS in 14m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/50965c9de34d40148d03e8321b62018c : SUCCESS in 11m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30b8b4a12a0c44159441d751cf96a2ac : SUCCESS in 11m 30s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c8c2df3f31524427a35d9846e00e3754 : SUCCESS in 10m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a854932201494eef934a3d8ecb8a2861 : SUCCESS in 21m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ca756122ae74fa7ba04d810cb951805 : SUCCESS in 15m 32s","accounts_in_message":[],"_revision_number":29},{"id":"15668a86950a8c47d4628638b0c90ede81532cd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 12:23:10.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"9ae86e2c0ce464acdb83c269f5632c26bff4eb49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 14:26:56.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffb589d1dd2c4fb1a3b0c113ec33b464\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/84cb6d6eea9044488667de2968ea5770 : SUCCESS in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fea7096ca6401aa176511229951a96 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2856b46887ed4c7da728af0800ac37ff : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8afa1bc3e4e24d989d8c422a9dd8bd66 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/005c7aee12584e8daaab16a58ae8c694 : SUCCESS in 2m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a4cdd2dc53d44ccb1e048454b298588 : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8b4215c8bff4a3187add8fe3fc564e7 : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5044ab4996d24ea1987b991aa0344dd5 : SUCCESS in 10m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5b37971e036744deb1fe675df6456d42 : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d122c88d1cfe400589c8e82bc349ecb8 : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e7d5dae7c630472e8d4f6afcba11b547 : SUCCESS in 1h 15m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a03d68f7756f4af0a6e484c583a062b8 : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f01ea5dc3d684c6f9dec09e74b70e0f8 : SUCCESS in 16m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d0bb4a8c05b14b5fafe84ca8b9273cca : SUCCESS in 1h 56m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9254cfb921f74b49b37c6da3a243484d : SUCCESS in 1h 50m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/13e690489a9f4ab28e5618e80d5aa7f4 : SUCCESS in 18m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2dd0c60fdee4365ac4ea2b7d4f66b6e : SUCCESS in 18m 33s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d76dcd08aa2840b5927966454648d533 : SUCCESS in 34m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c41f4363d59f43d0a8cc6c5f5d2ddae7 : SUCCESS in 22m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8c8a8da45c4d4e1a847902b3bb7f07db : SUCCESS in 15m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa25a4959ea342dbaf5f5ea2a84e8f21 : SUCCESS in 10m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/352c3ef4c2c94e64a3156a754e9f6c90 : SUCCESS in 19m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f47426a1896a430395d6f0a1ecb20c04 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":29},{"id":"1ef6c24e50c6305916d7bad0841d3c582e1eabd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 14:26:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":29},{"id":"c48f30cabbdce19605ab5f82a3dcf8016d98099a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 14:29:21.000000000","message":"Patch Set 29:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/006c0d8cb4bc4166967bcd214f9acefb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0956cd1793a04e2b8a44dbe792fcd293 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a725423804194ed388d566bc7e8f8fb8 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"f5d026badba96b4ecc597dcec2189fe3cb9220cf","revisions":{"d847edb5e80b24726fd114d2455c30ccb2d2b629":{"kind":"REWORK","_number":1,"created":"2024-08-20 13:10:14.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/1"}}},"commit":{"parents":[{"commit":"c726492e495f1c4f4ded8515a3bd3d0fcc9b383e","subject":"Merge \"Stop replacing dashes with underscores for Kolla interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c726492e495f1c4f4ded8515a3bd3d0fcc9b383e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:09:58.000000000","tz":60},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d847edb5e80b24726fd114d2455c30ccb2d2b629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d847edb5e80b24726fd114d2455c30ccb2d2b629"}]},"branch":"refs/heads/master"},"a321590ad747052bc0368dcf6cb560ea88bf820e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-20 13:24:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/2"}}},"commit":{"parents":[{"commit":"fcfb9ba332ee31c3a4e3b57fa2f0d78704e3a4da","subject":"Bump Ansible version installed in Kolla Ansible virtalenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcfb9ba332ee31c3a4e3b57fa2f0d78704e3a4da"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:23:55.000000000","tz":60},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a321590ad747052bc0368dcf6cb560ea88bf820e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a321590ad747052bc0368dcf6cb560ea88bf820e"}]},"branch":"refs/heads/master"},"97b2e54e7c2c3583acbce21d8e8192192c0e402e":{"kind":"REWORK","_number":3,"created":"2024-08-20 13:31:06.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/3"}}},"commit":{"parents":[{"commit":"fcfb9ba332ee31c3a4e3b57fa2f0d78704e3a4da","subject":"Bump Ansible version installed in Kolla Ansible virtalenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcfb9ba332ee31c3a4e3b57fa2f0d78704e3a4da"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:30:54.000000000","tz":60},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97b2e54e7c2c3583acbce21d8e8192192c0e402e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97b2e54e7c2c3583acbce21d8e8192192c0e402e"}]},"branch":"refs/heads/master"},"82e54ca6536220df62fbd1cca55489334cd36e48":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-20 14:44:18.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/4"}}},"commit":{"parents":[{"commit":"9d2e83095d00f4fd2a48e76a623ba5dbfae82827","subject":"Bump Ansible version installed in Kolla Ansible virtalenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d2e83095d00f4fd2a48e76a623ba5dbfae82827"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 14:44:18.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82e54ca6536220df62fbd1cca55489334cd36e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82e54ca6536220df62fbd1cca55489334cd36e48"}]},"branch":"refs/heads/master","description":"Rebase"},"463f48859144f2c33d680975daa31c658a873eaf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-08-20 17:08:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/5"}}},"commit":{"parents":[{"commit":"d38838b4a3a82d68eb8e630c20b247ccb67b05ca","subject":"Bump Ansible version installed in Kolla Ansible virtalenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d38838b4a3a82d68eb8e630c20b247ccb67b05ca"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 17:08:59.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/463f48859144f2c33d680975daa31c658a873eaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/463f48859144f2c33d680975daa31c658a873eaf"}]},"branch":"refs/heads/master","description":"Rebase"},"6e1a9526353ce63c6162bbc1454fcb6de532b259":{"kind":"REWORK","_number":6,"created":"2024-08-20 17:15:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/6"}}},"commit":{"parents":[{"commit":"d38838b4a3a82d68eb8e630c20b247ccb67b05ca","subject":"Bump Ansible version installed in Kolla Ansible virtalenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d38838b4a3a82d68eb8e630c20b247ccb67b05ca"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 17:15:47.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e1a9526353ce63c6162bbc1454fcb6de532b259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e1a9526353ce63c6162bbc1454fcb6de532b259"}]},"branch":"refs/heads/master"},"8c71307c4b5598e1cdc6eba274f1d861efe52934":{"kind":"NO_CHANGE","_number":7,"created":"2024-08-20 17:46:40.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/7"}}},"commit":{"parents":[{"commit":"dec43999f72fb29484388aee2ef0b25762c60769","subject":"Bump Ansible version installed in Kolla Ansible virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dec43999f72fb29484388aee2ef0b25762c60769"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 17:46:40.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c71307c4b5598e1cdc6eba274f1d861efe52934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c71307c4b5598e1cdc6eba274f1d861efe52934"}]},"branch":"refs/heads/master","description":"Rebase"},"469a9e9dac3ccdf91213a402f945088b7a9de6e8":{"kind":"REWORK","_number":8,"created":"2024-08-21 11:31:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/8"}}},"commit":{"parents":[{"commit":"dec43999f72fb29484388aee2ef0b25762c60769","subject":"Bump Ansible version installed in Kolla Ansible virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dec43999f72fb29484388aee2ef0b25762c60769"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-21 11:06:32.000000000","tz":60},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/469a9e9dac3ccdf91213a402f945088b7a9de6e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/469a9e9dac3ccdf91213a402f945088b7a9de6e8"}]},"branch":"refs/heads/master"},"2157f73ed8a7dd2c8c963733f789d61924dc233c":{"kind":"REWORK","_number":9,"created":"2024-08-21 16:02:12.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/9"}}},"commit":{"parents":[{"commit":"dec43999f72fb29484388aee2ef0b25762c60769","subject":"Bump Ansible version installed in Kolla Ansible virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dec43999f72fb29484388aee2ef0b25762c60769"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-21 16:02:01.000000000","tz":60},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2157f73ed8a7dd2c8c963733f789d61924dc233c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2157f73ed8a7dd2c8c963733f789d61924dc233c"}]},"branch":"refs/heads/master"},"5ab4844938b7c89ea9198e8a23e08b4b3ef76b3f":{"kind":"REWORK","_number":10,"created":"2024-09-06 08:32:58.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/39/926639/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/10"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-06 08:32:58.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ab4844938b7c89ea9198e8a23e08b4b3ef76b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ab4844938b7c89ea9198e8a23e08b4b3ef76b3f"}]},"branch":"refs/heads/master","description":"Rebase"},"c2ab92d9ecc917a42216df223cc7ef6865c76ac5":{"kind":"REWORK","_number":11,"created":"2024-09-06 08:37:10.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/39/926639/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/11"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-06 08:37:10.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2ab92d9ecc917a42216df223cc7ef6865c76ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2ab92d9ecc917a42216df223cc7ef6865c76ac5"}]},"branch":"refs/heads/master"},"c6acad5447a1f769182425956754fe73c3336077":{"kind":"REWORK","_number":12,"created":"2024-09-06 15:25:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/12"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-06 15:25:26.000000000","tz":60},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6acad5447a1f769182425956754fe73c3336077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6acad5447a1f769182425956754fe73c3336077"}]},"branch":"refs/heads/master"},"26fc4db3b0cb0726cbdb363aaed3c04aa5c0e0fa":{"kind":"REWORK","_number":13,"created":"2024-09-06 16:01:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/13"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-06 16:01:50.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26fc4db3b0cb0726cbdb363aaed3c04aa5c0e0fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26fc4db3b0cb0726cbdb363aaed3c04aa5c0e0fa"}]},"branch":"refs/heads/master"},"88586a0f335c259107746632ed529ad03083e2a1":{"kind":"REWORK","_number":14,"created":"2024-09-06 16:50:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/14"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-06 16:50:28.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88586a0f335c259107746632ed529ad03083e2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88586a0f335c259107746632ed529ad03083e2a1"}]},"branch":"refs/heads/master"},"89c5205e51d386d80b8fb08f3c522dafd4ee2c13":{"kind":"REWORK","_number":15,"created":"2024-09-06 16:53:06.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/15"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-06 16:53:06.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89c5205e51d386d80b8fb08f3c522dafd4ee2c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89c5205e51d386d80b8fb08f3c522dafd4ee2c13"}]},"branch":"refs/heads/master"},"61ca2f039c0a9320978bc2a211ea51dc0b11016c":{"kind":"REWORK","_number":16,"created":"2024-09-06 16:54:44.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/16"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-06 16:54:44.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61ca2f039c0a9320978bc2a211ea51dc0b11016c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61ca2f039c0a9320978bc2a211ea51dc0b11016c"}]},"branch":"refs/heads/master"},"c6c28c654436c0cf7173bafe7c69e80bb3706331":{"kind":"REWORK","_number":17,"created":"2024-09-06 16:59:42.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/17"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-06 16:59:42.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6c28c654436c0cf7173bafe7c69e80bb3706331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6c28c654436c0cf7173bafe7c69e80bb3706331"}]},"branch":"refs/heads/master"},"280ed0bb2903c0fddd2541a027c5638b105da528":{"kind":"REWORK","_number":18,"created":"2024-09-06 17:03:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/18"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-06 17:03:04.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/280ed0bb2903c0fddd2541a027c5638b105da528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/280ed0bb2903c0fddd2541a027c5638b105da528"}]},"branch":"refs/heads/master"},"0563092b4e0089268a936078672e25486c4a4519":{"kind":"REWORK","_number":19,"created":"2024-09-13 14:33:05.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/19"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-13 14:33:04.000000000","tz":60},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0563092b4e0089268a936078672e25486c4a4519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0563092b4e0089268a936078672e25486c4a4519"}]},"branch":"refs/heads/master"},"dd0c08697bb29c29f8ef7f7bd28bc1b6b1b2760b":{"kind":"REWORK","_number":20,"created":"2024-09-16 08:31:41.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/20"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-16 08:31:41.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd0c08697bb29c29f8ef7f7bd28bc1b6b1b2760b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd0c08697bb29c29f8ef7f7bd28bc1b6b1b2760b"}]},"branch":"refs/heads/master"},"6e7c5babced01454e748c627141e7cac428640d7":{"kind":"REWORK","_number":21,"created":"2024-10-24 16:14:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/21"}}},"commit":{"parents":[{"commit":"28d78297af96c6ea7c138cddd9a47fc72b387afa","subject":"Merge \"Support setting whether an Apt repo is trusted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28d78297af96c6ea7c138cddd9a47fc72b387afa"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-24 16:14:27.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e7c5babced01454e748c627141e7cac428640d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e7c5babced01454e748c627141e7cac428640d7"}]},"branch":"refs/heads/master"},"478347f9dc243aa2565bfbde98fb07d26caa04bf":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2024-10-24 17:22:26.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/22"}}},"commit":{"parents":[{"commit":"9b53c0ee95c02d5d945c7f690f77ce53bab95145","subject":"Merge \"Fix group and mode of /var/log/journal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b53c0ee95c02d5d945c7f690f77ce53bab95145"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-24 16:19:11.000000000","tz":60},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/478347f9dc243aa2565bfbde98fb07d26caa04bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/478347f9dc243aa2565bfbde98fb07d26caa04bf"}]},"branch":"refs/heads/master"},"08c9b0a920cc7f75f8a49442c68aaed1bcf32993":{"kind":"REWORK","_number":23,"created":"2024-10-30 16:46:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/23"}}},"commit":{"parents":[{"commit":"9b53c0ee95c02d5d945c7f690f77ce53bab95145","subject":"Merge \"Fix group and mode of /var/log/journal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b53c0ee95c02d5d945c7f690f77ce53bab95145"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-30 16:46:15.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08c9b0a920cc7f75f8a49442c68aaed1bcf32993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08c9b0a920cc7f75f8a49442c68aaed1bcf32993"}]},"branch":"refs/heads/master"},"eabdfbc180b3bff88a9ab5d801a2b50937e60cb8":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2024-10-31 08:56:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/926639/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/24"}}},"commit":{"parents":[{"commit":"ecb883c3bb540a5c18b62b0267a411f64d63f21d","subject":"Merge \"CI: Skip test_ntp_clock_synchronized on Ubuntu 24.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecb883c3bb540a5c18b62b0267a411f64d63f21d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-31 08:56:53.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eabdfbc180b3bff88a9ab5d801a2b50937e60cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eabdfbc180b3bff88a9ab5d801a2b50937e60cb8"}]},"branch":"refs/heads/master","description":"Rebase"},"95724ec11fdbf5ca32e0269d88e2fc7f1429e37a":{"kind":"REWORK","_number":25,"created":"2024-10-31 12:43:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/926639/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/25"}}},"commit":{"parents":[{"commit":"ecb883c3bb540a5c18b62b0267a411f64d63f21d","subject":"Merge \"CI: Skip test_ntp_clock_synchronized on Ubuntu 24.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecb883c3bb540a5c18b62b0267a411f64d63f21d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-31 12:43:53.000000000","tz":0},"subject":"Aligns Kayobe Ansible version with Kolla Ansible","message":"Aligns Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95724ec11fdbf5ca32e0269d88e2fc7f1429e37a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95724ec11fdbf5ca32e0269d88e2fc7f1429e37a"}]},"branch":"refs/heads/master"},"b58cf8cc35ab6531a3d01dd493c082e04ceeb3ae":{"kind":"REWORK","_number":26,"created":"2024-11-05 09:05:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/926639/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/26"}}},"commit":{"parents":[{"commit":"25d34ad3656a1c9f57b9d9815859c11010aa9173","subject":"Merge \"[release] Use OpenStack 2024.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25d34ad3656a1c9f57b9d9815859c11010aa9173"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-11-05 08:23:14.000000000","tz":60},"subject":"Align Kayobe Ansible version with Kolla Ansible","message":"Align Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b58cf8cc35ab6531a3d01dd493c082e04ceeb3ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b58cf8cc35ab6531a3d01dd493c082e04ceeb3ae"}]},"branch":"refs/heads/master"},"8deebe530cbf628ff1f3f772946c0c478a14bbba":{"kind":"REWORK","_number":27,"created":"2024-11-05 09:11:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/926639/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/27"}}},"commit":{"parents":[{"commit":"25d34ad3656a1c9f57b9d9815859c11010aa9173","subject":"Merge \"[release] Use OpenStack 2024.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25d34ad3656a1c9f57b9d9815859c11010aa9173"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-11-05 09:11:18.000000000","tz":60},"subject":"Align Kayobe Ansible version with Kolla Ansible","message":"Align Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8deebe530cbf628ff1f3f772946c0c478a14bbba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8deebe530cbf628ff1f3f772946c0c478a14bbba"}]},"branch":"refs/heads/master"},"47b8f81848a9c842cdca0e359c3c619d6f7e1e94":{"kind":"REWORK","_number":28,"created":"2024-11-05 09:20:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/926639/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/28"}}},"commit":{"parents":[{"commit":"25d34ad3656a1c9f57b9d9815859c11010aa9173","subject":"Merge \"[release] Use OpenStack 2024.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25d34ad3656a1c9f57b9d9815859c11010aa9173"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-11-05 09:20:36.000000000","tz":60},"subject":"Align Kayobe Ansible version with Kolla Ansible","message":"Align Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47b8f81848a9c842cdca0e359c3c619d6f7e1e94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47b8f81848a9c842cdca0e359c3c619d6f7e1e94"}]},"branch":"refs/heads/master"},"f5d026badba96b4ecc597dcec2189fe3cb9220cf":{"kind":"REWORK","_number":29,"created":"2024-11-05 10:21:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/926639/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/926639/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/926639/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/926639/29"}}},"commit":{"parents":[{"commit":"25d34ad3656a1c9f57b9d9815859c11010aa9173","subject":"Merge \"[release] Use OpenStack 2024.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25d34ad3656a1c9f57b9d9815859c11010aa9173"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-08-20 13:01:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-11-05 10:20:47.000000000","tz":60},"subject":"Align Kayobe Ansible version with Kolla Ansible","message":"Align Kayobe Ansible version with Kolla Ansible\n\nThis change bumps up the maximum supported Ansible version to 10.x\n(ansible-core 2.17.x) and minimum to 9.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Idfbf0eed0b2333d550f822e40a0d96ae227f6159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5d026badba96b4ecc597dcec2189fe3cb9220cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5d026badba96b4ecc597dcec2189fe3cb9220cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
