)]}'
{"id":"openstack%2Fkayobe~927002","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Ie15019f91b9e7b7f8cf97b628886f6b0beeb24fa","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-08-30 17:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie15019f91b9e7b7f8cf97b628886f6b0beeb24fa","subject":"CI: Add openstack-tox-pep8 job, fix pep8 failures","status":"MERGED","created":"2024-08-23 09:34:12.000000000","updated":"2024-08-30 17:24:28.000000000","submitted":"2024-08-30 17:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927002","meta_rev_id":"d9c1b9490e800f4ec748f5ad67791129d6d85af7","_number":927002,"virtual_id_number":927002,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 17:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-08-30 15:06:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-23 17:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-08-30 15:06:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-23 09:34:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2024-08-23 11:55:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 12:12:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2024-08-27 12:12:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2024-08-27 12:12:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-08-27 12:12:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-08-27 12:12:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-08-27 12:12:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"2871fb93274efd0cf6f17d65c987bd4263522af9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 09:34:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4222468e71ba1e6a292cd4f9d68711d66bb02aa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 11:55:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5c1d786c9aa4e3089bcab8a5eed8b64\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ad89590213542c7a1df9761b6188715 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa2367fbb234a868485d3a5e170fb91 : SUCCESS in 7m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65c345bd03dc498bab39a126929dbf68 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc9de266d5634d85ae21a282baf0397d : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e70a1e66151e42a8be91e1ba9d422971 : SUCCESS in 9m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a8f2438ff419421eb0bb399f09ab6e1d : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/137e716070a346259207e14790b78c5d : SUCCESS in 55m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7374e4cf9c149a1adf5b157c97f4b4f : SUCCESS in 49m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/22a555e26ae7482db938470ca369f917 : SUCCESS in 1h 13m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1fbf9680a27e4c599c283776f12dafd3 : SUCCESS in 18m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0ff517a889e146b6be340ea31ae997e7 : SUCCESS in 17m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1db3376c1645454dbb5d51fed655e60e : SUCCESS in 13m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3bbd896483ad4b70a8770d786faa44cc : SUCCESS in 2h 14m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83f2338077cd43c782191199e38641eb : SUCCESS in 1h 23m 42s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4c3ad8dde69e46eb9042d4ae809f082f : FAILURE in 1h 21m 48s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42c49ca958ef4243b635a44cdd96c2e5 : FAILURE in 1h 00m 38s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/45dc94803ee54d7c915037a60bc8cd25 : SUCCESS in 19m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97d096c51de14ccd95466b361349ec32 : SUCCESS in 13m 47s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3eac60a643dd41c3b10885c82c79a396 : SUCCESS in 35m 27s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bbbb490e082a4eb0a0ef7c529ab3ccb6 : SUCCESS in 30m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5774f66ce9794307bed5ec61263f8524 : SUCCESS in 24m 49s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b498552e7fbe4c9498e83dc3dbe55ece : SUCCESS in 35m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0469bbd7f99f4e9db77a6e86a68a3884 : SUCCESS in 23m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e51f86c29d0a41099d3026cecae563da : SUCCESS in 15m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0ecacbb08344029aa9ff406788ce570 : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3261d28e1ff94cb8987667eefdde2006 : SUCCESS in 22m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4b915e5df3f4310bb612fce66e65c48 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9706cc395b66416298ced00745fc325f : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d876d82fe36d4e77a154d220c3a70d63 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed352e0549774717bb8d64f2a30339d8 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc4b1d3cd39d4ef1a9b47f2db2c27e1d : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"081446e341e5aae218623fb6d99ec040f3b6f34d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 17:10:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b68c9ce4755f1aac3fbca0e991e2296eab8b8c8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-30 15:06:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"98f58baec0cc19b10a35414992c90fbbc10d12af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:07:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e85c764dc0f24a0fe32fd0693584eda0a547437d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:20:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8456106910487093f652fa1b12ade0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37c57f970fe47c19c6b985bb8f7cf8f : SUCCESS in 7m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/abee7eeacf634cb9b4c0d573f28a610c : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8fdf7f1aa3245039ebd506fae9446c0 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f1c81cddb71418caee7173b5f3df473 : SUCCESS in 9m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c75664722ea04cbf9bf662af6ffdaf02 : SUCCESS in 1h 14m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b63c5183019c4a41851e8519ce8e66da : SUCCESS in 58m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ded870e449074174b66cfd3d0c895dc6 : SUCCESS in 1h 18m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8f4b7ad260354b09ad8756395f119edb : SUCCESS in 17m 52s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83192e8278b340a5bbefa278a387ba29 : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f79e9df34c134cef829de9a387f15aef : SUCCESS in 2h 07m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4d3ac36c1d64157bc7c792a51a68788 : SUCCESS in 1h 34m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1beb422925154275b331fa4c7e102b74 : SUCCESS in 19m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/682338015da34d4ca1ce4d911eae8a3d : SUCCESS in 15m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c06f3ee8a3c94f4785d67217292ef82d : SUCCESS in 29m 19s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/903ea7d46e66467cb02c8311440dc461 : SUCCESS in 21m 39s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e664693f709f4a92b538e11c81de9a61 : SUCCESS in 35m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f4b748e41e145839ca76f9f664e23dd : SUCCESS in 23m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8f7b67c69f6f46c1bb7e1d415dd95c82 : SUCCESS in 16m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/990ff65f44e0456dac66e3cd05375817 : SUCCESS in 9m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f0a80ae03801493e987b03da05198ee0 : SUCCESS in 19m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3310cac96de4263a436374ec4d6e609 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2dde69ffbf8e14a29e2365d74ed551616b1a1bb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9c1b9490e800f4ec748f5ad67791129d6d85af7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b621f6b14f284eaba0ebbe2dcc223b27\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05acf10fcdf14965a23b9ec172d59dc2 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"022b2be9fc160c0731a9c97fc6eeb66992431ee5","revisions":{"022b2be9fc160c0731a9c97fc6eeb66992431ee5":{"kind":"REWORK","_number":1,"created":"2024-08-23 09:34:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/927002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/927002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/927002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/927002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/927002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/927002/1"}}},"commit":{"parents":[{"commit":"ec1febba7a2d0300ace2cf9f39778422b744ed23","subject":"Merge \"Stop replacing dashes with underscores for Kolla interfaces\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec1febba7a2d0300ace2cf9f39778422b744ed23"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-08-20 10:23:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-23 09:31:12.000000000","tz":120},"subject":"CI: Add openstack-tox-pep8 job, fix pep8 failures","message":"CI: Add openstack-tox-pep8 job, fix pep8 failures\n\nSince I8f5338b7eee31f1b8bc3e783d10666e3092a6540 we have not been running\npep8 jobs in CI. This change fixes the issues introduced since then, and\nadds the job back.\n\nUnlike the original change, we are not using the openstack-python3-jobs\ntemplate since it would use py38 on this branch.\n\nChange-Id: Ie15019f91b9e7b7f8cf97b628886f6b0beeb24fa\n(cherry picked from commit 610433b11ba3819869a892489546d9f8b2b743c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/022b2be9fc160c0731a9c97fc6eeb66992431ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/022b2be9fc160c0731a9c97fc6eeb66992431ee5"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
