)]}'
{"id":"openstack%2Fkayobe~927015","triplet_id":"openstack%2Fkayobe~master~I7d71689ef9b5a8a092fe78da4ee3d2a645a1316e","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-08-30 18:04:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d71689ef9b5a8a092fe78da4ee3d2a645a1316e","subject":"seed-vm/infra-vms: Use efi as default","status":"MERGED","created":"2024-08-23 12:53:16.000000000","updated":"2024-08-30 18:08:47.000000000","submitted":"2024-08-30 18:04:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927015","meta_rev_id":"e2c508e20db9a6d3f835fdac09f162aa4864122d","_number":927015,"virtual_id_number":927015,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 18:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-08-30 15:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-08-27 09:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-08-30 15:14:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2024-08-27 09:13:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-23 17:29:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 09:11:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-08-30 15:14:08.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"f9ad01e24f6678d48086d00da67b1195177232c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 12:53:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5601fd8c6f37d0574d4d4c94c21e74ee1804f580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 15:10:36.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b28c2eec9030fe3ec1d304946498c139dc62b82c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 17:12:10.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e0eb9019c14a385188125e6a6e04596f4f7ad7b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:29:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ffdf1375432411d9ac198a9dbd4ee42\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a19825300ef84910ae4943214988bb0e : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a17cdf2d7c8f43358c4beff1333bc099 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58d09d9591c5422fbe9794afadca3ae0 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce1eba0591b34701863a6a13f5c24113 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/925a7b5d2909492ab274417e602493fc : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507ba0c20cdb4c9eb697e0c2a838fb1e : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57da9c88a3df49ae892e4b3742d02418 : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f30fe2d38b143f6b2b9ae5903d022be : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e56a065d85b44172aa5ca5d4444fb02d : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3bfcf752376c4667ac50b1bdbac744fa : SUCCESS in 8m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ed1761ba2aac420c83b53f37a74a4412 : SUCCESS in 1h 02m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f8d90b5533da4ee1a51428073b9af383 : SUCCESS in 53m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/685449adf00e4252a3e5cc037823463d : SUCCESS in 49m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3bc0d07cc0ff4909803d1b88aecdf8ab : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d88a8945ae5447a28b55e00c3513992c : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e2806874b7e94cc99bfd2008fc336d7e : SUCCESS in 15m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a82120e7ddd24b7db1fa66b57f0d4161 : SUCCESS in 13m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b3fe01ccebf94c318536ac9ffad1dadf : SUCCESS in 1h 51m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3db7f5159ad43f3afadd35ccb04cf06 : SUCCESS in 1h 28m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3ece43b066bd40c5957b46d51433d309 : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8de1dbbcc3d4428280b560f4b963dd3f : SUCCESS in 14m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0f1afff974a4474a805877a005af8a75 : SUCCESS in 40m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e0bc2b4c2584d91918d2c8350a29849 : SUCCESS in 33m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e70351ec4e114ef9b66babfe2dfbee05 : SUCCESS in 24m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2de584fe95e744b58f7956fe704ccb97 : SUCCESS in 16m 58s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e87655d6574647858becf505630aef02 : SUCCESS in 17m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c58c7063dde74e78ae9f406c2c4b2fbc : SUCCESS in 10m 26s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/aef64a5781554ff9947c52ff091499df : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c2d89d3c4ab4486b9b823a710212adba : SUCCESS in 20m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3292582ec916432ea1c89fec7ec29929 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":2},{"id":"753d1e1fc48d718d059c554f5e7fdda18fcaf73a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-27 09:11:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c991139833d07f267880fa589db6f1a232df9486","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-27 09:13:21.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"2c616d3c169cb9a5a801c66fcd9eaff0fd33cd2f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-30 15:14:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"577d2f405a301b5a445d7c384524130b02a8baa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:16:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08b6534a13e497b4b72b51a397fd2acd9e0d991a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:04:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10a3c99d2e0f49b9aa3ad05b616e0184\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e59a5f946ed2475b93feba0123326ff6 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbbdfdec2d954093aa1efbd9cd769908 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba9ec794e9b14a9a8e63213c4c3aa320 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ac0785165b4d07ad1e9f9082a1e03d : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03109d04b8b74c5d93befc2cf872a491 : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/22b709acf661417b8715bf3a735cd5ef : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1ba53c60f8c48bdb4d36cf96b81a7c6 : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/571fc1363b00472fb40bbd68a85ae7ad : SUCCESS in 11m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3dabad19b19c49fca2a381f719524929 : SUCCESS in 1h 14m 09s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6dcf03ff3f146929b5cb2c107645d63 : SUCCESS in 50m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/551d7d9d3a2f49cb8663f4ae03d7306b : SUCCESS in 1h 20m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6781dae96e4d4185bd2213abc5d947c3 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a02dcd663c884a0cbb51d6e5ad6f1605 : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/356653a3d418457bbe0bdf828cb2c7b8 : SUCCESS in 1h 56m 23s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b288214e27e4d3c85ae60ed39009175 : SUCCESS in 1h 51m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/76dfacb829154ef19c4c2eeb1b821204 : SUCCESS in 22m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eae203e629294538aa4d81faa72bc7be : SUCCESS in 17m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b4394feaeb3943b08ac60304c821e430 : SUCCESS in 36m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/028473d047eb46f69e99a1d32b269249 : SUCCESS in 22m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/93a8584cff8b473fb543fd4caac8467f : SUCCESS in 18m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed19ebc8917e41f6ab63388c69f619f1 : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b9f4855fa38f4b6bad296769fb1ae815 : SUCCESS in 19m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f5417731e64437eb2733e07a625d7ba : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6d2d070b191e6bdcd04d182f5f31e4b42f269398","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:04:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e2c508e20db9a6d3f835fdac09f162aa4864122d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:08:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db2a674982974ac8ac9c821abe4325b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5daa95a3571f444dbc156d2ea1a6514f : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb113934ee3a4caa8dd54d4f66aeb703 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49c33e0a73643da543f042a6354c10566f81d166","revisions":{"fc0a92873d21aff25080ef44fdef45d498dd041e":{"kind":"REWORK","_number":1,"created":"2024-08-23 12:53:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/927015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/927015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/927015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/927015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/927015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/927015/1"}}},"commit":{"parents":[{"commit":"045c04f97c4e63048f4b1f8559c8fbda5dc7e18a","subject":"seed-vm/infra-vms: Add support for EFI and q35","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/045c04f97c4e63048f4b1f8559c8fbda5dc7e18a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-23 12:52:39.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-23 12:52:39.000000000","tz":120},"subject":"seed-vm/infra-vms: Use efi as default","message":"seed-vm/infra-vms: Use efi as default\n\nChange-Id: I7d71689ef9b5a8a092fe78da4ee3d2a645a1316e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0a92873d21aff25080ef44fdef45d498dd041e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0a92873d21aff25080ef44fdef45d498dd041e"}]},"branch":"refs/heads/master"},"49c33e0a73643da543f042a6354c10566f81d166":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-23 15:10:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/927015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/927015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/927015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/927015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/927015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/927015/2"}}},"commit":{"parents":[{"commit":"9702736b3ec3c0dc44657b34c90594ccc3c75493","subject":"seed-vm/infra-vms: Add support for EFI and q35","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9702736b3ec3c0dc44657b34c90594ccc3c75493"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-23 12:52:39.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-23 15:10:11.000000000","tz":120},"subject":"seed-vm/infra-vms: Use efi as default","message":"seed-vm/infra-vms: Use efi as default\n\nChange-Id: I7d71689ef9b5a8a092fe78da4ee3d2a645a1316e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49c33e0a73643da543f042a6354c10566f81d166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49c33e0a73643da543f042a6354c10566f81d166"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
