)]}'
{"id":"openstack%2Fkayobe~927261","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I88f20112ab4740fe114a6484f88f4fbfb66863e5","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-08-30 17:20:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88f20112ab4740fe114a6484f88f4fbfb66863e5","subject":"seed-vm/infra-vms: Add support for EFI and q35","status":"MERGED","created":"2024-08-27 11:49:02.000000000","updated":"2024-08-30 17:25:39.000000000","submitted":"2024-08-30 17:20:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927261","cherry_pick_of_change":921660,"cherry_pick_of_patch_set":26,"meta_rev_id":"dd65d04008c98791571d42f2ea8e860e47d996ee","_number":927261,"virtual_id_number":927261,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 17:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-08-30 15:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-27 16:40:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-08-30 15:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-27 11:49:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-08-27 13:41:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-30 15:12:42.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"2be7ba08d1aa606bcbc03572f916b09a73790900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-27 11:49:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"38259f5ae41fb4557585f50d5745c99e10434fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 13:41:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10207aa66aaf4b65970e632add9d45a9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba9fbf96e4df44728b1cac98967dd8af : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd50650c2f94b76aade06e72627bf55 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa41db2aab844ae6ac57e4bd6e8ffeec : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a75e05af001471bbe84b88d17c04cd5 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7bd3eba881d43c6977cb302187e2b73 : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/296d6667a00a41c288f86f8755149364 : SUCCESS in 8m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/acad9e2612ab432b94602f9c442ecc1f : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/13fd9ec4fdb64b4585fbaf0c986a27b5 : SUCCESS in 1h 01m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0d48a2fe08249c0ad9fa86555e727c6 : SUCCESS in 52m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cba9999b8946434eb8e06460b7cd1716 : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c54fe332426e488495e8178515439143 : SUCCESS in 20m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ec3548c3d76246e58b59c4b997835929 : SUCCESS in 18m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39fe4622d2ac4fd3ab0724bd018cc72f : SUCCESS in 11m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d68a48a539ac4cacaf4aeda927ae51ca : SUCCESS in 1h 51m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bea645378faf4c288625b79c8c3aec50 : SUCCESS in 1h 37m 40s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e4a99d23e38e4a56a31f157424fde4f3 : FAILURE in 1h 28m 20s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/369bf055f8f644c2878b3c79b0c22910 : FAILURE in 1h 10m 01s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bde76723c4ae4a1d9704a29e8c911add : SUCCESS in 19m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62a06247677f4d72a7be1382f2ae7bd3 : SUCCESS in 16m 45s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f4581cf0fb2e4bf8b8cd04aaa73bdf16 : SUCCESS in 40m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8877cd302d9f42a48c9b0d9b7485c93c : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af70c8f40997436d94fb8d2cdbb2890b : SUCCESS in 31m 10s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/843cda8475ea4740aa5e2ef79cbc3afd : SUCCESS in 35m 27s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ebdb0cbfa19479e86e19bfbc293301a : SUCCESS in 23m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5fa024d145634a5aa1dda69b1070f592 : SUCCESS in 17m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/32de4746d61f4766bec81a3cfa469965 : SUCCESS in 14m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/047dc950e4794db9adc5a171385f1bba : SUCCESS in 10m 01s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/6a0ed188dc0f41c1be77d4f17cd2f7bc : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c0e03a78c3b6421788d3f65156ce4614 : SUCCESS in 19m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36ccfb6724d549c69a2ddebf60f0a0f4 : SUCCESS in 12m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b7656e91ee54d61a1b5a662d0a2cb54 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00bf2523131645c3954336ecc07a7ad1 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c73ff91c4dbb4696b8a6677080de1318 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c72943be4740bf7d5096a6dbd74f643fb8312b26","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-27 16:40:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4e90cefcdc9b8e377c5182fed5082329c58aa67","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-08-30 15:12:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e23f84b3eb94bbdf2f170d3b73324141cafa4f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:15:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5f6642f78bb973a448bb6a815f5e93d4293119c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:20:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7f49432eb4d4dcfbc04169e13f1fad9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3246d9299a404318b3202300fe9274da : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f61828f224824eabae6c11bc91fe07be : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd3ef342583e4721b3e64d38a5758f9a : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b2f81f436374a00b4ea41bd11412817 : SUCCESS in 13m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f011687312a44ed0840dd5cd97689815 : SUCCESS in 8m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5fa1e8af79984bdab53471ebf01bf13b : SUCCESS in 1h 17m 31s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d060d450e9e48f4b25a205b71bb9409 : SUCCESS in 48m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e04d26ac2cbb4469a70b75fbf1a38fce : SUCCESS in 1h 26m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/17ae486e4a824f00b91b349c05e0ae04 : SUCCESS in 25m 21s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/942ceafd900f48f2aeb35f6432e02e81 : SUCCESS in 18m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/44290505aa894ba59e3b8403b399e765 : SUCCESS in 1h 46m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31c50df45c3947bfb5af2b976efe6c43 : SUCCESS in 1h 43m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/377c011a4640459798bb5fc401df1393 : SUCCESS in 19m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14a9c9099d774e249cead9b17667b91b : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b1331726a0e043f1b910a60f2b70ea27 : SUCCESS in 28m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6e4704338e9429db35a9ee769da4e46 : SUCCESS in 24m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/160d38fde1cd48c8b89d851fec198105 : SUCCESS in 32m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/437aff3b63e84135a9bc523a73ebf02a : SUCCESS in 33m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/56b6d66917ce49c8a617451a8e70fb23 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e29e3949ffcf4575b735b0b7ea986380 : SUCCESS in 13m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15015a6b34e64d72b0afe7ebf160d82b : SUCCESS in 22m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a73c7af33d947068f8860b7204a6746 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"09a6ec85b208e017b702c2c10cf9b2e013f2acfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:20:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd65d04008c98791571d42f2ea8e860e47d996ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:25:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/737fa99dd63541c4a6e907714f835de0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb677c077114998b6cc88c0f788fec7 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4e2159223754626903f3afadf70b7a5 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dda83ae0c8ee9cc7a7883786436b9f1a3fa6ce8d","revisions":{"dda83ae0c8ee9cc7a7883786436b9f1a3fa6ce8d":{"kind":"REWORK","_number":1,"created":"2024-08-27 11:49:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/61/927261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/927261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/927261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/927261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/927261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/927261/1"}}},"commit":{"parents":[{"commit":"ec1febba7a2d0300ace2cf9f39778422b744ed23","subject":"Merge \"Stop replacing dashes with underscores for Kolla interfaces\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec1febba7a2d0300ace2cf9f39778422b744ed23"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-10 12:16:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-27 11:49:02.000000000","tz":0},"subject":"seed-vm/infra-vms: Add support for EFI and q35","message":"seed-vm/infra-vms: Add support for EFI and q35\n\nBump stackhpc.libvirt-vm to v1.16.1.\n\nChange-Id: I88f20112ab4740fe114a6484f88f4fbfb66863e5\n(cherry picked from commit 9702736b3ec3c0dc44657b34c90594ccc3c75493)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dda83ae0c8ee9cc7a7883786436b9f1a3fa6ce8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dda83ae0c8ee9cc7a7883786436b9f1a3fa6ce8d"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
