)]}'
{"id":"openstack%2Fkayobe~928484","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~Ief37b7bee758e34dddb881a9e68728013861c0ec","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-09-09 15:55:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief37b7bee758e34dddb881a9e68728013861c0ec","subject":"arista-switch: Fix eos_config call","status":"MERGED","created":"2024-09-07 07:26:03.000000000","updated":"2024-09-09 15:56:43.000000000","submitted":"2024-09-09 15:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928484","cherry_pick_of_change":922631,"cherry_pick_of_patch_set":10,"meta_rev_id":"98000e2282cb6eb50337a5de6845e75e944ae714","_number":928484,"virtual_id_number":928484,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-09 15:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-07 09:34:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-07 14:54:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-09-09 13:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-09-09 13:56:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-07 07:26:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-09-07 09:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-07 09:34:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-07 14:54:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-09 13:56:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9ea2633b1a64f20ad99b9e0bc7e6c6aeec0a9956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-07 07:26:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"203d9336efaf392bfd6c94672d0763d7f88d1b4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 09:28:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e334ae73189644eaa4541d25116be400\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0fb7eced3554298b7b2865dad32c195 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3cfae29ecf04dcca9feaf718ebb34ef : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51f02177e6a940debe920f62d9105e94 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ddfd0f544240ccbc9a53f771b33193 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c67bee81daa94955b8ac8051807a96fd : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/543423f3beb94d8680679c22c68d5abb : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d74e7bbe4ed4b1cabb142b0e9c839bb : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16faf42ff4f54b218a55f2ebeaa98836 : SUCCESS in 11m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2bae05256e1040cfa01cbb8dccbf5410 : SUCCESS in 1h 06m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bb393192ae79418cb6d10112b9a33c9b : SUCCESS in 52m 32s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe5a5189c5f34cc69624f1194123eb98 : SUCCESS in 41m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a67959efe7c64cadb75c3adbb57b8a5a : SUCCESS in 1h 11m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6d138f769c004330aa32a3e405a51c5a : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0d2cf33f28a748ce8107b1c220f38502 : SUCCESS in 13m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9904a2e4037b438890ab1d69048eebb9 : SUCCESS in 11m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/25116d1b84c446358289ac64fb72ddc0 : SUCCESS in 1h 50m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c2cb0bf1fec4a9b91da91516974c54c : SUCCESS in 1h 09m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/df67a1f4f54242ce80e18394589696f3 : SUCCESS in 19m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e6bdec73cef45acbec30408a630c545 : SUCCESS in 12m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4616a68052af42f5a0333ca1e5e4d9a5 : SUCCESS in 38m 58s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a29183c717e94dccb5ab464146e63790 : SUCCESS in 32m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d881201b5dd4b88bab200d6f2180bbb : SUCCESS in 20m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9996ff7e3a97475aaba5a7c98b858e7c : SUCCESS in 13m 53s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0c09fbba8f764af6bb3fe3a25b307ab6 : SUCCESS in 10m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbe97c6f0cb74613adf83e8452b80edc : SUCCESS in 9m 23s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/41e6e2a2fea541848e38c7b8df8f7e1e : SUCCESS in 10m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/75aba4bbc8fd49d6902989427d69b672 : SUCCESS in 13m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6073adba58e4cbdae43a60b1de6e9f9 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d30cae626927d8722849ba82e691b7d3d872c25b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-07 09:34:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90b6d19d7fcbecd14ac885b8497a550bab238b32","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-07 14:54:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d48e678daad3f269f8c809d4aa7ed4e73a69df80","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-09 13:56:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13433afd26d2b54314be825fa6048812ece80793","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 13:57:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12dec8977886e7a5d26c82f7edfcde5c06c74723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 15:55:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5160e23495e456ea37f90943f6b5b72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76915d509bc46d9993814d07b07f807 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5837dd80d0d24f2ba608aef294d483c0 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65c067a6a08432997656aa0d89e1bf8 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aedc3f8a8d584200bd20e0ce70b3e46d : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbc8dc465a5e4bceb63b3ff3ae5b6885 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03e8f071ba0d4adf95f781afc5553dc9 : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d71263f593074b289df1a6602677b9d6 : SUCCESS in 12m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/22c88185a4de41b7af85f7b96b480ff4 : SUCCESS in 59m 44s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7392599f00f047ecbb138ca48b67f382 : SUCCESS in 53m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/494cb09ebf2c49929ca7973d381b1119 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0b17564b6670480795bb7dea069611c9 : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cf53a27674545feba7a4adabbef4400 : SUCCESS in 9m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8e7761410c5941eab024d51a65ee1c4c : SUCCESS in 1h 53m 09s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0e5b1716d3a49838bc2d8efa80c161c : SUCCESS in 1h 29m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e21cac55ba9e4822a9251ab04adaf81a : SUCCESS in 19m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69e295293e9142c8869b8d77bac53321 : SUCCESS in 10m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ef4ebc2db8ad4e208a0afe563fec9b21 : SUCCESS in 34m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d749c864a254390b987ad84707cd653 : SUCCESS in 16m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/04948f0a89df4147b268c4cfbb06e3f7 : SUCCESS in 11m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/697c5fb9fa044adf84de417740aedfb9 : SUCCESS in 7m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/27172fbb3407491b8467143a8488c4ab : SUCCESS in 14m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/263401c2d9dd43d9a1b89ca1a11ace49 : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"98104bf1d63159cd8b01bf6ca95b5b156d3ec38e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 15:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"98000e2282cb6eb50337a5de6845e75e944ae714","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 15:56:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432940443fee4073bb548fa4512f8606\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f79a9cea3dc24c7ea6dd77cc28285a2f : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c67a658cd5e43b0bd82548b92030bdc : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"afacd339fece3ed5694a00f3d009de77dbccb813","revisions":{"afacd339fece3ed5694a00f3d009de77dbccb813":{"kind":"REWORK","_number":1,"created":"2024-09-07 07:26:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/928484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/928484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/928484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/928484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/928484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/928484/1"}}},"commit":{"parents":[{"commit":"ace210f9b59b47faa51b5d6a15dc818fc432460c","subject":"seed-vm/infra-vms: Add support for EFI and q35","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ace210f9b59b47faa51b5d6a15dc818fc432460c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-24 14:56:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-07 07:26:03.000000000","tz":0},"subject":"arista-switch: Fix eos_config call","message":"arista-switch: Fix eos_config call\n\neos_config does not support provider, nor connection: local from at\nleast 2022 [1] - since ansible\u003e\u003d7 (ansible-core 2.14).\n\n[1]: https://github.com/ansible-collections/arista.eos/commit/588d94b2ed22817b63f996187e7d757d3ebad725\n\nCloses-Bug: #2058365\nChange-Id: Ief37b7bee758e34dddb881a9e68728013861c0ec\n(cherry picked from commit 32489a45874f34414fb442cc6009857648ff25f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afacd339fece3ed5694a00f3d009de77dbccb813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afacd339fece3ed5694a00f3d009de77dbccb813"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
