)]}'
{"id":"openstack%2Fkayobe~928658","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Ieec98826a393f75592254e7cb7b9d2c4aa903dfd","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-09-12 10:59:59.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-09-10 09:07:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-09-12 10:59:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieec98826a393f75592254e7cb7b9d2c4aa903dfd","subject":"Use Rocky Linux 9.3 images unless in efi boot mode","status":"MERGED","created":"2024-09-09 15:49:10.000000000","updated":"2024-09-12 11:01:10.000000000","submitted":"2024-09-12 10:59:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928658","meta_rev_id":"4cfe986eaabbacc0b67147eb4de8321a54e0d55c","_number":928658,"virtual_id_number":928658,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-12 10:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-09-12 09:54:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-09-10 09:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-12 08:51:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-12 08:51:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 15:49:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-09-09 15:49:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-09-09 17:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-12 08:51:17.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-09-12 09:54:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"39eb846d39d840ce761cad141851c60e71788fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-09 15:49:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f831a53f58a46e17ef78c8194de63f604e1b3811","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-09 15:49:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35af88b8d12af76810d3ed2c778d95d54b703e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 17:50:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c3500be775d4b10b89963a11be0ac69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2824bda0c73644afb7f2e4a791e00e37 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f378d9a55440afbf6d92563b0da2eb : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd5ccf8ac8f34b0f926d6728dcb380ea : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da639c93a0b54508afb0282e33408029 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d2d6bc515374e808e80feacedcd3c49 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40d7c795c5ae429687c398c5f70fd42f : SUCCESS in 8m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a7e253597bd0499d991c8c4d933f93a8 : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8e03b2f5bafd43d0b534dc1d9bf59781 : SUCCESS in 58m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/581bb99e6cd644e0942afdc8c61fd761 : SUCCESS in 50m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f56d69a2aaae4ca5b6c8f068a92448df : SUCCESS in 1h 13m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/26cab0c2e60545419d4135530faab743 : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5a189e9513f64dccb9a05fa22090a8cc : SUCCESS in 16m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21c1fa2f1e684d53933a6251d8eafde8 : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a4b9fce5f4c745f1aab4d9b079159a26 : SUCCESS in 1h 55m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bd1d19d3238420cbeb2d4e5ced9283d : SUCCESS in 1h 22m 37s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4db0b1ccbfb6422880ef5a618bfa5967 : FAILURE in 1h 16m 29s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/674c9d3aa438444196e5dc4686a280d6 : FAILURE in 1h 07m 18s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0a6f0846740646b9b42b96b47afebb15 : SUCCESS in 18m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d9b6fe1fa54411e92f1f3529b0c3e73 : SUCCESS in 14m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/539d544312ae4d95a6135f08aef0593f : SUCCESS in 38m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1df20d0d79f34ec78027b24a5ebd810e : SUCCESS in 31m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f65d14502d854cc2aee68936ddb8def8 : SUCCESS in 21m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/49cdf5b918904342b89416f59a773ccf : SUCCESS in 33m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/651a954eea4248929abeb3f60fa28612 : SUCCESS in 25m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/292ca8b369a3456a92d75fc6fb52d40f : SUCCESS in 14m 38s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/9911f03f7aa04feb80625cea0570d92b : SUCCESS in 11m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d96900fa35524913bd27bb618eeb9941 : SUCCESS in 11m 01s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/0d35576f877e4c0791905c22ab92eae4 : SUCCESS in 9m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/801f905b7d00422d92594d2817a66000 : SUCCESS in 18m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f76a8381fef44b2b46d42782fd0ed8c : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b5fef5eb3f45dda2f4e05e9989e802 : FAILURE in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02c0889898db491f9596835892546202 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ee13105c665405898ac11ebf0805898 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/532eaaaedd13449e917b2aa5c88c1393 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"75cf6b931cb7364b7f1229afb05fe88fb835a18c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-09 19:28:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"901870d21365d126216c9243cc8cb243af3690b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:46:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf06fd05dc94018ab610ccb6ad4a82b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2af2687f377e484992eddbc5f2a0943d : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f6ab957cf046219d02470b9f136586 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42d0cdce29244b0bad4098f2f716725d : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/77ace1180f16431d925226476a4c08a5 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f175016bab2f4607942bde407fbaa5b1 : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3c25c8162ea41b0acf2429adf2243ed : SUCCESS in 8m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/38363b881bf54a0693885b0ac1c1da9f : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cbbb3246119a4d4282f46484521e4b4c : SUCCESS in 1h 06m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1d8d308e1e94fc1b49b92fc30829316 : SUCCESS in 53m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d99f6b40942c4333a0e55caed7d68f89 : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/717b7d49e4dc43ab92ffbbf7cdff080d : SUCCESS in 21m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d4ed0bf097114bf4bf566ead64133017 : SUCCESS in 15m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8df3ac242f3943298d2db5ecfeb6e44e : SUCCESS in 12m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/619947442fab42aca23a8a584b84d376 : SUCCESS in 2h 08m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7e278b839764b41a70ad296c5a668b8 : SUCCESS in 1h 38m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/55d37e309702415eb94f6904f791f90b : FAILURE in 1h 31m 23s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4433db986672440a88af9106f573363a : FAILURE in 1h 15m 06s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/101156ac3bdb4a66b24827406eb6fefc : SUCCESS in 13m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1963fb171e9e46c58a9ff1878687bc75 : SUCCESS in 14m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e2f75c5e7d7f48ee95afb9dd24911677 : SUCCESS in 38m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b3d654d5e0b24774a0b52faa719163eb : SUCCESS in 29m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/730746fee9f54ff8ae0fc92ffe8fa13f : SUCCESS in 21m 34s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4f80d2449ea442319b27eddcaef5286d : SUCCESS in 35m 48s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/291d7558fb424c078e4e34d0a0ca4553 : SUCCESS in 28m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/47fc3e3693a043efa59dc9cf15eeb9f6 : SUCCESS in 17m 24s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/87eb067ccad7438d92e8b89a3a71628d : SUCCESS in 11m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41f092a2e07647a199e3285a3e0577e4 : SUCCESS in 10m 59s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/ed0330efa9a343d39b1d1cabb4d026ab : SUCCESS in 11m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ff6acbe5e00444b8c859921bc68e7b2 : SUCCESS in 13m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7851b164ee464ebb89b4fd6ea087457c : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e292bb861d4fc0bc25b21b07db6db6 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a57eec1d76db439e940094f91a75d163 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e419796ab53457a86b31bb4fc6a8851 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ed34259e18346049eaab8c73676dc08 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"fb5d461b7b4e6c4b985d32bbbdebe9f83847adad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-09-10 09:07:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4f1f7945e1053a2b494e04a9e3a6d15ba2c318d7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-12 08:51:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"08d41186d09d4b50f9b7aa21144b51cfdfd77550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 08:51:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9559bf4b0253ea37e2f61fcad34c3d3a5f0438ed","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-12 09:54:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2d223601af9c9fe0a0efe940384398142d68f895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 10:59:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09ece9baaa3b48f29cecee90a1384845\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4043e972388443d0ab0b7bd60282d960 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd47a7d32aae438091066a27ad8c7e51 : SUCCESS in 2m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2fe8d06428fe46108627c4675fb1092a : SUCCESS in 3m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/64775e8dd1344d6d8e9e5034f3ae0e1d : SUCCESS in 6m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/feb2e79df990417f8ef588e8569a1fe9 : SUCCESS in 6m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5c8add3e16c2422b9edaf2d705cf843b : SUCCESS in 40m 34s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f28c8bf0ac54c88980d92f06e02e08a : SUCCESS in 51m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fbd3bf1b6f9f442a9b776f66132c54f0 : SUCCESS in 1h 14m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/db729cdeeaa24f08aa71b3e1f8f4e313 : SUCCESS in 24m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bb2b81916af4a2691c46014309723ec : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c45ba7b892fc4f279872f13d69cc5201 : SUCCESS in 2h 06m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7074fa73ef394a469ae35d3491c972d0 : SUCCESS in 1h 35m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7a71bc2c243141058a6a02a1420877f3 : SUCCESS in 21m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/407ff52a75b84e6f8871f4626c04e5d0 : SUCCESS in 12m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3f36cadc4fb84574b23e344d6aa6f751 : SUCCESS in 33m 27s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c091731e61fd49c988d6862c5019b235 : SUCCESS in 24m 02s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7b708c5affff41cf8ffb1df215ae862f : SUCCESS in 33m 38s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9729185dc4849d7a1b08d2a38c155c9 : SUCCESS in 24m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/28fdc9d3cf1d46f8912242a984823125 : SUCCESS in 14m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6a0725371e543e98029e3d4de51548c : SUCCESS in 10m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/92f4717e18594301a106e1de46b622e1 : SUCCESS in 19m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eabc06ff8b66417d84559816ce49c091 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":2},{"id":"5bf394760b19e4e7fca93c5b51af9435ca763592","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 10:59:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4cfe986eaabbacc0b67147eb4de8321a54e0d55c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 11:01:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7daea57d12be4d94a86c3db6baad3f2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b09317ecee894cd39fd134685caf2cf6 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/689372c0271e47878314949eb9310291 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5274c4e6cdaf22e7e753363e78a6c69dd423d831","revisions":{"ac1bf2030edf69ee0170482ce8e95fa6dda8b95a":{"kind":"REWORK","_number":1,"created":"2024-09-09 15:49:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/928658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/928658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/928658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/928658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/928658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/928658/1"}}},"commit":{"parents":[{"commit":"d05484b56d6776b37282f64371e433dd41a22e87","subject":"arista-switch: Fix eos_config call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d05484b56d6776b37282f64371e433dd41a22e87"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-09 15:46:26.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-09 15:46:26.000000000","tz":120},"subject":"Use Rocky Linux 9.3 images unless in efi boot mode","message":"Use Rocky Linux 9.3 images unless in efi boot mode\n\nChange-Id: Ieec98826a393f75592254e7cb7b9d2c4aa903dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac1bf2030edf69ee0170482ce8e95fa6dda8b95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac1bf2030edf69ee0170482ce8e95fa6dda8b95a"}]},"branch":"refs/heads/stable/2024.1"},"5274c4e6cdaf22e7e753363e78a6c69dd423d831":{"kind":"REWORK","_number":2,"created":"2024-09-09 19:28:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/928658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/928658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/928658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/928658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/928658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/928658/2"}}},"commit":{"parents":[{"commit":"d05484b56d6776b37282f64371e433dd41a22e87","subject":"arista-switch: Fix eos_config call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d05484b56d6776b37282f64371e433dd41a22e87"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-09 19:28:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-09 19:28:08.000000000","tz":120},"subject":"Use Rocky Linux 9.3 images unless in efi boot mode","message":"Use Rocky Linux 9.3 images unless in efi boot mode\n\nChange-Id: Ieec98826a393f75592254e7cb7b9d2c4aa903dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5274c4e6cdaf22e7e753363e78a6c69dd423d831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5274c4e6cdaf22e7e753363e78a6c69dd423d831"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
