)]}'
{"id":"openstack%2Fkayobe~929556","triplet_id":"openstack%2Fkayobe~master~I46a71f7065b0da9c23cf094e76840e005ecd3a8e","project":"openstack/kayobe","branch":"master","topic":"no-public-openrc","attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-10-08 08:34:42.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-10-08 08:34:42.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-10-08 13:05:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I46a71f7065b0da9c23cf094e76840e005ecd3a8e","subject":"Use public-openrc.sh from Kolla Ansible","status":"NEW","created":"2024-09-16 16:27:59.000000000","updated":"2026-10-08 13:05:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c79f2a855700ff594288b7914185bd7a53fc62fb","_number":929556,"virtual_id_number":929556,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-24 17:28:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-10-08 08:34:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-10-08 13:05:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-10-08 08:34:42.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-16 16:28:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2024-09-16 18:07:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 08:53:56.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-09-26 12:14:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-07-16 21:10:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-10-08 08:34:42.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"e4a45bc234a768e503d59ef6ec0101fdc18ffef7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-16 16:27:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b76f56a4be7a17c4fe5f0db2609529d28987b9f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-16 16:28:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9b61980f8c61f72122f752f9f9b8e57bc36c41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 18:07:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7e17efda85442982e3ebeb570281e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/740a246ae267462982b6649ca546bf33 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57f74978843486f9995da17f20a3938 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bbe99cb8f5448e38cee7da06515fb81 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bba126bb84e4bc4bc1a75940ce41d0c : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7dd4674db4fb4bd290a1172c4ad3eac7 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a894543182c49788b0c2e6b141ba3ae : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0ab0d5e8d494f538294ddb54f6e7871 : SUCCESS in 2m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5b81cfdf4cd4d7cab7c92ec6a2697c4 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e69d95b2480f4b95960d3817bf3c43ba : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc2cdb9531a8441c913443f24f8446b9 : SUCCESS in 9m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1b6037c562a6489e8db6382b7986b364 : SUCCESS in 58m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3932cf8a999b45e7bccec99fcd9f451a : SUCCESS in 58m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2808ecdc205e449d97a765828a5b7e4d : SUCCESS in 55m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ab59edfc19334e0b9a6cc6a0e869c235 : SUCCESS in 1h 09m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bcc5d2edfd784f28bc8116f96271c94d : SUCCESS in 15m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8e1f3605b3a74e1c91078837434303b6 : SUCCESS in 11m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df25e5168c6c4ad6b2320915017e37c8 : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6d4d6f30fdeb46c3bbb383b826cde90c : SUCCESS in 1h 16m 27s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88f48350767c46ac8404d5462035a301 : SUCCESS in 1h 39m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/16f0b17305d94d20853bc7d4dc202ef7 : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ee35fce856a40c495af8cd7f4ba8a6b : SUCCESS in 14m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ece14652791d40f08b3f2d0be31da072 : SUCCESS in 43m 27s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/388987fd53f9400083edc71361cfe968 : SUCCESS in 31m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/109ffce058354d34a9898e450569649e : SUCCESS in 26m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87fafd4d4afa462e8d1250af00d975cb : SUCCESS in 13m 22s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f81a470ab62d44ada951fbf56cbd4be7 : SUCCESS in 17m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1084d1a59a344edba6decaf9f4e0d9b : SUCCESS in 8m 32s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/abd044bdf93e454a81da959b91654b67 : SUCCESS in 11m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99cb6000ec854708a05550531ea91a51 : SUCCESS in 23m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cb7fa2b7ace4b62ae150dd0685f580b : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"db1cd3170472fa9d027351dd701ef5ad6685cd66","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-17 08:53:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"118d143b9f3204fa78de6e645ca07035634a1439","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-12 13:09:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e211382afc080624d18a95d55212f1a1f6a5ba82","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-12 13:12:55.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"923c715c1b08d47f9dc82e02a8dcd596233f2bc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 15:21:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c72848df8f5e4d1dac130df213e02dbf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34f2f0eda0574a85bcf1e7f4d2c1a613 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709dfd2722b14aa5856fa2a7b9b7ae12 : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eed89fb1937e48f587624899df16d189 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c14d6dd955045b8905e078193649c6d : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f5d206bb9b542258da9b542ea28ec79 : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/50fe05d88153402a9980d1a35e755b16 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5937076116604ad3b2673852be17cd4f : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd1226cc52654ba1ae9fae16594b1faf : FAILURE in 6m 19s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/46791ed31b4b440eb5d725aeac6564b8 : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/37e454904561459e90ec3c24b4241026 : SUCCESS in 1h 15m 00s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ade32c88c8824dc28ce91be1df2de3bb : SUCCESS in 33m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/634db463257847c0bab346bbda5ba9c1 : SUCCESS in 39m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4aa0b2231c5545c5a03964d4c38d0ad3 : SUCCESS in 12m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1e2e0d16954f4a9d84243b97239efff1 : SUCCESS in 10m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/292a74066e9e423f8511fd9723e96d07 : SUCCESS in 8m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f2582511e88e40e39afad4c515b6f54d : SUCCESS in 2h 01m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66d8d0f5703e4461acbba3752d209cf0 : SUCCESS in 1h 38m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0562bb52cf63403e9204689d0ce168fc : SUCCESS in 13m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c63e02a9b1844f7a8c96e02b24afc1a6 : SUCCESS in 17m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/952f26c3f18c47b99fef5a878ab671fc : SUCCESS in 41m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c9e58048a3334f6994f21c8e1395ef57 : SUCCESS in 21m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acfc042c999843f58471317a23b0b664 : SUCCESS in 28m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/565a27a79fd947ceafc10f0ada5c8f30 : SUCCESS in 18m 01s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/51d0c68d1fc04693b6c4bdd3f000f8da : SUCCESS in 16m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eeb4c405aba14b02bdde1e75592fb944 : SUCCESS in 7m 41s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/bb10ca3fad1e424e9972c1c1d789193e : SUCCESS in 11m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f0d26504974845b0ab201f7bb7a08f90 : SUCCESS in 21m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e8fb7cbb479460e9883f1f4cbbf1786 : SUCCESS in 18m 33s","accounts_in_message":[],"_revision_number":3},{"id":"8b1a29cf905ee73bd4c768b1f6be6225df4a3bb4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-24 15:25:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c02bd2d1dc9a6860e74b92583231afd116f326eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:28:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e52e6565bd2460bada12c4ae976d693\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14863975a65a46f184f685ebb667dd81 : SUCCESS in 10m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fbac937b984448d58de3fbefcca45770 : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85897c6dbb3749a189c54211f3976538 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88dd72abaf6f4f9b9579e93e0524fd7e : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e46ccdc7fa240468d5140957aa96e70 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3c64dec5e264754a4a5b3fc7c284539 : SUCCESS in 7m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28fd6cce561546b7aecc0d577a6a6214 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c68aded3171b4de7aa5f7f76b9e420af : SUCCESS in 18m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a79d7672ac914b3ea18e5d4313aaf8e8 : SUCCESS in 13m 53s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2aa4662dc49e4505967ceb7891f1df3f : SUCCESS in 20m 28s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8585e87d767e4c94b5663d51277629f0 : SUCCESS in 20m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bb49f5fcc7a455bab0b238e4a0fec2a : SUCCESS in 16m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1cf7450f962444c0958601e889d3850a : SUCCESS in 23m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/62d3a2f5178546c48541e470b9e50aae : SUCCESS in 29m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d82ef7044c040008bef66484947c3d1 : SUCCESS in 24m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5bf3b5a963554a06a7bc577eaf0e3d58 : SUCCESS in 25m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8eddcd8607b4169ae7c0b969f0c7345 : SUCCESS in 19m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c5c69434baeb408bace9e13919aaf196 : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5a51c770b5114dce98f54a401881727c : SUCCESS in 1h 04m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/242afc5f9fe549f4a2b0c91119b0d2e7 : SUCCESS in 1h 12m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e493fcdd37a4556b0439c1f30e09c6d : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b374b1aa54bb40bb8190bc1b104dd157 : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cd3f8fb582dd4db8be35f731cd83c9d4 : SUCCESS in 2h 02m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1cd926e43a7e483392a7964cca6059df : SUCCESS in 19m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/80c30969cd724dc1916edf039a77c5ef : SUCCESS in 23m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a0b478a08204a378fff45566a1952a1 : SUCCESS in 26m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ba80c395d884a2eb1f2d982fe082521 : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5cbf769b5ce244acb539a2881cb643f0 : SUCCESS in 39m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43c4f35ea74746aaa0fc18e6f1e379e4 : SUCCESS in 25m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/44f8f0fc51434551b960aca47442991f : SUCCESS in 24m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/163210b300d44dacbdeb0362baca1a29 : SUCCESS in 22m 07s","accounts_in_message":[],"_revision_number":4},{"id":"3ff703bc15aefa5921d6461dc9fde81df02469e4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-30 08:36:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4c66f24a72ee63d8a4f21e14b3cdff1393dd6b55","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-10-08 08:04:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5cc768173a1fa6beca764071331a63cbbe368676","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-10-08 08:30:24.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"3ec3d4f6809e7fabaf07e5e5c2934dc919b8563d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-10-08 08:34:42.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"c79f2a855700ff594288b7914185bd7a53fc62fb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-10-08 13:05:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c7b4d7b615f964556c901391a488f635576c286c","revisions":{"8416737a04052487ca29343d8ff187f096fdb669":{"kind":"REWORK","_number":1,"created":"2024-09-16 16:27:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/56/929556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/56/929556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/56/929556/1"}}},"commit":{"parents":[{"commit":"6b37698527ca614aea1d262c0f56707274ea60c8","subject":"Merge \"Support skipping SSH keyscan for older switch devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b37698527ca614aea1d262c0f56707274ea60c8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-16 16:25:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-16 16:25:36.000000000","tz":120},"subject":"Use public-openrc.sh from Kolla Ansible","message":"Use public-openrc.sh from Kolla Ansible\n\nA public-openrc.sh file is generated by Kolla Ansible since the Caracal\nrelease [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/862159\n\nChange-Id: I46a71f7065b0da9c23cf094e76840e005ecd3a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8416737a04052487ca29343d8ff187f096fdb669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8416737a04052487ca29343d8ff187f096fdb669"}]},"branch":"refs/heads/master"},"848eccf59e99434bd700b967759b08b684f0dc3c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-12 13:09:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/56/929556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/56/929556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/56/929556/2"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 13:09:05.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 13:09:08.000000000","tz":60},"subject":"Use public-openrc.sh from Kolla Ansible","message":"Use public-openrc.sh from Kolla Ansible\n\nA public-openrc.sh file is generated by Kolla Ansible since the Caracal\nrelease [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/862159\n\nChange-Id: I46a71f7065b0da9c23cf094e76840e005ecd3a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/848eccf59e99434bd700b967759b08b684f0dc3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/848eccf59e99434bd700b967759b08b684f0dc3c"}]},"branch":"refs/heads/master"},"a57b0953c30a709bb7c2a58254bb3976bd40f557":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-12-12 13:12:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/56/929556/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/56/929556/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/56/929556/3"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 13:12:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 13:12:43.000000000","tz":60},"subject":"Use public-openrc.sh from Kolla Ansible","message":"Use public-openrc.sh from Kolla Ansible\n\nA public-openrc.sh file is generated by Kolla Ansible since the Caracal\nrelease [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/862159\n\nChange-Id: I46a71f7065b0da9c23cf094e76840e005ecd3a8e\nCloses-Bug: #2071817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a57b0953c30a709bb7c2a58254bb3976bd40f557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a57b0953c30a709bb7c2a58254bb3976bd40f557"}]},"branch":"refs/heads/master"},"c7b4d7b615f964556c901391a488f635576c286c":{"kind":"REWORK","_number":4,"created":"2026-09-24 15:25:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/56/929556/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/56/929556/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/929556/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/56/929556/4"}}},"commit":{"parents":[{"commit":"9e55adad9fe28100339ea2a0e31aecde80374017","subject":"CI: Prefer ipv4 over ipv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e55adad9fe28100339ea2a0e31aecde80374017"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-24 15:24:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-24 15:24:58.000000000","tz":120},"subject":"Use public-openrc.sh from Kolla Ansible","message":"Use public-openrc.sh from Kolla Ansible\n\nA public-openrc.sh file is generated by Kolla Ansible since the Caracal\nrelease [1].\n\nThis also fixes wrong values of OS_AUTH_URL when using a single external\nfrontend in HAProxy.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/862159\n\nCloses-Bug: #2071817\nCloses-Bug: #2168467\nChange-Id: I46a71f7065b0da9c23cf094e76840e005ecd3a8e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7b4d7b615f964556c901391a488f635576c286c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7b4d7b615f964556c901391a488f635576c286c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
