)]}'
{"id":"openstack%2Fkayobe~929739","triplet_id":"openstack%2Fkayobe~master~I62221df4650ddd02afdbc0993e783bf66b706a1c","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-09-30 16:26:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62221df4650ddd02afdbc0993e783bf66b706a1c","subject":"CI: Workaround rc:-13 issue","status":"MERGED","created":"2024-09-18 08:47:31.000000000","updated":"2024-09-30 16:27:58.000000000","submitted":"2024-09-30 16:26:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929739","meta_rev_id":"de22b326f17a33b9b891cf2674bc847b0f3d371d","_number":929739,"virtual_id_number":929739,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-30 16:26:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-18 09:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-18 08:49:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-30 14:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-30 14:12:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-18 08:49:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-18 09:37:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-18 10:41:36.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-09-18 10:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d364ebd17eed873ce3b61b5bb1504e2990c4c04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-18 08:47:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebaabf63d5d5d1d45cc0e9140e7a70f7a85cad67","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-18 08:49:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"706fe3158fe2be914703b6ad3387714418cc0cd1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-18 09:37:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e983441a68d9cc1bec56718267a21e02bcec331e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-09-18 10:41:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b98969cd027eeadd4085857e3404169ef9aa50de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 10:50:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11cccc89abcc41a892b3eeca56b9bac8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb94b0a34dcc411b922f04b19a3cbbb0 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb494d72f414874a2c0842292ec88bc : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60ab17540cb14d9588c65e1d72c75dee : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9fff4bcd71d54063b759616c6ebb13df : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5cc7f9b85434a3da8a9bb48ccb4f636 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbade5c3229d4382aecd747dd947f407 : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a48597d2c30045909fe4ae0afc200e78 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da09eaa327984804a57630b3c6e9e429 : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86ceca1ddbf94f89a8d18009550f81a7 : SUCCESS in 10m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/22a2c19cb56849e395687602333a822b : SUCCESS in 1h 16m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/de39718dd8fc459aaac40054596b1f66 : SUCCESS in 1h 06m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b704f2c4e159487d9c0486a15f2acd38 : SUCCESS in 32m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fb26b8360e874e0882c1a95bc741f73f : SUCCESS in 1h 15m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d37a89c4f62e4b1fa18980faaea25a28 : SUCCESS in 19m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/458abe85c9fc4730a4b148f45ba8fc05 : SUCCESS in 11m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5056e2844dd949deb7c9e01b68a1b263 : SUCCESS in 15m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/176609a90cf54df1a6ad6e7c02d6fd56 : SUCCESS in 1h 57m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81bed7f2db864187866618925bc3aa2d : SUCCESS in 1h 04m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/202dae44c6bc4701bdc7c41c92b936dc : SUCCESS in 23m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d27cf1424fe4e5fb869c8d38bbe5914 : SUCCESS in 12m 53s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/294753cf395e4553b238a8830413766b : SUCCESS in 27m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9d9500ab05a74d13b516c8d20849c01a : SUCCESS in 33m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a0f703f52174fe8af8ac30468336a4d : SUCCESS in 22m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b267de1c28154469a9aba8bed64efb05 : SUCCESS in 16m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d6f2be20802b471bb608465b24f20891 : SUCCESS in 12m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64b6c084628845ddacffe6411a9a8711 : SUCCESS in 10m 41s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/b31270f98ebc482f9926da5463479aa3 : SUCCESS in 10m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d6fa1586a2a448ff893fc77bc27413ed : SUCCESS in 22m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a6029be28e3474d934f5fc2c62d94dd : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"47bc338f206c46637ddc80175f6a4c90fea3df18","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-09-30 14:12:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3375446f1df81ac3756def4b1baec861ec3dcf3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 14:12:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c6cf69b227ad7a9c3ece2612bfadb3037501b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 16:26:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b05fb184e3c47a5a074af450fe73d7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98000e30bd0e4c088745bb91d27f46b6 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c47115f57c8c4a5ca7ffd7d884fafecc : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/894acc4adad8478d90575577319c227b : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b74c4650634de3b8151f1378b1aa51 : SUCCESS in 4m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21ecb210e3cc44e6a1de41eb85fcc47e : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea71b64713a3456c8b68a65c49f0c58c : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b418288995d4b56858efb605eaefd4e : SUCCESS in 8m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c9ae6bac71714d8c9a1803b512c4ddc7 : SUCCESS in 59m 35s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/def1d57cf377441eb601a34ee6853012 : SUCCESS in 29m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8fc044bec5f34ca3bf3614c89da10179 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/297bb16a12254284bcfa34cca4fc4c42 : SUCCESS in 20m 30s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b10d9adda947400893dbecbd1208462e : SUCCESS in 12m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/adbd2fc6275844ffb611b8d78f606828 : SUCCESS in 1h 56m 07s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f04b5b4b9da54b609b6e9588c18705cb : SUCCESS in 1h 43m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fcf23ad1218141d28768d45d710ba590 : SUCCESS in 19m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57c1bca311dd43028101db049af6d633 : SUCCESS in 17m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1f2b20b2984f464d82971e0582240fa4 : SUCCESS in 36m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1410377407a941f89994b6ce8e18b033 : SUCCESS in 23m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5b82860575f4deeafc5948d8950d370 : SUCCESS in 17m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a77f80ce3b84e1ab450fdca5deaad7b : SUCCESS in 12m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0c354a56c5a14a258981c9fdf4b987c4 : SUCCESS in 24m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ede84782612642b19838728eddb5876c : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a7d24dea7df3f39e8aea6fd5207e9093cf3b491f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 16:26:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"de22b326f17a33b9b891cf2674bc847b0f3d371d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 16:27:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2e734312c746ed9bce45a49f0eac30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b700cde6bfb44a091aee8d13b090ddf : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c86ba774ebbeba23e75df5c515c613662accf06","revisions":{"2c86ba774ebbeba23e75df5c515c613662accf06":{"kind":"REWORK","_number":1,"created":"2024-09-18 08:47:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/929739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/929739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/929739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/929739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/929739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/929739/1"}}},"commit":{"parents":[{"commit":"903126c01836f2404879291bd130060bdd75f336","subject":"Merge \"Avoid shared IPA image cache on Ansible control host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/903126c01836f2404879291bd130060bdd75f336"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-18 08:38:22.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-18 08:44:35.000000000","tz":60},"subject":"CI: Workaround rc:-13 issue","message":"CI: Workaround rc:-13 issue\n\nThere is a race condition in Ansible that can result in this failure:\n\n  msg: |-\n  MODULE FAILURE\n  See stdout/stderr for the exact error\n  rc: -13\n\nDisabling SSH pipelining prevents the issue at the cost of Ansible\nexecution duration.\n\nSee:\n - https://github.com/ansible/ansible/issues/78344\n - https://github.com/ansible/ansible/issues/81777\n\nChange-Id: I62221df4650ddd02afdbc0993e783bf66b706a1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c86ba774ebbeba23e75df5c515c613662accf06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c86ba774ebbeba23e75df5c515c613662accf06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
