)]}'
{"id":"openstack%2Fkayobe~929902","triplet_id":"openstack%2Fkayobe~master~I1c6befbdda7e684228065103feea7250a0ea3826","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-10-24 10:29:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1c6befbdda7e684228065103feea7250a0ea3826","subject":"Add podman support","status":"ABANDONED","created":"2024-09-19 10:28:41.000000000","updated":"2024-10-24 10:29:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96eecd05e913c1d817254505e54e15408391a985","_number":929902,"virtual_id_number":929902,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-19 12:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50c0a4c7ef2dedb992af89025d39ed4ce7767c2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-19 10:28:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8e20500aa4c6d2f644511e239d0ddf93621e600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 12:47:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3e920137fd246db917bb1041a1c5e3e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fdd175049374e01a0f73ff057b80e5d : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dce2668b084b4190b88cb2bd790dfedb : FAILURE in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17f2bc307a624fc28a1af175ba8c09ee : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15d12f8e06e7482681d05ab31f0e5084 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b65151df500f47d3923873b8ac8c904d : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878343d1a8a64df2b61bd5f0dab08db3 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84b81d2703d146749a0344416b9b1041 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/195fa0a43a9348bda8d700de7c479cb1 : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bcb7a3950d144657881f5ca0eaff6cfe : SUCCESS in 8m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/38c86c2229f940ea8b24868900be6b0a : SUCCESS in 12m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/220cb21752eb4185b5312eb5d6864ab6 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9b47e2e397c6467b876b0d7b95eb1584 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4ed0a3a45d8a4d9180afa3dc4984af78 : SUCCESS in 54m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ce6217c7881429ba94fb412790fa446 : SUCCESS in 50m 14s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/9e1a38bb7de1464b937375c3857161e2 : SUCCESS in 56m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dc61c1dc3e70493ab8ac5d7c82527302 : SUCCESS in 1h 08m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f447d8e773a843a4a8c2e8222f6ee0c0 : SUCCESS in 20m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5853c939523141aeb27a56dbb295c9eb : SUCCESS in 18m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fb90ef59d464e789e94abd7bd63c8a4 : SUCCESS in 13m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/336223fdfca2426f8047354e869dfecf : SUCCESS in 2h 12m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c4d2bfd22404704ab30f70efa78cf48 : SUCCESS in 1h 52m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b9412cae1ff142c28c1935a7b864c99e : SUCCESS in 19m 25s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/02b343612a3d45dabbbc7650689e2896 : SUCCESS in 20m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4e0c7b9e8174cfd86d4f979b99ff14e : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/ac08455ae37a4cb9bb760be630eea435 : SUCCESS in 11m 56s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7d51a167771b4919a9e05cf8cc3d8cc3 : SUCCESS in 25m 20s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/17b4ab4478d34f97b00519f3ba430dc3 : SUCCESS in 1h 08m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1f908b2e33fd4fce8791ada8a7b17161 : SUCCESS in 33m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8884a0af84964337a3d9870fb14bc5e1 : SUCCESS in 24m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/de990361c9f1440698b758e9119a5232 : SUCCESS in 17m 25s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/bfa6ccaf9cbd434796557f8c64a5ada5 : SUCCESS in 11m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/225b93c9a5854ee3b94a49b846ab95e9 : SUCCESS in 11m 26s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/65cad93b7ecc44849e3ad99bcb564944 : SUCCESS in 10m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f22124dc6b04435583db02a137b29765 : SUCCESS in 21m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7daf885587464fc0b7d0b100e219e7e4 : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3ea0ffb09f1f8b5863679632d56c238d8d712ed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-24 08:47:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cef389fadcd4d3fc6f8071be559987efdb9bf98b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 08:47:44.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Job kayobe-seed-images-ubuntu-noble-podman not defined\n\nThe problem appears in the the \"opendev.org/openstack/kayobe\" project stanza:\n\n  project:\n      queue: kayobe\n      templates:\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n  ...\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":2},{"id":"47b9ecb99dccb00ddc26b7749cfabaee2e7c041e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-24 09:10:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"96eecd05e913c1d817254505e54e15408391a985","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-24 10:29:58.000000000","message":"Abandoned\n\nThis is a dupe of: https://review.opendev.org/c/openstack/kayobe/+/909686","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"94d5018328b562d2be66f5815515ecb60722be00","revisions":{"6bc633b2a3e3af0d53b3e9bcc121a521ba5af80d":{"kind":"REWORK","_number":1,"created":"2024-09-19 10:28:41.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/929902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/929902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/929902/1"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-19 10:28:23.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I1c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bc633b2a3e3af0d53b3e9bcc121a521ba5af80d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bc633b2a3e3af0d53b3e9bcc121a521ba5af80d"}]},"branch":"refs/heads/master"},"74df718580b82861b6af5a304d801b773698c7e7":{"kind":"REWORK","_number":2,"created":"2024-10-24 08:47:22.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/929902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/929902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/929902/2"}}},"commit":{"parents":[{"commit":"9b53c0ee95c02d5d945c7f690f77ce53bab95145","subject":"Merge \"Fix group and mode of /var/log/journal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b53c0ee95c02d5d945c7f690f77ce53bab95145"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-24 08:47:20.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I1c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74df718580b82861b6af5a304d801b773698c7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74df718580b82861b6af5a304d801b773698c7e7"}]},"branch":"refs/heads/master"},"94d5018328b562d2be66f5815515ecb60722be00":{"kind":"REWORK","_number":3,"created":"2024-10-24 09:10:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/929902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/929902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/929902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/929902/3"}}},"commit":{"parents":[{"commit":"9b53c0ee95c02d5d945c7f690f77ce53bab95145","subject":"Merge \"Fix group and mode of /var/log/journal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b53c0ee95c02d5d945c7f690f77ce53bab95145"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-24 09:10:24.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I1c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94d5018328b562d2be66f5815515ecb60722be00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94d5018328b562d2be66f5815515ecb60722be00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
