)]}'
{"id":"openstack%2Fkayobe~930204","triplet_id":"openstack%2Fkayobe~master~If149418f509c687c3e5d1072cc39a80af33dec5a","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-09-27 13:05:33.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-30 12:23:27.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2024-09-30 10:30:18.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-30 12:23:27.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-30 12:23:27.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-09-30 12:23:27.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-09-30 12:23:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If149418f509c687c3e5d1072cc39a80af33dec5a","subject":"kolla-build: Add support for cross-arch builds","status":"MERGED","created":"2024-09-23 14:17:01.000000000","updated":"2024-09-30 12:24:33.000000000","submitted":"2024-09-30 12:23:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930204","meta_rev_id":"94c15d657b886411a9415c1cca78bcecd4574473","_number":930204,"virtual_id_number":930204,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-30 12:23:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-09-27 13:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-09-30 10:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-09-30 10:30:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-23 16:13:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 08:46:40.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-09-24 10:12:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-27 09:24:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-09-27 11:26:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-09-27 15:31:48.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-09-27 15:31:48.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-09-30 10:13:51.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"19618e84386c4b39eedbdd1667807325dfb7621a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-23 14:17:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08f65b4c7eafa7cc70524bd22e180ad5f4068f68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 16:13:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f1b887b58b546b8837cbc412d477066\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fe75857fbb84e98b090a74e76cc753d : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18cc9b89668546578bcea366f7abaf5e : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/927c3fd7d76b4acfa3d7cfdf2c29912b : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1262a80bdb2f4787b53e5ec4e79c1e7e : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4576fd605c3045eeb1265ed300dc6cd3 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4073ad8599e149b4acff0eda52ff5fa3 : SUCCESS in 5m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c1833cc71b249e9893a935d91d3d0f7 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19036bf7b8c24c85bd7296cc047017b7 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/11348e20f313483da82a3a970b2e52fb : SUCCESS in 9m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/14b1a18dfc1b4f87a2548e9df71dd11f : SUCCESS in 55m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f14376ea8d664786bf91ac8a176cff47 : SUCCESS in 1h 14m 37s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bb4f8f55e0e497f84d9bad55d0911d2 : SUCCESS in 50m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c66e9b1f5dc14dad9d79b4d9bce405f3 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/155ed839ad424b40bf5de78910fae04d : SUCCESS in 18m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5516c6fe65c84027ba064908ee3e061a : SUCCESS in 15m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4bcf499b7e44ee8a11d456896bf28f7 : SUCCESS in 10m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0ee92cbb304e4996a61b8e1c96d6f2bb : SUCCESS in 1h 46m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/926bd19334674028a8fb489f007f223d : SUCCESS in 1h 29m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/531a466ecf5840e2a11d39ee0ac1e250 : SUCCESS in 23m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a498108f4774b279a816b720d5f53b6 : SUCCESS in 9m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2686c0994634400c95d8d772bfc33503 : FAILURE in 34m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ab1935fa13d04826ad587e8ce2d7dec0 : SUCCESS in 32m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/238c4fc7cbfd4dd38653428c25d5d013 : SUCCESS in 23m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fe53e3439a0f43c780b37ca272121742 : SUCCESS in 14m 49s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/26fff60f2d1e4e5f8c574ca8e44a2240 : SUCCESS in 17m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16ba6555affd4de2b280cd0d463f4a3b : SUCCESS in 10m 21s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/3000eaaaee0f4590bba44a524621e6a5 : SUCCESS in 7m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a990c9e06174f2c81fc1b64537efaa6 : SUCCESS in 19m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31f0670150144148926e93ed2b0c3656 : SUCCESS in 16m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d62e716e1a55bf8c5ab8184e9d0ee33f31596bf8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-23 16:29:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7d0fe8fb3dc5daf1d065dbc5aa20b50473ca5b59","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-23 16:33:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9cefa8f4414a9fd016c20f453f7762d40e3f1ca7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-23 16:33:38.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a6da5f4cf588f97613c3018cc1b5fe324ff4823a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 18:17:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cf813337fae4f638109466b0702e02f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a06a3ff1c0840878425221905f160ae : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/992ad6b8d8ee4b33bc4925652e5cfd5f : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85d7ea8ddbfd48b1a6eb2b04bb62a0a8 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b9d3b8853f940d39bb183fbdb62d84a : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2ee8f2927ff4582a35185f15a3dea62 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bdeb78205744df9bcdd62d517e065eb : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dde7b3549c0e48f5bc69e42aca24b58f : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd99a2d5a6f14c3e9221cdba0400b31d : SUCCESS in 8m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/74d0518bdeca45e38199540dd2ae355d : SUCCESS in 11m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f456bc72477e46e3b34b7b25adfff504 : SUCCESS in 56m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6773784f47dd4d0eba6293f4cd525d2d : SUCCESS in 49m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99b408e5d9254c1780851d2645f8d047 : SUCCESS in 46m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/47b7f6502e2f437eaf35217b525fdcd2 : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fff88d1113ee4732bfa443d62bfa8ee1 : SUCCESS in 21m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/075931fa07e94db9aa5f86b6b878fd84 : SUCCESS in 16m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3070609f99894e839c1f477c0441f5d7 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/14141d2655af4e76a693879236d9a75a : SUCCESS in 1h 38m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b7a49e1454b4893b19d068970988c8f : SUCCESS in 1h 29m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/96a77152af2644959e6614102b5057bf : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89700556635e42baa1eafe9fe0f8fefb : SUCCESS in 14m 04s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0ea52ae57038432c985e776f87b4c9e3 : SUCCESS in 35m 58s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dc019e04bfee4817b3c9ac3ab3cf922d : SUCCESS in 21m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e51bcee101ef46d4b71205caea9b8416 : SUCCESS in 22m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15220b24c5da4a439ed35a3edefaf288 : SUCCESS in 16m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/41f937f37d594681be6ab31395dffb19 : SUCCESS in 15m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/407d33bd6f5942898c27ba4883d8305c : SUCCESS in 10m 07s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/331a1614ed9148fb99c818f861e0ffca : SUCCESS in 12m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/de89822f1483411ea54c2b8b132e9602 : SUCCESS in 19m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87eb90b66b924d868519daeb269a6df0 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":4},{"id":"7550afd1fe2739b30c83621635c31bb03fd76d3f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-24 08:45:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4104a9c423187658ddcb73bace0068f365e1563f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-24 08:46:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"2dd511fe5894974a2e5fe37421870f9d3d7b7399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 11:32:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7edac892cef144d69c0aa5a35ea2cea7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61356c1b22db4136a0c43d9fb4caf2ee : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201c5bab61854f5e9476168a5d019041 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0f24a2cc5cb49de96c017b06d39d03e : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c30161b160f94fa5a2278cc81fe25fd3 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24edb2a6c26a4cc3bdc4028fafcf4f73 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b815b4cf9c6a4dbdaed5deabd73419c5 : SUCCESS in 17m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8969a9d23bf84339b7111bbe880bddf6 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d60f3172d564a92b6a430cc6c80a3a9 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b1b28a786a44dcfabcd6766f77c89f0 : SUCCESS in 10m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/24bde336ba13499f8bbade605bd7d49f : SUCCESS in 55m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/594fc366089c413ca62c49add20fc6b9 : SUCCESS in 53m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d5839b771a24a538c0e954028b6b6af : SUCCESS in 49m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b95c61967d474e9695d2cb34986149be : SUCCESS in 58m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/80cfc8b4aa5848ea9fd3979eb115cbd2 : SUCCESS in 24m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8489eff3b3c649b6a47d6a7aad868797 : SUCCESS in 17m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33c1fff8274c4639b523665efdcdb57c : SUCCESS in 13m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f3d107ed4f06486fb0803cc909457fe7 : SUCCESS in 1h 54m 09s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbc07d509c01475d8d992809052c7f8c : FAILURE in 1h 04m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/65d234a604fe4870be8d2e9b0de0c33a : SUCCESS in 20m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0e1b32d5a4943269b7d1449d5b92f48 : SUCCESS in 14m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3e91ec3054614763aa2d184feaf3ca5d : SUCCESS in 37m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a1b893c8c494a5983241912a58eb1b8 : SUCCESS in 36m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ba4830321c24ee99cda8e4a2a55111b : SUCCESS in 24m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e517774d4c3d43b1942241736ed47f83 : SUCCESS in 22m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e79d7391e296480ab615833c657abaa7 : SUCCESS in 16m 46s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cd361c2c9134e1eaac294b7a84bb46e : SUCCESS in 10m 52s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/28fd29cc31af4f9889c4a9d24b19a44c : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4d323663c8644191bf1eaf96de6c9042 : SUCCESS in 22m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a74eb00545b84e129782608fc0d4ab5f : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":5},{"id":"2bb30e97266648ce3cca31cedea9af986450b46d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-24 11:46:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"59b15e3bc5744ea644c72fac94c61616daf40d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 13:50:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de814301b89847b691b6e4d6ce5ea7d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cfd209606fd4355bac465bda2f99dd9 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/601ab4865fcc4904b668f965bcbc1d47 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20e6611e2a224cad838a99a062609c2a : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef13f9eb8d324a99af6577bfcb876403 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aca6e1c31fc14ebda6155aa57e52b153 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11dbcfec18cb4791a3a286c1317942ab : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c6c940c1af404afd8970e69bcd22316d : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b54566800a94ff49ced4f5f31eb29e9 : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12ea25d3ba8746b5987d785b8889bd68 : SUCCESS in 10m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ff4334a00a2c448c8da51b3e8726582f : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d1807249590f496fa182226eba25cf1f : SUCCESS in 56m 59s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d52c6c888ad4c658b365892b99bf2eb : SUCCESS in 31m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ac705ef15cca4842b95bd462916daca3 : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/23ae59ca902a481484f546245756fe7e : SUCCESS in 14m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4f2183f83a264d51933edfb876675255 : SUCCESS in 16m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c0ee0e104f14f47b3730b6ec8948276 : SUCCESS in 12m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0fa1bc9e5a9b474f9070e06ac99f020d : SUCCESS in 1h 53m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/428e569acd254fe282abe237959402d5 : SUCCESS in 1h 33m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5ac38e9aecaa44b18c7f5ed8abb975c5 : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2eefaa7368a546d28d68f3e45c837bcd : SUCCESS in 14m 45s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1591b68eaebd49718b805bd3e35c5168 : SUCCESS in 45m 52s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/246c8875beef4485a2940f8246ec5dc1 : SUCCESS in 36m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f98405c683c4e9bb112daf9aa0cdc0c : SUCCESS in 26m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/17ac4bfb47aa435593bbf587f9998213 : SUCCESS in 16m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/cfc159f10de5427c898716fec54ac845 : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0685ce4bcc954e74ba730e110544ac7b : SUCCESS in 10m 12s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/995710f6b947446284f0f7f85316f945 : SUCCESS in 9m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/03cc7bf67fa44f168e96769a30e6a61a : SUCCESS in 21m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc282a5ca98843e7893c51aca2977bf4 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":5},{"id":"b077c13024523af92fc33dee34af0ba21c7ad66f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-26 10:04:59.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"50cc6a5c3cb41dddea3cedd43e0df94f68def957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-26 10:13:03.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"d336c45b2a3cf331014d56f67591f0361fef3da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 11:48:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407fabaf804e4a8582aae6dbd4b8ed24\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f91636e3d6bd4d2ead98b618720c2fc1 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915229f4c24c4c60aaeafa4b7e7237a1 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b10a589e8f14ff181cd9f4d5ffa278b : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa103472787c4205a4f96cd2a7413549 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ccb43c4ffe641cdbbc708b389748efa : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5e2b0a2ed644269f117c63b858bfa1 : SUCCESS in 6m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62460c80263c484d9cf537b5c5294e07 : SUCCESS in 3m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/595de7766e6f4821800e65a5f7a49b27 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a99145b891c04e4c812e6967649d80c4 : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/edc45b5d2cda4db5beb43eb81a8d5e45 : SUCCESS in 57m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/91ea220997ec4bc7850176ac58b4361b : SUCCESS in 34m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b136495a64024b42b15b84e6e49143e5 : SUCCESS in 46m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7f58312a39b74ee8b6d26ded0f90af5f : SUCCESS in 39m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/621917ece24e463d83064e6939336a6e : SUCCESS in 18m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e0168112a5f64e6e96f0325c0cfbd186 : SUCCESS in 14m 56s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5813c3df3f8f48a092f2be5352b3a2e5 : SUCCESS in 13m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9915cb70500944c494110b6c5eb5009d : SUCCESS in 1h 34m 22s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc68ff1771a24a86b1e5dcbdb45184b8 : SUCCESS in 1h 24m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/77f85bfe9c044a829d71aa9608e1a406 : SUCCESS in 12m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcf1c28b69d648358ff2dbcc5d0a018d : SUCCESS in 12m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/23f54e8c3333435285b5fc922f2e00ac : SUCCESS in 40m 33s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4c93539cf203481b927bec65c8388afd : SUCCESS in 27m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79fb9ba6c2694185abef5ba3fe7d59f0 : SUCCESS in 14m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5abeb221afb645848fa5f59bf6f6a034 : SUCCESS in 15m 42s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/521a7eb9edd1443cbe10dff02d0c57eb : SUCCESS in 14m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d797f0442a4b4aefb531725737e2318b : SUCCESS in 11m 03s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/574d1567653946e59a96025a76ea66f4 : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07059ba17de642fcb78f59a1646c1c0c : SUCCESS in 21m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cbdd373d4e94b708a3217688d6690d9 : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":7},{"id":"e2ef89bca5e7ffe48284385393239c89766d78e4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 09:24:09.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"70419ddd515b2fe88f85520baddc3aaadbee4eaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 11:24:45.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"fb950bd0af9e3b9b32d7c1d2fc89e678b7f5422c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 11:25:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5ab7ea62d86652718b7c02e32395a4c3d948f747","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 11:32:22.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7bf57084cfd25a669f98f64abf13fb432974c60f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 11:33:46.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"a8205a739dea16c0bf020563232676a4477a9afc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 11:34:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"448afc4ee6f83251c6ebeb3e340db16dd887a98c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 12:28:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d61e720efc4c14cc7a3682624ba659da2cc5572a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-27 13:05:33.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"acbfdf76d04fc0bed88607d7a759ac6056fb2a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:05:18.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99580a99b2334fcc9231947497ceb331\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c1cf27b6d88444493180188615c5cd3 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c969b0c3a4c04aaab73a2267c5835a47 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53b6e8b89e4f44afa765c8965ae456dd : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/015a300ba69e41acb8685dbd4fd226dd : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21aa5962c9824f6eb325452a7beb477c : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1510c795924f31b72b1220e3179cd4 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffa49f46030c4a599237ba92d7a282b6 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da8758d8d3774e98ba6f2c066f8647e2 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0c66055904748dbbbff5e1b34e30b5a : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8029bfbbdc394f129073f93d81621cea : SUCCESS in 9m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/83358f011de346bd8ecac2eae23a7a39 : SUCCESS in 39m 08s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b011039dbcf848c9b35f743a5bb1bbaa : SUCCESS in 57m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb24ded5d4904864961a913f9e8fa559 : SUCCESS in 32m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4e1a9e370ac64b77bbf8d6965742d6c9 : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/394ff2611ca640ffb433eac10c0e44d3 : SUCCESS in 18m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/444a2441108d444a98a06a0a6b7baadf : SUCCESS in 17m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8129d67e024a4de68b666fc6acad04af : SUCCESS in 13m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/01639fb7b6f24e64860593858379d40e : SUCCESS in 1h 15m 48s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c05d42822b954e34ab82e3dd1d0441d7 : SUCCESS in 1h 35m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/efa86dcdccb842bcacf04d295aa97a0f : SUCCESS in 16m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/171517de9095442c80e06f65b5d70704 : SUCCESS in 12m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/43616b0852a04e5ebb707070d72fa454 : SUCCESS in 40m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aaee35f1634948fa9168866f60f5b4cf : SUCCESS in 32m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8aef5da76f245a69a29f0a11286c668 : SUCCESS in 25m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f450bdd34db44fc398139cef665a6b4b : SUCCESS in 18m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/705829b1c14045249394fa2d73f7bbb3 : SUCCESS in 15m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33d150da7f8a4ee087207cf32b9d2df3 : SUCCESS in 10m 24s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/9a3a5121da344a088e67a81f5a364e3a : SUCCESS in 9m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/01682110e06b4913a2123ae9868a2f96 : SUCCESS in 20m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/440f4afaf19c46f18c0d456beb0320bd : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":11},{"id":"59d58569f0986168c122b1654a048f2d70f63977","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-09-30 10:13:51.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"60e065ff542bb59b9bcc5eadd2a9c8b955f37267","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-09-30 10:16:40.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"52342887d24be7a05e93c5e3f6d4a8a9ccb0a2db","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-30 10:22:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"62769f8234e871ea91d8866ffbd3a648706ff4cd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-30 10:24:11.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"52764a766d0a7dd70b857f2bee0d40cee13e9b0a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-09-30 10:30:18.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"77caa4ce860dd6e7e98496021252fbe15ac04577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 10:30:42.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"8c47d4782c8de9fbe12667dfa9717c94cb2f3633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 12:23:27.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02707dfa17f94ead83453aa4b21b7868\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0febcb24babc416eb20743bce397ec2d : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7477488106340dab8dc47e7c6ba3069 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd1a58ed29454e21b5cdd8c5b6775113 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/800e479272bc408182949de16c7258f2 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95419ac5e4874ae1bb59036615b044d7 : SUCCESS in 2m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c1810459d8734ed5a3879154f93b619c : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8919c5eba924cfdbb2c19fd92a25f43 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3bba82b23c534840a4d662c0ff6c106c : SUCCESS in 9m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/90ad5b0f0522409eb951085168297854 : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e140ff07411477999724016b4832831 : SUCCESS in 45m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b7196e5930aa446095d635c1e3818e84 : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3116c14554274f6c857f07958c92bd22 : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83d08fa7fa17495ab48eb1b416a4e720 : SUCCESS in 14m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e79221838b543568c62a9475e468ec0 : SUCCESS in 1h 51m 28s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c61b88189be6420e9897ea82bb72276b : SUCCESS in 1h 38m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/659c88d470e743d99674a127fe761fcc : SUCCESS in 23m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f88000880864b7989747a1ae0248c1d : SUCCESS in 14m 38s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6417fbb6abe14e0f809952c909162e85 : SUCCESS in 35m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3555360ed12846b99fda1a9100444880 : SUCCESS in 29m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d5e913f3567a437faea0d801608bc7d0 : SUCCESS in 19m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03657cd94d864d5fa50646586668e144 : SUCCESS in 13m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/60efce535b3c4ecaad65e26fe886f570 : SUCCESS in 21m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4698406da434458da4ecf189d3545cf8 : SUCCESS in 17m 35s","accounts_in_message":[],"_revision_number":11},{"id":"b7cc19bf0e810d989f26704528bd627cbfa60b0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 12:23:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"94c15d657b886411a9415c1cca78bcecd4574473","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 12:24:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e0df7e956434b5189d132c12325c559\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740daed7a6f64f13993ccfa36aacaada : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/383e5e5911fd4e05bd0d11caabd73fbc : SUCCESS in 40s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"996651b478bc2bdcb0c41e0f59ec55218bcdd87f","revisions":{"b1737f6845f057de44664c0873091afecf298fad":{"kind":"REWORK","_number":1,"created":"2024-09-23 14:17:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/930204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/1"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:58.000000000","tz":120},"subject":"WIP: kolla: Add support for cross-arch builds","message":"WIP: kolla: Add support for cross-arch builds\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1737f6845f057de44664c0873091afecf298fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1737f6845f057de44664c0873091afecf298fad"}]},"branch":"refs/heads/master"},"49dbe1b6039d2e9e4ebd3605ae1b0d353c305c08":{"kind":"REWORK","_number":2,"created":"2024-09-23 16:29:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/930204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/2"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 16:29:34.000000000","tz":120},"subject":"WIP: kolla: Add support for cross-arch builds","message":"WIP: kolla: Add support for cross-arch builds\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49dbe1b6039d2e9e4ebd3605ae1b0d353c305c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49dbe1b6039d2e9e4ebd3605ae1b0d353c305c08"}]},"branch":"refs/heads/master"},"667448305679d082ff80523402e16bf26304882b":{"kind":"REWORK","_number":3,"created":"2024-09-23 16:33:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/930204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/3"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 16:32:57.000000000","tz":120},"subject":"WIP: kolla: Add support for cross-arch builds","message":"WIP: kolla: Add support for cross-arch builds\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/667448305679d082ff80523402e16bf26304882b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/667448305679d082ff80523402e16bf26304882b"}]},"branch":"refs/heads/master"},"47c5d124f4c5104167cf64fee89526e085b19586":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-09-23 16:33:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/930204/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/4"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 16:33:26.000000000","tz":120},"subject":"kolla-build: Add support for cross-arch builds","message":"kolla-build: Add support for cross-arch builds\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47c5d124f4c5104167cf64fee89526e085b19586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47c5d124f4c5104167cf64fee89526e085b19586"}]},"branch":"refs/heads/master"},"8e6f00953f0a2520635c1ebcd552f6741865cebd":{"kind":"REWORK","_number":5,"created":"2024-09-24 08:45:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/930204/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/5"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-24 08:45:51.000000000","tz":0},"subject":"kolla-build: Add support for cross-arch builds","message":"kolla-build: Add support for cross-arch builds\n\nWhen kolla_base_arch and ansible_facts.architecture differs\nwe need to run multiarch/qemu-user-static image - see [1].\n\n[1]: https://github.com/multiarch/qemu-user-static\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e6f00953f0a2520635c1ebcd552f6741865cebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e6f00953f0a2520635c1ebcd552f6741865cebd"}]},"branch":"refs/heads/master"},"2272374ee6780b20dbf64061cc9a5265067a6d5c":{"kind":"REWORK","_number":6,"created":"2024-09-26 10:04:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/930204/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/6"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-26 10:04:59.000000000","tz":0},"subject":"kolla-build: Add support for cross-arch builds","message":"kolla-build: Add support for cross-arch builds\n\nWhen kolla_base_arch and ansible_facts.architecture differs\nwe need to run multiarch/qemu-user-static image - see [1].\n\n[1]: https://github.com/multiarch/qemu-user-static\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2272374ee6780b20dbf64061cc9a5265067a6d5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2272374ee6780b20dbf64061cc9a5265067a6d5c"}]},"branch":"refs/heads/master"},"da64b945b483c316b73238e6ed4255316782a6e1":{"kind":"REWORK","_number":7,"created":"2024-09-26 10:13:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/930204/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/7"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-26 10:13:03.000000000","tz":0},"subject":"kolla-build: Add support for cross-arch builds","message":"kolla-build: Add support for cross-arch builds\n\nWhen kolla_base_arch and ansible_facts.architecture differs\nwe need to run multiarch/qemu-user-static image - see [1].\n\n[1]: https://github.com/multiarch/qemu-user-static\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da64b945b483c316b73238e6ed4255316782a6e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da64b945b483c316b73238e6ed4255316782a6e1"}]},"branch":"refs/heads/master"},"41c690f48767c3bc312aaae2f84041d784da5250":{"kind":"REWORK","_number":8,"created":"2024-09-27 11:24:45.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/04/930204/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/8"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-27 11:24:43.000000000","tz":120},"subject":"kolla-build: Add support for cross-arch builds","message":"kolla-build: Add support for cross-arch builds\n\nWhen kolla_base_arch and ansible_facts.architecture differs\nwe need to run multiarch/qemu-user-static image - see [1].\n\n[1]: https://github.com/multiarch/qemu-user-static\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c690f48767c3bc312aaae2f84041d784da5250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c690f48767c3bc312aaae2f84041d784da5250"}]},"branch":"refs/heads/master"},"f6c8a0b1da5715801883b15d9c6f1f56954d9ed2":{"kind":"REWORK","_number":9,"created":"2024-09-27 11:25:36.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/04/930204/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/9"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-27 11:25:31.000000000","tz":120},"subject":"kolla-build: Add support for cross-arch builds","message":"kolla-build: Add support for cross-arch builds\n\nWhen kolla_base_arch and ansible_facts.architecture differs\nwe need to run multiarch/qemu-user-static image - see [1].\n\n[1]: https://github.com/multiarch/qemu-user-static\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6c8a0b1da5715801883b15d9c6f1f56954d9ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6c8a0b1da5715801883b15d9c6f1f56954d9ed2"}]},"branch":"refs/heads/master"},"78d15e2658dc60c7c02d8fd478599a4e51d866df":{"kind":"REWORK","_number":10,"created":"2024-09-27 11:33:46.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/04/930204/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/10"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-27 11:33:41.000000000","tz":120},"subject":"kolla-build: Add support for cross-arch builds","message":"kolla-build: Add support for cross-arch builds\n\nWhen kolla_base_arch and ansible_facts.architecture differs\nwe need to run multiarch/qemu-user-static image - see [1].\n\n[1]: https://github.com/multiarch/qemu-user-static\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d15e2658dc60c7c02d8fd478599a4e51d866df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d15e2658dc60c7c02d8fd478599a4e51d866df"}]},"branch":"refs/heads/master"},"996651b478bc2bdcb0c41e0f59ec55218bcdd87f":{"kind":"REWORK","_number":11,"created":"2024-09-27 12:28:32.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/04/930204/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/930204/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/930204/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/930204/11"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-23 14:16:42.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-27 12:28:21.000000000","tz":120},"subject":"kolla-build: Add support for cross-arch builds","message":"kolla-build: Add support for cross-arch builds\n\nWhen kolla_base_arch and ansible_facts.architecture differs\nwe need to run multiarch/qemu-user-static image - see [1].\n\n[1]: https://github.com/multiarch/qemu-user-static\n\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n\nChange-Id: If149418f509c687c3e5d1072cc39a80af33dec5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/996651b478bc2bdcb0c41e0f59ec55218bcdd87f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/996651b478bc2bdcb0c41e0f59ec55218bcdd87f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
