)]}'
{"id":"openstack%2Fkayobe~930277","triplet_id":"openstack%2Fkayobe~master~Iab3f982ae914c19d9c3d9ea6a1634179f85704f9","project":"openstack/kayobe","branch":"master","topic":"previous_release-bump","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-01 13:58:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2024-10-10 19:35:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iab3f982ae914c19d9c3d9ea6a1634179f85704f9","subject":"Bump previous release to 2024.1 in Dalmatian","status":"MERGED","created":"2024-09-24 08:35:34.000000000","updated":"2024-10-10 19:37:35.000000000","submitted":"2024-10-10 19:35:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930277-previous_release-bump","meta_rev_id":"8426709b9951bf1df406a162fc760c36324005f3","_number":930277,"virtual_id_number":930277,"owner":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-10 19:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-01 10:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-01 13:58:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-01 11:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2024-10-10 13:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-01 11:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2024-10-10 13:50:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-24 10:12:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-24 10:52:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 12:34:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-10-01 10:54:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-01 11:11:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-10-10 13:50:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"32abd8dd0478dcab167b5dc6e902a862ae7834e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-24 08:35:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"842ca7d7a8d4966fef4add67328f8ff3418554f0","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-24 09:02:06.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"70374caad77b7e4aa32f4ae2068d883ae1329f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 10:52:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f511022b3aa4cbdb9ae9ac0dfe5b56e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea856249fe294cd6ae47854abe4aad7b : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc8f0532e4e34375991f0869c56b6fe4 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1945ef1935ed43cd853a8506c625b45e : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/607f705c1702415babbf7ee3a7928588 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10fe89bd9f0a4c6490a7c403deadc1ad : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129fd4e569224de2b9da9bba96f51d0e : SUCCESS in 8m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/186e024851f640519711ac6e6cf6ec32 : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3f0401891f1a499891ebebb7ed46ac95 : SUCCESS in 50m 56s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeb43511c3ef494ba5d57bb0822cbf12 : SUCCESS in 50m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8a3892760d1641d4bc58ca1b3124707d : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/108acf3b439b40299653ca3b57d6f718 : SUCCESS in 20m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/352e336ae00945fcb6e9db321c18f916 : SUCCESS in 19m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ec6ae2cf4d94f9cb9021c82ff73033b : SUCCESS in 12m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bac417543d0541319a8eac920c589f41 : FAILURE in 1h 15m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e83f9b0acde84c998a9cc7a663b772a3 : FAILURE in 1h 00m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4f82218444bb4d05b6ff992dff315ec0 : SUCCESS in 18m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19d0d9b97b1841d7a611881c3b797cb0 : SUCCESS in 12m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/942be99186db4d30ae709c693ea3e3dc : SUCCESS in 42m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6467a802923341328357fde83ae8b15f : SUCCESS in 31m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c9f25eb63a04dc383611e121d17c71d : SUCCESS in 28m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2e1774f2ab74e3eb4f6bbd8c05f46b9 : SUCCESS in 17m 40s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4e95978429d04eaaa18fed4425f3f564 : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5eb1ece063cd4225bbac740e4f5ce156 : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/b32bf607c6ab4c63be34fa26ac60c362 : SUCCESS in 12m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ae9663641314cbabde0279d7dfb0446 : SUCCESS in 19m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b79f87c3239344e3b079830105cbac16 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3c4c7c2314ac8e6325b68bd3b5ed44fbd84178c6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-26 09:01:47.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"312f9d265be88746368a80f3a3cf51e1541c0fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 10:14:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/796fceb5d1564cf3af59ff167bfa4cc7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38621421179940d5b31b714fd2a257ce : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de798ba231a4498a9c8783950f80ceb : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/619e2718321942e58aa4a07728a19394 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e39fabda5654b13aa72cca52640c916 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b745bcc7b5cb41f2ab87cde706b240b2 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2d569bc2a3442ab211fca0ee9316fd : SUCCESS in 7m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f28256974949484aa3730b071a55b86b : SUCCESS in 59m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7d1f02bca97e45aeb6349998dfb116ad : SUCCESS in 57m 40s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/307f7bdbe31d40b08b6e8aa2eb278c56 : SUCCESS in 50m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3f6c1487903f431c856812c7441f39cc : POST_FAILURE in 1h 00m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c111e283d8654ad0b8c7fd0576562741 : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/671c07008f4f4edb8a63bf40c1540ebf : SUCCESS in 15m 49s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49bda47126724dc8aae43474bc5f9828 : SUCCESS in 13m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/29c526e0d25b454b874d183725044029 : FAILURE in 1h 12m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0453bea78a184cfe9d49a01689ea081f : FAILURE in 53m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e995c27184bf416d9b7a456991cf928a : SUCCESS in 13m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec287be59ce54bddbc898376ac9c8124 : SUCCESS in 9m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8fad16cc4aac4feca6ac9b335f1ff394 : SUCCESS in 43m 50s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/941f5834e2484bd58f19fc72c3c31f36 : SUCCESS in 34m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b5f8bbd30e94c529947d3f2d3262088 : SUCCESS in 17m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5d47cb97d9374182ae164133c80f5d19 : SUCCESS in 15m 36s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/7cb606dbd74f48d9b5f83d2f56de07fb : SUCCESS in 16m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7093723ae7440e1bbabacac2b8f07fe : SUCCESS in 9m 34s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/d61c856d513249a7ad48e24ed4a35eb0 : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/639848d3cecb43d1a07802ca1529c60c : SUCCESS in 21m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1457fd299bb74b02a756a719c56ec302 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"30dbd94805bd0fe6c77c8ab898172286f4cfc4f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-26 12:32:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"56714937a3b9b3ec4417dc13548c0f8ef090f89c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-26 13:22:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7a3f926cb5963a7d6364e97381919a4555159d8e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-26 13:23:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"d5ae902040a97a2d2e746e7ed12cce8c918fea06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 14:40:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfddc4794dee4be9a3a258588597a0bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e96be15fd9b24ce280eaa4703fa66807 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246a476ae3c8428788ac1ee8a5abee1d : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbf2b64f0b504f9a9987c7a52136dc2e : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4009158008e14569b50374cbbcea5ff1 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45d3251cf9a14225a5b4f65062e676df : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74313c6e82c54d5b94d3dea2ba57ef52 : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f44aea720ab442b6963be5c8a7d987bf : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/161b504a77394f28b35e473ef82aca13 : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f882169bba204631b4221065e78fb836 : SUCCESS in 9m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1ab4636c40f54fc7bcbc9fb76c9320f2 : SUCCESS in 58m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e8c6ba32a0d34c05b750a6fa7d85ab24 : SUCCESS in 53m 04s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fced9646933b421e8dc900fd0830e6eb : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/84ee9c91382f4a519a6de39b8307a3c8 : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fae1f8e5bde3403ebba3f2d2ad3844e8 : SUCCESS in 19m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/674e4bdbe6364ee8b421e59d6d1f580d : SUCCESS in 15m 56s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9d0010df16b4085976f2c943058291f : SUCCESS in 14m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9494fa149d7e48b2a397819538a6e3bb : FAILURE in 1h 11m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a6f5c4b37364f53bc170225bef0a36b : FAILURE in 17m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fc03bed1840343ae984b5d101b92d6d6 : SUCCESS in 20m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68a479edc82a47c8bed07efbe6220c20 : SUCCESS in 14m 59s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3d0f5ace36794c14bf915beb378c5a4d : SUCCESS in 44m 07s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c54957477df9406985ddd59ad4398599 : SUCCESS in 28m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2baab1abfa14e3b82664fd8728db050 : SUCCESS in 24m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec5fbba375714e1ebbcf55d3bbf14851 : SUCCESS in 17m 41s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4f64772d96ab4a38805d3d605f1701a3 : SUCCESS in 16m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc79692acdd4494db6630554ecacd094 : SUCCESS in 10m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/8d7b2dd2a5f04b2e96be7b63dfd3ae50 : SUCCESS in 11m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d8ade39eb58e4f3f95112631856b8ef0 : SUCCESS in 23m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eadc66498d234be8a525a94186878ce9 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":5},{"id":"930ebfefd460e9d30eb5ba085eb4ce05aac2c34b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 08:43:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"79b0b33e9544442624165bba8639e07f07b08a24","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 09:12:08.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"e32b0d0cac6ce13c320c1eb5ea33558e8cfdcb50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 09:38:21.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"542aca9f46001d5292df0c486f6644d8a80e3f99","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 10:16:14.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"ceb60dd390d6149cb0f7415b79a3e3ed3abb2ab3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 11:26:52.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"3a9e8f816b0d3fcff6355d4389bae2bde2c23662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:38:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/baeb93f7834e424aa8248a97d9c1e1b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa5908789f724bb9a16d1b00a11ea30d : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a296107743874d36ae21d30341de08b0 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29ebd8acf54c4be782ad1b4fb1ba87a7 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c413353193c84020ba385eb04fc6c8fb : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b985db02c97b486bbd6eeb28e912d88a : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b023287e17a4d1f90a36aa5f917c8d5 : SUCCESS in 6m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cd3db03e82c641f19f156ba87468ca47 : SUCCESS in 1h 00m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/aaa34e9f23af4878b18c4b0a857ce209 : SUCCESS in 51m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/284e8079fb4644f7acca73c99d09fadb : SUCCESS in 54m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e6958587b09342bb9fbf52917aae8e78 : SUCCESS in 38m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9ff9c5aca93849bd8025c9b41c0c1d0f : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1744bd91db1643e893587f5b165564de : SUCCESS in 14m 52s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88c08bc207e24748ab8691a464ea7f9b : SUCCESS in 14m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ac6ef5231e224a7dac0f707b77c1ee38 : FAILURE in 1h 10m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11a90866541b424d982720ed3061a641 : FAILURE in 31m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f2542f43d93d4f7ba103931449c742e8 : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20e28af7d90046458beea7957207507a : SUCCESS in 13m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9faec4c3bdd047f49063a9f041b4cc61 : SUCCESS in 38m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/281ba86073aa4761908cc852e5e92439 : SUCCESS in 35m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e41e60d469da426599f22a67750196e3 : SUCCESS in 28m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c73dd86efe104988a2bc054fdeb63554 : SUCCESS in 14m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8bb5d390b2a14c3d832d853ddcce52fa : SUCCESS in 15m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7396c5a6c66414fa79bfa840253bbeb : SUCCESS in 10m 01s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/913fa40043a7487bade7aa50c7ac2c98 : SUCCESS in 10m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7dbf0ff65de9445b8c339c1792b572ad : SUCCESS in 19m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c95a33e8c3aa45f88c69fde042fb42d8 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":10},{"id":"32adcab3cfa31884941532d21c1054d61bac89ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-30 12:06:47.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"702939ef7c22f555da63dcd4e2ddce4c8f2afdaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 14:17:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fee94c70cf149cba26423540922bc2a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a12943c073e747688cd8938e0db525cf : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eca76b36c9141f08b945ae404c33961 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10303028ddcf4bff8cebbd600af23fee : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a64b8eeba714f13a6f2eacb6e17b154 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f3dd8f2d754a0fb58b0dc7f850f31f : SUCCESS in 6m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/02b5be19e40d4a309eccd259669a50dd : SUCCESS in 54m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0be7ff86f87f4b5b85cdaab04c1de8ee : SUCCESS in 1h 07m 18s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d313eaed13a46adbddad9c7e75bb39d : SUCCESS in 46m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/53d60acb52a54a5cb8f38f914f9f3eb2 : SUCCESS in 1h 03m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/27f9cb5f16bd4daaad7717afbbd48732 : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b63c33280d99475da0e61e88be128145 : SUCCESS in 15m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b0c0b6dc63a4527b52cfda442b6fff7 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b809fa8f2f3d423ba26251a4233e26de : SUCCESS in 2h 01m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/037068b7324a4277b835a101b121a1e4 : SUCCESS in 1h 47m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a05cfa14b4e147c88db11aeff7ce0403 : SUCCESS in 18m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc169b77905045de88641e0e31684bed : SUCCESS in 15m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4aeca10c169e4b64a78a2427377160a4 : SUCCESS in 38m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3063013256fe443799cdb8fc99b6822f : SUCCESS in 34m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/086ae12436a544c6a7c86f8171cdf11f : SUCCESS in 29m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15588626e7d345d596e53e0d1dfd279a : SUCCESS in 14m 41s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0b854b7ffb964f3a96cd48dfa4a0db0b : SUCCESS in 17m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/086aca48275b4e6e96527b4d6ef570e9 : SUCCESS in 9m 43s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/b76b3b32a7884ee28b50574b065aff26 : SUCCESS in 10m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b089cf37c1a94ecd859dbd3225183a59 : SUCCESS in 14m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e82a01d546a04d968e5cb16d2684af8b : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":11},{"id":"33b586e98e827287bbbbe9faf030f82fed4cee85","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-10-01 07:55:10.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"b717a138ca2d37bc0257b747db1747b6cf4de943","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-01 08:10:40.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3c8d6cc79609edf766610765ae58925867659a6e","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-10-01 10:47:39.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d95d1586d1cfb8771fbad19c354b7ed84571e919","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 10:54:33.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"9fec05fde6736fe7661768d2f792a7563d4f7c28","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-01 11:11:29.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"2d35bd5dd8729a037992143517d0b7e2c2523f55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 11:12:02.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"c5ffa9d4088106a22ba525b5ffedc1f629ab5128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 11:53:35.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/930691 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"aa4962558dd3b6fcdaa384cc74338644f06a8f92","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-01 13:58:59.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"523f33076b2116c8acfc050fd9d0740c2707980d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-10 13:50:59.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9e53092f7ad6d4ec49d6e6d180c6f2b277640202","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-10-10 13:51:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"402f94382f1bf10a9cf20f43534c42bca1598421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:44:40.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1fe8f53963b4c0c92ff472402bb6ddb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3487cf8793f4cd994c4e0c046489a09 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da787e56f57a4876a67e1ca994ce4a98 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68590ea015164869b096c2783d25dfde : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d6b9a63ffd147be971f36242a184d51 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39810961534746d6a40b8297ce051017 : SUCCESS in 7m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9d66d39e2edc46c3b1115ba0193e36d9 : SUCCESS in 59m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/610324c45b694b5e99583e02697ba666 : SUCCESS in 58m 41s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a65aa4532cfe417ab3d46770d4574177 : SUCCESS in 48m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fc17802d1b5b4ac5895d53b7b49e3168 : SUCCESS in 39m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/087d4172f5544da9b0c578f0f9dad0ac : FAILURE in 10m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4714ac1af8624dd6a542464d53ffd4c6 : SUCCESS in 18m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d77923da1a1d4963bb3ba7baf96d865f : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ebf653bc4d034e11ac95c1506ccbe7a9 : SUCCESS in 1h 52m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/134b3f7f3e964df6961dbe2468431773 : SUCCESS in 1h 37m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a2f9bc6f6c264e72943be8b211286f3a : SUCCESS in 20m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6a9500e75ee42b193d74d03d4e4c5aa : SUCCESS in 13m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1607d65aec2745f7a7f68f8a862a08c5 : SUCCESS in 50m 50s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/496e9254b8dc428c9a6f962afdc9b1cb : SUCCESS in 35m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce1b0ca861964a93aa1752251eba1c6e : SUCCESS in 24m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4aa7b80a9690478faf02dab76f767966 : SUCCESS in 16m 16s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8f2ff9b713744abcb4eb75dd79c4b9cd : SUCCESS in 16m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0b1cb32619f49318be7e671fe468b04 : SUCCESS in 11m 32s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/53fb42bebd2f46338b6db416643facb3 : SUCCESS in 10m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f6e155b0b33d4cff9a5b970684180660 : SUCCESS in 22m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a758b845c46e48f6bd783823ada0ca83 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":11},{"id":"487cdb12faf8641b802606257e3129323bacf6ad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-10 15:57:02.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a35e760c65edaafc4414b63808bf866c7557bd83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 17:39:51.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee9a8b752fed439c8d5c7d93f3bedf6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a88264c3448e4c43a75264944e521a9f : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e6de390167442780bf7e481f76eaa1 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5be5bec6aba24e63bfa024df0ca1e0be : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04566ac35b074cfe9cca86fff694ca7f : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29fa77cd2fc471a92575e234029ce2f : SUCCESS in 7m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/59461f6a278844a080698a2ac914f827 : SUCCESS in 1h 00m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/371eaa2499454492ad0bfbf4d6984172 : SUCCESS in 49m 47s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b35b5522c75344e1a4e63618a445d514 : SUCCESS in 49m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/814e9692ba3f4bbea726c603c3d657b9 : SUCCESS in 58m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/57e8f915ac2840cea849b145c728a6cb : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1b25aee4f96d42b785e9d5aa8203b5d7 : SUCCESS in 16m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9301cb52742447f806e5f59123276e6 : SUCCESS in 11m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4098a616ef92463fa52fdbe1539ee5c0 : SUCCESS in 1h 41m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18721eb7e5034f96a22dc236007272e4 : SUCCESS in 1h 28m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0f86c0da29e54303b2d31f180c2fd085 : SUCCESS in 20m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98c5a6bfbc784130a670650276f3adae : SUCCESS in 13m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1858351b850f47588a36cbe8d26e34e3 : SUCCESS in 53m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/44362c66268641b185b8da30fa59b4f9 : SUCCESS in 35m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/deb321d283554d26a6dc18b25e1db540 : SUCCESS in 15m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a12eec4d148549cf95191c92456de814 : SUCCESS in 17m 08s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4778bda19f1940bb9d9e66332c849907 : SUCCESS in 15m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7255badd9d744757af3c74066b273809 : SUCCESS in 8m 08s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/da21b1d0a69643a09558ddf86fe1197b : SUCCESS in 10m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6ba0551919754ecbaa7f24ab62139876 : SUCCESS in 20m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1be93c97173a4cd9b5876d2906b9faf9 : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":11},{"id":"6c0695a63e225b5c5540500ceec0f99600f3b556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 17:40:23.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"dd0efdc50f2c336760f490c019fa188e18b9859d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 19:35:09.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc2b839d40344ce0b8aa0ed2774343d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e927435c1ddd4a8090f2a1170f70070c : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d172a684e85c4fccb69c1e58ff558bf1 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82fc079c26fe4eaebd5d759f63c79b1f : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fb6cde90a14340862a6db87652e2bf : SUCCESS in 7m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/04ff13de30984c1f912ca4f09abdc618 : SUCCESS in 48m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c89e6cdcd7f42b7845558dc6a8c4cea : SUCCESS in 55m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7dd09f3d98c743f78ff9e9142ab4c4df : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3f2a37415388479dbb1ad65af8ea68a3 : SUCCESS in 14m 30s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18098b05ee244c8a88441694ebd04f6a : SUCCESS in 13m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5b76125be2f54371a690fb3c8725a401 : SUCCESS in 1h 50m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f582c4ccbb5d4af792d7f3492fc9e6d2 : SUCCESS in 1h 03m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f79862f7055344baad8ca5ca897fccf9 : SUCCESS in 18m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9804c18861d4376b875b843fb69dbf3 : SUCCESS in 11m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/234adbff058844afa1df8bc26dc36d2b : SUCCESS in 34m 33s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57582b8625ca4062a4a596cb4112e9e8 : SUCCESS in 22m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a7d4acde903d4ded87151b275964eb72 : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08cfed85f96241a0adf8efb19ee22273 : SUCCESS in 9m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15ba6952dc2a4d70937106161882abd3 : SUCCESS in 24m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01759d60137e4bd4b47d7a90f1dfbb61 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":11},{"id":"e9197bbdc1230083e32ad2c37199ae9d12f4088a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 19:35:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"8426709b9951bf1df406a162fc760c36324005f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 19:37:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbea19d0796d47bcbdefad3599bfb13b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393ffc984c7e4cc2b79fec5687ca7a8e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"60d612828e506e4f37e1a7de4b7077b8043a276d","revisions":{"2c1b11693e3a8d0b41b84568f87d1c9adcf0b588":{"kind":"REWORK","_number":1,"created":"2024-09-24 08:35:34.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/1"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c1b11693e3a8d0b41b84568f87d1c9adcf0b588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c1b11693e3a8d0b41b84568f87d1c9adcf0b588"}]},"branch":"refs/heads/master"},"d00160cc0c03e13dfba12d89b8124389049d15d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-26 09:01:47.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/2"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-26 09:01:47.000000000","tz":0},"subject":"CI: bump previous release to 2024.1 in Dalmatian","message":"CI: bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d00160cc0c03e13dfba12d89b8124389049d15d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d00160cc0c03e13dfba12d89b8124389049d15d2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4e3d6405b20ce01e82c101a7df69142e296e1fda":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-09-26 12:32:41.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/3"}}},"commit":{"parents":[{"commit":"9c0a990117bde6a804d1645bd28cbd9e0310e3e0","subject":"Add support for Ubuntu Noble Numbat (24.04) LTS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c0a990117bde6a804d1645bd28cbd9e0310e3e0"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-26 12:32:41.000000000","tz":0},"subject":"CI: bump previous release to 2024.1 in Dalmatian","message":"CI: bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e3d6405b20ce01e82c101a7df69142e296e1fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e3d6405b20ce01e82c101a7df69142e296e1fda"}]},"branch":"refs/heads/master","description":"Rebase"},"1691c2e7c379c1cfefc5795b430428679ec9d6fc":{"kind":"REWORK","_number":4,"created":"2024-09-26 13:22:02.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/4"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-26 13:21:51.000000000","tz":120},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1691c2e7c379c1cfefc5795b430428679ec9d6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1691c2e7c379c1cfefc5795b430428679ec9d6fc"}]},"branch":"refs/heads/master"},"89e57de738b52a10e106a6182f9911282115681d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-09-26 13:23:13.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/5"}}},"commit":{"parents":[{"commit":"b9b627717038fd0a895e30d3babb00dee8d35253","subject":"Add support for Ubuntu Noble Numbat (24.04) LTS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9b627717038fd0a895e30d3babb00dee8d35253"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-26 13:23:13.000000000","tz":0},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89e57de738b52a10e106a6182f9911282115681d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89e57de738b52a10e106a6182f9911282115681d"}]},"branch":"refs/heads/master","description":"Rebase"},"030b9de1f75d95087435806d4fa68d0dce572ca8":{"kind":"REWORK","_number":6,"created":"2024-09-27 08:43:27.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/6"}}},"commit":{"parents":[{"commit":"200b955eabf034ccc6ee550e73d78eba68374785","subject":"Merge \"ironic: revert to ironic\u0027s default drivers \u0026 interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/200b955eabf034ccc6ee550e73d78eba68374785"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 08:43:17.000000000","tz":120},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030b9de1f75d95087435806d4fa68d0dce572ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030b9de1f75d95087435806d4fa68d0dce572ca8"}]},"branch":"refs/heads/master"},"0e2c6f359f26b6959f7ab4a2bf6243a9d79a1c9c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-09-27 09:12:08.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/7"}}},"commit":{"parents":[{"commit":"625c54f9f6fb1c7d3fd111d6946c799f524d08aa","subject":"Add support for Ubuntu Noble Numbat (24.04) LTS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/625c54f9f6fb1c7d3fd111d6946c799f524d08aa"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:12:09.000000000","tz":0},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0e2c6f359f26b6959f7ab4a2bf6243a9d79a1c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0e2c6f359f26b6959f7ab4a2bf6243a9d79a1c9c"}]},"branch":"refs/heads/master","description":"Rebase"},"f970ab3baa35434a2132d90077864bd74cc8e91b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-09-27 09:38:21.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/8"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:38:22.000000000","tz":0},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f970ab3baa35434a2132d90077864bd74cc8e91b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f970ab3baa35434a2132d90077864bd74cc8e91b"}]},"branch":"refs/heads/master","description":"Rebase"},"6b4a041dfad98c2527f98b9e39e33ca33a51c55d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-09-27 10:16:14.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/9"}}},"commit":{"parents":[{"commit":"1d120b598e8ce9af1ec44ec566a001132b481cf6","subject":"Get the list of ironic nodes - use correct scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d120b598e8ce9af1ec44ec566a001132b481cf6"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 10:16:14.000000000","tz":0},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b4a041dfad98c2527f98b9e39e33ca33a51c55d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b4a041dfad98c2527f98b9e39e33ca33a51c55d"}]},"branch":"refs/heads/master","description":"Rebase"},"22c7b1ec67ba7c3b4c3f2f2232fda771ce1a97c8":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-09-27 11:26:52.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/10"}}},"commit":{"parents":[{"commit":"76339a63d2fef37f502156c895760eaa2adb383c","subject":"Get the list of ironic nodes - use correct scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76339a63d2fef37f502156c895760eaa2adb383c"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 11:26:52.000000000","tz":0},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22c7b1ec67ba7c3b4c3f2f2232fda771ce1a97c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22c7b1ec67ba7c3b4c3f2f2232fda771ce1a97c8"}]},"branch":"refs/heads/master","description":"Rebase"},"60d612828e506e4f37e1a7de4b7077b8043a276d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-09-30 12:06:47.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/77/930277/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/930277/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/930277/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/930277/11"}}},"commit":{"parents":[{"commit":"479b4cdf73fd20d3a957862784481593019e8b8d","subject":"Get the list of ironic nodes - use correct scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/479b4cdf73fd20d3a957862784481593019e8b8d"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-24 08:34:34.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-30 12:06:47.000000000","tz":0},"subject":"Bump previous release to 2024.1 in Dalmatian","message":"Bump previous release to 2024.1 in Dalmatian\n\nChange-Id: Iab3f982ae914c19d9c3d9ea6a1634179f85704f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60d612828e506e4f37e1a7de4b7077b8043a276d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60d612828e506e4f37e1a7de4b7077b8043a276d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
