)]}'
{"id":"openstack%2Fkayobe~930335","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Id597daad620d90d30187fa74f9d21d220cafacd5","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-10-23 14:34:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id597daad620d90d30187fa74f9d21d220cafacd5","subject":"Remove duplicate description of filename key","status":"MERGED","created":"2024-09-24 13:58:51.000000000","updated":"2024-10-23 14:35:19.000000000","submitted":"2024-10-23 14:34:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930335","cherry_pick_of_change":930285,"cherry_pick_of_patch_set":1,"meta_rev_id":"e448dfc4466a70a1836cb1e90dacb8eaf562c6fc","_number":930335,"virtual_id_number":930335,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-23 14:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-10-23 12:45:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-09-24 19:05:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-30 14:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-10-23 12:45:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-24 16:08:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 19:05:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-30 14:02:46.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-10-23 12:45:23.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"3af162fd5d0c888fe972f513c40221b6a4c4e18d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-24 13:58:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"74ffd0aaac3f490b4096171b5f7c66246a34ac89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 16:08:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3dcfd9a742b740b788a48a922a8993de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f46402e7b4ab4b85a6b5d3e8ea3d3d11 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/652de5ab5f2347f18089a7b54313e165 : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23725b8e9d474ef482fa1eada883f745 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/90818a78df2b4d518c790f3116bced43 : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d4cca050e294599be725647c64d9cc8 : SUCCESS in 8m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ed8b9eb71623487e9bcb75af4dfe8900 : SUCCESS in 1h 00m 41s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e2e0af66a7654635bc75872c65920a94 : SUCCESS in 1h 02m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/207c1ab0edb04801a47fd9f265e0f360 : FAILURE in 13m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/959987c5277741ceaa28a4300f891dcc : SUCCESS in 1h 11m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/28c93ced6ebd40cc8d30bac0cf7bdae0 : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/06d054deea9b4c49b320eb546c232323 : SUCCESS in 16m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5c90338a4134f08a020d9c72bdd0c9e : SUCCESS in 13m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/235157cc8bf347e3b21f777e093ac492 : SUCCESS in 2h 03m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24407b93ea104e03990add2b5aecb380 : SUCCESS in 1h 24m 55s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ea813de60105443b85250b49426aca3e : FAILURE in 1h 16m 17s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d6ba798e2bd4c77a2a8a7a3621d8bea : FAILURE in 1h 05m 33s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c16b530f6f54409f86a8404bddfdd89a : SUCCESS in 19m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95471bb14fd846eabfa11d1a5b89269b : SUCCESS in 13m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/af0fa6a4720f4b5eadf906afd8e0081a : SUCCESS in 37m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c7bb4a6123094f6ba2c5ed14cdef9524 : SUCCESS in 29m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f84c8b6a3d46423a87049c7ac7df6724 : SUCCESS in 14m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0764de84be104aba90e25f81392b722d : SUCCESS in 30m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7288b7b9dc354e398ad0fc15fa7c3a40 : SUCCESS in 16m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/889502920bdc4472abad13e4242fe32c : SUCCESS in 10m 35s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3c810d06eab846e181fe92bd28614302 : SUCCESS in 10m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edbe1fcb2c834420ac2a498c81fe46ee : SUCCESS in 9m 34s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/84a20b22c49f4a5e842cb59f7cac594e : SUCCESS in 6m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3ee64a3b128749b6b768bc24d143f588 : SUCCESS in 19m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e92a2cc12ee4ff5a3bbbf40fc2931fd : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccedb494ac494667a9bcfe3aa3b23656 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bec4b8e78240430cb94603a848f924e1 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a05978a034142c6b7fe0da5dfa70291 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4e975e1719040979d73b019fc6e6c65 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3c7d68788713a0185837933ee42b238d78fadb05","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-24 16:12:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"240206a35cdff998f56be9d3cef9460b5d9098e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 18:02:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b9292f68674596bb088ad45a535b1e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d70a5e4de4d446c78c1844f115e24ad3 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb8171255fc4953b83e5e5211c880c6 : SUCCESS in 7m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/042058d2ec57439db9648d307928a5a3 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c93c79ddaeb044c887edddaad5a40a4b : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9aeb2800d08a4808babd5b8c13186b9b : SUCCESS in 8m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/47929138904e468a9abe3883f6d099d8 : SUCCESS in 38m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3b38a277b6ee4ba8b51653b97f32a7c1 : SUCCESS in 59m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9a795be0c3f414d9d40293872224576 : SUCCESS in 59m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9e1ef6608c9f400ea4dc7d0446a04d21 : SUCCESS in 1h 08m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/01446956a0754c52a88639e9013efba7 : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3e54aa3341ee4cd0b5d1afef7953ba88 : SUCCESS in 16m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f50ceb2510a64679b7697ba1bc94c026 : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c1fb7a673cbe4cc3abba27b60ae4227f : SUCCESS in 1h 45m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0de58f8d5c174e10a4c5069f3e91c5f1 : SUCCESS in 1h 36m 31s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c27558a2a4da486ca69cfc2760597917 : FAILURE in 52m 14s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8802f0c6f458446aa0b168c9d8b67276 : FAILURE in 1h 00m 21s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/804c819e9ad0435cb3b152ebf584ae9d : SUCCESS in 20m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3489f78601c2488885afb2cf925d32f6 : SUCCESS in 9m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6c6ba2c9dca44d4ca713be6b06dbc0e1 : SUCCESS in 36m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/873a58a077a44114bb9954057ebce879 : SUCCESS in 31m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7fd7b90264b4dad8f75257cfdff492f : SUCCESS in 24m 20s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8311d0ff0a6e48a9851aafb92a104dce : SUCCESS in 23m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38fa4365e8a649e19ea9aff6ae18af40 : SUCCESS in 26m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8e1644e9d7a547cebbc9b127f1f136cf : SUCCESS in 15m 45s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f471f6587e3b4289b79acd572670ffd6 : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdddd86b630942219b07633686c69575 : SUCCESS in 7m 47s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/ae17e13c933a4d6cb0d630a90398a173 : SUCCESS in 11m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f98917acede4350892f0216c5eb64bf : SUCCESS in 21m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fead1f1a2064dfb861856c75b488ac1 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa40e161d7c748b6bff2ad1d745a2844 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/143b6d4cfaa54ba8a1edc767e85cfb91 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae0df9155b7441539a3ec4e5d8b7db21 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dddd1da17cfb4191b175c310ea395d1b : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c38b4e376105115c501883ae8fdcd06a50533ea7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-24 19:05:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a748d2ad4909ffc8776c832caa9c1a237355bd4","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-09-30 14:02:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf1da479b21817b3b2874490df370f73cb899a8b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-23 12:45:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22b3e376d90cb23110080a204b08b36365f16d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 12:45:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"709f464364521ae5390cea00d603f9bccec6171e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 14:34:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd5276c33b654430baaeaeca74d30865\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63da47a40c24657a2d2fd9129d85735 : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e6dc6462a6c94aee80b73b08f23d00a1 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d67c4b24abc44fafac2775bb672f4cad : SUCCESS in 6m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6513eae13b7a43259e7665553873da90 : SUCCESS in 7m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5ae7ae542b8d4a318a19f35e39ffe1eb : SUCCESS in 49m 57s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35ab0cf3253e4481b6b7da023545df86 : SUCCESS in 49m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/087f6f3aff544a02889f84a5bf5d8d14 : SUCCESS in 41m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/59fcf85b565e480fb032a5334fa27165 : SUCCESS in 20m 09s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b4359182d9a4862bd2cebb316dd1ed4 : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d881999ce3644118ad23911d8ec7b65f : SUCCESS in 1h 30m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/623fe6c23cfa404283811ec643750216 : SUCCESS in 1h 42m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5672e6f3af5b4575b59401af1afa9b46 : SUCCESS in 16m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2415bf1a8c194351a422200614dfa165 : SUCCESS in 14m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/18e9823c64e84154b71f5b58bf0b9873 : SUCCESS in 25m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7498bc3fd73471ba585b35d8a1c87d1 : SUCCESS in 25m 52s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b18340693ac244c48d0c68203f118c44 : SUCCESS in 29m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3465fe00f4384629aa0e48388c62f836 : SUCCESS in 26m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b0c6e9da327d44eba97d290cea58d2e0 : SUCCESS in 11m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8c5bbbb5d2b4c79a0ffa6885041dd4d : SUCCESS in 8m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c67dccf4d2474cef92fa3af42138b55b : SUCCESS in 17m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db7c89331427423081d36292a9521e47 : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ab78d06189c5babc797963f0bc2e0af81e3c32ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 14:34:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e448dfc4466a70a1836cb1e90dacb8eaf562c6fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 14:35:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19c6ce1a5f9430e9b83176d4449f172\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb26ffa19ebe4b919d397b24bf7ec052 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0cf61dbbe62b9f0d9ba6edc8498a19566ea9aac","revisions":{"c0cf61dbbe62b9f0d9ba6edc8498a19566ea9aac":{"kind":"REWORK","_number":1,"created":"2024-09-24 13:58:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/35/930335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/930335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/930335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/930335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/930335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/930335/1"}}},"commit":{"parents":[{"commit":"5274c4e6cdaf22e7e753363e78a6c69dd423d831","subject":"Use Rocky Linux 9.3 images unless in efi boot mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5274c4e6cdaf22e7e753363e78a6c69dd423d831"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-24 10:17:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-24 13:58:51.000000000","tz":0},"subject":"Remove duplicate description of filename key","message":"Remove duplicate description of filename key\n\nChange-Id: Id597daad620d90d30187fa74f9d21d220cafacd5\n(cherry picked from commit 3a7bff6ea1ccf40dd4a72e58f9ebe91224998896)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0cf61dbbe62b9f0d9ba6edc8498a19566ea9aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0cf61dbbe62b9f0d9ba6edc8498a19566ea9aac"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
