)]}'
{"id":"openstack%2Fkayobe~930691","triplet_id":"openstack%2Fkayobe~master~I4da3a3739d57817ffbd094caea0fc80c3ad8eff8","project":"openstack/kayobe","branch":"master","topic":"set-driver-info","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-02 07:15:43.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-02 07:30:11.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2024-10-02 21:04:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4da3a3739d57817ffbd094caea0fc80c3ad8eff8","subject":"Get the list of ironic nodes - use correct scope","status":"MERGED","created":"2024-09-27 10:06:31.000000000","updated":"2025-01-09 12:00:50.000000000","submitted":"2024-10-02 21:04:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930691-set-driver-info","meta_rev_id":"3ce09c0ed430750ca76c8329034a4c4fc402cd3e","_number":930691,"virtual_id_number":930691,"owner":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-02 21:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-01 10:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-01 08:08:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-01 11:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-01 11:10:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-09 12:00:50.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2025-01-09 12:00:14.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-27 11:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-27 15:30:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-09-30 16:58:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-01 10:54:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-01 11:10:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-10-02 17:01:25.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-01-09 12:00:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"9daa17267238aae9b4c84ef278370af0140560e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 10:06:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e2b089d0f6805d22f8ba950e310431f4d721152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 10:15:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4206a9b87dec7100fd68358c1da2c97f4d2ad9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 11:25:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97a718bc6d4f418495fb0cad53b85438\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18b1b1c6224b46fcb94f8e77dbb0fe70 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4043317670485a890789371f115148 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b76d4269f1ab4944b7d9a39be2e83faa : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/145ac96fc21c4debbe901802df7541a3 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9074bbbe75245a7a5e8ed7613a9496b : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2816dd93829247b9a9c90bb28a02cd69 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a08b6cf7bcf44dec91917190ef6e7368 : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/914ced1cf84649b9b6fb3b6b9a3cc811 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d1ae7d09ee245b0b3810f3f1270d4c7 : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/57d32563205e447db0472b95d30a8124 : SUCCESS in 9m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4de25f6f5c0a4135a8e0e03c9c56d216 : SUCCESS in 56m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/31e4063dbde24f07932d6676b5ce8b0e : SUCCESS in 49m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db9d9cc022e1407b9fa7cb799069cadf : SUCCESS in 48m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ba8565903e1444ae989431d9f29fad09 : SUCCESS in 56m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/db97036a812c40629f8644f703b2accd : SUCCESS in 19m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/32db56502b63474cbf6af5e0af7a8121 : SUCCESS in 17m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c29a3ee7ab314a5c97c6991417a1f0a1 : SUCCESS in 12m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c66bdff3febf4c6fbbf97fe0c1f3aca3 : FAILURE in 1h 07m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae03c47ea77f4e499b1be777031a7be3 : FAILURE in 57m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b41fb1fc3a5147a8af358afca5af1bcc : SUCCESS in 20m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e7a382b40d044ccb3e23da6fe5b8869 : SUCCESS in 12m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0da5f32f905b49508f7851bba85c2ebc : SUCCESS in 39m 23s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/94fb15f9e4374cbc9b0e57f2b8523847 : SUCCESS in 32m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/641c8b0e9d54409d907dd47c0ce86fed : SUCCESS in 23m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/11c86bdd09d144d39dd47a47001fb6f6 : SUCCESS in 17m 48s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3355f092304746c09e94db7a9def1294 : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dab1f4185aa409fab051beac9896447 : SUCCESS in 7m 20s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/cc76024ca73e41118b783b9850c082d0 : SUCCESS in 10m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/547a1801500c49d6b4d1d41fd94a74bd : SUCCESS in 21m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65bbe5a73839400fb3ac2b58435ca9ba : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e33badf7e182c390644e19801733e6708581aea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 11:26:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fbc4c403c682783d5b0a857a6d31e6e172ab9935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:41:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0d4cff0376147a09b782eb4bf3479ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e8bdd099d484686bf68d44c4f058e7c : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e18e58fef9f34382b3ad2f6ef0d89bfa : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7c4817840da4fb79669fa6a2f49bf4d : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddd82b0ca6354a108a3d1b4cdf865313 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f4e6b2d930f47ada35815ab087bcd60 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd72b07c509488c894b08401fbef961 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abcf4a41afa54415a85a93dd2df3dd4b : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/38c5c8a3bdd3455b9f72ef040837d1b4 : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91d717b5689d44fcb1d968662b67bc94 : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7ffc011e494479dabab28a8abac6cd9 : SUCCESS in 6m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d3d6f4f996f8442dba3d2da5dcdb71b3 : SUCCESS in 54m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/78431084b6554e42a6f8a2b62ead0eda : SUCCESS in 1h 00m 11s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79ee5eef71fa465b8c3435011f6eb471 : SUCCESS in 47m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c85e811174ec4dcf89b29df9e8ebe617 : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c688f5ea5a9d4e77b4d0e8c577d8ef2d : SUCCESS in 18m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/261932af957141d59163f9fdea95fcdb : SUCCESS in 11m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a3cbee34c9640bea36ba9f6e33f81cd : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c874246415b745a6a045b9d00b2bf2d8 : FAILURE in 1h 11m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e2b6bb933f0446383f337897f1333b5 : FAILURE in 1h 05m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/475b9c9a7d6641239e662e4663acf208 : SUCCESS in 20m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3370cfa4f5e243bb9008c1e85c0d2885 : SUCCESS in 13m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c66a675b8f0d4925a267eb0d6996579c : SUCCESS in 26m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/41c6a1c52c9242d38eec3c9fe0053992 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d46c578b5c714a8bae0bf22ab5bf67a0 : SUCCESS in 27m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec9425a04d92448ba9a932fdd427f5a9 : SUCCESS in 17m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/025a828c59a6402e90c577708967bac6 : SUCCESS in 18m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa3beb164b4740f28c3f8027f406d435 : SUCCESS in 10m 52s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/8c4aab4ba2d243e3b4c9538adab11124 : SUCCESS in 9m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4478eb1610554651a8f39000ba122477 : SUCCESS in 20m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b27384d0cbc4a5c9552662effb38740 : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7316666ed0ec40cb10e0f6f63cb7db9cdecc0f1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 13:28:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"43d0c4c43698317007ce067aa292159be3492be8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 14:26:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bb4537eb52392f29a5e6ff4547e09cadc7634160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 15:45:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"98ad04eccef61dbde673217a1383d8f5a70c2ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 17:07:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4612dda0da94943804a598b7fb98aba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f872413c23d48b1b533b73f5a0278b4 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfdd8fb7a9fc4be08f8efc8bb3b9b9fe : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5baae1a7441f4a32a70424ab56ba8de0 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e90aea3c1564c0f9487a6ee31da2b93 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a241eeaebd8e476a8c65f93d9cc24236 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bec125aa5d34b43b4bfe9ab963cf075 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0a84e0aaeb24e94b5c78f243e860351 : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f813d6d2068404fa58eaced435400e7 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c2d1afc5fe11411ca0a92bbeb43d4243 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e143af603fad4b76991ef4608839495b : SUCCESS in 12m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6a7f4c09c671458eb6b16d78adb38ed7 : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ddc15a4a52c848a691ccb56fd643b3ba : SUCCESS in 52m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eea0a4d873fa4f05ba584acff90f4b14 : SUCCESS in 54m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/23064157f84c42d19b5275993658ce19 : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7a856e2a166342e8b516e0766e7f25e0 : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d2c9ca92effb485d99239dd19d62016f : SUCCESS in 22m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4646a06f6ad2443898e8fe92638b8f8f : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ffe1f3aaf5c444f4a2cbface277065d6 : FAILURE in 1h 15m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee6bec0eef5a40628acdf5b3d873fc56 : FAILURE in 57m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1ecef2fd5cb64338983e351c05c814eb : SUCCESS in 22m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dde1abc0b72048c6bdf85a9895fa05be : SUCCESS in 14m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e20dc83e8ceb4a69a85d4185acde57c5 : SUCCESS in 42m 52s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4d00664ceb3c4200aa7346acdc00e384 : SUCCESS in 20m 33s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33d6eb2933144a3a8fc2e20ce1f2c903 : SUCCESS in 29m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e1e0949aefbc41e6b24a7ad8e106e17a : SUCCESS in 15m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f0543ec85e7640059ad8a6c468c2f460 : SUCCESS in 10m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15e4c3ebc5144fb2a8b2fe2808488479 : SUCCESS in 10m 00s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/ebf8005152f14088aee384f59b3fcfe9 : SUCCESS in 6m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b54305efcf3b44b7b63041684ef05a78 : SUCCESS in 21m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43176c08daea40c7a725da0aa07a3762 : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":6},{"id":"59300f3ff0edff03a6c3658d8ed90794ad19d3f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 17:34:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"551dc2ff5b0c9324a7b75c2bec95bac0f6fbad03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-27 18:46:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2f64113d0edb7ba522b0077d2649084cfb664e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 20:17:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee63b447b92245be8b78581adea9d25e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d013bd14e04506a29d9e1eb82d87dc : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a20e2bbb284c5f97f3215b7d2d648c : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d0d13651a5240f98aa4933d04a71384 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03f497068056406491a54e9ec7e1a0f7 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d032fb390dc345c690609625068383b0 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e4279fcbd1b45f7865371e089a9d14b : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ad83474e65e469bb3da937d7c9fc807 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9fc8375c71804efa91b1f11506e48715 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8ed9b045d9e4acd804e2e34a5e9401d : SUCCESS in 9m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1eb4a8bc7a2b49eabd5a5d5a43e9e473 : SUCCESS in 1h 00m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1e55cd1e73114668a0358f1126c7cba6 : SUCCESS in 54m 03s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8907d62ed8f041a88d7740200e7174cc : SUCCESS in 50m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4e55a687fff14c8c84677040dfeccc13 : SUCCESS in 1h 01m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/68ac0768fa6046f19ee8dd2e99a00170 : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d37cc681e62e4fd68db757ebfa358722 : SUCCESS in 20m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f660275f7f3d4096b05533bcf59362c0 : SUCCESS in 12m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e2054ca1c2e4e5083a3b39c1217c1d9 : FAILURE in 1h 01m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/175a883d61774e69bed3f175146810be : FAILURE in 33m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2304f142851c4dac8dfe8f1ae77944c5 : SUCCESS in 20m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b34b35f29b048588c22bc4a5fc44118 : SUCCESS in 13m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5ec02a52d2134a3d91b8ba2350830cd8 : SUCCESS in 42m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d3fcfa0fb17f4859b3472922dc1ed6b6 : SUCCESS in 35m 43s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fce7c47e60bf45d8981b65dd4dc2d9e3 : SUCCESS in 26m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7f8a00d3988a4c4e93c3fd89940e0726 : SUCCESS in 19m 17s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f62ca83131d7457abb59211c1732e127 : SUCCESS in 11m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ce466e2534a4c1494183283ee408f3e : SUCCESS in 11m 07s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/4e35d988536444fdbec320ed7c8763fb : SUCCESS in 11m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b24118e5fef945e88909f9b5a5465cb5 : SUCCESS in 20m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df8c91563ccb4203a05d090b68e06a7f : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":8},{"id":"e3dbfdbf77e79641ef507da082a388e7f2b18a14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-30 08:36:16.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"e2acb5d0548851b6b24b4b98b473c67b9e660db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 09:41:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a5f537db7b1466180e2599d545c32e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/494dc4be0b4b4e5a9656383a481d695d : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbb7e2373dfd4613932d4296ffb1764c : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8845b4bb39dd4c498e5227006ba304a1 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d74d552720214ef3a3fe4082159ef0ab : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7d536ee8c743d7a571dc25562d5c18 : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c77c57edb753426da3b0b011c4b6d9c1 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28288bb2b7ba439ca83c3c82214c07fe : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a0c61d1ac3f541ad99835fa04e0912c0 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78b463bccbad4040acf8b6f7a3f7a98c : SUCCESS in 6m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1829535b7cf542b48ba1a0955a9f8d07 : SUCCESS in 35m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/758c38ed3da946b293d88510787ec11b : SUCCESS in 33m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a89b03d308c94b1ebfe8d81209ef716b : SUCCESS in 28m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/12167d017e034ffd9db175bfc57ed58e : SUCCESS in 37m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/52eb4cb62b2c4f9d872ed8de0b2915d3 : SUCCESS in 13m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1f157d3978334a30a3b24118b722b229 : SUCCESS in 16m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63f8faedc03844dcb664fde4150087e6 : SUCCESS in 11m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b513993cabe248bb87137b6ea6f4a662 : FAILURE in 1h 03m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e60e494f0404d9589764832ee67eb54 : FAILURE in 53m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/28a0a49f4ac1442eb98c3e347232a88a : SUCCESS in 19m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e46a4aca00334c7bad5aa8f4025f3aed : SUCCESS in 14m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7dc6fe2dcc20414993b0912ce93d22cd : SUCCESS in 37m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/834d022106984b26b93633757b57d558 : SUCCESS in 29m 43s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fba4d650e3a4e479678da9620a21994 : SUCCESS in 25m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c7de3bec947c495ea126104569fb8c0d : SUCCESS in 16m 04s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0b97b3dd42264b1eb31db66d45e60e3f : SUCCESS in 15m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19011d218d67456ca81f2f8dc97feef7 : SUCCESS in 10m 22s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/a25876a034694191b4302e8e59d38eb1 : SUCCESS in 9m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8be79a9518e541b88aa2b53d0f27fce8 : SUCCESS in 21m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd3390cd32d146b09a664a938c430139 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":9},{"id":"70639800cebb3871c27c63914445f5ecda208412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-30 09:54:21.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"d435866c14b68aba3abadef40d7cc6120251ac99","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-30 10:52:47.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":10},{"id":"39cdd2a152958fc190048fae8f0a80c8edc1806e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:54:48.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3104823a2f4441bf8e95ec3b23fc83\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e749b77c92f8478d83585ae85483e9c2 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfbd6c73aec041df9f1db9a470ced15f : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34771d811e514f118124986a28f6a977 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ea3ebccd11f448c930d87dd130cad4a : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2828ca1516a64f78bf950f6dab4f404b : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a76750cc39e44ad69dedd9c05fd4a4ad : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec5d72be47394d128e3bf79e4ed7ee13 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a706e3c1ba564385bc61c8978c68bc36 : SUCCESS in 9m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8534e8927ad346c4a107edd8c1135336 : SUCCESS in 9m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/22ba93de3e4c466392979f044f6fe0e6 : SUCCESS in 53m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8fa4d145048241409b1a382a98c8a084 : SUCCESS in 55m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d871519218f04ee0a17697db23defca7 : SUCCESS in 48m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/529fc095e15441d0a63390f799d13895 : SUCCESS in 1h 04m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/01906b90273f4b269ab340e801d842e9 : SUCCESS in 18m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/52e96c4b1dc041dbb47a993005d4687b : SUCCESS in 16m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c478656c0ed94061ad08c72aefdc71ad : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c1056f029c674237ac2dc356eef69c54 : SUCCESS in 1h 50m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec67a7900c2e4179b360b4f2721e6ed0 : SUCCESS in 1h 42m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b9e45fd0233447ff92f5de77e84b7f1c : SUCCESS in 18m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c759d13794447b08ddda42521544932 : SUCCESS in 14m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/47dcd587126a4bdba738e2439bd9facc : SUCCESS in 44m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0a797506ced7410abfe5ac277f36dcf0 : SUCCESS in 32m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d92c41c069b540e89622dbe3d7b1c990 : SUCCESS in 14m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e6ff1df453d4083a7e2670dd1b97bb1 : SUCCESS in 15m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/71595d9e29b641bd823c28a577c66291 : SUCCESS in 16m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a004adcd12aa4f2c82c89a0a2efa6b03 : SUCCESS in 10m 43s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/da31b480641b4a1e80ec0ea98eb81877 : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/38c3fc0b884042d7af60c2bf6c3b0529 : SUCCESS in 22m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8306ee527c5e4fc090a89584228c5ee7 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":10},{"id":"f3346d890300e471ed65a6719f131965eaae7bbd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-09-30 12:04:30.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"996642211456042fcd3148c74af80675ae259bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 13:51:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38dd98bf99014d3fbfe1946ddd7d5a60\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32ea22742db24423b8b1f25559a160f3 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f09b395c2e87479ba85115d0ebea73f9 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ecfea96fac54892aa14bb52a7012897 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83048462821f49f2bd8235d7e2d8d141 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3addea4423754d1f801d85acf7b72ace : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a58205efa3b4bf2aa58eaaa4293bdd1 : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0187b0420f2494abeb3c284ddd079f1 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/90b28dc5a985447991ebf592c14ba761 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52411b73aa964d479621c01aef0bf734 : SUCCESS in 8m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4a15e2a8db3a42cc8ae30ea0082c28a9 : SUCCESS in 55m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f863a231e37b4026afe5250e3dca89ba : SUCCESS in 1h 01m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05098e70b59a4b8286d6907a5d9fde6b : SUCCESS in 48m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f4e82f7dc6404f2a8c378edc5d9484d3 : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/612e586cd9674d01835356f7a7ba3690 : SUCCESS in 21m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6eff3f029ad949c4a055f5044344208c : SUCCESS in 10m 13s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abd4887383fe437db55cffe55ba21c09 : SUCCESS in 14m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b12a968807604924bc730cb9f894cc31 : SUCCESS in 1h 06m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/424b479cf97e44e98ac109a565a5d28f : SUCCESS in 1h 44m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e64d6575db0f45ba9ae57f18e6d04df5 : SUCCESS in 21m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7ad1b81b9bf402fa885e7cdf7bc73bd : SUCCESS in 16m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c1fae89ab3d749f99a5c5ca1c85409a2 : SUCCESS in 44m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b8b2fd0bbe984ae4877bea895fb86a05 : SUCCESS in 33m 42s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74858f5ac5bc4ebfb49184a720732832 : SUCCESS in 24m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c97dad5319054142ac124d423a62ed19 : SUCCESS in 15m 58s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/7e45a0fa6eda4ae78aba1e7022fb653b : SUCCESS in 15m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f06606ad0f64b03b681d8274d6068ec : SUCCESS in 10m 34s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/9732bca1127f4f12af99cda0aa055d48 : SUCCESS in 10m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/37739325cdbf44bda0b91c536f9babe9 : SUCCESS in 19m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ddc6bbe9f5445c7b05bc8e387c3abad : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":11},{"id":"cbb0e32efb5ce142b5be084fec61cd56964e5500","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2024-10-01 07:54:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"d955dcbc8b0d4ad67ce0efc408547cf4962fca62","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-01 08:08:42.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"75d156dd6a77dfc775564cf686ac242aa7b1525c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-01 10:54:22.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4d967e4189d31748f7b3af789f592b9bfdfa473d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-01 11:10:42.000000000","message":"Patch Set 11: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"6d5460b7080975c0946900201809daeb5b57e750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 11:11:09.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"24355e235788ca3b66d476af44acc9eecbbd1576","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 11:53:33.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d79427ef407446bb12521056acb8243\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e64f7e4ba6440badf1bcd6deb1c40b : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d0110610307463598fa817cb7c7dfc7 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3112b064cefe47edb0e2dd0fb75fe27f : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e83e669e6e6418ebfb2f4c5100a7a9f : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75de60058c3543fc89da4fafac93cd73 : SUCCESS in 2m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b32636b47a1d4b4eaa74984d47359c9c : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bb1b4437165145d2a4b02afc694dacc4 : SUCCESS in 9m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4df130b6fd274fa5b1ed4e6ecb6292cb : SUCCESS in 9m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cc0cd4853c4b40a1b9824c3d01e165bf : FAILURE in 24m 00s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00cbeb9d80c542c4afa5a6fbc43d7d3f : FAILURE in 17m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4eedcd1103fd4ba39da5626c4d029cd2 : FAILURE in 17m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d076c1467ba44c39a6849fbf19a6ac26 : SUCCESS in 20m 31s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17e27010e9f24a358039c5a4779eb1c3 : SUCCESS in 8m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4118ef91352245d8a3f708eaf7a1f73c : FAILURE in 21m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68bd4ca6822f4f9ea968a188e23fa35f : FAILURE in 14m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f58326fd16b04e57a5ad88879826fc80 : SUCCESS in 12m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/574c997145e3472f8a447d043d71fe1a : SUCCESS in 14m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/621de971b91741228b21bb41da113fd4 : SUCCESS in 36m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/673e99dbd01040d7a0cf8fa4934713c5 : SUCCESS in 26m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6d610e6bef5545c9a65fd8cd730e0f81 : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da26492a99bd478a9881914724bde0ec : SUCCESS in 9m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f056f3ce440742828096c388e596c6dc : SUCCESS in 21m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32f3f7bf315f48e9a0610b67e76a5a08 : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":11},{"id":"c78506138a603e1020975713dbbd47b079538459","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-01 21:01:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dfd7ff16f3d6fcfa93f582ad08999816923c59ff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 05:55:03.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6c6b5c6a7e52fbf53bd1032aeaa00a9e2461b4bc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 07:09:59.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ef1e2f4f22322a7c71d7141cc58656a43ff0b439","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 07:13:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4fdecad74b9563d7a00f1fce292b38a65a7f4c62","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 07:15:43.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"98907a85fdc0b93afe3f262ff70a742cb217815d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 07:30:11.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"47d58164e46c5565f15cc275aed71eab2dc0eef2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-02 17:01:25.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bd4e1565c8c0c935b6784b2dc08eb31a16b7c599","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-02 17:02:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7ae3b9e0495133f159ad8756a29d4c6e2649ca9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:57:38.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd7c2a034e7041e88484b9644fbfd8b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68eb45aecdad4b0599eba65e7e2269d4 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08004b0d75b545888ae5724649a18bf9 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abcbfe1614804c1a81f93874a015c5b2 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afd02f31c8f147b58c870e163c447f0a : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bc8d9efa32a45c29d17c521d17d1197 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8f22a7116ba4c28a1afc75dde1370d5 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/93dcd4e473824766a87de82d064a083f : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/90e60720ea97465e91218a93fc7545ae : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b0a22e57dc3246df8d61b9b1ea837875 : SUCCESS in 9m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/68ee1f0ae91e4a5982635d9a30993062 : SUCCESS in 59m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bb2958eacc5a43978da0d0a4a169bb78 : SUCCESS in 46m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28a7abba42764394a15a3a1bdc6c9494 : SUCCESS in 29m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b8f28ce4e1854ce988cc62bd0c907b1a : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/05a5c5588e7f4196b577be554c05230c : SUCCESS in 21m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/76ac7b5ccc384ef9969e6a9001629d14 : SUCCESS in 17m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ee8d1509f4c4b2598ec445027d87bae : SUCCESS in 20m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c9e3b8fafab84c28b9ff8ff198dd9ad6 : SUCCESS in 1h 41m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b0ed577157b4611b4b21bb10ad49e44 : SUCCESS in 1h 43m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/402417cfe982409f8c23cc4c0e72834c : SUCCESS in 22m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebabcd578ec54bd2873ede94cb2499ab : SUCCESS in 20m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b0f618ae91924043857b5a72433cee2c : SUCCESS in 41m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/744d00d21ad846429aa7630b02f1fd59 : SUCCESS in 33m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2bbf175855c46d8b7d24076cafc3147 : SUCCESS in 35m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b26b567208164778917e9b24b8fac0d3 : SUCCESS in 17m 14s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/08c8c4faf43f4ec7a7367aaae728d31f : SUCCESS in 15m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9423312c87547ce8d95e2983de0c96a : SUCCESS in 18m 14s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/a2d8363021634a7b9f7203f45e5641dd : SUCCESS in 17m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4c53c86465874a64ba53c9622965c33b : SUCCESS in 23m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ea4616c9dd34d1db21be76977a82e4a : SUCCESS in 20m 58s","accounts_in_message":[],"_revision_number":11},{"id":"53571f3744f26a504b1a4cb36bcba5b4806708f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:58:02.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"0454d4feefab95143643d95fa61ad6755d07d66d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 21:04:34.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cee3a628e1840048f8520d2c593c1c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbb60a232e5d4e8297bfba9e759d3188 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7d72a25d5bc49b7b5a10324558de816 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57b7b61dd8c043f5a3d53c990d3cd638 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37974f2066a4835a635220a613003e7 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f8483897424643aba164cb155c7c48 : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2e49b574f1e42c8a5e5c38747552f02 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5392b8a87ee141cdaf45fb2b16feb48a : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7c7b9f78564f40ac84df9cf0321cb1bf : SUCCESS in 8m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/865677bb44184f9d9233bd10d2e1a1ac : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd0de3a04d22405db7ba8d8727286345 : SUCCESS in 47m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5fbb02629a6744ebb7c153647fef1f11 : SUCCESS in 57m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/34605af2362142dca24e0e79df88567f : SUCCESS in 18m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1001b005ca634a698f03ec5c4ee3c192 : SUCCESS in 15m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6980d94a17374040a394feb9685c38f7 : SUCCESS in 1h 59m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3b25a05da37445eaaed8dfbd6b704b2 : SUCCESS in 1h 38m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/898c9482f3f64183af5295b76035f1ea : SUCCESS in 19m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/299a0476b5f9417f9fdfdb5502251800 : SUCCESS in 12m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2de60f3de41e4f72b3c811f4293d6b39 : SUCCESS in 32m 29s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb3a95caf11a40db935af62de5f071d6 : SUCCESS in 23m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a7b2fca092b8491f9edb555f7a1c99ae : SUCCESS in 18m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/078c96af6d0b446891058112f71249d0 : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b75f73eff40d4483a5fd4c9ff660b43f : SUCCESS in 24m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/234f416666044b8cbc4f7245741d841e : SUCCESS in 17m 55s","accounts_in_message":[],"_revision_number":11},{"id":"70dcffb83036fa4b8a8455bb0e17976169463d53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 21:04:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"763e5eceebe11ba753568faa5867a76427b5529e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 21:05:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2986f4debe54f799ec0740fe084ef72\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f322b5fd85c04c9084e7a5a2bfde7b67 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/161ab4f760294dddb727d8a95328962e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":11},{"id":"e8419150a07f6d16081b22f8a22fdc77c62f9d67","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-09 12:00:14.000000000","message":"Patch Set 11: Backport-Candidate+2","accounts_in_message":[],"_revision_number":11},{"id":"3ce09c0ed430750ca76c8329034a4c4fc402cd3e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-09 12:00:50.000000000","message":"Patch Set 11: Backport-Candidate+2","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"479b4cdf73fd20d3a957862784481593019e8b8d","revisions":{"155206a3d52d683ce7551baa1de039e73bafbbe2":{"kind":"REWORK","_number":1,"created":"2024-09-27 10:06:31.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/1"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/155206a3d52d683ce7551baa1de039e73bafbbe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/155206a3d52d683ce7551baa1de039e73bafbbe2"}]},"branch":"refs/heads/master"},"1d120b598e8ce9af1ec44ec566a001132b481cf6":{"kind":"REWORK","_number":2,"created":"2024-09-27 10:15:40.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/2"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 10:15:28.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d120b598e8ce9af1ec44ec566a001132b481cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d120b598e8ce9af1ec44ec566a001132b481cf6"}]},"branch":"refs/heads/master"},"76339a63d2fef37f502156c895760eaa2adb383c":{"kind":"REWORK","_number":3,"created":"2024-09-27 11:26:10.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/3"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 11:26:03.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76339a63d2fef37f502156c895760eaa2adb383c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76339a63d2fef37f502156c895760eaa2adb383c"}]},"branch":"refs/heads/master"},"40a16597b17e2cbe0e4e8bba2ec9edc76212a881":{"kind":"REWORK","_number":4,"created":"2024-09-27 13:28:03.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/4"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 13:27:52.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40a16597b17e2cbe0e4e8bba2ec9edc76212a881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40a16597b17e2cbe0e4e8bba2ec9edc76212a881"}]},"branch":"refs/heads/master"},"b3a70f510ea4b3093debc1abba7cd787f5a3cf2f":{"kind":"REWORK","_number":5,"created":"2024-09-27 14:26:31.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/5"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 14:26:25.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3a70f510ea4b3093debc1abba7cd787f5a3cf2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3a70f510ea4b3093debc1abba7cd787f5a3cf2f"}]},"branch":"refs/heads/master"},"5a0f0fc0a5725e969540bd06f7cc928d1bc30045":{"kind":"REWORK","_number":6,"created":"2024-09-27 15:45:57.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/6"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 15:45:51.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a0f0fc0a5725e969540bd06f7cc928d1bc30045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a0f0fc0a5725e969540bd06f7cc928d1bc30045"}]},"branch":"refs/heads/master"},"3c732011518590e95e373eb62da0ccfca77605e4":{"kind":"REWORK","_number":7,"created":"2024-09-27 17:34:10.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/7"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 17:34:04.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c732011518590e95e373eb62da0ccfca77605e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c732011518590e95e373eb62da0ccfca77605e4"}]},"branch":"refs/heads/master"},"876b519d07d4cb1774cddeaaa056252a27244794":{"kind":"REWORK","_number":8,"created":"2024-09-27 18:46:29.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/8"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 18:46:22.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/876b519d07d4cb1774cddeaaa056252a27244794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/876b519d07d4cb1774cddeaaa056252a27244794"}]},"branch":"refs/heads/master"},"4b60a3c6a56d47cb5ff393487e83c832b2a3d2f3":{"kind":"REWORK","_number":9,"created":"2024-09-30 08:36:16.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/9"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-30 08:36:08.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b60a3c6a56d47cb5ff393487e83c832b2a3d2f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b60a3c6a56d47cb5ff393487e83c832b2a3d2f3"}]},"branch":"refs/heads/master"},"24bb43f20e033596c56e2d7850cee05546a8f4ec":{"kind":"REWORK","_number":10,"created":"2024-09-30 09:54:21.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/10"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-30 09:54:16.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24bb43f20e033596c56e2d7850cee05546a8f4ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24bb43f20e033596c56e2d7850cee05546a8f4ec"}]},"branch":"refs/heads/master"},"479b4cdf73fd20d3a957862784481593019e8b8d":{"kind":"REWORK","_number":11,"created":"2024-09-30 12:04:30.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/91/930691/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930691/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930691/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930691/11"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-30 12:04:17.000000000","tz":120},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/479b4cdf73fd20d3a957862784481593019e8b8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/479b4cdf73fd20d3a957862784481593019e8b8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
