)]}'
{"id":"openstack%2Fkayobe~930891","triplet_id":"openstack%2Fkayobe~master~I650286298450432df83733c74a2bf143116134b2","project":"openstack/kayobe","branch":"master","topic":"tenks-uefi","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-30 16:58:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I650286298450432df83733c74a2bf143116134b2","subject":"CI: Adapt Zuul jobs for Tenks using UEFI as default","status":"NEW","created":"2024-09-30 11:01:52.000000000","updated":"2024-09-30 18:44:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"b79f3600665fb9f61a35863edf6d7b11992e2366","_number":930891,"virtual_id_number":930891,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-30 18:44:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-09-30 13:11:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-30 16:58:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"7f89b5665d0146907de50c53fa7343cf4beaf5ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-30 11:01:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f469d996275d8f1865ea7bce04944bd29f163e5a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-30 11:02:27.000000000","message":"Topic set to tenks-uefi","accounts_in_message":[],"_revision_number":1},{"id":"fd20a309587a8960cd361f078b36a7550dd6de7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-30 11:02:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e38f2093d352492439360d731fab1d23d918c2ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-30 11:04:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a8b0ba69cb0ee1af0f0c5e3831f3af6a67ca515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 13:11:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/341ba9f0849e4357b7050c66703fe6ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7109a84503ec431a83c35d5ac1d65a6b : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60551eaeddff435db1865933194263a9 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66924d5f138d44db8ebff15af4ce9f8f : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7f73cafd82b4d61944a4839b76689fb : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec3afbd3dcc46998c4a74b3ab9751ac : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0dbc16efab1246829888b5076517680d : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8161caffb99e4cc085e86394c0b67efc : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1e991dd68434d90806b27858a58e060 : SUCCESS in 8m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e11f8dd374594c39805acda96ddb10e5 : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/de6832e1e9fb434db998c96c8793e2cb : SUCCESS in 50m 34s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e373c7a238ae43fdaeea154091d24e40 : SUCCESS in 48m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0f56ebdfb3774373a6e033bb81910442 : FAILURE in 1h 37m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8e835be68461401ea537224bbffd34f3 : SUCCESS in 22m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/877d3246d8e8474599672314b13c9536 : SUCCESS in 15m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65fc909e43ec498aabbbf31e89d1637a : SUCCESS in 12m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6b7dd02a3e724f9cbc8651cb90e17fa6 : FAILURE in 2h 00m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df5a5e42d7ca48b0bcaf31b14f186bfa : SUCCESS in 1h 34m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6866ac2e2e0f4a2c947f5b3c4361c48c : SUCCESS in 22m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63fb1d618bdb401dbbc0a0508e25c785 : SUCCESS in 12m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e4c6efb9089845c68b8b4db10c7e1a9d : SUCCESS in 38m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a854594cd2da4e5e99b5220bc28726f6 : SUCCESS in 32m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26a53f736015416ab3b39b814c12d3c5 : SUCCESS in 23m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/44b256a5ad49437c84ea9834e8beb98b : SUCCESS in 15m 38s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d88e34fa646c4974987ee0d7bb360b17 : SUCCESS in 16m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06df1b8890454d82a409760962811018 : SUCCESS in 10m 27s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/2fa20fab83d346c6bc6c0e52e30cf94a : SUCCESS in 9m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/02e7e36208b34b388f298f11211643d5 : SUCCESS in 19m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b03064bba5b412f88a963bbafbf6439 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":3},{"id":"cde64b5a529922768400ef9a8aa34418fffa528a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-30 16:11:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b79f3600665fb9f61a35863edf6d7b11992e2366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 18:44:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c14274ee987240a7bb031b97e0fd382a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71fc79a4fe5543aab38a8f3483e5d9e7 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9fbae80c9a2432cb57aac10a0a67412 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38da5f82c9444531b48fe4c6bafad87a : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0120cbae44324b27974890d4a3070404 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63925f9ed5f5401298b8d9fd59e51a34 : SUCCESS in 4m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/770ce4a035f342c7a69d17c3ed7321df : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae72032fd44f45718763207d355525c2 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/411caa6ef1cd4d00b0ba7e35cf3823ff : SUCCESS in 6m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/41ddadcf626b424e9fa10821256bbed7 : RETRY_LIMIT in 5m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b05d0f9cba004002b09c7df213234f47 : RETRY_LIMIT in 2m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daa72c9c9ea246ebb1803b7511187742 : RETRY_LIMIT in 4m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/181778ee532e4700aea70562b3c6b920 : FAILURE in 1h 39m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e74d6844224240a192aa8914074b3466 : RETRY_LIMIT in 6m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/74ff457741264d6b853883868958e2d4 : RETRY_LIMIT in 5m 09s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4c7686c49fb44a691f8702a9ccff718 : RETRY_LIMIT in 4m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8f9d93eb1a9942deaec022c7969fe0bf : FAILURE in 2h 26m 28s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1650c6ad298403a89ada8884cda24c5 : SUCCESS in 1h 41m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b8c362cb603d4044904a214651489431 : SUCCESS in 18m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05d4d56f96ca415394c2a4ab71095b17 : SUCCESS in 43m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/634b3488df184086bdc5aacf62ab828c : FAILURE in 36m 54s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/79e81bd65c594d87896d34a24906d317 : SUCCESS in 58m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a57656c5c46410f8f2e1005c6b894fe : SUCCESS in 31m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fa765d0b528647099b15f546be3ed87a : SUCCESS in 14m 10s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f9cca14872344215bf0cb8457d5e4b40 : SUCCESS in 19m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e4e5b351ef447b5b8b1b0bc79fb2622 : SUCCESS in 9m 58s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/bb776ffe298141faaba188cc06a0abb9 : SUCCESS in 9m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a52770d542624b8a9753e137887195e8 : SUCCESS in 21m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d893c3e394c4f7d956d0814cfdb3feb : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1801726b39b93d225a629d3339d004dbd28d3b19","revisions":{"3029472ec3aa8b331c0416814effb07764c56b80":{"kind":"REWORK","_number":1,"created":"2024-09-30 11:01:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/930891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930891/1"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-30 10:57:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-30 10:57:23.000000000","tz":120},"subject":"Adapt Zuul jobs for Tenks using UEFI as default","message":"Adapt Zuul jobs for Tenks using UEFI as default\n\nTenks has switched to using UEFI as default boot mode [1]. Change the\nkayobe-overcloud-tls-base job to test bios boot mode instead, while all\nother jobs now use UEFI.\n\n[1] https://review.opendev.org/c/openstack/tenks/+/827553\n\nChange-Id: I650286298450432df83733c74a2bf143116134b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3029472ec3aa8b331c0416814effb07764c56b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3029472ec3aa8b331c0416814effb07764c56b80"}]},"branch":"refs/heads/master"},"c2df51744b9638b2b187c3c2bd4bf5fae3a6cea1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-30 11:02:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/930891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930891/2"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-30 11:02:32.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-30 11:02:32.000000000","tz":120},"subject":"CI: Adapt Zuul jobs for Tenks using UEFI as default","message":"CI: Adapt Zuul jobs for Tenks using UEFI as default\n\nTenks has switched to using UEFI as default boot mode [1]. Change the\nkayobe-overcloud-tls-base job to test bios boot mode instead, while all\nother jobs now use UEFI.\n\n[1] https://review.opendev.org/c/openstack/tenks/+/827553\n\nChange-Id: I650286298450432df83733c74a2bf143116134b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2df51744b9638b2b187c3c2bd4bf5fae3a6cea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2df51744b9638b2b187c3c2bd4bf5fae3a6cea1"}]},"branch":"refs/heads/master"},"4d862bd3eebd9ba4beb6171b93b30d79a7e53eac":{"kind":"REWORK","_number":3,"created":"2024-09-30 11:04:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/930891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930891/3"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-30 11:04:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-30 11:04:34.000000000","tz":120},"subject":"CI: Adapt Zuul jobs for Tenks using UEFI as default","message":"CI: Adapt Zuul jobs for Tenks using UEFI as default\n\nTenks has switched to using UEFI as default boot mode [1]. Change the\nkayobe-overcloud-tls-base job to test bios boot mode instead, while all\nother jobs now use UEFI.\n\n[1] https://review.opendev.org/c/openstack/tenks/+/827553\n\nChange-Id: I650286298450432df83733c74a2bf143116134b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d862bd3eebd9ba4beb6171b93b30d79a7e53eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d862bd3eebd9ba4beb6171b93b30d79a7e53eac"}]},"branch":"refs/heads/master"},"1801726b39b93d225a629d3339d004dbd28d3b19":{"kind":"REWORK","_number":4,"created":"2024-09-30 16:11:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/930891/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/930891/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/930891/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/930891/4"}}},"commit":{"parents":[{"commit":"3a7bff6ea1ccf40dd4a72e58f9ebe91224998896","subject":"Remove duplicate description of filename key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7bff6ea1ccf40dd4a72e58f9ebe91224998896"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-30 16:11:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-30 16:11:50.000000000","tz":120},"subject":"CI: Adapt Zuul jobs for Tenks using UEFI as default","message":"CI: Adapt Zuul jobs for Tenks using UEFI as default\n\nTenks has switched to using UEFI as default boot mode [1]. Change the\nkayobe-overcloud-tls-base job to test bios boot mode instead, while all\nother jobs now use UEFI.\n\n[1] https://review.opendev.org/c/openstack/tenks/+/827553\n\nChange-Id: I650286298450432df83733c74a2bf143116134b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1801726b39b93d225a629d3339d004dbd28d3b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1801726b39b93d225a629d3339d004dbd28d3b19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
