)]}'
{"id":"openstack%2Fkayobe~931155","triplet_id":"openstack%2Fkayobe~master~I15323b95199b9ef19504c7408dd89720f9ba9c32","project":"openstack/kayobe","branch":"master","topic":"journal-group-mode","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-10-23 19:00:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15323b95199b9ef19504c7408dd89720f9ba9c32","subject":"Fix group and mode of /var/log/journal","status":"MERGED","created":"2024-10-02 10:11:48.000000000","updated":"2024-10-23 19:01:51.000000000","submitted":"2024-10-23 19:00:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931155-journal-group-mode","meta_rev_id":"8fa62ee6998fe0022c81ed388c99e55418d39e8d","_number":931155,"virtual_id_number":931155,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-23 19:00:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-10-23 12:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-10-02 10:30:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-09 15:41:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-10-23 12:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 10:30:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-02 12:13:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-09 15:41:04.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-10-23 12:31:00.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d4f11dd96fa3d8a862565f75148a6caf63a76bbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-02 10:11:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03d77e794223a3a50d8d6019d453ef9057634727","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 10:30:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c58278a02e8efc9816e14da42a8629fcc310f8ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 12:13:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89fb00d563274547849e001206c64bfc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01afdf10e1df4d9698e179c94119ad1d : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5208ab2605cf4701af8783de466db3c9 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0079f3aaa184384a51e3f82b6e25546 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7f51e8b8e194319a1b67946c186041c : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc8ffe511744dc2b3dc27b41476a7aa : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e68d5c2d59e4dd7ba51566c5b66e72b : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cabf0467490348619aa560ed17929d21 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/89dc3a36031a453e83477c7dc4a314e8 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae30f6cd79b041e88f88eb551e366768 : SUCCESS in 8m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/395f9338f77145d3aeba401e0e07340d : SUCCESS in 1h 00m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6528f8e6995d4c82b72fedf1643b443f : SUCCESS in 1h 00m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e70996cdd9934f7f826013fccc232ef7 : SUCCESS in 58m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/eaa7e65ef1164810b5d0725cdaf937dc : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9cd688a55c434b6eb1ec786ed394ac7b : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d363aae645f1464e84f851a85bad7ff5 : SUCCESS in 16m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6323ad69fbc94ddda61eacaa4921a9c3 : SUCCESS in 14m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a069f644db6a493babb636b938081a40 : SUCCESS in 1h 51m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34bd35e1babf477f9b670b120b196486 : SUCCESS in 1h 45m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/93858d9c431e41c8b31c15bd953672b2 : SUCCESS in 25m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57cdbf4a77e54924afb0748dd404e5e4 : SUCCESS in 16m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3f9a6c96820044c1bf0189631a257dd2 : SUCCESS in 46m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b8fdd23403d84f668c0b4a4fb7f6577a : SUCCESS in 37m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eccdc0f5bf72491ebc9b7200f24e2901 : SUCCESS in 15m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f229323e0e544a42a8c5857ecc59ffc7 : SUCCESS in 17m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/15f79b7ff2c04487b275322ad0f22e02 : SUCCESS in 16m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/197b933b74534336a8a7b4923753b35b : SUCCESS in 11m 22s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/c0d8984334e5471bb3663423b11dd9db : SUCCESS in 12m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c5fd7ff8eac04755b12052ad82d7a90d : SUCCESS in 20m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d859d6a28a246f6a245c273c7fac141 : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"35ed38125aeb3c6c2e837478063b226ffab62fb6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-09 15:41:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"34ad5770a7a906c5b5ce9ba962862898490b1ccf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-23 12:31:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46d7fa57ba8f1eaf2a2a71561f5b4e1eb8333271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 12:31:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca9429d61f027b78adb6d80f5b5937a1f0c3f4b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 14:27:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78c640e1003445ec949ebe987f65c409\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ba7057f6b94b49987bd4cf0e87be3b : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13b2caef39b14082ad2b7f11d076e895 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16fc7f91e11e41d3a2c1d910915d5af6 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd3dba110e540c5891aa2383dc5bba2 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d8a076b65664e709a17116dd9c6934b : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa87f36996494e0ab6f36c4ca4cb31d4 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0fd0b53d7c1a4848a7a8d25ef7ea9647 : SUCCESS in 6m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f9ec3f6d74ea4dffbceb5aa1d17a2707 : SUCCESS in 9m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dabfc04e8e154c1ca51deb61091fc6c7 : SUCCESS in 49m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3d73abe825f4e7d8a683762c91f5915 : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c128e8715b7d4547883796a45b551f8b : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/61fc67344566463dbaa352a09923a250 : SUCCESS in 19m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd63d116ca8f4c73b48d3780dc0003bf : FAILURE in 8m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e20f5514b5034351adfb7eb31a0d8468 : SUCCESS in 1h 53m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae8d2c02cabe47efb8378ca7d42d009a : SUCCESS in 1h 20m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9f760b5eade840dc9cc87bf02f9bd673 : SUCCESS in 16m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8dec765ca3b4100ad0f081bb8ca41a9 : SUCCESS in 16m 54s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bac429c2bc494db4a4405c6bc5597e85 : SUCCESS in 33m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41d57cd7bc3a4c05830a6a2ffcc0234c : SUCCESS in 16m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f420a6ea1be2493c98ba22166227badd : SUCCESS in 19m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/794c4565fe934efbbd69eecb27f348c8 : SUCCESS in 9m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b35d67693b674a428813202c54690614 : SUCCESS in 20m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51aeeb5e98c347af8ab3ff1a13d3aa10 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1b57b46cb6f644b0b5b62c9805ec914a20317aa5","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-23 14:50:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbb9b635726b846601206fb502c03e3dfbc09b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:56:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4399290bbb6348deab4a8dcfeca2f9f1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a8e057535a0466ebebc2e6a4fd4d84e : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8561949821f4d11a9416c660f6d92f5 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccc5b501d71d4446ba522b94d22c403f : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/116bc20766ec4029b86277b2a82a59ac : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357af5afbaf948da9286624c39c6c021 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c2d0094bd9b422d874e13737db3e38e : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9df69ec230b04dd2944b2151b25172f4 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ed9212218bc4715a2b6a897bcff492f : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09d408f6c5d7412a8fa15322a8997c68 : SUCCESS in 8m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/726790be5a3f4c4493042d638ffae034 : SUCCESS in 59m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/493a1260129440558562e57a574a3632 : SUCCESS in 55m 24s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e166acc0fe4426cb4cb0b2be7a65de7 : SUCCESS in 1h 07m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5ef84e0e25b24fc794ac100613df7f9d : SUCCESS in 1h 12m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/61516b8fe9f640038588589670bb5de2 : SUCCESS in 22m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/571e81c96c5c4935bcfd259ac2f5c2a6 : SUCCESS in 19m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16220f37b57f435c8ee3f39a46821182 : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/46696eb76e6f4254b07953ec4dae703b : SUCCESS in 1h 59m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/917bc717a36443a884a383d9963e595e : SUCCESS in 1h 08m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4d2722af11924c32a2f3f52858072211 : SUCCESS in 15m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06606f73798942039a409de14796469f : SUCCESS in 10m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/266e8c6163c6407795cda6528a4f0ce7 : SUCCESS in 29m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7aeff7885a0a4ec9a58840800ce88361 : SUCCESS in 24m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b1bf2df781d4aa1b6a3ef1db6be367d : SUCCESS in 24m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/526692415934487088b218f260807d69 : SUCCESS in 15m 36s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8be04cb7d826444184e7b8643521bdb3 : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88d447f451cd4b4682c6b3543549fd50 : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/5780d1b6dcf84d08aecb94faa88f3897 : SUCCESS in 10m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/92d08ec88cca408db32faf81aa29e0dd : SUCCESS in 22m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/146597813fd34cbebe70cf6b96e20110 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"84c1ac46dab984fed2a8f97f3846185b282b77a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:56:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18429afb53c1bd2a7c0076b5dff9ca868d208443","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 19:00:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b1893b5560e4093a9dab13d66b0fd67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593a279e270c4e389bde2af43a665ae5 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de4701beb53e4b2e8211746794e8159e : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45511ca6d2944a37beac934badbf7f01 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb928e1e518496999726ba76945f652 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7301cb338b942688893d5d2bc907904 : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78fd3d8d4816403e813415375b9e09c1 : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/738fdc9269a044ff847b5af83fb568a3 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7b1507289554052b0eb71c28825ed43 : SUCCESS in 8m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d279b3b276c14f27b49c7c94152c3afe : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afe4edd5f1084b8784c14e034f09a96c : SUCCESS in 37m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f825e4d1b642421192bc81c6fe47a23a : SUCCESS in 1h 05m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e008c3b7a20a430f9ac4d3b23bcc7868 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c68341c4ca1747e4b3dd1f22ed65b6e9 : SUCCESS in 10m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a14a71946cf484cadfe914990147d50 : SUCCESS in 1h 58m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e97238262724e3ab718007186c6f665 : SUCCESS in 1h 27m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4ca9c739a6d244488b808db6012fb1dd : SUCCESS in 18m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f34458f42196482190cc8cd4712a187c : SUCCESS in 20m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c4c7c1199935406ca240a7dcb361ece3 : SUCCESS in 31m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7716376540441be978d7c22bd5dba99 : SUCCESS in 16m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ef21d3f4d5d41808e886a7e90449196 : SUCCESS in 17m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dee9ce31742147cf9a93e899e225ca5c : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1cf45d1b378b49019cd9a88c4e3e49c9 : SUCCESS in 20m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6e337a280fc4527af426468aed9311e : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"158074c95629bc75839aaa66d2c05ec1afa46ba0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 19:00:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8fa62ee6998fe0022c81ed388c99e55418d39e8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 19:01:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f779447a00cd4491940b259911939283\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe87edc3dde4f008808d4320ecf79b7 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a0461df5a404b4ab26970dace01949b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ed077cba4d1c5f627c3d5a2d803105ef2416240","revisions":{"5ed077cba4d1c5f627c3d5a2d803105ef2416240":{"kind":"REWORK","_number":1,"created":"2024-10-02 10:11:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/931155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/931155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/931155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/931155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/931155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/931155/1"}}},"commit":{"parents":[{"commit":"8d58ad0958eace90e5af342ce099a9202c34b174","subject":"Merge \"CI: Workaround rc:-13 issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d58ad0958eace90e5af342ce099a9202c34b174"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-02 10:11:28.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-02 10:11:28.000000000","tz":120},"subject":"Fix group and mode of /var/log/journal","message":"Fix group and mode of /var/log/journal\n\nThis updates the group and mode set on the /var/log/journal directory to\nmatch default ownership and permissions used by systemd-journald.\n\nCloses-Bug: #2083494\nChange-Id: I15323b95199b9ef19504c7408dd89720f9ba9c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ed077cba4d1c5f627c3d5a2d803105ef2416240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ed077cba4d1c5f627c3d5a2d803105ef2416240"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
