)]}'
{"id":"openstack%2Fkayobe~933145","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I15323b95199b9ef19504c7408dd89720f9ba9c32","project":"openstack/kayobe","branch":"stable/2024.1","topic":"journal-group-mode-stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-10-25 14:02:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I15323b95199b9ef19504c7408dd89720f9ba9c32","subject":"Fix group and mode of /var/log/journal","status":"MERGED","created":"2024-10-23 19:07:14.000000000","updated":"2024-10-25 18:13:46.000000000","submitted":"2024-10-25 18:12:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933145-journal-group-mode-stable/2024.1","cherry_pick_of_change":931155,"cherry_pick_of_patch_set":1,"meta_rev_id":"f1836e1b5022929eded2cf227a92153217b4997b","_number":933145,"virtual_id_number":933145,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-25 18:12:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-10-24 08:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-10-23 19:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-24 12:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-24 12:17:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-23 19:19:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-23 21:50:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-24 08:34:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-10-24 12:17:25.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"6709d390347a67b391f1435e7433185a0220bdb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-23 19:07:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d9b61f7b65fb45edb0273f5c950a2a25c110bc38","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-23 19:19:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4cc850c34099abeb9038ef7b241ab1b86806022b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 21:50:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54207bde5bb64e3d9094e8efdd99d136\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/165eb6f978874209b8948e1d54d8c085 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ef2de6b18b41f2862608a5aef685f4 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ffb54ad04df4dfba998067275a943f2 : SUCCESS in 2m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/912ccb5c36d14178b8321d98ccec9207 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9624429efcc04b6899224a6fd37d7eb4 : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/038da480c59149fcafac39d513c38b5d : SUCCESS in 7m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5dbd2fd23a4f431285b7cd3d77529580 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1eca1f31481a4f61931041d42094c6a9 : SUCCESS in 51m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f458cddc3064c88afd728b06308141a : SUCCESS in 55m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1ce248a2ce9147559e35f8eba0cbf04e : SUCCESS in 59m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/79935cfb88b14bb685b2e414720181bf : SUCCESS in 21m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/03d00f49243a41c3811b49ecc11ed5da : SUCCESS in 18m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2902ab0514cc48b092e44bbce5838eeb : SUCCESS in 12m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a1cdc396d98491fa5f13cbc71559743 : SUCCESS in 1h 46m 27s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97dce653bd4140ceade487374393ef80 : SUCCESS in 57m 10s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a0a32a010a9546319aec0a98be6c32fc : SUCCESS in 2h 32m 54s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6990daa7d8848dc95170ceff33a0d1a : SUCCESS in 1h 48m 42s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5a8cfb83a74548d282339c57c794e4ef : SUCCESS in 19m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5de131eac784b80bcd1ab582ec0f10a : SUCCESS in 7m 56s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ce6e5c439d334fc8ba730188a15761ea : SUCCESS in 37m 05s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dbec59e8b88f40f69834ccce04ae20d0 : SUCCESS in 32m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa8bbacbc241407f814a2ae9194780c5 : SUCCESS in 23m 07s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2281f396c39341a5a50e67e1ac43b978 : SUCCESS in 23m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03fd90bdad0f41c9a9cb2a33620538de : SUCCESS in 16m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/def7b7831158474c8330e071eb51faae : SUCCESS in 16m 49s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/6b75f4580270487d8091dd64a863d1fb : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2639f318cf124e829464637d79b54c8d : SUCCESS in 11m 45s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/6fa01a93f8744cfab678c918a861ed06 : SUCCESS in 10m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/78bf20e293c34a17a3274bec886c6398 : SUCCESS in 19m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f05fb334ba443abba750f263af63377 : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3dc98e3f094a14a242200b232a4a6b : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/527ce314bcc24b1ca32a0cf55275fd31 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ff2887771b946b689b26c487142365c : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c24e82a37e6489dab80a3f95b57d0ab : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"83ff3d2344636b7e2b182480378ecedc4d0fd636","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-24 08:34:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98d2f430db4fb01d6333b063af8a599bcb4e89cb","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-24 12:17:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7b4a1b9574ee5564c159d959a389b7ace121b5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 12:18:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"593911882c7f6a09e6626f5538ca4df13b8c1d31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 15:21:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce7a803e9fab46a187d4d7bc7ce7e3be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e112b19d911458a9d61e96e3e6e54a7 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dc94e1453e94c8ab0f01a26766eca2b : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f259f2db37c4247903685a67e8bcff6 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d78fa80d82543ec80e5c7f82b152eff : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c46e07b47cc149daabc06a9cb481e2f5 : SUCCESS in 7m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fa5c0417ddf742d2bec2e1f26cf979a0 : SUCCESS in 41m 25s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d37d20b1fa3d4a869ae319adab2b9d73 : SUCCESS in 32m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a46196b2b57247bdb84a8575d0fe7245 : SUCCESS in 46m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c2acbba967ae483092bc9aca09ffdf5e : SUCCESS in 15m 16s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e28dcbd6f22e4fd9a1cd1a8df9bd19fe : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08274209e50a4349963c8efc73d13038 : SUCCESS in 1h 57m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bca98f8f7d994593b8fe0ed070b3e472 : SUCCESS in 1h 45m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/74be2287deb6427f909f5250984d7563 : SUCCESS in 22m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51a08069d6754b1a81c6c2f1bd785a95 : SUCCESS in 9m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/063b53a7f5f04852bd6fa56a124dfc06 : SUCCESS in 34m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00d66925099744ee94d5b1e444964b61 : SUCCESS in 21m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6b7a244c686048ce9f86970f4039d7ad : FAILURE in 13m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a1a088c3d23436780b1770dfc682ea1 : SUCCESS in 25m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6cd8b957aaf64f0d890237905cd915b4 : SUCCESS in 14m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13633be96e08469fab74462187139720 : SUCCESS in 10m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/88efcc2b99474af5bef9d9f2ee9a3e68 : SUCCESS in 20m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a106f528d334f67a1a45e9f4e34c286 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a5761cc68b2bc3aa53be957dcdb6b492efb905e8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-25 14:02:49.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f684527727b6b92838930bf5351201d48c4b1c30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 16:11:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/649a851eaa7c4cb1b3d6ac41cb9a4d43\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8fed859a7d242db80e81da49e411513 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a94e5fa5440b481db6fe4fded9b38f5b : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/685c434faa4940c18ca39c197dcf3704 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8fff3551d7e243ba8a36c849ca09f098 : SUCCESS in 3m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dad8c89e96174e948227668eb5fc44ca : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c23d2c5b748427da89e425721c110ac : SUCCESS in 9m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/275bcd9d9cd046639572d49fbd329028 : SUCCESS in 1h 01m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f2cae34d15694b7d99bc2f84a07294d3 : SUCCESS in 59m 39s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a63685df5ec4638b6da55ae3849ea8f : SUCCESS in 48m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ec7a2e1af507495b9fa5b71c6f5fa79d : SUCCESS in 1h 10m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bf0800ea247f49cfb41ce7dec8742a76 : SUCCESS in 18m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c048c4fb63b844a089e2847f7a42e1f9 : SUCCESS in 16m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7953c18ca5554c50833b55b960252a9e : SUCCESS in 12m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bee89fddcf834b0fa7bb2bdd06fe910f : SUCCESS in 1h 41m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/caedf2dfc4f0495fb54f98e522a3768d : SUCCESS in 1h 33m 33s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c533f99370f94ab7b522fd4734611522 : SUCCESS in 2h 07m 04s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6492c52dc85246d6a11a7815f4b9718e : SUCCESS in 1h 50m 54s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/728b2f00924f4f21892f0972ad080449 : SUCCESS in 18m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ab16d91006c430c923b4d9fd6effc6a : SUCCESS in 11m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/85c0e1261f644cebae198a56ca3c2e20 : SUCCESS in 42m 33s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cbe94d1f4f0e4407a715d3ec0a0c6e0c : SUCCESS in 28m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e23b6a9941be43b5ae0d5ba85c74284e : SUCCESS in 27m 37s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c2f7d2813fc94addb4e392abb920f89a : SUCCESS in 34m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3fa2ac6b0a5474d8dc029959c69bb20 : SUCCESS in 28m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99d0b062209e4777bad9e408c9d2092d : SUCCESS in 17m 36s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/78fe42a63c954b1abe5162cb31c06941 : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c6e337188014711bba082906f21dcac : SUCCESS in 11m 41s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/a87523e0c10b45c3baa68986d2922c6e : SUCCESS in 10m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e129ff82afd94c43bc92634b8aba6f49 : SUCCESS in 21m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c98a8c4f979b4f73bb3686f53856f554 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/effbc5814bfa4cb2b1dd8f2164b33fd5 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ad11e798b3f4e0e8d9a676f81db0093 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e50fc08b251745d3a747e99e0d78fbd0 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8081ba503940406f8144ea58cc9137e5 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c10ddfaf8c1350f55899c71f343e2e9d8822caf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 16:11:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ca1065a4d41b17ef9e791f11b84079b86803822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 18:12:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55ea99168b314ec6a00551f20c57a416\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a047c09d014ca5bba237f5ae29882f : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5e5378940204c9abaf70ddbdce60d0f : SUCCESS in 2m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ec16e76e2ba44e4b11aad6a152422cb : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e4fbeba29694536be52b3ff5d2b86c5 : SUCCESS in 7m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e1ba6fb1fd7744bc8275473a7a4b92ef : SUCCESS in 10m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/338c1efa42cd4417ba2560a33cbe45cf : SUCCESS in 50m 31s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f87c168d63c4529b022e67eef7ea365 : SUCCESS in 48m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/db1bd300196d41c7aacf4863091a66dd : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6779120fd0a24902abcbd4569970d446 : SUCCESS in 19m 19s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69cf9d2082684b58a862414114cc9e61 : SUCCESS in 15m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/04b81c758a554b29a8fcb940affc5ea3 : SUCCESS in 1h 55m 22s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9bdd5d0109e4383b47d2597052fe6d2 : SUCCESS in 1h 26m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/21aa956ee1e64832888b1628f3b621fe : SUCCESS in 19m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e69cc3ab201403d821f7060649f5269 : SUCCESS in 14m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/159b621c132b43689ddc3b1bcc1c7abd : SUCCESS in 28m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff34e8413bcb4cfaaf8ad9c3155a2d6f : SUCCESS in 25m 39s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b5380e76e63f4f43a94e4972a0644e8a : SUCCESS in 36m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0134c42196ab490396d022d2bce66c14 : SUCCESS in 26m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f12a44feef894bd09ef0110376b17334 : SUCCESS in 14m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c04a74533b6c4c39a6e983126590d30c : SUCCESS in 9m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/399da5590c5f43dd8b5cdcadca554a56 : SUCCESS in 18m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3adebced4e2c4e4ca160e9d953f3ad4e : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"755fe2d67d596ded98824c4ca56313307a5456df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 18:12:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f1836e1b5022929eded2cf227a92153217b4997b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 18:13:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9194c0f437948458322283290ae4d1a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d500fa498334b0bb3e29bc36caba4cc : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a49cea07888a43aaa4e05c4572e880c2 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd47742c22742875fd86305804431537d4224501","revisions":{"bd47742c22742875fd86305804431537d4224501":{"kind":"REWORK","_number":1,"created":"2024-10-23 19:07:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/933145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/933145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/933145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/933145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/933145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/933145/1"}}},"commit":{"parents":[{"commit":"d8c9a9bb59bc3368429fb3d0871ae2a465dc45cf","subject":"Merge \"Remove duplicate description of filename key\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8c9a9bb59bc3368429fb3d0871ae2a465dc45cf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-02 10:11:28.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-23 19:07:14.000000000","tz":0},"subject":"Fix group and mode of /var/log/journal","message":"Fix group and mode of /var/log/journal\n\nThis updates the group and mode set on the /var/log/journal directory to\nmatch default ownership and permissions used by systemd-journald.\n\nCloses-Bug: #2083494\nChange-Id: I15323b95199b9ef19504c7408dd89720f9ba9c32\n(cherry picked from commit 5ed077cba4d1c5f627c3d5a2d803105ef2416240)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd47742c22742875fd86305804431537d4224501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd47742c22742875fd86305804431537d4224501"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
