)]}'
{"id":"openstack%2Fkayobe~933464","triplet_id":"openstack%2Fkayobe~master~If0f3ab6f26aa774cf6176d992ec1180f2c0d7838","project":"openstack/kayobe","branch":"master","topic":"replace-distro.linux_distribution","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-11-04 15:10:08.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-25 23:25:00.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-11-04 13:15:02.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"If0f3ab6f26aa774cf6176d992ec1180f2c0d7838","subject":"CI: Replace deprecated distro.linux_distribution","status":"MERGED","created":"2024-10-25 21:15:23.000000000","updated":"2024-11-04 15:11:12.000000000","submitted":"2024-11-04 15:10:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933464-replace-distro.linux_distribution","meta_rev_id":"6acbf62ff4fbf243b1abf8389c1f69294f149439","_number":933464,"virtual_id_number":933464,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-04 15:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-10-31 13:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-10-31 13:30:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2024-10-31 13:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-04 13:15:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 21:41:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-25 23:20:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 09:37:06.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-10-31 09:44:29.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"3a105754a37f6d30b08e2c54625b923a13314521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-25 21:15:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"782fb271be4de1c3c4a9a926d442de54e748b81a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 23:20:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0851bd88116949909bdcbe0cb8770784\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06823c4fef0548f5a2d62efb2c099e22 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11061213c454a20ae169bb3c30e350c : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64d2cabcf30f439e8b3a37faa89fdd90 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53b2cd63f5da48d6843da4504a1dda74 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a74ee12f73ca4304b47ccf95cca0fa05 : SUCCESS in 7m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab1d93d27451479faae911ceeb8f1d77 : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e77899ffdc6d472eaf56e9cda41f1f3d : SUCCESS in 8m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84a6807fefcc42899325ab9876cb16f1 : SUCCESS in 8m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c80f97c3e85b422a9187366f47b876e6 : SUCCESS in 1h 06m 37s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/627a22a9f52b48d390107e5fb1ed37c3 : SUCCESS in 46m 40s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38cfc1e231e64b74bac8144f6a50ec2c : SUCCESS in 50m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e6581ba1a5cf4a53a0b5c95a40902e35 : SUCCESS in 1h 12m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/428f90f72a604bbcbdf8d1ad3b3235e6 : SUCCESS in 19m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/40deedaf6e3643fdb7fa711c09905a2c : SUCCESS in 15m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f29e640db1645f091a80794895f411b : FAILURE in 8m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e418639e98644a39cfa1652e12c79a4 : SUCCESS in 1h 56m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a3fda68928b4f43b224cfb9baf6249e : SUCCESS in 1h 19m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4af86378b4c34ecf848445f9dc58062a : SUCCESS in 22m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c05c1f23c5fb465d8d99800fe1edcad6 : SUCCESS in 13m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4619e711770a4e78ae31acedbac2d37e : SUCCESS in 33m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bff4a4a9d18546e08ef7b8286649d10f : SUCCESS in 20m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff4b5586be654c7794df936cd28bcbbc : SUCCESS in 24m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/54a12dd7094c415fa559603db134e639 : SUCCESS in 15m 00s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4bab94f67a8a4d1280b3b0f576a933e1 : SUCCESS in 16m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7910e0c9bb746f8abb7a5fcc280553e : SUCCESS in 11m 03s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/926df08aa03847e08a45befa13a6e91d : SUCCESS in 9m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/402be72018a44be080aacd7544b035f6 : SUCCESS in 19m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9ee282461524d9698af05c62ea74b41 : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9ad4ddaa6a622fef9c4ecff488f925e4aed7ada9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-25 23:25:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"799d2cd786f7a5cddc309c7f96816abc0fb09150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-26 09:42:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eeb9a0e9d7eb312000ae91f79b98b130c6efe0bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 11:32:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e76bab994fbb4ec985e77a174e8c7321\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad6c391fac65418eba1d690f605fe244 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b970b9a60145208a5dfc4f55faafcd : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd739dc65e1249ec8902302fa0e7298e : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3b49fbe7d994136ac2546341c9a4c97 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27084fa3d76a4038a8db17ada4086298 : SUCCESS in 7m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f9bb77d74c14f3ea24ea6ba695573b6 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23a00f4a4dfa4325938ef539592265fa : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/052ea9dedbf04bd6b78d6eead1b5fbe7 : SUCCESS in 5m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/876a2c23a1044b8dbdee17976948c4e3 : SUCCESS in 46m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/55b03a3ee1b241cd89c771c34ef91104 : SUCCESS in 55m 45s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b6f320bafe547c69fa4d3c642b07d1e : SUCCESS in 31m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a7a8fe5fdf364398b87e49839189e236 : SUCCESS in 1h 13m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6859b2b1c4314057874af784d8ffac8c : SUCCESS in 17m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5e0eb66bab154acfae04d375f2555e39 : SUCCESS in 15m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b3ee4caab8e40fb8d314cdb6ff060cc : SUCCESS in 12m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/86d51e0fb2a044d4b349c52b8dd310d3 : SUCCESS in 1h 49m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a551e5ffad2143e885d94f21b276ddc8 : SUCCESS in 1h 28m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/65cfd3eb03a048f3babb31a2b8961745 : SUCCESS in 12m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82f1451a9a784c7d83e803f461a9fca3 : SUCCESS in 15m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4bff5fc2aecc4f11b8c9fa8597e19913 : SUCCESS in 32m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8bed831abf054424acf1d5a5624c00e5 : SUCCESS in 29m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e0a0e3aa73a4d1e8f7787512d54aeae : SUCCESS in 24m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b574ece09a3e4cfdaed6ac0ba4404eb2 : SUCCESS in 16m 48s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f874e40770fa4c8285212ff61e2341fb : SUCCESS in 10m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69dcd3aef3a2487d864b604cb68d2ce8 : SUCCESS in 9m 10s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/fe50f1e890cd4f958fdf459346ea7731 : SUCCESS in 9m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f4ccc939dfb4459c8b09cc3a5e945fd8 : SUCCESS in 19m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d90037fb038486fa663610d2e8b3c36 : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3903757b749a669ebd8245754c807bb6853fa2cf","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-31 09:37:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c33625be2b080669fb001728aecf387f3975761","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 09:44:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"28ae9462a80ba72a7b4e7efd742a99bced65a895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 13:30:22.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"668586de254bcfdd6fe305fc50de6090e1b85406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 15:26:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a13482c308bd48e7a760e80cdb96db5c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc482d17fb1c4a82ac23824165f20262 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47fc4385645341f2af201ab551e6a16d : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c1e96ea23b44a2fba09100adf674045 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37ab9a9a32c9405f929f862bda7fc147 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5172f4179181402fa62255806b0be6bd : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5481456f9fa240db9aed81da57038c5f : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fab7440d08f9439da56a25fda627bbd5 : SUCCESS in 7m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b82bdb21eb244448ad234c08ea833c7 : SUCCESS in 9m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c019c4f8d36e4bbfae8e40f2fc301cf0 : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/486bb1f1a5424a09b4d453f5e7461618 : SUCCESS in 59m 46s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84ae5f65367343acb39974413db07ce6 : SUCCESS in 54m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f20429a89f4c4e529400ab622303c0a1 : SUCCESS in 1h 06m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/49155c2732584ba18d95474148f74687 : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4da1620f2db54ceb834253f38395d490 : SUCCESS in 16m 36s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e06147ca4cdf4c9a9ec62fea8050d0d9 : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3f4bebd524904174b1a66dd84e115353 : SUCCESS in 1h 47m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f2d8048f9ec45f69b0b2962f34b5a34 : SUCCESS in 1h 36m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4f0ec516340f4494bb14e8513b8f23ce : SUCCESS in 24m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76434615653840be862039f6f3512eb9 : SUCCESS in 16m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e914ef2d928e40ebaea1a72ff8be4a55 : SUCCESS in 39m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd7dfea941b045078e4f9bde02e6d6e7 : SUCCESS in 32m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac0b7231ae174355ac4706da4f7389ec : SUCCESS in 24m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/370cd14ba1cd4de1a3ea82a9e8cf0edb : SUCCESS in 17m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1cb0b2d5c1824d3dbbee498e131a578c : SUCCESS in 16m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a183f2947c204137b4a63d97f54b9e68 : SUCCESS in 10m 34s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/221ec162aa644fd6bf47fed83e212d60 : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e1c5cace303248d4ae31b84eb15f5da3 : SUCCESS in 22m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e0ba4081eca48698c85b9e6e46970a2 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":3},{"id":"a3d388d5c5e8cd61348e633175e027a7c2e53374","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-04 13:15:02.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"15e26656ebb6cd1f8738a1a059fc6fc8855cc583","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 13:15:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2efbd5495400415a2443b30701292494828e3d48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 15:10:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7503cd364954c8984c1f250cbafc977\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe13c9ed1614928bc674cdbdce1fe92 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe526d47f54748db87f17bcecbd716bd : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4055c3a800e54e7b8bf1d437caa1b48f : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b7f68664c6416b9a9f267898638b90 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e9b87345d87423a83ddd2ad2055dd9e : SUCCESS in 3m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1eb60d423684492893c46a03aa1b000 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14655dab64c944208941b1328bf87147 : SUCCESS in 10m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2c0cfc5741ce4278bf64acabd05902c9 : SUCCESS in 57m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fee207e7d632473eac6f3c6cc1c83d73 : SUCCESS in 53m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/62285d422688427384fa11d3e939f992 : SUCCESS in 1h 14m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/654bf29fff6e4ac2b5f6a350099e5d46 : SUCCESS in 20m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16e8bd6b08fd43a3b96a0a0afa1640bd : SUCCESS in 8m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/24c234224c0a4e7a93c87e329b0b0c16 : SUCCESS in 1h 43m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79860c973b7344ae9fb1b7444a731c76 : SUCCESS in 1h 53m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f17147cd95104129b90c484a4b04e981 : SUCCESS in 17m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bc06616ca6443efb06007e5ff04934d : SUCCESS in 14m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b1a0d0cef3fd4a86916135d5ff4eabef : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/def1d17363b04fc89e1368d18e310707 : SUCCESS in 25m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3286991b378843ed9625eb7e9abcfe7f : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/953244f8fc6444f294444278513013e3 : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c68f27ceef754fc895adb4a285cbc2c2 : SUCCESS in 22m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a524bd0546d42e9b66f687b5f2a312a : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":3},{"id":"75bd2c688830fda0e4bbf67064fe49fb81d47a0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 15:10:08.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"6acbf62ff4fbf243b1abf8389c1f69294f149439","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 15:11:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be868f1b3494868a45df51abd3c2fc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329e575460444488917c3eba3d06abb0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"70742ebc27ed7b39c36eb26a7f6d63d07a753601","revisions":{"a68cd077e3ff8de75fd06d626bba419f39320fae":{"kind":"REWORK","_number":1,"created":"2024-10-25 21:15:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/64/933464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/933464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/933464/1"}}},"commit":{"parents":[{"commit":"9b53c0ee95c02d5d945c7f690f77ce53bab95145","subject":"Merge \"Fix group and mode of /var/log/journal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b53c0ee95c02d5d945c7f690f77ce53bab95145"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-25 21:14:32.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-25 21:14:32.000000000","tz":120},"subject":"CI: Replace deprecated distro.linux_distribution","message":"CI: Replace deprecated distro.linux_distribution\n\nChange-Id: If0f3ab6f26aa774cf6176d992ec1180f2c0d7838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a68cd077e3ff8de75fd06d626bba419f39320fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a68cd077e3ff8de75fd06d626bba419f39320fae"}]},"branch":"refs/heads/master"},"db8a9932f182358509e9e832131448577ab7e098":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-26 09:42:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/64/933464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/933464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/933464/2"}}},"commit":{"parents":[{"commit":"9ec381913895143de35fc4b79ba95413ba1d472d","subject":"CI: Skip test_ntp_clock_synchronized on Ubuntu 24.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ec381913895143de35fc4b79ba95413ba1d472d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-26 09:42:13.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-26 09:42:13.000000000","tz":120},"subject":"CI: Replace deprecated distro.linux_distribution","message":"CI: Replace deprecated distro.linux_distribution\n\nChange-Id: If0f3ab6f26aa774cf6176d992ec1180f2c0d7838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db8a9932f182358509e9e832131448577ab7e098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db8a9932f182358509e9e832131448577ab7e098"}]},"branch":"refs/heads/master"},"70742ebc27ed7b39c36eb26a7f6d63d07a753601":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-31 13:30:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/64/933464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/933464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/933464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/933464/3"}}},"commit":{"parents":[{"commit":"2fb57a72a7cdd24df5f52a7cde47b5e6e68a08db","subject":"CI: Fix Ubuntu 24.04 version check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fb57a72a7cdd24df5f52a7cde47b5e6e68a08db"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 13:30:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 13:30:08.000000000","tz":60},"subject":"CI: Replace deprecated distro.linux_distribution","message":"CI: Replace deprecated distro.linux_distribution\n\nChange-Id: If0f3ab6f26aa774cf6176d992ec1180f2c0d7838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70742ebc27ed7b39c36eb26a7f6d63d07a753601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70742ebc27ed7b39c36eb26a7f6d63d07a753601"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
