)]}'
{"id":"openstack%2Fkayobe~933538","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I683b685bf4e2f625144e4656bcd69c03eda7aa51","project":"openstack/kayobe","branch":"stable/2023.1","topic":"proxysql-stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-11-04 08:23:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I683b685bf4e2f625144e4656bcd69c03eda7aa51","subject":"Add support for building the proxysql image","status":"MERGED","created":"2024-10-27 05:45:14.000000000","updated":"2024-11-04 11:57:48.000000000","submitted":"2024-11-04 11:56:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933538-proxysql-stable/2023.1","cherry_pick_of_change":933406,"cherry_pick_of_patch_set":2,"meta_rev_id":"0fc511c094d9bcda2f994f028ae539bd12766c3f","_number":933538,"virtual_id_number":933538,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-04 11:56:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-10-31 09:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-10-27 09:43:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-10-31 09:52:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-27 07:43:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-27 09:43:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-31 09:52:21.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"4036aa5383be38e97d3fc9bb6b3a53b7915fba30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-27 05:45:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2e66d292116d1a0859ecf6cc7b7b08113a9c835f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-27 07:43:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98573f656cb94312b25feeb792c19a5e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e63df877e4b4babb8dd6ed1eab74501 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ad5423e2a74d35b4453336b5268987 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0b45f83d3264f468c98592b3b74f310 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5efe89f76354d48bee79b551dd2cc3a : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df3a07baff3940e5b8319ade0a6012d5 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/50083f2baf654f6887b33c1547e06bc8 : SUCCESS in 3m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f499f682582476392bc433f11192190 : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c24b338bdc04097a00438777041ea42 : SUCCESS in 10m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/19b6e0377f164881b5789461ac6facb0 : SUCCESS in 59m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ccf5ca18f5dd415ea772404646ac113c : SUCCESS in 56m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f891120c26a442b682c4c7f78c215c0c : SUCCESS in 30m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7df7716aeb2f4c9e8f5fc6ec7818a978 : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2006d2fec29543faa7054a4549b884c3 : SUCCESS in 13m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ed320bbc9fb1436287597af2813f2a8b : SUCCESS in 16m 40s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dfbe4763fac4b30a399a028d01ed35a : SUCCESS in 8m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/43d8524999054794b6e53e2c261d0502 : SUCCESS in 1h 46m 42s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52216b82e38c4b048c4f26fcf1155028 : SUCCESS in 56m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4466d8158b2849619c6536d17528358f : SUCCESS in 21m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6289627f6535477c9cc7d326848e0505 : SUCCESS in 15m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4b2fd920ea3a493bb89a47bddebefc4f : SUCCESS in 39m 45s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9359fda45ab24c5597a84a620c4f20c9 : SUCCESS in 26m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/badb01e5bded4ba5b899d5eea17e327d : SUCCESS in 28m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c5439371f52a46c7a8fb9401ec0a4fc4 : SUCCESS in 15m 05s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/da83634ff11b4662a695314dc093cb0f : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8fa422b721644e1b169fe9b1d2e472d : SUCCESS in 10m 26s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/87b2c1114d8049f8aa0217ba320c67d3 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d7012e005e814b7e8de61dd0a28f7314 : SUCCESS in 20m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0bc1848095347fb904c59a252bdc599 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"628fc5cc9d468275e724212e14d9018591f2e07f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-27 09:43:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"03512e2dd71c83163a1d1acff0807add8a46644e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 09:52:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ffea609b485f01bf2251c077daa14c05772a3f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 09:55:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58e2a84f74886198bfb022620b3af89d81a1c381","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:17:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6138789dadfa42ac82d7f10364abf1a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f336c09c35794deda35a7657afbd7007 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1b73ccfb5cb431589bc56d34274e990 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542240b21e6d44bfaf85749054c15ef3 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/affd9d683ac248a5b49e0449e6bec89f : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75a7afd815284ae5adc3e21fe32c7ca3 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a9a60300ab74a0daf14fbe86ab8cae1 : SUCCESS in 10m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/152e840e4f604ce39d474393e598beac : SUCCESS in 12m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b31d73c94fa3481097722dfa4a08d839 : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c11ca29cf2b2461fa3ee829674c05e00 : SUCCESS in 55m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/80c42421e0a04c2a936dfb3b05b6f5c1 : SUCCESS in 59m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9910c43abaa842b29e9b334c3039f403 : SUCCESS in 19m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dab01bab27b4136b2e260ccb7742968 : SUCCESS in 15m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2b78913ae0b8406eb70f20a32f34a1d3 : FAILURE in 1h 27m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/321ebfe18ce44f739d1b427d19e396a0 : SUCCESS in 1h 31m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d28970b333a242ca89ec86bd3d52a830 : SUCCESS in 17m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8530c6a52fc94771b0b77711508e092f : SUCCESS in 13m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08384bab9772450da7e0bee80b0f7adc : SUCCESS in 36m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77cbce2d548d48fab15a2f3c99b36256 : SUCCESS in 15m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6ae11ab2fda746269520a9b4c0d5738c : SUCCESS in 15m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/151e10daaff042908298b1ac31901a08 : SUCCESS in 10m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/520d2e92760e44be97d89a6ec9cb35c7 : SUCCESS in 21m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a5d34e2388e4fd2b7a30fd14c63224d : SUCCESS in 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4da364321a23607a59f999bb740e51d0a007102a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 12:25:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9b79769f77cb0fe3bf73cbc81ed27d4b7e6e469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 14:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b23e92891ccb4c448525b4ee7155b94f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af2587dab0dc4774afc1f825a8544cc7 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930779bb44ea4bf99b2a31708258f526 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe080369e5a54f348520ccacacafc02b : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc892e5683d24a7c9a5043c046672c3b : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f05db052746f413799dddbbc9552cec3 : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/518316228e5547e284be746bab513a0c : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b8e8b8c8b12433b8cbcd4fdb144f896 : SUCCESS in 6m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c947b7dcdd5418d979de5c18c0a5984 : SUCCESS in 14m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2627664ef5324620bdaaaa32669a7227 : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c7bad2cd2b1e46aa8e125a49c11302e7 : SUCCESS in 53m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c244efb8fa844936aed4efb27cc0f79f : SUCCESS in 55m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/15ac3eaa8f9d4bb095e34b82f4aa04b2 : SUCCESS in 1h 11m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2a01917c41d84f27a3aa4d8d0f5010c6 : SUCCESS in 17m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d91670c7e2de421eb9a2c23e04bcac38 : SUCCESS in 16m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acaa778571f54c6c947f7c562e2bcbae : SUCCESS in 14m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8067e67da2614996ad3234b88c12f235 : SUCCESS in 1h 38m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e2c970fa43947d6b6f682d6a76e0db3 : SUCCESS in 1h 37m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5c36de089b584146813c9b3ae2c4ef6a : SUCCESS in 20m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87ef183add8242e5a57262286c026902 : SUCCESS in 14m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/abfcb11d1e6041b29e7be7778b67ed94 : SUCCESS in 40m 22s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/053b1cd7660a48ed9ccb1aefd57b8910 : SUCCESS in 32m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f40bd489be4b48648a092b30b7cbb7cf : SUCCESS in 23m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fec40a869b5441c293e63d972355fcdf : SUCCESS in 15m 09s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1456b3498ae34eae9c92625a850e6389 : SUCCESS in 16m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6a7a37c2bd541ffb1b1f2bd3b8a58f3 : SUCCESS in 10m 07s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/52c301738db6408881acabe77274be14 : SUCCESS in 6m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5bb7923711144b868dac535026d1c27b : SUCCESS in 12m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/137209b350f74c4da4a826d45430b337 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d81fc8f47a50848a0121857d5f84f05a0459f86b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 14:09:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13108ba1cf595678f6a483932d9695a9ccfe7e8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 15:56:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/356647ed89574dc889f5a7fedce52c62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a44b35ced26c4d738e5dc7d9a069d858 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b48238aa24a4c0c9ad03b22584c2ebe : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3dc8ba739f247d29b26c3e7cd5ba4ba : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91f1c4ddb6ca47cfab15b6f96c566675 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/065c972eb3714896942dd380425dc16b : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/626bb70ca74e4ad4bf484d307d52b00d : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5eab264b60f34bf897f7c1f8197dfc11 : SUCCESS in 8m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/99fc89d5d9754fcbb45abba2c66a3ade : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88a6c5ac728048068e8ba17b2ed4826c : SUCCESS in 54m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1360e1d5473b44a2abcce42991ee7969 : SUCCESS in 1h 07m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3d51aca6f2a1429abfa36a6234e5e890 : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9489ce6754034a3cab34816238a8f681 : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/341bf56ce9864933a9c7fcb3c430df69 : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4a7b045f5284a5e97a521586d5adcb3 : SUCCESS in 1h 39m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/973c1506254f44afaf9793d17b43f044 : FAILURE in 16m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26a6bdbfdc524fed96b25c73c8b2b35b : SUCCESS in 14m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ccf203a5489040abad37b367f8155544 : FAILURE in 27m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c53d73d0be9a489896110d18f7cc1e3e : SUCCESS in 23m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/27913c0f01fc491390dcbd59e834f9a5 : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfff9f21903b45e0be2d21cf4dee650c : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/da11cb8d15d14fb181560211bf862e3f : SUCCESS in 22m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e64c4a7424d54c79b055d0b178ffc3fe : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b492162838c33b8c969b92b51fbba627a2c1b6a7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 16:17:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8df2ee05227e5160a72fbe72879e37d88d8ff4b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 17:55:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93b4af02fb2a4323afcbc11cddee9b40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ec250dc6f904439b7951847af5ee23e : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eca3d761737491fb8a8bda48b9af788 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf28080722034dde9a803ce6b0b022e6 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb75fc36162468b91e7ba0605638bff : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fe6dbb1f7454065ae8f988b195eb071 : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3526d1bde7694befa3be3a594b6c0247 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b059217705df466f95b6ed2240e484ad : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/963668b0857b4405a7dfa5ba13ced861 : SUCCESS in 11m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/62e10b183bc0472b8deb8fb34b82e413 : SUCCESS in 1h 04m 31s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/68311efd63e3422ca5f517eecd0c9358 : SUCCESS in 53m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65be2e3f752d42e2a6e3a048ff491578 : SUCCESS in 52m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fc0c725beb1b4677b8dc98804b8ac406 : SUCCESS in 59m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d591f64d57e44144b7ac444e7f85ef14 : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d2e315c10d4844dd845ff85104b6cb2a : SUCCESS in 15m 56s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1474dfe238a24d659aa7d06a370e693a : SUCCESS in 13m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a5db237f12ea4b4e8d53e8bbe36a2f95 : SUCCESS in 1h 35m 44s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a99df6c553b445d9ac8a7da6db59010d : SUCCESS in 1h 24m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/73f5be5963b34fc5956d9c88e613092a : FAILURE in 14m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b250089c29348e7bae2184bc251ba87 : SUCCESS in 14m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/892cd93de0f247ba9ca882ef0d427016 : FAILURE in 10m 56s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/97c26229c18a41fda74491cfe05b9be5 : FAILURE in 26m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d13d2fc4b3841ce9e41e3823bc8671d : SUCCESS in 25m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c8b033702ec14b3d9d772f7b572cea65 : SUCCESS in 16m 24s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/dc386bdd61914957918c58368c31af14 : SUCCESS in 16m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd635e10a9784d23b72bbf1a3873e628 : SUCCESS in 11m 52s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/55549903a235452b9b7f4be58a887017 : SUCCESS in 10m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/45dec93c4cac42ab8c2ac26e15053f63 : SUCCESS in 18m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38e3b77fc61040c1b7d1e6bb0fe56cf2 : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c07a5b58ed759ce8abc43019ad66bebe85d290c4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-04 08:23:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd274725e12a775e1c2f7222911399da6512fbf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 09:59:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72ed486ba8b45529523f0381d2000c7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/479c7efee1714825af95084d3e4a8bd6 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54d8f0d6ba0245e6ae1d03d592f492de : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dc186098f354ecda254e92273a572bf : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ebd434ed0d40459e51baae9ae82d2c : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffcb7d2829b24a3ab617e69ab646bacb : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/733f2439d17e430eaec1b4250ff2020a : SUCCESS in 3m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c244dd6f52e642239e8b68adba5d0f59 : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c4d1a7febea436f8630155dac409714 : SUCCESS in 14m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4692573764d44bc8a006f440a86d8345 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/55f1201db8c743cda9d6cf60d0fc0942 : SUCCESS in 56m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7403b860d90e4a3b971821e1c3a14ad3 : SUCCESS in 1h 01m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8782a50923694ff88516dda919e7e735 : SUCCESS in 1h 08m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/64a28432f73b47679fbf6ec1e2b353d5 : SUCCESS in 18m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f2155c5ef442429d9f89d4925f5d5b9c : SUCCESS in 10m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cbae491fbed4fe29eaf9087c1678402 : SUCCESS in 15m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2d3421aaeb144dddbcc2666d5aea5823 : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3eda9674bf564b79ae0bdcb39f43b9dc : SUCCESS in 1h 30m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e292010fa90d4f2db0b862536eb1289f : SUCCESS in 11m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/599dd61456ff4417967a1882fbe96d61 : SUCCESS in 13m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0a927592c71d413d86d077f081c1bc5b : SUCCESS in 22m 27s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/73518feb9cc341d5afd9f6be30b35351 : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13ab5a608c704366be7960d1c6dd983e : SUCCESS in 16m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e15b30c15619428cb5f672b64e144754 : SUCCESS in 15m 13s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f45c733f184049be9978bcabed1589e0 : SUCCESS in 9m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/261c1e8bd8544b4cacbc48877352d39f : SUCCESS in 10m 19s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/f7279221883a426bb04bbb4e37d3d9b8 : SUCCESS in 5m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/54cf096821534df5b46fb31d43510362 : SUCCESS in 12m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ea14c914bc04d478e167651f48e1a0d : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e543953d5f3f22f2fe1b7b6e9c9813c4110221fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 10:00:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f5133b5b5ee40f21c986f1306c4d72ae89ac2ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 11:56:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a16492888f344908352e90ae282c81c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8193cac82ab34d3abc48b1e0311ad7b8 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fbcb8a9ee1f49baa3b90486752820a1 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccf61432d9647aebf7afd5b79627062 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10f835d6c5c14a138df5e87111858ad0 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d3168667c86406a9e17709ac75a6914 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd66c07906ba4c3d8301492b323357ce : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca3c38085b384a05ae8bea87c17734ee : SUCCESS in 12m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c48864ce38e843a7bc434c11e179a475 : SUCCESS in 59m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65a78af10d774640bd20813d572e87cb : SUCCESS in 45m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5a0b4ba8ff8b43de9a1c667fe3b70142 : SUCCESS in 1h 01m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/09872f55cbfb48cdbfb0b6e52ae85006 : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acc79594756a496ab95ee05c7fc7e7b8 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f61e53d00535417fbfe4bec0c1480fa7 : SUCCESS in 1h 49m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50ce6fb0f7e046aea9cc36f5b21d78cb : SUCCESS in 1h 34m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9f6ab33248d74e2a91e8cb1d7bb23566 : SUCCESS in 20m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/374188bbf96145558f09c6a7f18d2578 : SUCCESS in 14m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ddd4d20196b64294a50ee447edc6f9e8 : SUCCESS in 32m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd520d03cc7c4143b55fbd99a601c5e3 : SUCCESS in 24m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/11a2cdaf9cbd46d0b7a85bd7f4f750e7 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f55ca39379c244b88aff89a91ca3f976 : SUCCESS in 9m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d3eb6e325cb148a1aa32679b4e65aa10 : SUCCESS in 18m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/415a53ff5c6e4423bfea54881dcaf7ee : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a745869dc1ca10a669af61d2d3ccc1fcde6eb919","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 11:56:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0fc511c094d9bcda2f994f028ae539bd12766c3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 11:57:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3809b9153e04192b41da173ab711148\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758d13fe8bfc4f6b952ace3f38be67d8 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0be1ab6040c405a9762e26de890710c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7782d598d75588448e46283aaed7cb171c82283f","revisions":{"7782d598d75588448e46283aaed7cb171c82283f":{"kind":"REWORK","_number":1,"created":"2024-10-27 05:45:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/933538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/933538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/933538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/933538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/933538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/933538/1"}}},"commit":{"parents":[{"commit":"cf32a832e669828640daa5860ac22808f2a08569","subject":"Merge \"[2023.1 only] Add a script to upgrade RMQ to 3.13\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf32a832e669828640daa5860ac22808f2a08569"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-25 10:09:14.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-27 05:45:14.000000000","tz":0},"subject":"Add support for building the proxysql image","message":"Add support for building the proxysql image\n\nThis should have been added when proxysql was added to kolla [1] and\nkolla-ansible [2].\n\nProxySQL is disabled by default in this change to allow backports. It\nwill be enabled in a follow-up patch only for the master branch.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/769385\n[2] https://review.opendev.org/c/openstack/kolla-ansible/+/770215\n\nChange-Id: I683b685bf4e2f625144e4656bcd69c03eda7aa51\n(cherry picked from commit 3b1b81b3a65c615cf10f1f49c017e879469dd2b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7782d598d75588448e46283aaed7cb171c82283f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7782d598d75588448e46283aaed7cb171c82283f"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
