)]}'
{"id":"openstack%2Fkayobe~933640","triplet_id":"openstack%2Fkayobe~master~I027c48d7ff6fe06e1898720273fef3da571aabe4","project":"openstack/kayobe","branch":"master","topic":"fix-editable-install","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-10-31 07:45:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I027c48d7ff6fe06e1898720273fef3da571aabe4","subject":"Fix detection of editable installation","status":"MERGED","created":"2024-10-29 12:24:43.000000000","updated":"2024-10-31 07:46:32.000000000","submitted":"2024-10-31 07:45:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933640-fix-editable-install","meta_rev_id":"24a46effd9a133106ace3841c8fea274bc35a3f3","_number":933640,"virtual_id_number":933640,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 07:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-29 16:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-29 19:03:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-10-31 05:46:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-29 12:55:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-29 16:11:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-29 19:03:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"d795ba41d27c8deda2c5f899d815811bdf30e4a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-29 12:24:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc9142adc103bbe1f01789e0618486ef0ef5b242","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-29 12:50:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5ebcef1c163b255c73dd02a1819db5c6b411e617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 12:55:55.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/933595 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"c8da8d6c661c0cbd477c68cbd41561bf6a939134","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-29 12:58:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7f7714c52c818cd76f3ed8777ca21e97b105290e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-29 13:47:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"615bf7d8e35d6047330019676889c965e2da60f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 15:08:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b79a2989d24141248a7ba5ba7e7ee12b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e1f4b1311b54916ad0a919f418a13ce : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/673c2d86f74a4828a40f283828f5e4c2 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47fd753b438c4f609449c5abc4a46c05 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3729c524d6de43ab94e956b4ab0c00b0 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aaf154302ca435c9aadc12b379d3102 : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82600ee14b734a22bdd98c77ad6ec755 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25be72040e4e416294af7e4e183264d7 : SUCCESS in 6m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7942c19e46545ce9297e6d1f6a51859 : SUCCESS in 8m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f880cb85f3ad4ba58e54b0312784211b : SUCCESS in 44m 04s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3e8250bb44004ff986947106b3fc9e26 : SUCCESS in 56m 40s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baad5abaebe74e91b30578420d152943 : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d6e6ce71e0c54419b68d14ed55531021 : SUCCESS in 1h 08m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4b514dce038e4564bca08cc72f05b9c6 : SUCCESS in 21m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/391bc343d40f48bbab33e9bf12b45796 : SUCCESS in 15m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ef0c72893f94a37ace3ed92d1b9f245 : SUCCESS in 12m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/214416326d4d46d4a241c41cddca68c6 : SUCCESS in 2h 02m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e4da1cbbf7b4834a35020b86df862f7 : SUCCESS in 1h 36m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c90882d399574f33a66c0a15377ba16e : SUCCESS in 17m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5738396ab4e1473089851f8faa8389c2 : SUCCESS in 18m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/416ffcbee17946699b225ff53d30e52b : SUCCESS in 39m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ff6e9ad02965491abac818bc08417d91 : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/625ff362eaf24e6b8f86c3eab11710ff : SUCCESS in 24m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e5517080558f46bb853d2ae42504b6be : SUCCESS in 14m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3b0cf22e102a4797b9e0fc2c7a1341b4 : SUCCESS in 16m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7bf1faa3a3f4a64aa69c9d4f14a7eab : SUCCESS in 10m 56s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/591c8798c2de41a8b4f6a6a8f415a51e : SUCCESS in 10m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/84c67e01d7b145b2a6e92fdb40bae14a : SUCCESS in 17m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43ac3b0b308042888237ee0154357d57 : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":2},{"id":"06c1c94bf832215d9f2a35a2dd9ffd022c89838e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-29 16:11:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"05322dd10f9fbaa5cd2c623c88d45248a0f9cde9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-29 19:03:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f6553f8d99ee3eaa23431728f13cdd7199cf76b7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-31 05:46:12.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6bb80ad26f8168e0b9f3ec0749518d96fbc0fd1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 05:46:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5edc1243852daa2fdf7519a62b0840f44dbbd996","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 07:45:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71d86eda631941309ecc2312a68b878a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3513f3760b0848c9863835dd13ada837 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7608bc18b87408a84dfc1709b55d2be : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c22e98f321a49088fc1c501dda40cf9 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1430bc0e07a40ffbe96a5d209e27d65 : SUCCESS in 5m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/56a6e2033d474d369b9d6471099c2350 : SUCCESS in 3m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/303239467a6d49f7b9c593bebd5679a4 : SUCCESS in 9m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e369c1e4c01c4f819c0d32e650857818 : SUCCESS in 8m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0924e4ade3ee4c2dab4becf373879b7e : SUCCESS in 51m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d18d0e8eef34c65bf9bdc80e0b0d0b2 : SUCCESS in 53m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e43a5b530e6a426a97a0b874c4729384 : SUCCESS in 1h 09m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0b90028025644f5d8903212bfe963198 : SUCCESS in 21m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c29293177c17472c8bf82ef41429e267 : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/373dfb153d434904934174a9dc00c5e5 : SUCCESS in 1h 54m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19a28070f2ee4b6bbadabeb1a8d284a9 : SUCCESS in 1h 35m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/02a6ade239e8450fab18b5dc8b2c571e : SUCCESS in 12m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e714da86b68945fd8396577e01fdba06 : SUCCESS in 12m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f046941ac42e4f47a4925a966f94b343 : SUCCESS in 36m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3caae0e96f604835956cf41ea56fc487 : SUCCESS in 21m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b1ff1811e7cc4bce96d3a25237e80130 : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c502a599fb145a9b3854b9ed95ac24c : SUCCESS in 7m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d881ee494c5d478fbf584c75e8f52b64 : SUCCESS in 18m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79a977da19234510b3efdb15e2199975 : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7eef39cb0c5a359958aec8a9421f12f04e8090ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 07:45:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"24a46effd9a133106ace3841c8fea274bc35a3f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 07:46:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf589fab06c4c5694e21b93f3246168\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57825334953c42ad969e4151b4963f90 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec1d3365299a0214db448e41e70fd8b80809dfb9","revisions":{"7fb365bec0dba39a89e827f9b421a13244cbd4b0":{"kind":"REWORK","_number":1,"created":"2024-10-29 12:24:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/933640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/933640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/933640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/933640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/933640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/933640/1"}}},"commit":{"parents":[{"commit":"3abc25fbf0eb02438c9de46598e159b9bedfe5d8","subject":"Replace use of kolla-ansible --key argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3abc25fbf0eb02438c9de46598e159b9bedfe5d8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-29 12:23:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-29 12:23:21.000000000","tz":60},"subject":"Fix detection of editable installation","message":"Fix detection of editable installation\n\nThis synchronises this code with kolla-ansible [1], preemptively fixing\nissues with Python 3.12.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/933633\n\nChange-Id: I027c48d7ff6fe06e1898720273fef3da571aabe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fb365bec0dba39a89e827f9b421a13244cbd4b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fb365bec0dba39a89e827f9b421a13244cbd4b0"}]},"branch":"refs/heads/master"},"ec1d3365299a0214db448e41e70fd8b80809dfb9":{"kind":"REWORK","_number":2,"created":"2024-10-29 12:58:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/933640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/933640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/933640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/933640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/933640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/933640/2"}}},"commit":{"parents":[{"commit":"3abc25fbf0eb02438c9de46598e159b9bedfe5d8","subject":"Replace use of kolla-ansible --key argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3abc25fbf0eb02438c9de46598e159b9bedfe5d8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-29 12:58:28.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-29 12:58:28.000000000","tz":60},"subject":"Fix detection of editable installation","message":"Fix detection of editable installation\n\nThis synchronises this code with kolla-ansible [1], preemptively fixing\nissues with Python 3.12.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/933633\n\nChange-Id: I027c48d7ff6fe06e1898720273fef3da571aabe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec1d3365299a0214db448e41e70fd8b80809dfb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec1d3365299a0214db448e41e70fd8b80809dfb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
