)]}'
{"id":"openstack%2Fkayobe~933849","triplet_id":"openstack%2Fkayobe~master~Ifcf827b33b6d8f7df7490bca753e098e3073ab22","project":"openstack/kayobe","branch":"master","topic":"fix-bifrost-variable","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-11-04 17:37:28.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-11-04 17:37:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifcf827b33b6d8f7df7490bca753e098e3073ab22","subject":"Comment out variable in bifrost.yml","status":"MERGED","created":"2024-10-31 10:56:57.000000000","updated":"2024-11-04 17:38:58.000000000","submitted":"2024-11-04 17:37:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933849-fix-bifrost-variable","meta_rev_id":"e862360b2fe7c97c6fd10bac95212f6a0b959e78","_number":933849,"virtual_id_number":933849,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-04 17:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-10-31 21:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-11-04 15:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-31 22:55:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-11-04 15:32:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-31 10:59:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-31 10:59:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2024-10-31 10:59:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2024-10-31 10:59:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-10-31 10:59:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-10-31 10:59:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-10-31 10:59:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-10-31 12:58:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 22:55:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"fd572cc2dc31748a7fb5dec456bec82adb3cee6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 10:56:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd694355782356dc093a9604edd990ed1d8f24b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 10:57:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e3de8e0082d3de3cacc146a29001dc2eeb37e58d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 11:05:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bdf31a7affc7120291f889b7bf6dd74f875f1d74","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 11:05:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f6f23d517a235e5d7785bbd96f696b0ab641e544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:58:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ac769d333541cea6d5656dc9513fe3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6befffb894014b41af180c6316687d5a : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8d9e7514a44a0dbcab5c65cded14c7 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f7959f32ee743489106e1de86866f7f : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c16b386eebd546129e395870f0ee9de1 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2a92eba72c437a8cb909acdbd8e82e : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d8f77ae16da43718529204a3f44f4e3 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/981e483845ae461db784ec2c089738b9 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71b512e5c2fa4bb194af7bbb5ea29ade : SUCCESS in 8m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/253193447c694eeebb9c7bf6b854e907 : SUCCESS in 56m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/154ebfde659345868080d38b9792d1d9 : SUCCESS in 53m 37s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24940141e8004ee8854f12ac263ba423 : SUCCESS in 1h 00m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/acbba57116f249a6a957d7c5de164e2c : SUCCESS in 1h 12m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2caffc5a405b409bbd17a90a4014b2bc : SUCCESS in 18m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/395d9b3d359743e2875f31a7e20adf20 : SUCCESS in 20m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1808f63b00354ccbab02d827e41191ca : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bccc20d9bf0046bc8b80269da0e6c20a : SUCCESS in 1h 11m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cd944a2043b41bf96e71c067188a92b : SUCCESS in 1h 49m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7424d6b3f4e4474abfa546217b73325a : SUCCESS in 19m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31bb53031d7e49c5aa7155a8dd6b4b2b : SUCCESS in 8m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/63e4fbed04ea4a54b48323ba485ca2fa : SUCCESS in 26m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d39da905805d4af99709aa96864ea909 : SUCCESS in 32m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54bd5ceac0ca471b81abb207e81ef6d3 : SUCCESS in 23m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/275769b7c61242baa92d14649c7f0b3c : SUCCESS in 10m 53s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/600863e31f4e45d3839c401b643d6bf6 : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94861d8713fa43ac9f5a0ce065ffabf4 : SUCCESS in 7m 46s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/907c28182b544dd3be7079db8487eecc : SUCCESS in 10m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1d0a6060ed394db0ac371bf26ecd428c : SUCCESS in 21m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6171737e177f490caa838532c55b8d4a : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c4b023669fbc27bc22e6ae7dfc337119dee7a2f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:59:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00911ea3cb3bd324fbcd2b6ec0d532a4eefcc54b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 15:11:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b950487129f4413b9421b42a8b166c34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fea7977789d4d118e40a4cee7c994f6 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2658683d8bb1487b9945b996e44922de : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28ffd6cf17d646be9f51096ffcdb8b95 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b5de5316d048e3bac4e81d0a2d6aee : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a6e435f16a2248e6b5fa7f48b3346788 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77c76f2445a541ec8e435057f3106acf : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8405454fd74471aa5aafb41264aeb05 : SUCCESS in 9m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7e6131e8a17a4390850a9d912510736a : SUCCESS in 59m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/829eb2bf01a24c56a4ad59d2652f1e6c : SUCCESS in 54m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e49297815ece49ed97d444978378b722 : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/66bfbdb553194587abd2540d3c116bb0 : SUCCESS in 21m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f5dd2017a52468da7d4941ca6c5bb47 : FAILURE in 8m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ded04f04ccb8417095cd18bacb13ba12 : SUCCESS in 2h 04m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a00dac3e5abe4eb69d71bf983e69b62e : SUCCESS in 1h 32m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8c991a76543a4cc3b0d0a397404a51a2 : SUCCESS in 20m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea22bdefb5cb4698bff5105b1cc958ce : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/df53e3b2d0bf4efd801b9f53a3eb5d6d : SUCCESS in 32m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6da2288ffcc44eb1aa323d73e5ae4799 : SUCCESS in 15m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/38a5bb3f1082416bb6ae7ba4ddf33e6e : SUCCESS in 15m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f3e488e6e24d36a65c684ca71d0aa4 : SUCCESS in 11m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76dae8fcf52f44039d7d46644e7dce93 : SUCCESS in 20m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d41a68b0fd964cb88cf4233bfa2cc1d9 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4db5fcdedeea6dbb968ffa7ad0540e8eebb00d22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 21:25:55.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"91c4f524eecdb66dfbb40e977d15916439200b49","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-31 22:55:54.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4262bd17e24cd683216fc02e143d572915a024a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 23:28:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/948c4a7c07b84f59a360b490f9615c26\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56ff901fdc7f4e1593cfebb237c3a0c7 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8a4e953f4224a9c97a8ecf08c030cce : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c00a2d689e7b4658817e796086bac545 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3661f50afaba4f4fb3cafc39da9c5145 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4367764e02174495b688a2a686e4e480 : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c04134d06d4547cc870a8c96ba7072e3 : SUCCESS in 3m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7c62e4e3bce048689471b481f3e835ac : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73652b31828a4854a36bd3209ed38602 : SUCCESS in 8m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4a008eb0d48c4b5da30ea90a5aea58c2 : SUCCESS in 54m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/dff2ece2ce844b28b6fd93f430d1dd09 : SUCCESS in 46m 08s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b647e79a30a44438f3951e862a41351 : SUCCESS in 48m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5e1d6211f2b6455c9f94ee010668f621 : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/73698f24a40a40a3a56fedda97792818 : SUCCESS in 15m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2de4a38cb67d438fa4e211bc37882e58 : SUCCESS in 17m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/073b3af7b4604a09bf1e5f3a8f217596 : SUCCESS in 9m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4a15f666c4be40b4924c219a371f5a36 : SUCCESS in 1h 53m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1af810a71104a55859901af05fea3c1 : SUCCESS in 1h 17m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c0b46fcbe17d49ff840958b714169b72 : SUCCESS in 12m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6c1cc59f175481a822e4f29c987b21b : SUCCESS in 16m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/27092b1315fc468382939d088312a07e : SUCCESS in 34m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e3593f9a69fa4bf0859af5e925575765 : SUCCESS in 28m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdf71bd9effa486b8a5f8d41827fa664 : SUCCESS in 17m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4546c520079244ceaa4ee2b957e668ff : SUCCESS in 16m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/c58e6c2920b6429188f81966106f3d23 : SUCCESS in 11m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f314020c8534f82ad9e7ccddb06e63c : SUCCESS in 10m 46s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d31dd40ff31448249e970f58984b67da : SUCCESS in 10m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e84dcb2bc15f4756961421d2b74bf993 : SUCCESS in 19m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e553c46bbed48be9c9e2396f658339c : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":3},{"id":"4711bae5f2e5a82720de81d1df1276bebbcf86d9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-04 15:32:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ebed675949752e1caaf0c62a3288505467f7ac5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 15:32:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aa40e7e3fb1827d7eb2aaacd7fd07f6d659e4aa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 17:37:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d9c1bc3291b44839d44b8a7341fccda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9056b09fa89948e383b50cbf5a3ff651 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83f55c492cfc4fc2918d19b0c7cc7603 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96c6938f35614fe9b8a191929995382a : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0195e42db1ce4d549ae98422f4438ef7 : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aaea4c54556c4977a991522f0577d147 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ebf1054926f4ea68c63cbd5f5867ce9 : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bcdcc2269ca645498de41e4ac3331d8f : SUCCESS in 9m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8e57c6eed0904096b63e5d22c1fbdf10 : SUCCESS in 1h 03m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6b771ab6b4945788e1f9f0ce5681133 : SUCCESS in 33m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/817ada08f3214570ab8b8396f3536301 : SUCCESS in 1h 10m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ec9a5e66e7c5494fa0461673f84621f3 : SUCCESS in 20m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddef6b2a8e8e4687be88977d81944ccf : SUCCESS in 15m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1103f6f56c58436f893ad3b434741566 : SUCCESS in 1h 59m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ec877e147cd40e3a9f966976fa25d0e : SUCCESS in 1h 01m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cd86221c0bc343e6aeff57ca53770b59 : SUCCESS in 18m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ef45b659017445ca90a2294d7c1e74a : SUCCESS in 9m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c306f698ad1c4ace8c0313b75707e571 : SUCCESS in 21m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0e36bd8fc24428cad10984e244e1430 : SUCCESS in 25m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec367e54b5d74827ac4669ead4dbe4cd : SUCCESS in 16m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c42368ad5084fe88c9beeac878a0368 : SUCCESS in 14m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bd891cfd810944bcbfefa2dd307cda3d : SUCCESS in 21m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac18c1761e214cd3a0ff439ae3cbd49b : SUCCESS in 17m 17s","accounts_in_message":[],"_revision_number":3},{"id":"f3bb402d1e5bf187aaff0a75ba29369ca1cce8df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 17:37:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e862360b2fe7c97c6fd10bac95212f6a0b959e78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 17:38:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d6c59ccd50547dea3bf3a7e1c6f6926\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698c99384a734cde8b0fc7f96f4324b8 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3d380f1a0f2f1b1f79f5df1302eba76408798485","revisions":{"9eb08ce28693b4573941e1dd4bc350915225dda0":{"kind":"REWORK","_number":1,"created":"2024-10-31 10:56:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/933849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/933849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/933849/1"}}},"commit":{"parents":[{"commit":"4bc3f4e5d974808d00fc45e2d83900e05b1cacbb","subject":"[release] Use OpenStack 2024.2 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4bc3f4e5d974808d00fc45e2d83900e05b1cacbb"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 10:55:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 10:55:35.000000000","tz":60},"subject":"Comment out variable in bifrost.yml","message":"Comment out variable in bifrost.yml\n\nChange-Id: Ifcf827b33b6d8f7df7490bca753e098e3073ab22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9eb08ce28693b4573941e1dd4bc350915225dda0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9eb08ce28693b4573941e1dd4bc350915225dda0"}]},"branch":"refs/heads/master"},"20007516238d48ff8aa11b74bab809a351b20374":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-31 10:57:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/933849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/933849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/933849/2"}}},"commit":{"parents":[{"commit":"ecb883c3bb540a5c18b62b0267a411f64d63f21d","subject":"Merge \"CI: Skip test_ntp_clock_synchronized on Ubuntu 24.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecb883c3bb540a5c18b62b0267a411f64d63f21d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 10:57:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 10:57:37.000000000","tz":60},"subject":"Comment out variable in bifrost.yml","message":"Comment out variable in bifrost.yml\n\nChange-Id: Ifcf827b33b6d8f7df7490bca753e098e3073ab22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20007516238d48ff8aa11b74bab809a351b20374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20007516238d48ff8aa11b74bab809a351b20374"}]},"branch":"refs/heads/master"},"3d380f1a0f2f1b1f79f5df1302eba76408798485":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-31 21:25:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/933849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/933849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/933849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/933849/3"}}},"commit":{"parents":[{"commit":"2fb57a72a7cdd24df5f52a7cde47b5e6e68a08db","subject":"CI: Fix Ubuntu 24.04 version check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fb57a72a7cdd24df5f52a7cde47b5e6e68a08db"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 21:25:50.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 21:25:50.000000000","tz":60},"subject":"Comment out variable in bifrost.yml","message":"Comment out variable in bifrost.yml\n\nChange-Id: Ifcf827b33b6d8f7df7490bca753e098e3073ab22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d380f1a0f2f1b1f79f5df1302eba76408798485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d380f1a0f2f1b1f79f5df1302eba76408798485"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
