)]}'
{"id":"openstack%2Fkayobe~933862","triplet_id":"openstack%2Fkayobe~master~Ife01620ec75bce441004562865a321721491be5d","project":"openstack/kayobe","branch":"master","topic":"kolla-ansible-cli-mariadb","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-11-04 17:37:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife01620ec75bce441004562865a321721491be5d","subject":"Update kolla-ansible mariadb commands","status":"MERGED","created":"2024-10-31 12:39:27.000000000","updated":"2024-11-04 17:37:30.000000000","submitted":"2024-11-04 17:37:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933862-kolla-ansible-cli-mariadb","meta_rev_id":"6e1b6a163c7ed8f57d9682458a2b75c433d1b36e","_number":933862,"virtual_id_number":933862,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-04 17:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-10-31 14:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-11-04 15:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-31 22:38:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-11-04 15:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-31 15:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-31 13:52:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-10-31 13:52:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2024-10-31 13:52:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2024-10-31 13:52:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-10-31 13:52:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-10-31 13:52:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-10-31 13:52:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-10-31 14:06:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 22:38:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"01b06f851d6f67e5947960c1f4d2545a6a88dd79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 12:39:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df74fa9e635bac5fb42612087e48377cdd7dd52c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 13:52:09.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"d6baa1685d9046198030cba7d15b88929729137e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 14:06:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2416b61ce82545ddb002388b633c3a56\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c6f0ad2761c49faa5a2030b57f7f55a : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a76b3eb0665480180e0a02992896beb : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c35790caac9a43259c14e262ed139048 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7eb3b92c6994171add6d5a57be679ee : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c464ce2bff3a4333b3d54d1451c4c74c : SUCCESS in 8m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2ec18ee7dd74252ba68a1b22fdc9394 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bdbd225730b438d869d9ba63b455bd4 : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e1c1c27468b4a4e8952e58bf529da2e : SUCCESS in 9m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2e156859a98e4bbeac44d9ac435817d9 : SUCCESS in 1h 02m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/af68c5a7d9dc45e19e3984a00e9ec8e6 : SUCCESS in 35m 30s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/471b466131ff41cba2e9ebb2735954ef : SUCCESS in 59m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6b4391c1e99f4e5c9c7d0fca02d74108 : SUCCESS in 1h 16m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/64562eb8b739483dae6d9f45bf5c4f1d : SUCCESS in 21m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b79228fd7d3b4e958d52335b09338442 : SUCCESS in 20m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1e3a86efad34baebfe11a9fc9e8e4ce : SUCCESS in 13m 33s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e431a8a4ef7645c1b1c08631207ff501 : SUCCESS in 1h 12m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b4b333ce90c41d68ae8b6411c3621a4 : SUCCESS in 1h 26m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/629b535483da4342b635bc39d204d0a7 : SUCCESS in 20m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f8f967d0f3e4b259375b5852e8114da : SUCCESS in 15m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/18e9ba835a4c4fad8c5383a7cf3e1c22 : SUCCESS in 48m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8c927c9818774b76911ed043f20b035d : SUCCESS in 21m 43s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40dc0617e41e45599f784062d87582b1 : SUCCESS in 24m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/be17eb14ada64c7ea7675b4ccfde1935 : SUCCESS in 16m 56s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3214fadadc5645f2be5d7d5f30d6dde3 : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/737aff6d0d1b4ad9bfe33824eeec9923 : SUCCESS in 10m 33s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e8ffec83aeb04da99c7107489fb9eb24 : SUCCESS in 10m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/68d3e837b57d4446a4e25d8f60bd536f : SUCCESS in 20m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7205ec689c749a6a69ce91434ff2cad : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"900ae43afbc2e3094e7f67b2038fe980203dda35","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 14:17:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff32b63196ce2e29a6d50ca4f235d75a34d2620c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-31 15:27:27.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6110016d61365ebbe0b1c70603238b4bae60bcd0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-31 22:38:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2146230a1e94e8bc8094b7148d9ed1cc5957f2e8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-04 15:32:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8882355c160c5fd16cb73fda6bb7721716eab6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 15:33:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad0bf41d815685645106a8a4e8d7d566017183b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 17:37:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4eacd8096d64c8ab49cfd37dd5e3ba4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3a9726a1d343a087b340df0927fd82 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9498f139c1474e7b9e87960da8041dc6 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3831482142db4a8d91e05b724209ab56 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adeadb89b9154193a5390202f80c5a2d : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0b26d683a004e95a9c0cc5c2ff3efa7 : SUCCESS in 3m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00122f90dc864923b06fb444f066451d : SUCCESS in 6m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c079c0c7d3a4a2d9c5566d084a75eef : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9168027587c744acaf6476e844703a8e : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5652f4cef154c45b273a1b9935a8296 : SUCCESS in 54m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5368322765f94c7fad654fd858b90c28 : SUCCESS in 1h 02m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/952a93949a15497a8086391a863a4ba7 : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/948ab74f91514966862e7604589800b4 : SUCCESS in 9m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e1ed2c6919e74610815a829bf073671a : SUCCESS in 1h 45m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84718596d3e64736a0279dd2f257d35b : SUCCESS in 1h 33m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b2429c88234443a2bfd636b3ef122a07 : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1539f2a0189b42aa817e83627c41af56 : SUCCESS in 18m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/00893a3723d1419296374248f99204f2 : SUCCESS in 35m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65e8194d499e4d938d99545220da2805 : SUCCESS in 30m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c16ff78582914eb688726639ce3d4318 : SUCCESS in 18m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/081c35f8e142456394d66f9edfedf088 : SUCCESS in 13m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4c339bb35754342b8d80d896f433226 : SUCCESS in 23m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/043f178a6c944b57b5bd1ba115d46f56 : SUCCESS in 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6e1b6a163c7ed8f57d9682458a2b75c433d1b36e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 17:37:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1710af579246f6f7ee08dc25a55b2ebdba4232a4","revisions":{"1710af579246f6f7ee08dc25a55b2ebdba4232a4":{"kind":"REWORK","_number":1,"created":"2024-10-31 12:39:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/62/933862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/933862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/933862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/933862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/933862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/933862/1"}}},"commit":{"parents":[{"commit":"186137e82dd2823e0bf0b37a8698a0952e5e6ffc","subject":"Merge \"Fix wording of confirm_deprovision docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/186137e82dd2823e0bf0b37a8698a0952e5e6ffc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 12:38:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 12:38:10.000000000","tz":60},"subject":"Update kolla-ansible mariadb commands","message":"Update kolla-ansible mariadb commands\n\nThese commands were renamed in the kolla-ansible CLI rewrite [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/923110\n\nChange-Id: Ife01620ec75bce441004562865a321721491be5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1710af579246f6f7ee08dc25a55b2ebdba4232a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1710af579246f6f7ee08dc25a55b2ebdba4232a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
