)]}'
{"id":"openstack%2Fkayobe~934057","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~If0f3ab6f26aa774cf6176d992ec1180f2c0d7838","project":"openstack/kayobe","branch":"stable/2023.1","topic":"replace-distro.linux_distribution-stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-11-06 13:50:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0f3ab6f26aa774cf6176d992ec1180f2c0d7838","subject":"CI: Replace deprecated distro.linux_distribution","status":"MERGED","created":"2024-11-04 15:21:48.000000000","updated":"2024-11-06 13:51:18.000000000","submitted":"2024-11-06 13:50:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934057-replace-distro.linux_distribution-stable/2023.1","cherry_pick_of_change":933464,"cherry_pick_of_patch_set":3,"meta_rev_id":"87ec382e2d03d0a8682f894a27b47f5d05253e03","_number":934057,"virtual_id_number":934057,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-06 13:50:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-11-06 11:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-11-04 16:06:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-11-06 11:18:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-04 16:06:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-04 16:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-06 11:18:41.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"c7085d3a112dab4a73d7db656d3a4c548b5a7e2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-04 15:21:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"abc691fccca4b640fb49b377394621333dddad09","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-04 16:06:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e8f8f7d5063d8d6961c4d3f90b7c4d12a8566c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 16:53:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08e7add9e3f14c739c246c24e1fe8faa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/593003694a6149bf9c26299610211090 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d4a715dc65e4986b60600a03c82b592 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb7d18b8de9749abb6dd26807bb2494b : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64e1d83989a481e90025fd45e4d5b2c : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21f5c14eabc04abc99e80a74da4e5116 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e4f06078a3b46ad8d7742a9dd2129bf : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5ae5e2fc0a849648612d53ee0e26795 : SUCCESS in 11m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/815421d83d134ac49b94ff8ed1cc4496 : SUCCESS in 59m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8a6af29f4b8d4a65a90b7f8f4b35bfeb : SUCCESS in 58m 06s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c38f43a393a4b1aa9919ebdbc58cafe : SUCCESS in 55m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fe524650a56946b58deafe52aca5d47f : SUCCESS in 42m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ffa4b33fd14a4deebeed5818fa79785d : SUCCESS in 18m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/caac8d3928294f6382b18fbc35173fdb : SUCCESS in 17m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99070c830cf143b48fca4a26799e382b : SUCCESS in 14m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3e22ad19be5b43719b9d4c0afd6c4e47 : SUCCESS in 1h 29m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42b4a1f898be4b4c967eeed087c2a6e5 : SUCCESS in 1h 05m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/553d4d0d4bf34ed28bb403d21e86d888 : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f503f24132054a2ba98e72c49a4acad4 : SUCCESS in 15m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0aa4cba61b1d4c9d93eed67e1f930f40 : SUCCESS in 41m 27s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5f5b2b3f0b884c7995e0d20bfd848f7a : SUCCESS in 20m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f5a17cb0dad49b8831887660b554ca3 : SUCCESS in 23m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/33abde1c030047cc8088482195dd48b7 : SUCCESS in 15m 17s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/145c9cb500ef4afca991d68507540f15 : SUCCESS in 14m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ac37cfe1c0f4dc0897c21444995c0d0 : SUCCESS in 7m 26s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/249ddd5fe1a748e2bb58f6a89be1b397 : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/12b7cd2df9b947729ef150fe024a1ad5 : SUCCESS in 20m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1855aa6d625d4b52885400f2cb643849 : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e9a6bfc1fe6309284cbef0ab9f61ad8957117276","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-06 11:18:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62b221ef9c2c426bb47aa26e9bcee0f9d184566e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 11:19:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ac8bbba3a9679c839e4b0dd18df50ab67821e4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 13:50:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6bc047ff7f41d281429c3c4130a65e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c51e0daa44d64c0fa9921a5065d6213f : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39c39ccfd1ad4a3ea2536821519f5353 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382cadc3d7d94c40a5b4a999b8ce061a : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84e4dc3e8d6a4b599564669b71fd70b1 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a2bb91b52d3a41fc826f16714c5a8c33 : SUCCESS in 10m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a08eec39288445308c2914d7aee65239 : SUCCESS in 14m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5e5c46b249204c28b6194006c7b7ba43 : SUCCESS in 51m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/637c7d6f597f495ba051cebc77f4a70e : SUCCESS in 56m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c7e6338cae1a46a286bd2439642be3bf : SUCCESS in 55m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2cb57515003848c28cbacf2cb1770298 : SUCCESS in 21m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc4b5f51adbc4b2b91848a9157f88fbf : SUCCESS in 11m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3e94818f7f1845338fbf7517d4865f93 : SUCCESS in 2h 00m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b2ff66508f642749ad2bed3e21abf55 : SUCCESS in 1h 15m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4227faf108fd4219aec21848078eb6e5 : SUCCESS in 20m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2701dda53da945898274f802cf890737 : SUCCESS in 14m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5879d8d830614b3585f09bd4fbf461b7 : SUCCESS in 36m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6413f76074d54f708690a857a7f6b5e7 : SUCCESS in 24m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5efb212a2d3d41039d2835a4ff47a90c : SUCCESS in 15m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/775aa3b0a6d646fc8e48b15e7d0c30fa : SUCCESS in 10m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ef65f051684c4d0c938f9f77d738079b : SUCCESS in 20m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b32c163bf5b4488d9f72262eb0f89bef : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":1},{"id":"dee9c96525e12da926d41c1d7df052e1fd69d6da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 13:50:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87ec382e2d03d0a8682f894a27b47f5d05253e03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 13:51:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d84084320c4562b1c3bc90bce386cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880781a091324b02bfe28b9832cb7f01 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fff9878e7b702593084f807e5412ddebeb9661f","revisions":{"0fff9878e7b702593084f807e5412ddebeb9661f":{"kind":"REWORK","_number":1,"created":"2024-11-04 15:21:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/934057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/934057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/934057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/934057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/934057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/934057/1"}}},"commit":{"parents":[{"commit":"7782d598d75588448e46283aaed7cb171c82283f","subject":"Add support for building the proxysql image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7782d598d75588448e46283aaed7cb171c82283f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 13:30:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-11-04 15:21:48.000000000","tz":0},"subject":"CI: Replace deprecated distro.linux_distribution","message":"CI: Replace deprecated distro.linux_distribution\n\nChange-Id: If0f3ab6f26aa774cf6176d992ec1180f2c0d7838\n(cherry picked from commit 70742ebc27ed7b39c36eb26a7f6d63d07a753601)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fff9878e7b702593084f807e5412ddebeb9661f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fff9878e7b702593084f807e5412ddebeb9661f"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
