)]}'
{"id":"openstack%2Fkayobe~934643","triplet_id":"openstack%2Fkayobe~master~I2d7a8f1368d4f0e02d199029c15c453a56582b41","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-01-14 15:16:32.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-14 11:39:08.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-01-14 15:16:32.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-01-14 15:16:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d7a8f1368d4f0e02d199029c15c453a56582b41","subject":"Bump version of MichaelRigart.interfaces","status":"MERGED","created":"2024-11-11 12:23:27.000000000","updated":"2025-01-14 15:17:46.000000000","submitted":"2025-01-14 15:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934643","meta_rev_id":"2d5a4e5e9d622d742ec783dd23ae8691dc387925","_number":934643,"virtual_id_number":934643,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-14 15:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-01-14 11:39:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-14 11:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-01-14 11:39:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 12:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 07:21:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-14 08:23:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-01-14 11:34:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-01-14 11:34:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"f06612ca152a51daa51ccc1cb9f0093172998db7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-11 12:23:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2ec4d99336f3828163066e768d3419f43a55341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 12:23:49.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Job kayobe-overcloud-ubuntu-noble not defined\n\nThe problem appears in the the \"opendev.org/openstack/kayobe\" project stanza:\n\n  project:\n      queue: kayobe\n      templates:\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n  ...\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"fcd4c6af473c762a25ab133ab84aad59a96ab30c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-14 11:36:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae28c3895ea6e21eb08275d9a6dfa54bc2920b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:36:57.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Job kayobe-overcloud-ubuntu-noble not defined\n\nThe problem appears in the the \"opendev.org/openstack/kayobe\" project stanza:\n\n  project:\n      queue: kayobe\n      templates:\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n  ...\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":2},{"id":"30e56dc26195fd6296eb636f97b948b2796c52c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-14 11:40:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d13816e7426ebb07c1dfd382b28acbfd4a8c73f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:07:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5b64f33c4e74bb2a9451cc3cc893909\n\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e11771d13ee34bd198cb6cee084544de : SUCCESS in 21m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0ca7a90e91804e5786364561975dbd70 : SUCCESS in 18m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3bb7ad650dd4f938061c3df1f2b06f6 : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":3},{"id":"5133af0fded6c485b1f30e2aab6a00d731aa86d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-14 12:13:15.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8449e4cf841456b62fa9489d63da371f1bf48314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-14 12:18:28.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ff169f791b0fa623791e2e885d3840e51eaf08f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:58:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf20352f07534fc78001d0596a024627\n\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/459980c37142440e88a4712094d66d39 : FAILURE in 19m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/92f5f7a213774b429310ea9649018c2a : FAILURE in 20m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a6837977b3b4e18a5768feb20124701 : FAILURE in 12m 57s","accounts_in_message":[],"_revision_number":5},{"id":"14a38dca140709c20e9a48e4197ec96b0ac72cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-14 16:25:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6ef04dff8af0294f0432a07550a5a0b5240ef5f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 19:22:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30d46ab0d723497d92749382138183cb\n\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9aae3ae630484efc8c03086404268213 : RETRY_LIMIT Host unreachable in 34m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f175a0be66784d3c8aba5f4bbe8c9e43 : RETRY_LIMIT Host unreachable in 32m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a03f82aeb6404c13a4fd23b86075ccd8 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":6},{"id":"4ff8bd5e34ebb86096135155d0bfba33867763d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-15 10:30:22.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"242df566c9c9b7afe9e25b7d8edb07a0d42eade9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-15 10:31:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c27583082a3627905674378d450b61cbe0091e9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-15 11:05:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"24c8b9522868bfbb4df4ffe1972a35b637cf6dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-15 11:07:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f7bddd68dc185ea9aae23e8ceb3a95c46b06cab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 14:03:09.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69d312abe1a54b3ea47873c4606c6eda\n\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b44a5a6a23144bb99f16c9b256fcad7f : RETRY_LIMIT Host unreachable in 37m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a13c44378a0343d39fcf4f3a04864b9d : RETRY_LIMIT Host unreachable in 26m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbbd3ca378b54376a174996501fbab69 : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":9},{"id":"2e3ecec02cceda46eb08c8d85971434d04832e8a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-06 17:14:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"00e741bd2df1d7a941d29b4109ce6bf69295ca9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 20:22:05.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4e08bec0a7b4f068c21663e3f88f4e0\n\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5152479044364547a8ceb60dfdc93110 : RETRY_LIMIT Host unreachable in 37m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cdd2576993e746158bab6ace64fb5f3c : RETRY_LIMIT Host unreachable in 30m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f75f2831e8949f4bbe272c777a6ef55 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":9},{"id":"a84f0a19e9690eaa4e54ef60bea1b7a782b05849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 12:07:40.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"0230e8430740f2d034eb2d8a3ff9463caacd650c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 13:17:42.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"aba71f3d9ece4a54a8acbf3405087073be0e3648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 13:19:23.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"6defd0cc62101b5bdedf686a7be876cd2d7ff1c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 13:44:07.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"dfcdbea8703990f90091ccfcbf21661646909a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 14:17:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f22949391ccf4a0791e469ef5d5ae4ed\n\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ee09061cd3854be1ba078340f821fc76 : SUCCESS in 26m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/95084d2bee094f678d58a32e9049f7d3 : SUCCESS in 24m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91ba6faf3d694972be8960515c10d0fe : SUCCESS in 22m 14s","accounts_in_message":[],"_revision_number":13},{"id":"c6ed5e58b93bd73ade75107345a3e63c4fb74262","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 15:32:51.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"6650a2633918bb53d9e636bc7d97315077d94769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 15:53:56.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"0a10ef8036db8524584b12f940987a697f2a50e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 16:21:47.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"b87e068c648df67763f861defffda9793eb7f1a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 16:56:34.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6d46bced02c42ef9fbea757b44948e7\n\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6fdb01eee5844e84a53a7e7d1dfe2871 : SUCCESS in 26m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1406b3d82e70471bae161280759f780b : SUCCESS in 22m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfa431daf4a6430198f97c44ac13fb09 : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":16},{"id":"2c7d679effcb1dca647cdd91e2fa08eb69d44583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 16:57:33.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"505aad46ac5c57b9b7f81aa4b715c6704b1d954d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 17:32:25.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6138b2de3f664e878ddc3156f821e60a\n\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b1accfa7d5754090b54872b7af639efd : FAILURE in 27m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3b15c4823239425bba185b7b4363409c : FAILURE in 17m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48403ab74cc444979fd3c36d058d2503 : SUCCESS in 15m 16s","accounts_in_message":[],"_revision_number":17},{"id":"95c2baa91e487107df594e1b314d7b408528aa84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-09 17:54:17.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"157d2ae54abfd815129ebc9606dff05266f606ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 19:53:15.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b27a37d651fb4d24a770da500a23cec7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c949415856b348e8bc68704f6a9a9adb : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8a336a184d949e0b3d6faac03a00327 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0bfefadd350e45d9bd3117788d63eee8 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a729bfde0f6a4650ba0b81ed907598ca : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e29d4edeb479401eac04cb280871a35e : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c57d050633984434ab150e912014dceb : SUCCESS in 13m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f295004c78074b789e231ca157c0afa8 : SUCCESS in 8m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e8661e2195fd48c780fb31088c1ef740 : FAILURE in 37m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f0729937c51549f7a42980322ff5627e : FAILURE in 41m 37s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3f9dc1d371f43f0ac33469655e90968 : FAILURE in 22m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b624c907fefe43e3bb819bb8134d39ef : FAILURE in 52m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d750403deeb04266834677de8a44c76c : SUCCESS in 22m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a96ae0c1e13243c3bb10f4941df0a1cd : SUCCESS in 24m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f146e8eec5a9482eb1688e3ed333d9ef : SUCCESS in 13m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/380ab404e1e54d978171c7dc26205c3d : SUCCESS in 1h 52m 09s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/492fa3a214e540128e96c7cae8a37121 : SUCCESS in 1h 55m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b8551dba62a54efdb311e7c307363b14 : SUCCESS in 18m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bb15cc478ca4d4bad8e4bda4f757ed2 : SUCCESS in 16m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/aad34a96a35d4d1996aef2636d61eeff : SUCCESS in 25m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f860b9b86b3944fe8bbdb1f127da0a0a : SUCCESS in 31m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e978d0392cd24a538b1492b0b25fe527 : SUCCESS in 24m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb984178998446b09e3b3056119cfce0 : SUCCESS in 17m 02s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/85bc015e8c084ec7b851b6cb9123e1af : SUCCESS in 17m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/171f9dcf12b144cbb140d45d60544462 : SUCCESS in 12m 03s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c728a31b79744fbead22da96489e4055 : SUCCESS in 12m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f4faa07308c147978b5ae2569465ff08 : SUCCESS in 22m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39c3ed608e464884a886d96d28476183 : SUCCESS in 18m 57s","accounts_in_message":[],"_revision_number":18},{"id":"72f492f19f7716a77508ae9395f4bbdfbf1a96b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-13 18:22:03.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"5acd48d58a8d16d54e73bcc4a206a1a4a1ef993d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-13 18:52:55.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"0f007dbf20d9ebb4c1afddd87061d81721ad46c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 20:40:08.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d57fb02463094983ac3437cfc9185936\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ba019dd7bc4decb4e847d6d7fc13e6 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23982bcca26048cf96e4f1fcf0a3a5aa : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cb8dd2d59e9487e8cc8bb6d26733600 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97366190ff6f4efd985acd5eeffa793d : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2242cea3047a473baa1fb861203b65fc : SUCCESS in 2m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ee69add5c564709b829fab3808b63ee : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/395c96f283524a51829b632b4b2c3eea : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c74d12417f124b308908b8346d7f50ad : SUCCESS in 6m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ee93a2deed634ae28a9e904313a679cc : FAILURE in 44m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c25115691ce244d5a1a528b20a3979ed : FAILURE in 47m 13s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0756db1f9c2541d69a6ac48264c4cc9d : FAILURE in 35m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a47ca7e5a83d4e10832970ef1f231d6f : FAILURE in 49m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/df16c3c438a24f778897d101a13d977b : SUCCESS in 25m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/74d7dbf57d3e4e1aa7556714087dbf0b : SUCCESS in 20m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89bf27aa2df046edafa9914dfab0cbab : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d6c0a1a3775e43538fd32dc9da8639d1 : SUCCESS in 1h 45m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4314a9157cf043cfa8debfb519e77d2e : FAILURE in 1h 15m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/90197db46c2643088c995c8a1432dee4 : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ed79dd8add84b1cb478f4cf95ba79ad : SUCCESS in 14m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/96ab376cb74147e59dda344597dffb36 : SUCCESS in 43m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b9f9879b205f4c7fa7cf5ba5c2cf16ac : SUCCESS in 35m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dabf43edc4546c8be77b8db3307d593 : SUCCESS in 26m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e09240b6fc244873a399ef6299fe8ef7 : SUCCESS in 16m 19s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/79d34683bb52412593d4ede8026e219a : SUCCESS in 19m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8df842f82884191aea856eab2e7fa71 : SUCCESS in 7m 39s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/723fe4708c7c4dabb46eb5183e22464b : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/884d17070be44dccb193381b673d629c : SUCCESS in 14m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/433371ab578d4cb5901235dcad1b9f81 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":20},{"id":"dd7286612edce5e4f640a88434aa589e13339267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-13 21:29:09.000000000","message":"Patch Set 21: Patch Set 20 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":21},{"id":"d1c1af1fc425259e501feb3fd6034613179724a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 23:38:52.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5af0b7179c547eaaff106b7f69d8bbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98d4e3fc672c4069a607d5f2b8542443 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d639cae2ec24272a93516ee96e0a48b : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/090715d1a71f4c2688906f0afc34c536 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efebba6d43994964b769ef09e7759e7f : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4a5a61a9e5d4e69921e5eb6f977ed7a : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/45aaa3fe6cb44feb81a59a1cb256f901 : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/496d3efc0aac40168b153f1151ebf501 : FAILURE in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1338e842e4aa4d2e800b1a095b414db0 : SUCCESS in 11m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5b4d99b1c6464f90af73869c67ff240f : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cbea01c6a6274bf2b3bff78ec0b53a22 : SUCCESS in 1h 06m 27s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e81829ea3894cfbad70a802d218dd81 : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e3e7c9118f774a66a3c5025ee470eb5e : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b246a0a060774d309ea3ef65f47a0cb8 : SUCCESS in 24m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9a35e58f8dfc4a4b8d778f3fff51ce53 : SUCCESS in 18m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a67810126178449ab3c469c0d5bef4c4 : SUCCESS in 17m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/430edddd99534ac6a1320acba497ac77 : SUCCESS in 1h 39m 23s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33f9217e89b3412fa0c1dbe0767be94c : SUCCESS in 1h 52m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7f1b5ba77bbf44d19e432f5c96ff3fcb : SUCCESS in 18m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a3bc81d43934bbf8202f217c77aefb3 : SUCCESS in 17m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2476b0cdf84c4527af7ccb5af85e9c93 : SUCCESS in 42m 21s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6bcc9eecd4b249ebb0e414dd129de437 : SUCCESS in 37m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e570a7215ab45d19fe276c8e4a56c98 : SUCCESS in 26m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/46c1abb7e2ea4559b7c7e1ea01a9d2a5 : SUCCESS in 14m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/08f4f43c43644d12a15a93804f74fea1 : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ceb5ba61fc214eeeb6e9eb78d7434aba : SUCCESS in 10m 19s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/52686d313bdc4165bcb6f855ace69dc8 : SUCCESS in 14m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/47cd8efea3eb44e09c13b0f6ebd735ca : SUCCESS in 21m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f54888f34811487194ebcdb09418c13e : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":21},{"id":"e3d03477d1dac5769b324dd5215fe483e548706a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-14 08:23:38.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d212bf832e9fdb250d7fb7130bb5e3c5022e697f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-14 08:55:28.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"46d61ed5c6a5901d812245b2097dacd1215e5fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 10:35:55.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb5967e3ea8b4c4383ca8cf12ef50d12\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/175ae12429ff4b288afa15a9f463ec58 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbd9457058948db955ae3705b60f24f : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd1ded369ece4aaab2a8f2651cc3900a : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2315126c3e4a238954a159f20af390 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3b685a60eb2451ab9a26e0637f6db11 : SUCCESS in 2m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a28d0ebec3e48cab1dd83e00ef74768 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c53ea01116a4436f8eaaeeadd32febee : SUCCESS in 8m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7cefa7ee79a4478e9172e09cdd611af2 : SUCCESS in 10m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b48f0358d6354cd0b03fd5d6a96c016d : SUCCESS in 39m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3775668e8b89412e8326676ed44f0bf4 : SUCCESS in 57m 08s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef67de9b483a45acab9e35abd5adef28 : SUCCESS in 53m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2108407c2f8b48beb2a0ead0bf1e8080 : SUCCESS in 42m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f847fabdaadb4847bd6679d3e7bbba98 : SUCCESS in 21m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7b21626073de4fd19b11a1b253e68a5e : SUCCESS in 18m 13s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc0e83bdd1114ad2a6033a688e18f713 : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/242ecc78a59c4762a6b8949011dc7244 : SUCCESS in 1h 54m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d406f9aa0a1a46849fc4e43cc5118bf8 : SUCCESS in 1h 41m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fb6f71023d61442a892cd79875f9f5c9 : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8eefca4d03540169663ad95cd9fbaf6 : SUCCESS in 15m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d77e7c9253a64a90988389d038db1e2d : SUCCESS in 39m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7774b1a7f7a241eb9be2da5a4269a832 : SUCCESS in 31m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95d1184e525e4f569d82cdb37f0e69ed : SUCCESS in 26m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1957e2dbf1ff441da987171808b2fa7d : SUCCESS in 15m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/630a143e5fb744b8b37071c8e6ecd356 : SUCCESS in 17m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42d6be2365aa4200bd17ff8f328f5575 : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/443b83e2415c47f7b8872a6db2faf191 : SUCCESS in 12m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e931e3cdd209491b979e91b49caea124 : SUCCESS in 22m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d10a8b872d0846f784a9b0b96dc7b05a : SUCCESS in 17m 21s","accounts_in_message":[],"_revision_number":21},{"id":"39669052760effe393e9cef346956a1fafe9f7fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 11:34:03.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"c957e8c32d3fd0edd19df7c26a10b51e7e69b494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-14 11:34:13.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"b3335a2860eec1fcc18bdcbd051720b6deb4ac31","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-14 11:34:28.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"8ba70b1e2d4a662c0a9611163aed5283779a8e26","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 11:39:08.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"452a57d9fec7c67dbd1f01b41f75fdf4fff4d0c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 13:07:25.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7488637bbcec44a8bdd2051738145871\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dfed237bcf448b989858156be7ac0f9 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91fdffe23a494b1ca3a27fa2b949a09d : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97d8f61508bc4f3ba57345e4b8e14564 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97eb976ef38e40e4b5af6d25d0235510 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b80f840ef564c0cad9412482d5a933f : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4306635d8d464498b626027a4ad608cf : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bef8cc5fc22a40199aedddf94b46aaae : SUCCESS in 12m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/695c1a7641d043bba1ce5a3fcd7bc1f4 : SUCCESS in 11m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d5e0646bb521437db9d6a3988390cbfd : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c178b36ac1cc4360a545085b7d8d025b : SUCCESS in 59m 43s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/550116cb0581441a933b20b988dd56af : SUCCESS in 56m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/667b4b6478f94c78af7badfd9c00bfb6 : SUCCESS in 1h 05m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f44e482284eb4c15a918b6bbdff00d3a : SUCCESS in 23m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5a1ec4fc979641a6831efd9c5173102d : SUCCESS in 19m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba89fe2b80064829b971c551246e6545 : SUCCESS in 17m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a273285d139a4053affbeff53a1ed6f2 : SUCCESS in 1h 12m 23s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f809d5b92a40404c98f25040daa7089f : SUCCESS in 1h 32m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2527a970263f407aa3319c3aa42898bf : SUCCESS in 13m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66036e3ce600470180b05c4b564ad441 : SUCCESS in 14m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d9359b3b58bd4b05ac6dfc72c2854f7b : SUCCESS in 43m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/29e508f4758e4d7e993a6bea6e511676 : SUCCESS in 20m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/212365eeb3c142d082427b92399365e6 : SUCCESS in 23m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c02d79ad4e4c4a3aaea1f82905a40574 : SUCCESS in 16m 18s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/971238256f8c40669d2249645ee0a217 : SUCCESS in 11m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f592c3bff054784ad01bc971e7bd170 : SUCCESS in 11m 21s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b899a2e7d17847cfbbfd361656f9bd8c : SUCCESS in 6m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e665664a0c84ee2a8c6f7cf34f721f6 : SUCCESS in 21m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa4e3d7e38804a28b63b446ba5de8ed4 : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":22},{"id":"3bbe613b1c7537572db83eb6906a3b97d263c1d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 13:07:57.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"8ff97a4c54d9b250828f871ff5626a311e10bda3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 15:16:32.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b26f8a64f6843e699252dc8ef5fdd8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645991de575e4a24bb63d740ddff69d1 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e62ae1f76ac472c826663fbbe8f8844 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f131b11a583f45b99a0ca4a0169750fa : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b57fd048249b43a79a01a041c498fc32 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1ec35d1896045a5a52b2947bb2269d5 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa35ff4324bc4fd281439866ec8d8010 : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6851f18e800849fe9cca05e94b5be470 : SUCCESS in 11m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9d8f0bf7ec8b41cbac80300081fe305b : SUCCESS in 1h 04m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d78bccdc00145a2996bd7d550cbce3f : SUCCESS in 56m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/30956288f7794e20b219efebdbe5ee8d : SUCCESS in 41m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/404fb2a3559841719ff95418637754f6 : SUCCESS in 20m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a7a7b7ce06c4b189d9e9ccd4024a4b9 : SUCCESS in 10m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fc9e56b492fe408aaf925bbf6c390fd2 : SUCCESS in 2h 00m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac070b6a4f1843faaafac388621c09c9 : SUCCESS in 1h 43m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/799a592f1af2403bb03bdf2397165ce7 : SUCCESS in 18m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e2056412a154f1286b8b37c8d1193f4 : SUCCESS in 13m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/574ba7bd2b064858a441ffd7c24602c5 : SUCCESS in 30m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f3df0d9016144c89301a6cf9dc1e0c8 : SUCCESS in 26m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c211677abac41c2a50fd2134bc89d4a : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8b3a29b54794eb289e7d18060d64f18 : SUCCESS in 11m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cbf11ae2469e400e9fb7e5f10af71a47 : SUCCESS in 14m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd2359ae734c4ab8a3748235cb5b0998 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":22},{"id":"c5d6ef61b36744522049e1bf1801ec53f0ce34d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 15:16:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"2d5a4e5e9d622d742ec783dd23ae8691dc387925","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 15:17:46.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df9bba0a771a46fdab0507c3fde0ba7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d9f9a4c1b147fcabba3a8060630754 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1251d74e0206494c8ec31cc48da9c593 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"447635ac5448e0d5d6c6f34a7e5a44f298e266aa","revisions":{"b2cdf33077b1aded0abee116a93eb5d935062920":{"kind":"REWORK","_number":1,"created":"2024-11-11 12:23:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/1"}}},"commit":{"parents":[{"commit":"6fb47e2193d912f85c977923ba0e06aea9d6c689","subject":"Add internal VIP address to no_proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fb47e2193d912f85c977923ba0e06aea9d6c689"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:23:06.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2cdf33077b1aded0abee116a93eb5d935062920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2cdf33077b1aded0abee116a93eb5d935062920"}]},"branch":"refs/heads/master"},"d004dc5d96986d0934bfa0c3243ace05d2be508d":{"kind":"REWORK","_number":2,"created":"2024-11-14 11:36:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/2"}}},"commit":{"parents":[{"commit":"6fb47e2193d912f85c977923ba0e06aea9d6c689","subject":"Add internal VIP address to no_proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fb47e2193d912f85c977923ba0e06aea9d6c689"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-14 11:36:19.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d004dc5d96986d0934bfa0c3243ace05d2be508d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d004dc5d96986d0934bfa0c3243ace05d2be508d"}]},"branch":"refs/heads/master"},"72c49fff99270c0b0e091b561f07295187c9605f":{"kind":"REWORK","_number":3,"created":"2024-11-14 11:40:16.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/3"}}},"commit":{"parents":[{"commit":"6fb47e2193d912f85c977923ba0e06aea9d6c689","subject":"Add internal VIP address to no_proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fb47e2193d912f85c977923ba0e06aea9d6c689"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-14 11:40:07.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72c49fff99270c0b0e091b561f07295187c9605f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72c49fff99270c0b0e091b561f07295187c9605f"}]},"branch":"refs/heads/master"},"6615f638fe3650b5efe8295edcc807b0787f2d29":{"kind":"REWORK","_number":4,"created":"2024-11-14 12:13:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/4"}}},"commit":{"parents":[{"commit":"6fb47e2193d912f85c977923ba0e06aea9d6c689","subject":"Add internal VIP address to no_proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fb47e2193d912f85c977923ba0e06aea9d6c689"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-14 12:13:15.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6615f638fe3650b5efe8295edcc807b0787f2d29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6615f638fe3650b5efe8295edcc807b0787f2d29"}]},"branch":"refs/heads/master"},"d8fc76bcabebcd8b9720169911e17183139eb8fc":{"kind":"REWORK","_number":5,"created":"2024-11-14 12:18:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/5"}}},"commit":{"parents":[{"commit":"6fb47e2193d912f85c977923ba0e06aea9d6c689","subject":"Add internal VIP address to no_proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fb47e2193d912f85c977923ba0e06aea9d6c689"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-14 12:18:28.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8fc76bcabebcd8b9720169911e17183139eb8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8fc76bcabebcd8b9720169911e17183139eb8fc"}]},"branch":"refs/heads/master"},"b74c5f47acd52a7aba2977ed9fec7b555beeec6f":{"kind":"REWORK","_number":6,"created":"2024-11-14 16:25:13.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/6"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-14 16:25:07.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b74c5f47acd52a7aba2977ed9fec7b555beeec6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b74c5f47acd52a7aba2977ed9fec7b555beeec6f"}]},"branch":"refs/heads/master"},"f868e5d1fd617873571a010c98a5d96b102adade":{"kind":"REWORK","_number":7,"created":"2024-11-15 10:30:22.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/7"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-15 10:30:22.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f868e5d1fd617873571a010c98a5d96b102adade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f868e5d1fd617873571a010c98a5d96b102adade"}]},"branch":"refs/heads/master"},"d2515f8e6daa2c1380a5b8698fcfa6fbaa5bb5cf":{"kind":"REWORK","_number":8,"created":"2024-11-15 11:05:11.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/8"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-15 11:05:04.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2515f8e6daa2c1380a5b8698fcfa6fbaa5bb5cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2515f8e6daa2c1380a5b8698fcfa6fbaa5bb5cf"}]},"branch":"refs/heads/master"},"65307ed0535f3a455ebc65d574be0bcbf4f315a3":{"kind":"REWORK","_number":9,"created":"2024-11-15 11:07:17.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/9"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-15 11:07:11.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65307ed0535f3a455ebc65d574be0bcbf4f315a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65307ed0535f3a455ebc65d574be0bcbf4f315a3"}]},"branch":"refs/heads/master"},"18491ec5f861e3a95fcea38617f523d782d3fe81":{"kind":"REWORK","_number":10,"created":"2025-01-08 12:07:40.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/10"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-08 12:07:40.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18491ec5f861e3a95fcea38617f523d782d3fe81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18491ec5f861e3a95fcea38617f523d782d3fe81"}]},"branch":"refs/heads/master"},"3040a576745d72e7406c1abcb1ca34fb68737486":{"kind":"REWORK","_number":11,"created":"2025-01-08 13:17:42.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/11"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-08 13:17:42.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3040a576745d72e7406c1abcb1ca34fb68737486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3040a576745d72e7406c1abcb1ca34fb68737486"}]},"branch":"refs/heads/master"},"d72d3009002c4d07eaca0adf5918d3541623d4bc":{"kind":"REWORK","_number":12,"created":"2025-01-08 13:19:23.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/12"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-08 13:19:22.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d72d3009002c4d07eaca0adf5918d3541623d4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d72d3009002c4d07eaca0adf5918d3541623d4bc"}]},"branch":"refs/heads/master"},"cf7ae89f604deca8fb1f5a02a9cce6f72d2cbd21":{"kind":"REWORK","_number":13,"created":"2025-01-08 13:44:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/13"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-08 13:44:06.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf7ae89f604deca8fb1f5a02a9cce6f72d2cbd21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf7ae89f604deca8fb1f5a02a9cce6f72d2cbd21"}]},"branch":"refs/heads/master"},"014ebde79ac361971c2cf32063a010dd9b55661d":{"kind":"REWORK","_number":14,"created":"2025-01-08 15:32:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/14"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-08 15:32:35.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/014ebde79ac361971c2cf32063a010dd9b55661d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/014ebde79ac361971c2cf32063a010dd9b55661d"}]},"branch":"refs/heads/master"},"5f9c4b4eddc6936e8f1b6d5e0a2e6ecdb7558254":{"kind":"REWORK","_number":15,"created":"2025-01-08 15:53:56.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/15"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-08 15:53:47.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f9c4b4eddc6936e8f1b6d5e0a2e6ecdb7558254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f9c4b4eddc6936e8f1b6d5e0a2e6ecdb7558254"}]},"branch":"refs/heads/master"},"e156676285bbfb412cc5b8ba7a7b024b53b398c7":{"kind":"REWORK","_number":16,"created":"2025-01-08 16:21:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/16"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-08 16:21:41.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e156676285bbfb412cc5b8ba7a7b024b53b398c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e156676285bbfb412cc5b8ba7a7b024b53b398c7"}]},"branch":"refs/heads/master"},"ccf9504f919aed8020d990ede52ec2f913d0a69f":{"kind":"REWORK","_number":17,"created":"2025-01-08 16:57:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/17"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-08 16:44:50.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ccf9504f919aed8020d990ede52ec2f913d0a69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ccf9504f919aed8020d990ede52ec2f913d0a69f"}]},"branch":"refs/heads/master"},"547e52967a2e32cb086cb0b8f73434bf49062f97":{"kind":"REWORK","_number":18,"created":"2025-01-09 17:54:17.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/18"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-09 17:54:16.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/547e52967a2e32cb086cb0b8f73434bf49062f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/547e52967a2e32cb086cb0b8f73434bf49062f97"}]},"branch":"refs/heads/master"},"8ae9dc6b3975ba1ccdcb7b8c67033cf80100d31a":{"kind":"REWORK","_number":19,"created":"2025-01-13 18:22:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/19"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-13 18:22:03.000000000","tz":0},"subject":"WIP: CI: Test interface bounce","message":"WIP: CI: Test interface bounce\n\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ae9dc6b3975ba1ccdcb7b8c67033cf80100d31a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ae9dc6b3975ba1ccdcb7b8c67033cf80100d31a"}]},"branch":"refs/heads/master"},"2b2710e1ee7a84ba31a39dd6f42929c8f00c7009":{"kind":"REWORK","_number":20,"created":"2025-01-13 18:52:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/934643/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/20"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-13 18:52:47.000000000","tz":0},"subject":"Bump version of MichaelRigart.interfaces","message":"Bump version of MichaelRigart.interfaces\n\nThe newer version sets the connection.autoconnect-slaves\u003d1 config\noption[1]. This will ensure that an slave interfaces will also reconnect\nafter bouncing the parent.\n\n[1] https://github.com/michaelrigart/ansible-role-interfaces/releases/tag/v1.15.1\n\nCloses-Bug: #2072340\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b2710e1ee7a84ba31a39dd6f42929c8f00c7009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b2710e1ee7a84ba31a39dd6f42929c8f00c7009"}]},"branch":"refs/heads/master"},"344a0a038e5a8bb5770a9c1d2bdd1d8090e0de43":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2025-01-13 21:29:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/934643/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/21"}}},"commit":{"parents":[{"commit":"febe06c06a196fe94eb647dc0460583ac1301251","subject":"ironic-inspector: Add kolla_enable_ironic_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/febe06c06a196fe94eb647dc0460583ac1301251"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-13 21:29:09.000000000","tz":0},"subject":"Bump version of MichaelRigart.interfaces","message":"Bump version of MichaelRigart.interfaces\n\nThe newer version sets the connection.autoconnect-slaves\u003d1 config\noption[1]. This will ensure that an slave interfaces will also reconnect\nafter bouncing the parent.\n\n[1] https://github.com/michaelrigart/ansible-role-interfaces/releases/tag/v1.15.1\n\nCloses-Bug: #2072340\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/344a0a038e5a8bb5770a9c1d2bdd1d8090e0de43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/344a0a038e5a8bb5770a9c1d2bdd1d8090e0de43"}]},"branch":"refs/heads/master","description":"Rebase"},"447635ac5448e0d5d6c6f34a7e5a44f298e266aa":{"kind":"REWORK","_number":22,"created":"2025-01-14 11:34:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/934643/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/934643/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/934643/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/934643/22"}}},"commit":{"parents":[{"commit":"febe06c06a196fe94eb647dc0460583ac1301251","subject":"ironic-inspector: Add kolla_enable_ironic_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/febe06c06a196fe94eb647dc0460583ac1301251"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-11 12:20:53.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-14 11:34:13.000000000","tz":0},"subject":"Bump version of MichaelRigart.interfaces","message":"Bump version of MichaelRigart.interfaces\n\nThe newer version sets the connection.autoconnect-slaves\u003d1 config\noption[1]. This will ensure that slave interfaces will also reconnect\nafter bouncing the parent.\n\n[1] https://github.com/michaelrigart/ansible-role-interfaces/releases/tag/v1.15.1\n\nCloses-Bug: #2072340\nChange-Id: I2d7a8f1368d4f0e02d199029c15c453a56582b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/447635ac5448e0d5d6c6f34a7e5a44f298e266aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/447635ac5448e0d5d6c6f34a7e5a44f298e266aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
