)]}'
{"id":"openstack%2Fkayobe~934734","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I4b3e4630942d860c8c4b8e85a794ba6ae920e6e0","project":"openstack/kayobe","branch":"stable/2024.1","topic":"test-database-backup-stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-05 12:16:37.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-12-05 12:16:37.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-12-05 12:16:37.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-12-05 09:51:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I4b3e4630942d860c8c4b8e85a794ba6ae920e6e0","subject":"CI: test database backup","status":"MERGED","created":"2024-11-12 10:18:01.000000000","updated":"2024-12-05 12:18:16.000000000","submitted":"2024-12-05 12:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934734-test-database-backup-stable/2024.1","cherry_pick_of_change":933863,"cherry_pick_of_patch_set":2,"meta_rev_id":"727c94760e82a57df0835cc2e4e8e6a3e54ccd81","_number":934734,"virtual_id_number":934734,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-05 12:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-12-05 10:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-05 10:11:57.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2024-12-05 09:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-05 10:11:57.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-12 12:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-29 11:24:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-11-29 11:24:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-11-29 11:24:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-12-03 21:21:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2024-12-03 21:21:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2024-12-03 21:21:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-12-03 21:21:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-12-05 10:30:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"ff83f3cf9b381b5e706b3cbba1bc7150973cb6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-12 10:18:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a88b698a873972811218cd17d480b93b4b19dcfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:40:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d24b1252e314ddd9a598126d3a9dcc8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4ed3777e1294fadb05a854cd79eaa95 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1386e75a76c4a1daa9fc3282ea76e17 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/afdd4d77f8364832b0940631ec78f3d1 : SUCCESS in 3m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8cb0754b766a41839b2f2b2269c0ee82 : SUCCESS in 7m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8495311359d345578f25d44c19fdb1ef : SUCCESS in 6m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/722ed9ba92e24024b3e1f88333b9f754 : FAILURE in 59m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8ca635e86d5045beb04d707bfa04c073 : FAILURE in 51m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6bd6889eb9c41f89ef904b0cbe5a2b7 : FAILURE in 46m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/75170d51fba84870833d965c7f44cc3a : FAILURE in 1h 07m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1eab176d6f9b4f349ffc0d9cb2a90c93 : SUCCESS in 20m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0b629b6efb51428ab80fe4e89efd6dc8 : SUCCESS in 15m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34fc8a42fd5c460891a96cc29ec7dd27 : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/852d718f958c471683b9704ab30452fe : SUCCESS in 1h 44m 07s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f20ff2588eb4b63b805c73041a712d3 : SUCCESS in 1h 33m 01s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a0478abd787e4163841f51b293b911f9 : SUCCESS in 2h 16m 38s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4d97141b3254048a7c375f3590d5d75 : SUCCESS in 1h 45m 55s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6074c612419b4a38b69cd84a5c01e671 : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f280f107eff94103baff3c446bb02670 : SUCCESS in 13m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/211843d1f26f4d6c84703510d53f7f1f : SUCCESS in 37m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/426f341f22ed4b1bbeef78927f013a4e : SUCCESS in 29m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bde9e7519d2245a797ae6e9e6ecfb282 : SUCCESS in 22m 11s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/201b4192061944488a7e51d596ba3039 : SUCCESS in 36m 02s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0840c27faccb4e8bb2bfd6f090974f59 : SUCCESS in 23m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82dec16761334664b482c30d4ea0394a : SUCCESS in 14m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/22aa6898783d4ce483c94a90e3b4bb7a : SUCCESS in 11m 46s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b86d7187549487399221ab20f1a2939 : SUCCESS in 10m 14s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/611d23fa22c8405c845a153ca7062e72 : SUCCESS in 8m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/799497b3b4f24917afba15616f06aabd : SUCCESS in 20m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d94c7dc4dbc54c60b3f3e3976b155fa7 : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e98b5cd5a74c0fb1e5272d1818a2c1 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1119cd0595ad4fdf979606ff08fee57c : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ba29b2ece5643309b7d430da022faa7 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33bd048a4c924f2badc2af2b0cfe0b69 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"752170653e54a7f0f77a8cbd550a16e9d48a3648","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-12 13:13:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1253eca3e2b4c5b7c7ee3aa98988e95c60506643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-29 08:21:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7a40b2c0b2ad37911ef7ef97a676a595e16c48b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 10:40:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c1c43b40004632a51a9a670e36bb9b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c05b4016f6742f7b3620e36b1568725 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ebae68916aa4d9c8def82500d26449c : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aabaffcddf7342bf92e80068609c8e18 : SUCCESS in 2m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d01e910b778344b2b45f6c5373cbecf0 : SUCCESS in 7m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/11042938a3bd4892983ec23090dc5782 : SUCCESS in 5m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/112b7e86df0e46caa82392de05647cc9 : SUCCESS in 1h 08m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9519580410bd4e10bf626f6784f1ecb9 : SUCCESS in 49m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43f9e34c7e5f43e69a4e4f1496a75905 : SUCCESS in 48m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e09ee9531b08458a9a79beb3b7dac31f : SUCCESS in 1h 03m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5bcf9dd475374407989f8e876d0df720 : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/447a84e7d3ee438fb5288ddc5fb86aec : SUCCESS in 16m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/293265d675bb4234a50e5b98a998a9be : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9bb4c217d4484a7f9cef1589741195dc : SUCCESS in 2h 02m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3086b61b8d4a4aedb49be8ef9455b557 : SUCCESS in 1h 31m 33s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/671b5a124f7647d9a42b16001d3f6137 : SUCCESS in 2h 16m 23s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe30c5daa3f9466fa6c29a494eef2cb3 : SUCCESS in 2h 11m 34s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f92ca8a619314e20845cd940d6ad5389 : SUCCESS in 20m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e333f49ed99d495b81cb1f01d0f4fbcb : SUCCESS in 15m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fa830b5520fd4473a7bb90ed9dcf1e24 : SUCCESS in 38m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f87b77bb7d2c4ce5b81e9a583c05a9b5 : SUCCESS in 31m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8202156e1a7f40fbbf63e96a8998f0c3 : SUCCESS in 27m 34s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e1822c0c5d5e42dc9e49d3b6241abd50 : SUCCESS in 33m 12s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/746944c073f14b8cb3c3556a11dee23b : SUCCESS in 26m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab78d916ce534f288dede1806b9e8e4e : SUCCESS in 16m 45s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d95233c2e1c041e0b63c335330180ebd : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90b0ca6496164ae18e6504318c5845d2 : SUCCESS in 11m 00s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/7f202ed20267444ca9f0893de32ab8a6 : SUCCESS in 9m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0aa903506aea442f805aa25469065704 : SUCCESS in 20m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3904b2ea51024180ab1abceedaff7643 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fdebb73b1c148ba859ae1c3a5c0f695 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e40052e32f2149e1bd535ce25e6bc6e1 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78311fd792a048a2840fd47cab39df19 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ddd60e504ce4d3db6d40e2171dc2831 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"647e94ff716f08746195159e2139a4b891e47a74","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-12-05 09:51:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ffd06643f91bd7b6dfa27f280d5ad37c1865173f","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-12-05 10:11:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3dbf459260b7102381d5e111b2cbb90797f6c077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 10:12:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08e245f1013d7ff548e547bf254d6a15afa2c8f0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-05 10:30:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a1aa1a76987c387cab2aa60faad0256e1c1ef53d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 12:16:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c606e90deccc4d218d6b4bd41edbca96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825b88f0efb349d986ced472ecf00c28 : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e1bb1dee41f043c69f0879a6611391f1 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/edec8a28bb794c52bd2c4826f9c4ab19 : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28625c6c591c4315a7c309756f3832e6 : SUCCESS in 9m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dc4636d2aaf94b5788fc65c0c99b7695 : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a3b625a616c4404b3f35d62969c8d86 : SUCCESS in 59m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/de22c0db80cf44728ed5721949195435 : SUCCESS in 1h 13m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d4ec8c1159764530a30032ebb31b77a6 : SUCCESS in 23m 06s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2179f02c2ba348f8b994b025771b4a61 : SUCCESS in 18m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/829515ca63b74685bb018d9e0f5ef16a : SUCCESS in 2h 02m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13d89f2c838e42cc9b1d03b046674623 : SUCCESS in 1h 02m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7e32ad285cb04405987e12bccc687a0e : SUCCESS in 13m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/034e01f45fd44512b72b824e1abdbe58 : SUCCESS in 9m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5c9443dde68f4fbeb7dab5b12727901f : SUCCESS in 22m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38f325da40474df48537e8c36884bbe2 : SUCCESS in 15m 07s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fa009167dab6412da03bc048d9666801 : SUCCESS in 23m 12s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4a3b96c2f8544f19e834a85ab2a135f : SUCCESS in 25m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/414c2da3a9aa4216877d791800efeeca : SUCCESS in 14m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f99a4802dcb04db4996b57f69c1c355b : SUCCESS in 12m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/db6d7e6b83cd43f7b3acaf1e1f16068b : SUCCESS in 20m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e3804ecf7da474480a20d79d1a8227f : SUCCESS in 16m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f255d1d91f31f413eab7b1041c1e43ffd6daf602","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 12:16:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"727c94760e82a57df0835cc2e4e8e6a3e54ccd81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 12:18:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5adee0112bd74617a01ab4fb8edd73b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/259c8bf092de45989113d7603be2189b : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d20514b7e24505e891da7ec6ea731c43fb96d7d7","revisions":{"bafa46b060d7732f9dd55a4151c72e99825c1ee8":{"kind":"REWORK","_number":1,"created":"2024-11-12 10:18:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/934734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/934734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/934734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/934734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/934734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/934734/1"}}},"commit":{"parents":[{"commit":"2c12fc6f4730449675e4aaa5e933cbec5790a976","subject":"Merge \"CI: Replace deprecated distro.linux_distribution\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c12fc6f4730449675e4aaa5e933cbec5790a976"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 15:27:03.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-11-12 10:18:01.000000000","tz":0},"subject":"CI: test database backup","message":"CI: test database backup\n\nChange-Id: I4b3e4630942d860c8c4b8e85a794ba6ae920e6e0\n(cherry picked from commit 034f5d9979bd6cbbde32e3bf45b4de8690297563)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bafa46b060d7732f9dd55a4151c72e99825c1ee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bafa46b060d7732f9dd55a4151c72e99825c1ee8"}]},"branch":"refs/heads/stable/2024.1"},"d20514b7e24505e891da7ec6ea731c43fb96d7d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-29 08:21:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/934734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/934734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/934734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/934734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/934734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/934734/2"}}},"commit":{"parents":[{"commit":"4c2b79fd5e8b469b7399da8b32fd3f09504f6e83","subject":"Merge \"Ensure ironic-python-agent matches ironic-lib\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c2b79fd5e8b469b7399da8b32fd3f09504f6e83"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-31 15:27:03.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-11-29 08:21:29.000000000","tz":0},"subject":"CI: test database backup","message":"CI: test database backup\n\nChange-Id: I4b3e4630942d860c8c4b8e85a794ba6ae920e6e0\n(cherry picked from commit 034f5d9979bd6cbbde32e3bf45b4de8690297563)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d20514b7e24505e891da7ec6ea731c43fb96d7d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d20514b7e24505e891da7ec6ea731c43fb96d7d7"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
