)]}'
{"id":"openstack%2Fkayobe~935669","triplet_id":"openstack%2Fkayobe~master~I6f466305d49031da4d048f8fa7d2625b261a6fa0","project":"openstack/kayobe","branch":"master","topic":"drop-tags-limits","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-03-16 11:21:58.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-19 16:51:43.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-19 16:51:43.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-19 16:51:43.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2026-03-19 16:51:43.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-03-19 14:01:49.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I6f466305d49031da4d048f8fa7d2625b261a6fa0","subject":"Deprecate kolla-tags and kolla-limit","status":"MERGED","created":"2024-11-19 16:11:42.000000000","updated":"2026-03-19 16:53:16.000000000","submitted":"2026-03-19 16:51:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"935669-drop-tags-limits","meta_rev_id":"92367c84e2cfa5e48d2290f0ec6359e82993e7cd","_number":935669,"virtual_id_number":935669,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 16:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-03-16 11:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-03-19 14:13:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-19 14:13:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":-1,"date":"2026-03-19 14:13:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-03-16 11:21:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-03-16 11:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 18:42:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-19 19:08:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-14 12:52:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-01-14 13:45:25.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-05-09 09:49:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-05-09 09:50:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-08-21 13:51:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-08-21 13:59:50.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"CC"},{"updated":"2026-02-11 08:39:26.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"CC"},{"updated":"2026-03-09 13:16:13.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-03-19 14:13:30.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"7a5d198ff9d31f3e828fda581265be202a693318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-11-19 16:11:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9254f3e44275006dcd37fd0bc1124bd7f603fe5d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-11-19 16:11:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ee39c644f796525afb765797bc1a3f54c3041285","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-11-19 16:47:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12cab11f22132f0555032013540cfb932cbe68d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 18:42:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7734ac7d46024bdab7202f41fd800088\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1c931b5303f4e16b882892dc9810348 : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c3ddaf07914caf8f6390f4bf57f567 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/894fac6995bc48a0bf1c0d8b1cba568f : FAILURE in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82dce882cc554d2ea54cdcc247eb461b : SUCCESS in 7m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2d2b1e494a049e08d51339d235ff49c : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc956acc6f75442787d5a15ae5b07b3f : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f43ce99895824c40b4a0fea566c555cf : SUCCESS in 10m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c8cdd2170893431a93530654cddf8aee : FAILURE in 12m 55s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c0b1efad62cc453db4edd3752d199e94 : FAILURE in 14m 30s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e038b41096b8466590297383c2b6db40 : FAILURE in 10m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/591e034220084de0a21d369c53d3c670 : FAILURE in 11m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/368a20a27d564d6090a60b5f6ce19673 : SUCCESS in 22m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/876c42750d4d4fa6a1e04aaf2367cf90 : SUCCESS in 21m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac286e5663654419b5c09b317002401b : FAILURE in 16m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/096db49bb7254c42aff0b23727c68c37 : FAILURE in 9m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e574ef556684f6db0ce17b34e4159df : FAILURE in 1h 03m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/628ddfc90f6c412cb8e49a56cdb5c2e5 : FAILURE in 15m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06d423fc705340cc84e28ea6712982dc : FAILURE in 13m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b933f6bba64d4b7ea7abb4535ce1c9fc : FAILURE in 14m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae9838bcc6a34864852eaea38c793e9b : FAILURE in 24m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8023cc48168462fb410a3fe14dae249 : FAILURE in 20m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/38282d4d987e47958d01b86ac09bcf9c : SUCCESS in 18m 46s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e327b04e3578430ca40d905a3693b83e : SUCCESS in 21m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b92682dd0f84ff69b55d28bc6dbb572 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/90d57f428fad4d7a8d096399f0e6ebbe : SUCCESS in 12m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/40e5582e26024416a3a4e9ff4c42fb2e : SUCCESS in 20m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52254b682e5644e989b5a19a92ff5575 : SUCCESS in 16m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ad0117fd43110b0c478633c1d5d79e2e985e2d8f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-11-20 09:18:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c2a5041c5dcbb31ee01f6ec0983c53b0ab791b2a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-11-20 09:45:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"45d21521cf19ae9365479fe1ec63685468f53010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 11:42:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94dabc74211a4a688f4d1b23c42be149\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f62be483e034dc2b9575c7db208c7a3 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f660b8e77e7b46bc8a6a483173db95f0 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91ae176c632a4f65836101c3212e4709 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f458567820d478b89ee06a767abef47 : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36d236bc82c84583ba5c065fa4b6d0c6 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ede5cedf348542f8a2fe7f2799571ede : SUCCESS in 15m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/26555e11d3a54c91bbf7ff8501979de5 : SUCCESS in 8m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/432560d49b32400384042610949294ce : SUCCESS in 1h 13m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c43d2bdc6a5243ec908d82cbc9971a7a : SUCCESS in 1h 09m 41s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/339df426f96645078f2fec2cdf4dcc69 : SUCCESS in 1h 09m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ad4d2a6e68f041358b6a87023ccae4bd : SUCCESS in 1h 26m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ba4d726d19fa4bf8b920073ac4a37c32 : SUCCESS in 25m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/594c3c0f41c342769a8bd2e0d0c54dab : SUCCESS in 23m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d7ec5b4736b46ef85e05fbc230e4a02 : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c7b17c05839c4817a32b041c5e2c1811 : SUCCESS in 1h 56m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a579d53b319f4b65a563649248a9271f : SUCCESS in 1h 45m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6afa47fef33b45e3b2613d86dd320703 : SUCCESS in 25m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/921bf578b9cc44c7a7bc80986e83b457 : SUCCESS in 22m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0e3df43ae8ac47a289bbd44667777811 : SUCCESS in 50m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7c8e82d2443f4dd6ad7d4c488fbf94dc : SUCCESS in 33m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/230c4012bbfd47d69551969d19ec4c13 : SUCCESS in 24m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/35bfc99fc9c34b6db85184b7636a54af : SUCCESS in 22m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f9f6fbf34d9649a6a591e4e5365a8c43 : SUCCESS in 21m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/115a16ff87ce4433bebcacbab0718fcd : SUCCESS in 16m 28s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b384c3011a7d43d5b6e9cd254cceb4f6 : SUCCESS in 12m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dc5c17729ae2412c909887c4928771c3 : SUCCESS in 18m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15d8c383937c4ac3933e8f2ec07b7226 : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":4},{"id":"99e50fd58e00c71155563fe01113b6bfc1715eba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-14 12:23:23.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"59c2f7cd3f36345ad0a63dba63a81324462ea6bd","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-14 12:23:50.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"a1a9b1234aae01cf30df759adad4624aa8a66d4d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 12:52:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e515a64fac16e4e34c965504d1e51605b3038b38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-14 12:59:59.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"d565504dbb3a80654f9b3b8a049a711151d9456e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-14 13:03:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8a2c9df17da7a2bec76f02cb8ec98b1521a56788","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-14 13:04:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dddc9e5256b0457417ef4b9e71ea8beff3cdd9f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-14 13:20:28.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"069f3b264e76c61bdea7dc925407be82c15a9128","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-14 13:45:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4890479aa3fd5ba34e203be337633e17b9cd3403","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-14 13:49:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d2ffce95ba475d5e119e1e4cea12378322cff629","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 14:17:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"816c6434b09849697f1258cc788ddd61194dc42c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 16:03:04.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/634c076bc5514e5b94e247b6d5d6ae07\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3168780a9bb4fe183c85996473aa623 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f035f7f98f544f4b9a28f4b617c71d5f : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c46bebf02f9d438a91b1cbdd10d80a84 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54c4005ca444aa3a3d12f8d4c89ae34 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/feac5b0176ed4dcc947f96da68f57572 : SUCCESS in 2m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/037374ce2e394a0ca11fa531234cd0c3 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ac09dc970ca4e139fa101b1fc84cac8 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3bcf7c58922a4f92986c35f2901c0f58 : SUCCESS in 11m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dec1fb96a7164c7087cfa18ce66ffc27 : SUCCESS in 1h 01m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6b8b33fa1eb341ada4eed8c3f5647184 : SUCCESS in 1h 04m 20s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d46d7ec3d2ac46de9060d9b76a946c48 : SUCCESS in 32m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4b32459a99824df8b2b1b2eb9b5ca496 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7958ba9e63b34b50bc59e176f3e47d0d : SUCCESS in 21m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b2b7529208ef4a3a848db77931bdbc75 : SUCCESS in 18m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc8026fd84444b2bad2a2a03c51ff8b2 : SUCCESS in 14m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/50490cf28cb34ee6ad8c75e128817aa4 : SUCCESS in 2h 24m 23s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/471d3dae679d4f19a783ac5d4428f414 : SUCCESS in 1h 36m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f28b99ee0e5243ef89b69a28858dd0d6 : SUCCESS in 19m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ac718e04da246a9b9b1386451c31c31 : SUCCESS in 16m 23s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e4263f47b21848fbb91e8cc528f22b9c : SUCCESS in 45m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e36f8ba5546d46118a57ac75ad74d421 : SUCCESS in 37m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e86da2ad9d84e75bd715a26618fe207 : SUCCESS in 26m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/13454b71ea1a46678e1fe86e3f04ebe5 : SUCCESS in 15m 01s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/fc42e8827830449abcc538c3bf48185b : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b6b01a5944548b29cdfdf7b6e0f0159 : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/6f471595b6714dc5b0a8699cba50dd60 : SUCCESS in 11m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/25692c4cf7e04be2b8a90ffb51f6c754 : SUCCESS in 22m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2c570a5b70548dc91b760bacf3b5e00 : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":7},{"id":"338e09b4d2b02e6fa3019a9aa8bb49e1d3b29ca6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-22 13:28:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e2142017e8eaa6ab6bff9be11ee37559861b5e98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-22 14:00:29.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"ee08c2652f36d0eda4ec1dfa97940d47d0b52db9","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-22 14:01:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"34c71e4e141bb69525363e09edcf9740ea5d2dbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 15:33:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d38e6b158dc4e53b4db77874bd166f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80c90cfd87b54e8287b08417085b9fdf : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be96725d08d4a11ba223b0740314a5a : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82262ae1d16c4924832325221386dcb6 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e19c15908214f6da041f7c981b01344 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be0c50fc1ec444eca5006dca1d475246 : SUCCESS in 2m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3badb153d7554a6db74312b3887f850b : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b621bcf9968f44d88daa6263fe9022e8 : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d57346d757524d9e8e76eaea303a2bb2 : SUCCESS in 8m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/918b487ab82847a3beabd8f714470d66 : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f75de4f1cdb844eaaac64e836354a32c : SUCCESS in 1h 26m 52s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5065a5e5871e4bfc9bcb95eacd09b8e8 : SUCCESS in 57m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b66211c13c144b1285fced38f4280d0c : SUCCESS in 1h 14m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e5f9ab7f2de54f16beed4a22a9274b59 : SUCCESS in 56m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/57d3323493b540ae8c004271870211e6 : SUCCESS in 56m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69cab7e036484e649e6996ca8b0678f8 : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/eb014c777cb3405682e7215664a0caf6 : FAILURE in 1h 11m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93cd0b11d48943e79b33bed107866d8d : FAILURE in 52m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/732e4d21000f4fa48da922a278f94ff0 : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/054c03239dfd4e36b45286aace93d24e : SUCCESS in 15m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/346182bebf1043e683433b69e97c13d8 : SUCCESS in 39m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6345dcc21d0146ed9a316fb78286015e : FAILURE in 21m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0242555dfa76467a8a7674622cf6c69a : FAILURE in 21m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b13294616eef4a90926ca2c74cf1a9d0 : SUCCESS in 15m 44s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/80204297cd304f63b96fcd4f71f7420c : SUCCESS in 17m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/538b02fe52fd49f5a8fc48671c5eacbd : SUCCESS in 15m 10s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/91f9641425c545b49a58e50a1f4b662e : SUCCESS in 12m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cc0a5e44bd5c4424a07021cc8f190923 : SUCCESS in 13m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94e541ef4c564848814912d2e72a0915 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":8},{"id":"3408cb025600c2fa928c72437b8bf895789aa2ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-23 10:29:45.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"5866cb5c773ab77c424cfaea31370c76d632e285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-23 10:54:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1c6d9e575d0eb9b64cb6c47e9ccda8b440babc13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-23 11:09:52.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0fa147a578cd3521d0a986763648b7ac8a14a293","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-23 11:25:39.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"54c351938c4afac64ffc634d9a952d4f181e12da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 13:33:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a9f807bab6d47368ab0c71d5e848c8e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81dbfa8cc9ea4405a2639746a27171ac : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ae52675ac843b28414473f11dc41b9 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/caf33b5ba03c404f9ba10d7253bda870 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a932daf597294caaada53c19174bc8d2 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/546957676d944e7d82d62209b063e977 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/87aa0355071c4b4da005923cd1b403b8 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/859432df5ea747e0b47ccaa9232d68f4 : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b38cfea7f062421da356f73f76c44ded : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5dfd63904f4543479cd5710399ae0cb8 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3f12ee30f4bd4804b2f57169941e6953 : SUCCESS in 58m 31s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f68ab8a7f164f64aee1bc004f2bcb2b : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/715d933740a74fe4a52da324c2a611de : SUCCESS in 1h 02m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/60d982ebbadb4ccab574beb4eb582694 : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/eb8e48c5ba7b475889d74f25f5053195 : SUCCESS in 23m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4faa9f83d284f3eb7550a793e037a90 : SUCCESS in 14m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/15d33ef142fd4bab97ce2440e7d4b51d : SUCCESS in 2h 05m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6df1a8fff9624c8cb61ac1ad44a455bb : SUCCESS in 1h 38m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d3511c2a61884a1b974308cb156b4525 : SUCCESS in 18m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9108fea8b25439cb0786b4ee2bc66e3 : SUCCESS in 9m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a2de3c9578584a01ad4d6cb5ac2f8e4d : SUCCESS in 46m 15s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5c6139e65e494b28a3ea55a168fc1fb9 : SUCCESS in 34m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c15f03a12ef484ea32b41a4b38fc7b1 : SUCCESS in 27m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0f5b5bfbd50d4ff881204da3f8d276e7 : SUCCESS in 9m 31s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/55cfd782c09d4540a83e3a50369ce952 : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8736819c845e4af49bdb97cef8d06d76 : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7fce9ad370ae47999e05c5f1f0f4f3af : SUCCESS in 11m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/92e96007c3bc4eb888fd810dc0ec53fb : SUCCESS in 13m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/708b47abdd1241639b6abac808ac6086 : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":12},{"id":"1324eb5325eb9ddaa461c62ad064e6b127efa9e8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-02-05 17:35:34.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"59bd2134656980ee3d7a2510245ca91fb9f4f744","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-06 10:27:33.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c63c376507109d6b02e3b408fcd58111647a280f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-06 11:06:47.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"d3eccd83d8f30ccbe9858a0b56a36185a0ff5dd2","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-06 11:06:57.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4c0dd72836eaf90d3c68ee420d285d4e59f1848e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 13:09:53.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecf3292f590c4313b09830bb4977e592\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a08a57158c34c5892dd0cf529a6880b : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c7d7bc2db4480cadc2523058f35f76 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa2a279d889842699180329c98c2134d : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97275f7635ea42b39dd7e60495bbacbc : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7a15c20d853497c882e1d1844bc3a07 : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89034c076c2b47de90afc09c7c6e703d : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03f9777de3e042f58519c21e9136a0e2 : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b1581d24ffd48cb99abef7e6d2361db : SUCCESS in 10m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cd02ddf7b4c245a385a3a85e7ce67fb1 : SUCCESS in 1h 02m 52s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/83439fa45c3b41e2aaeec2b19540c3ea : SUCCESS in 37m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c8eb47ab81cf425c8b24f402094b7ae8 : FAILURE in 11m 36s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cbf522ac7fc4f40b14ed2403df97e83 : SUCCESS in 56m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e06740cbdb7041768badaae6f0d62f5e : FAILURE in 57m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/298c9d9218c9482eb4e64e6851cb4b61 : SUCCESS in 1h 18m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a4e9a5dec86b490d93f14ea011c44025 : SUCCESS in 15m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7cec2261fc834cbc9cb5a9375e9b5d2b : FAILURE in 11m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5276a8dd75da40da80c4e97af8c9554d : SUCCESS in 14m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f90481f908ee49a5b1c0a3b2356e9e33 : SUCCESS in 1h 29m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa5914f7ded2421eb02b82c9259b4418 : SUCCESS in 1h 46m 56s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/679b3e1f6b144983979b1ec02aa8cd28 : SUCCESS in 1h 56m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e3b5431523b7478baefc2c30c4f5c954 : SUCCESS in 16m 38s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ab3750940c7440c0945b89deef035821 : SUCCESS in 11m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c6ec70839c848e990b0ef6287dda5b0 : SUCCESS in 16m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/679c5686831544bc81f22829dbf2dc95 : SUCCESS in 18m 40s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a46576bc55254a3ba8d06ea3515d6b6d : SUCCESS in 43m 46s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/b341a939868446b9b78692d128b3048a : SUCCESS in 1h 13m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/183e4fe2b73e4de3a949291b96732499 : SUCCESS in 27m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66e599f548504291a3f62cd3eb13cdb3 : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1a176bfb8b9d4b0084fa57564cb6536e : SUCCESS in 32m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/75d9afb6f36442fb83298db30a05091d : SUCCESS in 14m 52s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/64b6a40e998c4e7fa50906beb9ab2380 : SUCCESS in 16m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/732a1c6290684f948e90250884438902 : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/2c2b3ac3b4ed4943a3cb4e4a35ad90f0 : SUCCESS in 12m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3b7a355b5c874de0ada1ed4c5b57ee34 : SUCCESS in 14m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eef0424419e149cd9e7b528f64816b93 : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":13},{"id":"aae77ed2e3705357c3ccd018448d2e531b9b43e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-10 09:37:08.000000000","message":"Patch Set 14: Patch Set 13 was rebased on behalf of \u003cGERRIT_ACCOUNT_35263\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"_revision_number":14},{"id":"2dd98d4bb95e4712d5028ada92b1a52ada69e73b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 11:42:06.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/117f4e9f4901446b9c3e37a5de6be319\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ada64cae860d4f879314a6b32c60f9ee : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fbae52b94d04490b46f9b6bcecf3226 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80494c726bc74ceaa4a4218512357bca : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f34ba25b75f041a883076aeb8bc36df2 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1496c62dd8734da29a8da1128fcbbff1 : SUCCESS in 2m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d4da81b872d40d8ab665bf7fccfae4b : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56025213c9c04b90acc99b284b7ebc09 : SUCCESS in 5m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a33404be544b4ec8b7f0894032863e96 : SUCCESS in 11m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a8f3bdf8785146a19c0acf4734d8a008 : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a46b46f42b447178c5cb0d62417615e : SUCCESS in 39m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58f9e72d419f40ae807c5418b9ea32ea : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e16b1af99c5d4f23915c998bfcd69740 : SUCCESS in 59m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/660d909c719f451b91cf447b3ead9b13 : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/baca608241974aea86380c82a613113d : SUCCESS in 19m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b880aa6fe87d441198f538bba392eac7 : SUCCESS in 8m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/87ddeea718da4006b899efd88813bdf3 : SUCCESS in 1h 53m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b26c801d1d784e6ca6ae6dc102f02c0d : SUCCESS in 59m 42s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5599d86a20ff4ee487259f1173c223b0 : SUCCESS in 1h 56m 49s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/70238ac10cf242fda011c9ca8e0650f8 : SUCCESS in 14m 33s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6537e03bb4da458cbd123723f7855adb : SUCCESS in 25m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ba1cafc55d240179d4032ceaffbce8f : SUCCESS in 25m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/067bba719d3f468a80a3f1b1b38ceecc : SUCCESS in 24m 35s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/542348d7b5334c8ebdda4213ac2f8f5e : SUCCESS in 25m 21s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/1cccf567abb24f4bb212c062af7f4168 : SUCCESS in 1h 15m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/855b0cd93d1a4e93a0d3718f18936699 : SUCCESS in 38m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6d2d97988ed445cab0a2d956b65b9df : SUCCESS in 38m 11s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/df63f98f85b745c79838907938495ee7 : SUCCESS in 36m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1025e47100214dcc92fe0f9f2173d741 : SUCCESS in 16m 03s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/6ab5bd6dd07f4b00bf634c76aa424a17 : SUCCESS in 19m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b74025c46e524bc79f4e95a7054c4965 : SUCCESS in 12m 18s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f5b870d4208a45a6ba26fc23b63f356f : SUCCESS in 20m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cedd45c09c1f4d56acd9b6eacb3459f0 : SUCCESS in 16m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/214b87fd4dfa4576beb1bd8323cc79f1 : SUCCESS in 16m 58s","accounts_in_message":[],"_revision_number":14},{"id":"01348b90915bab9ee5d1c85fddc2014fe1ed7cbe","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-09 09:49:24.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"802ebfcd6877c0991cf7adae19ebef268328a633","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-09 09:50:51.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"6a37599085874bae641ea49097160478c109b855","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-21 10:23:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0fb08405572ae3f7f9e1849980a00d104c4a94dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-21 10:28:19.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":15},{"id":"5684e638492420ab5ece84897648464f6ecfd8fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 11:25:43.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fcd63713c4644279b70c2f6777edbe4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b95e9f1178cc4e7083fa911c43915daf : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ecd4fa54ac2b4000bf875af69d8489db : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8273fad9b9dd4e5386e2ddcf129e3fe7 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a8a96bd14b5482da4a1a8d08aadb3c8 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3445b9803b024d6ea7cef0c8ad9c4381 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94a028c7e7de4beba45977014583c155 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/728837d47fb043d18a37e46a442e5c78 : SUCCESS in 7m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d51915797f624babbb8506c21d72ff29 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39795589a9a14673acfdb2837a747c0e : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3dd4700140c1467e8cc54c36f38acd74 : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e524d870c6e4b5988cbcabde9ecfbc1 : SUCCESS in 10m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f32fdad98bd4453d86f1ca03a98acccb : SUCCESS in 37m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/65ae5ecb846044ee9419305b44e769cc : SUCCESS in 38m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f6b7b3aa943f4ee5bc58a3405a109cb0 : SUCCESS in 9m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8db6843982e348e38fedb85a8171bbaf : SUCCESS in 14m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdaa3fd5d7d44a5e9bee576ceea1c053 : SUCCESS in 56m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f2cc131b124b49dbbe2ad8ebd5b7dbd4 : SUCCESS in 17m 56s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/daa1f9890d5547aeb64e524474cb7958 : SUCCESS in 18m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df96ad8edaef4ef5a37dd7be50562443 : SUCCESS in 8m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/622fd007c18c40bfab31aaf102d130dc : SUCCESS in 9m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c65c9f1333054c0fafdb00070d4bf3fe : SUCCESS in 16m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f672eede9eb047f3927596a2ff7d0789 : SUCCESS in 19m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6853cb6b53df4082bbc5e57601602dc1 : SUCCESS in 7m 41s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/7a75011b313c46818abb7c0725d4dda2 : SUCCESS in 7m 30s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b761e24e5a3e44a1a9f214b2896e0c1f : SUCCESS in 9m 59s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d714ee58283b4ef6b5b4bdcdccf6cb79 : SUCCESS in 14m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4acfaedc468649978f20d1332895ec0e : SUCCESS in 7m 03s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a576951b1da74534b959820678d62c7a : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":15},{"id":"48259dbe36e6f4a49518463795b932ca75af9027","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-08-21 13:59:50.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d801abb6954779f98dab73333f9df146186483a8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-27 09:06:05.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"137eac2f4d1855909389dcb255ca9fa6ec9135b8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-27 09:14:15.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"0c543929dc194ccd91ee701f3109c49b273ca585","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-27 09:29:30.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"195184da0f18195c818ec886d7e594d87c436472","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-27 09:33:57.000000000","message":"Patch Set 15: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1bbec99385478c21839c786ece181c10d90329c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-28 08:34:17.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":16},{"id":"e8db838793b12256becee963cc58fea52df1783d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 09:47:19.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd6ba7bb347549da95a37ef7946c0d57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df392d361e704bb283ce887fdec8fc80 : SUCCESS in 4m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/276c0934cd7f4c03be203e2513c50259 : SUCCESS in 17m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47c35e95758b46be8ace8f25fb088037 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911ce447b22c43f2a7b8dfd0ab4ecaa1 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e38009d7ab54fd38b53792eaa55c674 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db7e281dc7714f2e80c828bde1a6ec6a : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cc0d28ba8324132baba308371179863 : SUCCESS in 7m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/720ab19b7cb04b97ad577823195c01b6 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0cc74c66415844889caca5595516c559 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cb10455818a4ebd96e5b058668e7a86 : SUCCESS in 13m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c7a4d5862e7842afb8006ecd5ca60426 : SUCCESS in 12m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4eaba9d57a044e0691de4e42d1d85a43 : SUCCESS in 59m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/61ce1b4874e8473a9a5b9daa9101bcc5 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/876872812f62435298925b478e0e720f : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87c528239a8e4e04a947d2713ff2bf1b : SUCCESS in 37m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14af568dec4441a69bb2a842f3add166 : SUCCESS in 59m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ac27457dd0f042c3aaa485399996b79e : FAILURE in 9m 19s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c878636bedc04bd692d83974e0a0b0a8 : FAILURE in 21m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cba7de622fd64bc9a50a04cfebb81df3 : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e9e5755b9f2344899b3e5b555990cf80 : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3b4484cefe084a2fb6e8f02ed090222e : FAILURE in 36m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef4e8b0f166941b9980941382b75d581 : SUCCESS in 27m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b1b1aa96ddc746bbb0ce15b17ee7c9a2 : SUCCESS in 57m 42s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/bee77aef2d684e17b0e3b02ef4f0ef4b : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c4e44b359b44757bf3266835a2d4d70 : SUCCESS in 12m 03s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/abbb7d4899784fdeb2f4ea5804d34d37 : SUCCESS in 59m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a5b1acff80943bba513f896b2399d38 : SUCCESS in 1h 05m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce6ca7d6bd924ceda345a00a44c52ee1 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":16},{"id":"3072f88adde6ab0b1c38267734000317fff08666","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-29 16:20:56.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"2a01ac0a3a08137d7f1114342f4b1b3989809f82","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-29 16:27:49.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"a0fb188cbaa7ee8dd563bb84f19ddc051c4b6f91","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-08-30 06:47:25.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"a0b83994bdca69024b76ddae67f7da51d7c4b5d3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-01 15:15:55.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"08c0bfbf8d7ed75d8bba5b34fadf35b2bfd1a4b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-12-10 13:44:28.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"12a564e8ff70744e0384077a028eb0f82aad343e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-12-10 13:45:14.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7400d987668a416214332029285cf44a0b4e83d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 15:55:29.000000000","message":"Patch Set 17: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bae7300bff54fbeb2bc2fbfb56ea111\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96cbd929ded6453db673ab8629228e98 : SUCCESS in 8m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e8c3b6c4703148f89703e62f13440c18 : SUCCESS in 7m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/308fac0908a542a988219023c0589815 : FAILURE in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a51007ec9194ccd9006e62ab6afe51b : FAILURE in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ac1ed41a2ae45fe8556b997904abe19 : FAILURE in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9295156867844dcbb664d270e162c58a : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/046c1c0464c04b84a6c56457ff0e5872 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c3df256a6a51447db5a95c123c0a46c3 : SUCCESS in 13m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d1cbce8cf87946f78e0161ad7516970c : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/42b6642ce1bb43859c3af205e04dba3d : SUCCESS in 20m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/160e3c8700364fca8ea22d2003b191f8 : SUCCESS in 7m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ab4c1b9a6cbc4b2f89a2bbc9e0648e3c : SUCCESS in 21m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a06b91811c594d7eaf859f33d92152c6 : SUCCESS in 7m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6db4b8aa2b96482aa1a6207167a6291e : SUCCESS in 37m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ee5ca51d52cc448e85ffe6ba0f8e5e2c : SUCCESS in 1h 09m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8667e92e956e498db3443456dc31ca7b : SUCCESS in 1h 32m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ab33e887098481e814512f8ab28ee58 : SUCCESS in 1h 20m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c22b1daf6d8f49e7b91790edeae51e3c : SUCCESS in 36m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97b126177ead4fdba07735d456216303 : SUCCESS in 2h 04m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ddab656fc0f04283b40531c337d2ad72 : SUCCESS in 9m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/85de99ee3344489aaae77660d75c699e : SUCCESS in 10m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c12493c1abf402a96a6a2717f2bb01c : SUCCESS in 20m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e0e503ef8550474c99247780e154b47a : SUCCESS in 15m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c606961ca73465abd1d75a10776e4f1 : SUCCESS in 28m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9fe82599f0964687a71e84528bdb1739 : SUCCESS in 9m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20f16319fc4b4dd9851f21de4dd09f9a : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":17},{"id":"4504b5920e347906057c7a078672b1474a22f693","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-12 10:33:11.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"09a13aac614f7f8f963ddd97e6a35dd49740ca0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-12-29 14:14:40.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"d525e29632105fa217493d29d18550f81e9dd668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 16:04:03.000000000","message":"Patch Set 18: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef547d29e9dd442da11e4fad7f1127d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31707f75111048319c30923dffd63c0e : SUCCESS in 9m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/afa377c69b3240389827bcb7e7123f33 : SUCCESS in 6m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5c9753fa9f04e7680d926735980d64c : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbce95aff9c5413d8797c43f4dfdfc1b : FAILURE in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3687a31eed7040f481f782f9555b5b3e : FAILURE in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dba0c6f36cb46e89dfa33014bd15f9f : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf19b048a4ac4711b66f4f11ebb8c65b : SUCCESS in 2m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48ab4277c6254d12b946114108389bf4 : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4294df7f622148f19a98c64a6ea3d774 : SUCCESS in 9m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e54466e1aa5046438d9dfc964a23eaac : SUCCESS in 9m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4990475c57674794a3d68d864c573528 : SUCCESS in 13m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d9685dc753c64c37acedb1b26c72ce87 : SUCCESS in 25m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec7ab82a33ea4d84a468552eaee131de : SUCCESS in 8m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fea22e8e1ea34b2c946aa482c6eac52c : SUCCESS in 1h 20m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/65301568e62b40359947ed15a34ffc7e : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3012a897876b4330a3b94bcef52297e6 : FAILURE in 12m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/694cbd23e55f43c39cfd13cdd957afbf : SUCCESS in 1h 13m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d5489df3e1844abb894bd10bcf40cc7d : SUCCESS in 1h 15m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f62fc332c3e4dffadb30bf5ae4c324d : SUCCESS in 1h 44m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1b00f4c61f904de494cb281167d804d2 : SUCCESS in 14m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/237035f96d1e4810ae8c01a38cd71f8a : SUCCESS in 14m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9eb8ed78885a497d81c6663eda00ae6b : SUCCESS in 16m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/84713ff9106444449cd746bb16751fd5 : SUCCESS in 15m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ceece9e21cb547e99d4383186d3416df : SUCCESS in 33m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/25e699ce50fe4f0c87db445f66386d31 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aacb6bbebc7d4a7a8d30d8ce36c75753 : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":18},{"id":"b6ef88e2ca77a3a49d5e3bd21c720898624438aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-15 13:32:42.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"9ed4a1dba42f15bdd4741d6ae313ff04ef0c6071","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-15 13:34:22.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1a3ba2e3cbdf1511cc48da1ca5fb691b52517bd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 14:53:37.000000000","message":"Patch Set 19: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/265fc872b78a45bcae14775e301a64c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7bd0f8c272b416ca8a090b0feed3ee1 : SUCCESS in 10m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e33d6d0d6ab44a39bcfa8d245dfbf03d : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3ca9d2c67cf463884edeee440601b75 : FAILURE in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85c7d87e18a4320baadf428bd88d6f0 : FAILURE in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/521fcd4d76544f2ca2a1358ca67dbf1c : FAILURE in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29d42f4c92af468192de543743ea7f08 : SUCCESS in 2m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/16cb337ad488475fac0c7cc387c167f0 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3ae4dc66f7ee49dd9b8780e2e75013d0 : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a9ea60b20c2470fb47aafbde745704b : SUCCESS in 13m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c7aaad88070f4987869df86d31595b09 : SUCCESS in 19m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b01f8355282e4226afd2800cdbf60407 : SUCCESS in 13m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e7d492991c8d4f9083438b3f255c5a52 : SUCCESS in 24m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed3fbdca36464ed7bb0dfbf6f11ceda1 : SUCCESS in 20m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e477cea6dfce4858a2379d6e3131b8a7 : FAILURE in 39m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4af21e9a2d8949999a87016c81702ceb : FAILURE in 40m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f9cc1c83c45c4f319d356e412c458e2c : SUCCESS in 1h 12m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb45694d3ec94b77ab1e06fdfdf5b1bd : FAILURE in 40m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eedffaec114d464a88a9ed2a14126cb1 : FAILURE in 40m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/76bd6b0fea4c470d89c313202426c375 : FAILURE in 57m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e605d3c53454b549fa00fe4ec4b656e : FAILURE in 58m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ec93daba8c24486e97fcedb870a74499 : FAILURE in 10m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6f3b8c644b5c442abf647f0b957715e5 : FAILURE in 17m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b9c63a6e45a456d88120b41742d8093 : FAILURE in 15m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/11ac193d04834f5e927ea6e76250121a : FAILURE in 15m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f9282fb6e6c743218e6cdd09d3eaa792 : FAILURE in 19m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8a44280e2f14e7cb838a9d999e1ecf8 : FAILURE in 19m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/65225151f59b4c31a721c30369b7fd47 : SUCCESS in 21m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a7bbafd8319454fb6c009d7d793d824 : SUCCESS in 19m 00s","accounts_in_message":[],"_revision_number":19},{"id":"f939ca1e6707ccd336087a7ffa05dc2ecf755f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-21 13:30:39.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"0f4a90e9eacc239da17ac89fb5fe61dfd645f3c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:38:27.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ddd4329fbc14f6eafb073022023d471\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e9932937964ebb8022915ad2e31c6e : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/965d60a785bd4985bd56bf05a0b64a8e : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e039935379284a1db1f95b03b38a28c6 : FAILURE in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d82cd392774567a3cf5505f7212958 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b91fe85ad78748089258c653dd372560 : FAILURE in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88654679858349d0b5eedac0b70d87dd : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f9d7e367af94499a9ff8d75f6e5d312 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6847d6e312a646b09fdfff985a8fea01 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc197058f1144e4d94542d715c63bd2f : SUCCESS in 10m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/086dcd6dadb94004a834b23b94ade19c : SUCCESS in 9m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b686ac79ec44bc5a5ba52d8f79f4a0f : SUCCESS in 13m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7231e12bb6ae483cb4f221e607b18448 : SUCCESS in 24m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/959e21376b954a20ab97a22ffc08f0c8 : SUCCESS in 8m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b15123dc0a0f4c379198b73b4121d911 : SUCCESS in 1h 14m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/105ac049e8b8461f91b16e44881c03ba : SUCCESS in 51m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a3966e37fc394160b436ead03d501bf7 : SUCCESS in 1h 59m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cdf7d7add3a44d38e2868aa4cb6c360 : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/348d44533dd24d5ca02ef34a9767090a : SUCCESS in 1h 00m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d28c2cb31d7140058cbb96543bfa09fd : FAILURE in 1h 10m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/671277708aa14fb596d3eb2c3f1ac0bc : FAILURE in 1h 03m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a31c8466ca5742aaa3141f318d27832a : SUCCESS in 13m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd9b3fa75e4847649b0cd055200fcf67 : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":20},{"id":"727f93204578fe9d05742e78387f8e930fa782a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-22 16:03:28.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"7e9c2d23d292a38e4399de99b7dc76f6ff463fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 18:22:15.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/700488057f464a41b24d9523e724f0fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f87f16524fd4a198d9741f6dde14083 : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/17dc1e9191004b709a900545d63b5e0d : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4acdad0c58c45edaf0e83b82033c0d6 : FAILURE in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67863146859f426d8e1ccefd48423c5b : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ce126ef687849cb8a3d2e252d898d02 : FAILURE in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f7e08be7d8846358be2c0258915aea1 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5cefb80da1dd414587343a49f1915cce : SUCCESS in 3m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d5b2f29e59d84ddb9e1c91ff98a0b002 : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c61d3219dd3b42a2bb199188e3e9b6fb : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5acd96a960a042b59ff1f9a6e16946c6 : SUCCESS in 11m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3f2963f1ab849fb9d78e2bf6c2d7a84 : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/006f3814e4fb4871937902e405700e10 : SUCCESS in 12m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/834933a7c3164ffead058b3db87656cf : SUCCESS in 13m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/aa97cd1659ef4ceab6b009eab1138848 : SUCCESS in 1h 17m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/250e79929a264645b3f7c1d8755651e2 : SUCCESS in 34m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3f7cdbdc1f7c45f480f3a66b4b3a6947 : SUCCESS in 40m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e655b71b2f8447068257b0e6789c8073 : SUCCESS in 32m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a367f7fd378487e8a9ef9dc07697756 : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fada8bea149043be81fbbf865b06e6a8 : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e529d3f77ddc4029990d0e579c96bf60 : SUCCESS in 2h 07m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bdbfa063d985466ba8aa960f4691fd85 : SUCCESS in 24m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3e409bfda95a40bebda6ce13226411fb : SUCCESS in 11m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8009c4f528564224a3f8282412042020 : SUCCESS in 18m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa626259a8c94c7585fb72394b2571cd : SUCCESS in 17m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4f301a98865849dc99fdec6b4dc09648 : SUCCESS in 16m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3db323cfe0f04c66b0beebff253b0399 : SUCCESS in 16m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1685f434a7c1423898ec1f85638d2c0a : SUCCESS in 10m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2d56da2a42d4a0f8159cfa1d0be5fe4 : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":21},{"id":"18db93b004b475d0dff2c1f761b5cba956d8a56e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-26 13:30:38.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"6f483d3fcbda815486994a63b1e5ce5e6be4e301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 14:14:56.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cee6d395cde4b32bccfa2151912e896\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b6947165b44ed9afe610449200575b : SUCCESS in 5m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dad142acec64451da2f6c998885e92f0 : FAILURE in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6381a80637bf4a069dbb833a3fb0bc8f : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75eb0d9d6c504111a445bab98681c8a2 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94d43fb480c04633bec6ee0526d6178d : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb05fd2bd78d4047b479255b92a4409f : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bce8ad348a6845499dff8cdf8e6c1d87 : FAILURE in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bdca822d54b74e429e29cc75a3b5e2b6 : FAILURE in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b20e95fd51a425d8d4b39b8f10db5b7 : FAILURE in 3m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bc53b09ec44e49aeb811f1da780aff02 : FAILURE in 5m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85db0845df864b5f8bbb4e1bb96434b7 : FAILURE in 6m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c3773de17ddf4f44a05825a43e32929a : FAILURE in 4m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cc217923c004a2dac672d1b6f445e35 : SUCCESS in 14m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/901131c325b2420cb81e219760159ff0 : FAILURE in 10m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0898fe01157a4e6d9315eeda4c659ba3 : FAILURE in 4m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/693d7abb28d64f0895896d0f4272585e : FAILURE in 4m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d625b32c612e4461a61b3bec49996dfe : FAILURE in 5m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/802e9f2cc3aa463f8ef745e2598ed0ef : FAILURE in 5m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/756da7d1608b43958fdc75233ec7b063 : FAILURE in 6m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae2b367925db48bf8a6e2e52a6cdeaa4 : FAILURE in 7m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8e4e1d8abb114d57a25bb992cb577e28 : FAILURE in 4m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/251127743a5f43c18372bf6ca9d85d39 : FAILURE in 5m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2c0f0b15d7143d3a2fce4698523ec3f : FAILURE in 4m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/41c9317c1a874b5d83cf8772c8883821 : FAILURE in 4m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8664194297f4408799dd05fd534b8878 : FAILURE in 5m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42327e5bd49a40969cd62e3906eb0fd3 : FAILURE in 4m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bd51f285a0b44f358021b05e9516990e : FAILURE in 4m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0e8a3330294425b99580290c59fe59b : FAILURE in 4m 35s","accounts_in_message":[],"_revision_number":22},{"id":"df9e2791274d158fc2382ce5bcdd293201724570","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-26 14:42:17.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"1078ffe3c5d07e344d6cfb2c1a89a68f1dbc9f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:40:45.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81f2db14dc67420590f1a26e44962bc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820c166cdb2b4e428d14be8bb707a700 : SUCCESS in 7m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/163223d422324dc9bbdaa695eb507ff0 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bebaa47259f4e2491aea884989dea5d : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821bbe59e4504802a43667059a10c28c : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ebfb17e0acc47fa8f5b4f86f2bc10af : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9d338432a264c1499799b562b918152 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/453a80b93f59494bbf5bea77b9f36bbb : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d02b22b31244f2db50f08d618c8eca5 : SUCCESS in 14m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce6fba4593e541d9ad7314b59d8b954c : SUCCESS in 6m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/99baab19f41b4557aabdacd2beea4969 : SUCCESS in 15m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d33a33821aa241918cc51d85c22f9dc3 : SUCCESS in 13m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7ec9aeb6e68e4fc9bc155a34161794b1 : SUCCESS in 20m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c46588ed1404b5c93d48c0b198d5f2d : SUCCESS in 16m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ff45c9c0ee12452b9799487d956e1b0c : SUCCESS in 53m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b72d6ee4f26a441a86b5b1ea33c70670 : FAILURE in 15m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7d8eaee0e41240ea96286f6e98a8fd7f : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d95ccb76df904f188d8cdf5145a27a7a : SUCCESS in 1h 16m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a918147354c64a7da1de3615226b2dcd : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/076845027526423991cd4caec1e9377e : SUCCESS in 1h 37m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce8ac6a9184149e481538efa7d25efe4 : SUCCESS in 1h 51m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/20f8c8139f5c4f5e9d330851a124b7d2 : SUCCESS in 18m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0af6319f11a7432592435fbdf7e0ceb0 : SUCCESS in 16m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14ee26f5118f4f54b13b4e8a937553a8 : SUCCESS in 8m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6bad90b7148443b2b0b06ff740cdf428 : SUCCESS in 17m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c4c5fe7d457e472db9d7fee6ff9bd82d : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05e1d841d13b4b0488d6ddae1b05a0c1 : SUCCESS in 32m 10s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7cb965a7d2d645878532ff02ed240c7b : SUCCESS in 16m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5886a8ff65dd4efe96dac8f5a2f03132 : SUCCESS in 16m 38s","accounts_in_message":[],"_revision_number":22},{"id":"3e8f6dfc4cb076b6dc0c0ef65f37ac5130088e13","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-26 17:22:22.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"d5e69e87e8c94a56907deaf821964a9d5a038e68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-27 15:09:34.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"ebe95539f17e5e34f9f11681b80f193b53e8e40c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-27 15:09:50.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"2d72cdb0da81854151ec3a94bbca7df3e9028f0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 17:01:09.000000000","message":"Patch Set 23: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e8542feab9142538a4789b9f0531f1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7625c9772248958dae33be96e1e049 : SUCCESS in 10m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/636f8205fac04aaa86048c29638135e0 : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85b9278bf34c432f8e3d207aec6fedf5 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b05ed0a72d84d65b02f6015dcbbb5fc : FAILURE in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a888e9c1f7d64bdea4f4a2923df8f2a3 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8ff670995744e4b9bc5cc43d255cfe2 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9af2c47ffda146748290d46c760f6566 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a973d04b3bba40b2b87aa7c7dc68344f : SUCCESS in 6m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4c413b31e634cc79d866ea71f972455 : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/618c6b103c2f4dc6841518e4e45b22c2 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6528704231634e0b8b00f19dbf1a6cdf : SUCCESS in 17m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a3234c105a2c4f859757802fa6176d1d : SUCCESS in 25m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/087cfdd5848540dabbd464922773b77a : SUCCESS in 18m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7b0e145a8fcc435a97629454a2413b8c : SUCCESS in 54m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ebff622c3d644a18ac87b115609f14d0 : SUCCESS in 56m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f5baa8933b8248f9b79be44db00f1b10 : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6df0cf090a4544ff9a774196d358eea4 : SUCCESS in 54m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ef2bcaaf70345069a2d5c77f537b541 : SUCCESS in 53m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0a1f53974f3c43ee94579f3d4be403de : SUCCESS in 1h 37m 22s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/44494561691c41c29297a63215048a29 : FAILURE in 53m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96dd303685e647e99a90ab538ce36c09 : FAILURE in 52m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a01ecedfbb34e1986870ba334513695 : SUCCESS in 1h 37m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a580e639865b48cfb1a50e84dc398de5 : SUCCESS in 19m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ff36eb5da3af42fb9ffa231931afcb95 : SUCCESS in 24m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bb74dd310f34aab9e8b8cd5911ced42 : SUCCESS in 17m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e819b3225389429fbebbb3280a3a1a92 : SUCCESS in 8m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0845e84d4ce644939693cfd2f5dc79c5 : SUCCESS in 35m 24s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2fcf75287d5d4c9eb2d5b3ebfc878d04 : FAILURE in 13m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb49ed18735b49a9bc57a9708b201af3 : FAILURE in 25m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e02b8ac78de54e7a888a55a91b2ef522 : SUCCESS in 15m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b50045c4a38141be834221098165465e : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ce00ff0904c40bdbd71d9ef7e7658b5 : SUCCESS in 17m 51s","accounts_in_message":[],"_revision_number":23},{"id":"ab6c673494553d9d7abd84e47939b116d9f11ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 13:25:55.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"98d51d8f1216f854a3a740377c8bdf4efc1b8b7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 14:58:53.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/094fc43fc5844d16a168b33e0fcdfcb6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c6bd036be74a4e9e63c5c6edf570f8 : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4788ef9a2fe482da76f703d2ab46487 : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58a0dde8a59d4c0996d6c7283f063d10 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5baabe7a0dfa4085999cb5d003a37dd8 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e6eea0dce8d4f3aa9c09463675961a3 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0919fc4dee584ae9a0d5e4f0cd254597 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5563ea484a224a46b59093ccb523f475 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e4dff4e9bba74fd78e4cd9ab47d020e8 : SUCCESS in 13m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/792c3e7d06a74d588d32c3de20c5f51a : SUCCESS in 9m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9ab761cbe2f34dffbc07092306183a48 : SUCCESS in 9m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/014711d1a6c64c2ebc7ce0c8989e60a7 : SUCCESS in 7m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7b8f19e221274f2894b9a07e3f7c2a22 : SUCCESS in 24m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dd43a0686874a969e18e4b99f634f0c : SUCCESS in 19m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0e3e1a26186c465a8cffc785d6a112e9 : SUCCESS in 56m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f25281c847d64cb9bfd3855de0b78a36 : SUCCESS in 56m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4fea652bf36b4424833153fcca6ad267 : SUCCESS in 58m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd390dc2b30942a7aeb896877573b566 : FAILURE in 17m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7566f9eb1eda4c8d88c88a8a0683dbd0 : SUCCESS in 57m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/11c7b92a827d4107841a99a7a798cb42 : SUCCESS in 1h 32m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b8ac8dd16fbd4dbca05ee721694a7d93 : FAILURE in 46m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7ce972246f74006830c196abb02a8a2 : FAILURE in 45m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a768112dbe164a7281c925dec5bd8f5e : SUCCESS in 1h 29m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dad11d16df7f41e8847abc0da4d2af95 : SUCCESS in 11m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b2393f33b4c54969870f561c6b68c052 : SUCCESS in 22m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5015a5008cc742f99ded64296440ed3a : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e1bf4d8fe6fc4daa95bbb2fe88537664 : SUCCESS in 17m 09s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/52020a5fdd90446dbb8eff154eafe13d : SUCCESS in 34m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cd11879f7dab47a5b074296a1c651f8e : FAILURE in 18m 51s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1701d82e66f04ecca39167e0c37dce74 : FAILURE in 26m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/157982032a0a4fafb6ff6dd9623972c8 : SUCCESS in 31m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5ab7d5ca8850401b87c1ae8aacc98c13 : SUCCESS in 10m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/765cfec1494a4c6d83c38798f8a9449a : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":24},{"id":"061442ca2ce6e31866bd54cd5752b9065da73a52","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 16:54:39.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"4a75aba44ad2c65f9dc984d77a85f15d9c37c4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:49:56.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/785a2f17c4c14c4db5d40013548a2a60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741d0b69e04844a68b821a19612ea1b0 : SUCCESS in 11m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4821d74a4f724830841b1c60d656eafb : SUCCESS in 8m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdf2eec96bf940b496b4e96f6f887c42 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2166db938845f88a97f3fe8609449e : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d9309fcb3ac4ce582c49c826ccc281f : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b14952474fa4143a409bfe34ffbbb75 : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/33b95582f6694daaa83cc0e70c7ee580 : SUCCESS in 3m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20737c81c7364064ae09d632ba33f0cf : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12cab60cf24749778cd2073f997feb87 : SUCCESS in 12m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a89a5c3bcd1541af8ae3403c25d1a98d : SUCCESS in 8m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1292eccefb674064b36d052874a240a1 : SUCCESS in 14m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f5839ba99cca41b0a10ad32ee5d77c5f : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d67e46645234525bc75c6b679b09612 : SUCCESS in 18m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/128ff1bf87004c4cbcc14f3389745c46 : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/21ba117b39b149b387097182ab417b5f : SUCCESS in 1h 13m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6783dfa1cd2a4daab3c9b0b0ca8e2bb1 : SUCCESS in 1h 13m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21cba3e5503947c3a8360f3ce3b8476f : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/625f83af711b44eda6069892ce769e07 : SUCCESS in 1h 10m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c2d68498fde94a4d936d9c0a40012d53 : SUCCESS in 1h 53m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7de35052c5034b4abd4bde1ad7e30c99 : FAILURE in 56m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13e73345305a4f3eb981cdde5a6c1298 : FAILURE in 54m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb17651a79ec431c85cbd73f8395d7b9 : SUCCESS in 1h 53m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c5a9766010b6406894ab430199b18243 : SUCCESS in 17m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b93d7dbcf0a443a488f4f7f6a085bfc7 : SUCCESS in 23m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b71b8e6297764811ba13544ae7a0a733 : SUCCESS in 18m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c2b6ddbcb8c843168b39ff6468ba4697 : SUCCESS in 20m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1bc5ffd3d0bb417aaa2c6d3e2ebf8048 : SUCCESS in 32m 51s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1dbcd38bc87c4ca98b23706aae1cafbd : FAILURE in 25m 15s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84ff8a14fe7a4e6599f39808db995a2c : FAILURE in 20m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/872eb1ef983a46f4aceb83b766142e8b : SUCCESS in 43m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/76e19c90322744c0a0720eb165678153 : SUCCESS in 20m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9fdd99202194a03af01293a5d841965 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":24},{"id":"3d53f459e7cb3238d8c37687000553ad574a7035","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-03 09:02:07.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"a958c6c7ec459395e15e7e9643607e1f31115b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 10:51:18.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33f893e32c1b4045b18c4a4565ab665e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee15dd157c774c6582362909486476b3 : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ec66df1ba6994e38b2f54ca66ca5ce1a : SUCCESS in 9m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7ff51253a3d4c03b5e2b7532a1d7564 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f5bfbcb07949caaa5e37ecdf81973f : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67882d23e58c4597b571c9159c470a3f : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48b047d2ffdc40e1bcb8cb2794da116f : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe814d4e7ab740d6a1467cd2499fae1b : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e54a903756d94c349972f6b1020225c1 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c7476ca02c04ae880717272bb998e93 : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/08e5b22145ea4c6d949311f639f86e18 : SUCCESS in 17m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e108fa098ba749e8a3f2ac820639f098 : SUCCESS in 14m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f034cc30207b47d78852a72ea9fb17c5 : SUCCESS in 25m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af233b6404d14f069abde0b1a716629a : SUCCESS in 19m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4ceefa956e6149c0b2637e4469b4366a : SUCCESS in 57m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f37a03b4538c46e19f02d47648d5414f : SUCCESS in 1h 00m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b223fcdf2cb94d0aa115210c9bd2c963 : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecd498c0bbd24dd98fa6450c2de480b1 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/89248438793c454a8b59b83f8031baa0 : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5f4ea176e414847b94bd2b60228ca2a : SUCCESS in 1h 38m 17s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c26d1b5e88504bf98540af3dc2b9f1ca : FAILURE in 55m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aefaabe82f174ea492c4c969e7d264ad : FAILURE in 45m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/707101ff18dd4da18b76c14d64bffbfa : SUCCESS in 1h 39m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/22c2ad1c24fa408891c7ed1b74423f6a : SUCCESS in 12m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1378ead5d92f4451b1f7582355e1e8f9 : SUCCESS in 18m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9f6745a6a0c4c7d8bbf9c4c9276886e : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d1be5a15532f4dc0b1189ed065e5e9c0 : SUCCESS in 17m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0e0b3be26751473aa4c8462b63f91791 : SUCCESS in 18m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/527f0ea2d7b948d9ba2a571ce0a18b64 : FAILURE in 13m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f20d3717056e4fcbb4dffe024cec25cd : FAILURE in 34m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f93e7a5af02443b983476b7d65adc043 : SUCCESS in 29m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2a5d94ff160a4f19a55b74ed1ff1ecd1 : SUCCESS in 15m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3aef6438d2e4beeb4addcaab50585bf : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":24},{"id":"0cf224610e266a4e685747a2102f2abb79b9b3d1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 15:45:48.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"40c6a25321fc7b9be137d8b157c74ac6554d22b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:31:57.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db8436a7c0e446008a57e43132e0e2ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d635d1c6534ae4ac3e6410d91a42ab : SUCCESS in 5m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d318b42ee0c540ea865c790d48654e03 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cc42da00b0a466e9e67eba234b8277e : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cdb89c84d8b4d68a3be4ca3dd946282 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/631359c01d314a399b8096b9e01bb9d0 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/682485eefcd94967a54bde9d2fe652e1 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc74dcd4275c482abb19b7c6e5f7bf00 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7baaf48b648f44e181e8b1a912fb05ad : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/525266bb3666439bbed6518517f5b54f : SUCCESS in 11m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d52f886769154814bb58d0e401d664e0 : SUCCESS in 8m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/682972ed311b48ae9f5b0afe66b71bee : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d45becdef34748ceb1f632b5b5b93455 : FAILURE in 10m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e871a72807e46528f7d9e45a5169924 : SUCCESS in 16m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d7f0e9f6b6b244e787099480e57bdb2b : FAILURE in 1h 03m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/eae2c86bc8c44a66a080b3c0f5c0aadd : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0c3a50584b1b420695491e0263ac3e34 : SUCCESS in 1h 00m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fe4bb4d9ac84b37acefe053dd66eea5 : SUCCESS in 59m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/167c839d49c54dcf8f9246b381c08239 : SUCCESS in 1h 05m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cb1e076fd3b44745a5e85a55f70ed12b : FAILURE in 1h 44m 57s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f29689d195074aadb901233fdad21ad9 : FAILURE in 55m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b51a361116c5410e8e8f9c2a2b1ea55f : FAILURE in 52m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6af87184867d409d963fbf7125e37e6f : SUCCESS in 1h 38m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3bd2932bf0914948a4422133a50df8d4 : SUCCESS in 10m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/24c6b6d5d91f4107b0aad1f114e562e7 : SUCCESS in 9m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87d9588f519c4518b1b8bc5637213aa3 : SUCCESS in 14m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0331ad82b1bd485da47565c3ae5ec13c : SUCCESS in 17m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3651ce3b576b43508e3cb666d7cc610b : SUCCESS in 20m 07s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2bb4dede25554ee98e3b928c4f0d3ce5 : FAILURE in 20m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e016d958334466d98b3f77432c72409 : FAILURE in 20m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/403799c33b754bc3a688eb30dcbca0f4 : SUCCESS in 33m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0cc6844afb9a4b0397b293940396782d : SUCCESS in 10m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c808febdf8f4894899e0db3b9756ba3 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":24},{"id":"e63af59e50e26ffa5613d40cbea47a6f1183d17a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 17:36:01.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"44ddaf1445cd0ba8fbdfdeeb265b23f25f5e4bc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-03 11:33:42.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"2bd7ee936fd2f3cf3e0f58d12c3cdd184b612aba","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-03 11:38:45.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"a2b288e31dabd83a2a8c61225157124573f2b532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 12:37:46.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd75f1a493c74353998c30c117dca26c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518ff5d137d9469497fba5655b18b8c3 : SUCCESS in 9m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df187907944c422faabb5e0804dc829d : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e644471132ef4276a7501c9f37a40b89 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df1ed8d7dfa45438375f38ff24d1e09 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42dbda65bcd3496da864e059cfe669e8 : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bbadd4828ae84599bcaa75633f5f1165 : SUCCESS in 6m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84d972c35e0b4ee5a728a9fe7671a2e7 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/88a433d0b39146a2881233371221c4f8 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f12d7cad132d4b0ebd9a1e9df4ca0410 : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ace1d7f98cb14b62a2eccdeadcebac8c : SUCCESS in 13m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e3c0d7193f0f48a99ced0bb028806872 : SUCCESS in 17m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04222eedda8549cf82f7cdfb59176747 : SUCCESS in 13m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fa27a52b69e7474485cfc1999006af2e : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d03d9fa4099540cc954d00f978908c6b : SUCCESS in 20m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a2dc06122fe44479863185131ddd08c6 : SUCCESS in 58m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3e4a604daf3843c7894adc93eff47eee : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0d07a656e73c4898b12ad87af689d33e : SUCCESS in 1h 00m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0188d191617e4d669fb214bc05754c9b : SUCCESS in 59m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5f91bdce0384462d8123303ddb073ece : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a426877fa8fc4c3d96c823556a8ebfb5 : FAILURE in 52m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/af190ec21e01403196b7bf1d33d5a611 : FAILURE in 51m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/491f7b78f58740d98334558bc0346956 : FAILURE in 46m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cff9519bc9140e2ab7caf01bd0c60ad : FAILURE in 52m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/978bfe6b7491434293d2b6f7d7d6f594 : SUCCESS in 21m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4f5d6bad4a44e88b756f9c9ee0d85ce : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":25},{"id":"4fa6655292a5d4e8493cd1614736f06180457e8f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-04 09:53:11.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"29365c72dd09b294ba4d10a62c26056470323e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 11:32:32.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f913d0bd32a484c9a4609ea127eb58f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba477e4203f4859a19deac2d82fda2c : SUCCESS in 8m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2975f894190d4234893eb3cfed15e212 : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16fdae1bd4a042e6a4bffff624da7a23 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f461d854a2242669759a3354ebfdc44 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adb98ab967564f5bb3e963087f4c0f84 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6f7542771d84282858e648d03f980d8 : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bf99cb4b16e441aa4a53b1e307727dd : SUCCESS in 1m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/644a607ba51543d0907b4c98f34a0f4c : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b089aa103ad14f7dba84ebe23820067e : SUCCESS in 12m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce640a69b4604c79bf07a7d3f2b177a8 : SUCCESS in 11m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0e25ba7e4c254b4b85bec267472e259d : SUCCESS in 21m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e61ffc11ac4445da63dfabac307dc3a : SUCCESS in 13m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/838542272efe4417897b99060fb7f624 : SUCCESS in 22m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5b27715eae84845a67ab1faada4fd59 : SUCCESS in 22m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e81c38d440d24f6099e14b3e84862b47 : SUCCESS in 59m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ad80d96c42a944d78dfe5db44ab961a3 : SUCCESS in 57m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9fc4c8a894bf4fe39d3bfc2789840f15 : SUCCESS in 1h 02m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90efe7ca9e434a9c94ff425eff9864db : SUCCESS in 58m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc1d5fe31c4645448cc22393810785cb : SUCCESS in 57m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0272a973847649e1b87d881275a5ed67 : SUCCESS in 1h 38m 22s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f54617cdea9b4341be20f198469eb68f : SUCCESS in 1h 36m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/157d0de2419e450384c3c6864dc661dd : SUCCESS in 1h 30m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e44aac12d564b86bbc0684e335a3a65 : SUCCESS in 1h 35m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aaab2ac1b2f84db3892656a9c092c9da : FAILURE in 22m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64584b5a9dd043fbb99b42e4ad705c94 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":25},{"id":"31988e67b315d57889be5eaef15466581808b022","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-04 12:06:38.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"0e3f672fda30ba0ca3531682813306065f207f3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 13:46:12.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f51c5575bf2b408196daa193c840b401\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995b4b9b3aac4a30a464865d9343de4a : SUCCESS in 4m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/883da442ac7f433c8a55a7e05fc68a6a : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e4de5589d284e23a1af5fce44c8d2f3 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a83e0e0c6974fbc92ef435ecae5134e : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e17e89e050c546938fea1b9aeedebabb : SUCCESS in 2m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e202f6665b34e838e985df075c28159 : SUCCESS in 7m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf2147b5ec42486b910eea630ffc034b : SUCCESS in 2m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f237d7fdc3a402f92bd477a695595f7 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e888ae67cac348d4a598eaa5cbce355b : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7232114095864c319112676daa334899 : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/73dac35e725d46f6bbea1cfbbd0d5341 : SUCCESS in 15m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2d804ea759747abb0cc733a7605a558 : SUCCESS in 15m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a37d87daa0224ab0807e651a90d1a2bc : SUCCESS in 11m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebcfe15acdda475a8c6990203cb3f540 : SUCCESS in 17m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4daedcdd38eb407e89847847df53d5fa : SUCCESS in 59m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/680b70b408124c9788b6c5644131b6ee : SUCCESS in 58m 07s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b966ab3065664965b77a721501568ea3 : SUCCESS in 1h 05m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62feace2dbb9499eaff332be163dbad5 : SUCCESS in 59m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0dde6f80df96430ea6d097d85bb57ec2 : SUCCESS in 57m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/167fb0b739ae4e4aa2ee78f8d9ee1dd6 : SUCCESS in 1h 37m 52s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8e537b2c0dcd4d7a906818642055239f : SUCCESS in 1h 34m 25s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc7b500330bb4ca88a1547cc56db83f8 : SUCCESS in 1h 31m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72540d104166425ebb9c3fa5da6fb539 : SUCCESS in 1h 34m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6db920ed48b447ea85bde33ffad438ed : SUCCESS in 12m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6d15ac5b3f01498285e54f916d511ebd : SUCCESS in 18m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e388b42b6ec46879227ea9a6872f51b : SUCCESS in 19m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b2ca43f77dd647d3921b56e34bbb2d4b : SUCCESS in 18m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/22c3a20e3ab74fcfb6ed703aba6dd4ed : SUCCESS in 18m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f01bc310a1c647b9b9f8ebc0d21a69ac : SUCCESS in 26m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fddd2dae4564cefa08e772b4e6219cf : SUCCESS in 35m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0bb5489c4ed4f729482b576b6a0b645 : SUCCESS in 28m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/41065721a4594709947271ed374361c5 : SUCCESS in 11m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a40e579ca1714c63a0aca25c8b674aa2 : SUCCESS in 16m 50s","accounts_in_message":[],"_revision_number":25},{"id":"c4132e849b25beb689faf0d172a0efd26d2c7142","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-04 13:54:09.000000000","message":"Patch Set 25: Review-Priority+1","accounts_in_message":[],"_revision_number":25},{"id":"84ba34876b31253d4be18f12f584fd6ba67945a8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-04 15:57:47.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"d36f8c83abe79dfe86c34a0a0f3979d6194f25ff","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-09 13:16:13.000000000","message":"Patch Set 25: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"64cfb4b04049c660c7901392bfa8b0af3f2a7c30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-16 11:14:14.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"7e7c1765aadb0aec547f6515b3df40e3a96f468b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-16 11:15:03.000000000","message":"Patch Set 26: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"d038c197097438a9181079007ea4f15802360f07","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-16 11:21:58.000000000","message":"Patch Set 26: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":26},{"id":"d98cfa17280bcab5ce4c755db964427a9412aa23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 13:03:36.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16640f85d7a84a1294364ac5a2e06532\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b394073e77848df8d9010e9692e5c4f : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dc284cdc5f2b4e45be152b074f3425d6 : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21de6d34afc14b3f87e8ccb3194d715c : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30dce8657564bcf87cb365246c4a369 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8f10a00fa55432fa374511f152aff6e : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6217f36ed00944dabfafd6cf595fb401 : SUCCESS in 8m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7399b9e707a54885bcb1874c77c47229 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a608188d223e4b99b1db03ba5d17d8e1 : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87add118a72044b2b82c778958ca3cc6 : SUCCESS in 14m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be2dd1d3f01c422f9479577f8163e4ba : SUCCESS in 6m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c154eeb973504f6fa57a03a8848fbd26 : SUCCESS in 10m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f756116f730043729c38100327c80f26 : SUCCESS in 13m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ef282a9e63bc49b39587a90a1b8cf350 : SUCCESS in 25m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d226ed34fd74565bda6a6131d4805c7 : SUCCESS in 9m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9dcdff3cad0745e1b42581a15d7e0c2b : SUCCESS in 59m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bf74684119af4fff97329cfdbfc5496d : SUCCESS in 55m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/39ef90dbf43a4f9fbb6a25250123a678 : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/894a706d08d54e48848a3ccd4914b5aa : SUCCESS in 1h 06m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3388d41443574b11a939c7171500fb75 : SUCCESS in 57m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/762b287167844ead994ec1008b47f236 : SUCCESS in 1h 47m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4b4de7d0f6034578bbc80f31d7c4f28b : SUCCESS in 1h 41m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cf30df401324b77909b3b901fb65aed : SUCCESS in 1h 40m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/889de5f4cb2d4769b54e28217b64984e : SUCCESS in 1h 37m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bde89ce050d049e089c0329a0c9fd7c8 : SUCCESS in 11m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cd841a56e9bc4529b155de02ccf82c7f : SUCCESS in 22m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e1a326bab7540a7a02e45b235a070a4 : SUCCESS in 22m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5ad63545d3b43f79c748aeb4f713d26 : SUCCESS in 9m 12s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f6deeccb754e4fba8416f51de2d7ff4c : SUCCESS in 34m 34s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fbb48008ac544024be6038096151a7c0 : SUCCESS in 31m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7988762db33742ccad1ae1a4b9bc0d15 : SUCCESS in 29m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f004e9d48a1417d8fc9485a20c805c4 : SUCCESS in 28m 10s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6aa6988912184a8096f1a16561487d36 : SUCCESS in 17m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e535705f81e4a83a6b55d6e1c7fa29b : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":26},{"id":"3e4b814825b88d96d6378efef51467a20279a283","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-17 09:33:55.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"5812220a3555d814df64eececc71c8658117b7db","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-18 10:31:59.000000000","message":"Patch Set 26:\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"519ca7acbaf200cdebeeab7b36f69ef765e76f10","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-19 14:01:49.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"6af6ed0b5142193a25a5ddb30947ba945f7e928f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-19 14:13:30.000000000","message":"Patch Set 26: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"e57679a3919fbd0283d385e73d83e83d64698ea0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 14:13:49.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"b0dae3c79922c849ddba3fdbea5afbfd6db91836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 16:51:42.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53eb7bfb530242c5a9f7a6b9e87c24ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163ff7c4a2974c21a75a7b2e94a10957 : SUCCESS in 9m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c5eeeb61c894088a1533d357e64ed84 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a03a39ac42424da885ea278ffc692b : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db16f15b5de243d19d1cda603e98df66 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b04f5cf8c34ac5984f0092b79a3225 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b2b1eccfccc47fc8005afcfae55478f : SUCCESS in 6m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bec80b172af84e3a9861e750385411de : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50780a3d707a44d3a864c2564ede7feb : SUCCESS in 14m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d8c8cd1a16db4e09938e14eb3bb72103 : SUCCESS in 20m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d149e2d255ab4aeab10abff343510886 : SUCCESS in 16m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4823d6c9317743a09fb5e5063194f898 : SUCCESS in 26m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9576084a3774447a20d2087add23f97 : SUCCESS in 22m 30s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2d1022efb9e14a98979828c2e0b65f04 : SUCCESS in 1h 14m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d46adc8d05ad405c8194376bb44b20c3 : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d4b3807049934c17b80bd720aca4c648 : SUCCESS in 1h 16m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae2d315e400548d6a4d0bd516f7d05db : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b8ae6aa0b014019a896dff777d0aff1 : SUCCESS in 59m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/57795d91eebd45e4acfb3d70ed882084 : SUCCESS in 1h 43m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/afe078ff47df4e3b86ba21db2cc4fbd6 : SUCCESS in 1h 42m 02s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4b61e41b8e1451f8484b8bfd5c394ce : SUCCESS in 1h 47m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48e28f03b61644789f5b4e5268d4017d : SUCCESS in 1h 53m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6bbac8ea971e4be9b1dc4fd876a74a7c : SUCCESS in 19m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e71f5cb7aff4b92bf6b4c425b389be7 : SUCCESS in 23m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2eb799713f840ecbd01b4e862427aea : SUCCESS in 20m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6358263c754b4cf09ecf8bf13a4637ab : SUCCESS in 17m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cf55804867754a42931f44edc9a4afe1 : SUCCESS in 30m 34s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b5a11648979f4b0b803337c932c233e1 : SUCCESS in 35m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32ffd69ad5b14d7081055251a714a33a : SUCCESS in 31m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c274f21f5fcb4ee1a572eface0aec7dc : SUCCESS in 34m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6beb0575f69f4eb69c721b0c2378b8a8 : SUCCESS in 21m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e311821d1034a6cb4d86cace458103d : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":26},{"id":"ac2d903d5aa626e84cde63a0ea76a8b61f32ff4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 16:51:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26},{"id":"92367c84e2cfa5e48d2290f0ec6359e82993e7cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 16:53:16.000000000","message":"Patch Set 26:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89cbe4f68a1846ef8d7545f01858e9e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf29d5fe9b7d4207ae94db0db2a4fe68 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d31dbd5a88b74a519e8c6c8c0e982389 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"fb1767ad7f42235f576ec5ae3fd6f623f207e512","revisions":{"353d7c60f97e35b6b288a2adf4df327bddc0db19":{"kind":"REWORK","_number":1,"created":"2024-11-19 16:11:42.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/1"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"subject":"WIP: Drop kolla-tags and kolla-limit","message":"WIP: Drop kolla-tags and kolla-limit\n\nInstead we always run the kolla-openstack role.\n\nAlso adds the bifrost tag to kolla-bifrost, so\nthat we can easily limit to bifrost in the seed\nservice deploy.\n\nStill todo:\n - Test more thoroughly\n - Update existing docs, including bifrost tags docs\n - Reno\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/353d7c60f97e35b6b288a2adf4df327bddc0db19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/353d7c60f97e35b6b288a2adf4df327bddc0db19"}]},"branch":"refs/heads/master"},"fb293d7ae8ae72a97d94fbe2bb751f7821f32cb1":{"kind":"REWORK","_number":2,"created":"2024-11-19 16:47:46.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/2"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:47:12.000000000","tz":0},"subject":"WIP: Drop kolla-tags and kolla-limit","message":"WIP: Drop kolla-tags and kolla-limit\n\nInstead we always run the kolla-openstack role.\n\nAlso adds the bifrost tag to kolla-bifrost, so\nthat we can easily limit to bifrost in the seed\nservice deploy.\n\nStill todo:\n - Test more thoroughly\n - Reno\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb293d7ae8ae72a97d94fbe2bb751f7821f32cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb293d7ae8ae72a97d94fbe2bb751f7821f32cb1"}]},"branch":"refs/heads/master"},"41746bcd077765993f98d6865cd499424938aab1":{"kind":"REWORK","_number":3,"created":"2024-11-20 09:18:20.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/3"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-20 09:18:13.000000000","tz":0},"subject":"WIP: Drop kolla-tags and kolla-limit","message":"WIP: Drop kolla-tags and kolla-limit\n\nInstead we always run the kolla-openstack role.\n\nAlso adds the bifrost tag to kolla-bifrost, so\nthat we can easily limit to bifrost in the seed\nservice deploy.\n\nStill todo:\n - Test more thoroughly\n - Reno\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41746bcd077765993f98d6865cd499424938aab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41746bcd077765993f98d6865cd499424938aab1"}]},"branch":"refs/heads/master"},"9424e4a36d9cf11692ea253b5c31c3037ab8a8a6":{"kind":"REWORK","_number":4,"created":"2024-11-20 09:45:52.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/4"}}},"commit":{"parents":[{"commit":"a36bb614c092c147da11c20334dff6839d3c29b2","subject":"Add loop control to custom DNF repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a36bb614c092c147da11c20334dff6839d3c29b2"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-20 09:45:43.000000000","tz":0},"subject":"WIP: Drop kolla-tags and kolla-limit","message":"WIP: Drop kolla-tags and kolla-limit\n\nInstead we always run the kolla-openstack role.\n\nAlso adds the bifrost tag to kolla-bifrost, so\nthat we can easily limit to bifrost in the seed\nservice deploy.\n\nStill todo:\n - Test more thoroughly\n - Reno\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9424e4a36d9cf11692ea253b5c31c3037ab8a8a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9424e4a36d9cf11692ea253b5c31c3037ab8a8a6"}]},"branch":"refs/heads/master"},"425e39318d522c0acdb23b0b0133bef0fc9658b8":{"kind":"REWORK","_number":5,"created":"2025-01-14 12:23:23.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/5"}}},"commit":{"parents":[{"commit":"aaa77476f43903a93c7d70888c1ca3e22cf33286","subject":"Merge \"Fix release note markup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aaa77476f43903a93c7d70888c1ca3e22cf33286"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-14 12:23:02.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nInstead we always run the kolla-openstack role.\n\nAlso adds the bifrost tag to kolla-bifrost, so\nthat we can easily limit to bifrost in the seed\nservice deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/425e39318d522c0acdb23b0b0133bef0fc9658b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/425e39318d522c0acdb23b0b0133bef0fc9658b8"}]},"branch":"refs/heads/master"},"3532c15379c1989da84c782d7fb291ccfd6c9d92":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-01-14 12:59:59.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/6"}}},"commit":{"parents":[{"commit":"aaa77476f43903a93c7d70888c1ca3e22cf33286","subject":"Merge \"Fix release note markup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aaa77476f43903a93c7d70888c1ca3e22cf33286"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-14 12:59:21.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kolla-openstack`` role is now\nalways called,to allow for limiting to OpenStack services with just one\ntag, e.g. ``kayobe overcloud service deploy -t nova`. You can still skip\nthis role with ``--skip-tags kolla-openstack``.\n\nAlso adds the bifrost tag to kolla-bifrost, so\nthat we can easily limit to bifrost in the seed\nservice deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3532c15379c1989da84c782d7fb291ccfd6c9d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3532c15379c1989da84c782d7fb291ccfd6c9d92"}]},"branch":"refs/heads/master"},"73a45f7c47911e2bf1cef2fd548f0813530e8d72":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-01-14 13:20:28.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/7"}}},"commit":{"parents":[{"commit":"aaa77476f43903a93c7d70888c1ca3e22cf33286","subject":"Merge \"Fix release note markup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aaa77476f43903a93c7d70888c1ca3e22cf33286"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-14 13:20:14.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kolla-openstack`` role is now\nalways called,to allow for limiting to OpenStack services with just one\ntag, e.g. ``kayobe overcloud service deploy -t nova`. You can still skip\nthis role with ``--skip-tags kolla-openstack``.\n\nAlso adds the bifrost tag to kolla-bifrost, so that we can easily limit\nto bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73a45f7c47911e2bf1cef2fd548f0813530e8d72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73a45f7c47911e2bf1cef2fd548f0813530e8d72"}]},"branch":"refs/heads/master"},"713a2707dec1839b4e3ae49e973a99c7a4720720":{"kind":"REWORK","_number":8,"created":"2025-01-22 14:00:29.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/8"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-22 13:59:45.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kolla-ansible`` tag is now\nalways called,to allow for limiting to OpenStack services with just one\ntag, e.g. ``kayobe overcloud service deploy -t nova`. You can still skip\nthis with ``--skip-tags kolla-ansible``.\n\nAlso adds the ``bifrost`` tag to kolla-bifrost, so that we can easily\nlimit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/713a2707dec1839b4e3ae49e973a99c7a4720720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/713a2707dec1839b4e3ae49e973a99c7a4720720"}]},"branch":"refs/heads/master"},"51a7ea142a1ebead8ef76a0173f15697ed85b921":{"kind":"REWORK","_number":9,"created":"2025-01-23 10:29:45.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/9"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-23 10:29:37.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kolla-ansible`` tag is now\nalways called,to allow for limiting to OpenStack services with just one\ntag, e.g. ``kayobe overcloud service deploy -t nova`. You can still skip\nthis with ``--skip-tags kolla-ansible``.\n\nAlso adds the ``bifrost`` tag to kolla-bifrost, so that we can easily\nlimit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51a7ea142a1ebead8ef76a0173f15697ed85b921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51a7ea142a1ebead8ef76a0173f15697ed85b921"}]},"branch":"refs/heads/master"},"27546c5880a03412e28fac8d56d6a57e59af4d79":{"kind":"REWORK","_number":10,"created":"2025-01-23 10:54:19.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/10"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-23 10:54:12.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kolla-ansible`` tag is now\nalways called,to allow for limiting to OpenStack services with just one\ntag, e.g. ``kayobe overcloud service deploy -t nova`. You can still skip\nthis with ``--skip-tags kolla-ansible``.\n\nAlso adds the ``bifrost`` tag to kolla-bifrost, so that we can easily\nlimit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27546c5880a03412e28fac8d56d6a57e59af4d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27546c5880a03412e28fac8d56d6a57e59af4d79"}]},"branch":"refs/heads/master"},"f5a9bd13cf4aff78d659a97c0385974509c2e793":{"kind":"REWORK","_number":11,"created":"2025-01-23 11:09:52.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/11"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-23 11:09:46.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kolla-ansible`` tag is now\nalways called,to allow for limiting to OpenStack services with just one\ntag, e.g. ``kayobe overcloud service deploy -t nova`. You can still skip\nthis with ``--skip-tags kolla-ansible``.\n\nAlso adds the ``bifrost`` tag to kolla-bifrost, so that we can easily\nlimit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5a9bd13cf4aff78d659a97c0385974509c2e793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5a9bd13cf4aff78d659a97c0385974509c2e793"}]},"branch":"refs/heads/master"},"a30f1d71e7957953401e0eae0940e06d915bf2a2":{"kind":"REWORK","_number":12,"created":"2025-01-23 11:25:39.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/12"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-23 11:25:20.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kolla-ansible`` tag is now\nalways called,to allow for limiting to OpenStack services with just one\ntag, e.g. ``kayobe overcloud service deploy -t nova`. You can still skip\nthis with ``--skip-tags kolla-ansible``.\n\nAlso adds the ``bifrost`` tag to kolla-bifrost, so that we can easily\nlimit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30f1d71e7957953401e0eae0940e06d915bf2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30f1d71e7957953401e0eae0940e06d915bf2a2"}]},"branch":"refs/heads/master"},"87a7ed0750c1d5f428c3ae11f6b281a01fbce264":{"kind":"REWORK","_number":13,"created":"2025-02-06 11:06:47.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/13"}}},"commit":{"parents":[{"commit":"fad3b7f1da951f85b214ac9b7a573f6ff11c608f","subject":"CI: Add Rocky Linux SLURP upgrade jobs for 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fad3b7f1da951f85b214ac9b7a573f6ff11c608f"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-02-06 11:05:08.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``--kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kayobe-generate-config`` tag is\nadded to ``kolla-ansible.yml`` and ``kolla-openstack.yml``. It is always\ncalled, to allow for limiting to OpenStack services with just one tag,\ne.g. ``kayobe overcloud service deploy -t nova`. You can still skip this\nwith ``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87a7ed0750c1d5f428c3ae11f6b281a01fbce264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87a7ed0750c1d5f428c3ae11f6b281a01fbce264"}]},"branch":"refs/heads/master"},"3347ef57dd495c5e49be53df959f40f077986829":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-03-10 09:37:08.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/935669/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/14"}}},"commit":{"parents":[{"commit":"80494bf7da3a060194370b9782b54851cb89f135","subject":"Bump Michael.Rigart.Interfaces version to 1.15.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80494bf7da3a060194370b9782b54851cb89f135"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-10 09:37:08.000000000","tz":0},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``--kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kayobe-generate-config`` tag is\nadded to ``kolla-ansible.yml`` and ``kolla-openstack.yml``. It is always\ncalled, to allow for limiting to OpenStack services with just one tag,\ne.g. ``kayobe overcloud service deploy -t nova`. You can still skip this\nwith ``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3347ef57dd495c5e49be53df959f40f077986829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3347ef57dd495c5e49be53df959f40f077986829"}]},"branch":"refs/heads/master","description":"Rebase"},"99d3f0e35316ab6e796756b3b46bf82bd5e199c5":{"kind":"REWORK","_number":15,"created":"2025-08-21 10:28:19.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/15"}}},"commit":{"parents":[{"commit":"3dee45436f5e5da6046a51d35ea736ea3c9f9b83","subject":"Merge \"Fix known_hosts module issue on centos/rocky 10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3dee45436f5e5da6046a51d35ea736ea3c9f9b83"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-08-21 10:28:12.000000000","tz":60},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``--kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kayobe-generate-config`` tag is\nadded to ``kolla-ansible.yml`` and ``kolla-openstack.yml``. It is always\ncalled, to allow for limiting to OpenStack services with just one tag,\ne.g. ``kayobe overcloud service deploy -t nova`. You can still skip this\nwith ``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99d3f0e35316ab6e796756b3b46bf82bd5e199c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99d3f0e35316ab6e796756b3b46bf82bd5e199c5"}]},"branch":"refs/heads/master"},"cd7dc6949e61535f6e6cae2197ce463ebcf1eafe":{"kind":"REWORK","_number":16,"created":"2025-08-28 08:34:17.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/16"}}},"commit":{"parents":[{"commit":"3dee45436f5e5da6046a51d35ea736ea3c9f9b83","subject":"Merge \"Fix known_hosts module issue on centos/rocky 10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3dee45436f5e5da6046a51d35ea736ea3c9f9b83"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-08-28 08:34:07.000000000","tz":60},"subject":"Drop kolla-tags and kolla-limit","message":"Drop kolla-tags and kolla-limit\n\nRemoves the options ``--kolla-tags`` and ``--kolla-limit`` from all\ncommands. Regular ``--tags`` and ``--limit`` will now be passed directly\nto the Kolla-Ansible invocations. The ``kayobe-generate-config`` tag is\nadded to ``kolla-ansible.yml`` and ``kolla-openstack.yml``. It is always\ncalled, to allow for limiting to OpenStack services with just one tag,\ne.g. ``kayobe overcloud service deploy -t nova`. You can still skip this\nwith ``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd7dc6949e61535f6e6cae2197ce463ebcf1eafe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd7dc6949e61535f6e6cae2197ce463ebcf1eafe"}]},"branch":"refs/heads/master"},"abb5fc4c6c542373526ff7fe1de87712e96afbf2":{"kind":"REWORK","_number":17,"created":"2025-12-10 13:44:28.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/17"}}},"commit":{"parents":[{"commit":"ec066451766f43873081b6f1120791ddfe05ef09","subject":"Revert \"[release] Use OpenStack 2025.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec066451766f43873081b6f1120791ddfe05ef09"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-12-10 13:42:46.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb5fc4c6c542373526ff7fe1de87712e96afbf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb5fc4c6c542373526ff7fe1de87712e96afbf2"}]},"branch":"refs/heads/master"},"aa6a29a54330896aaab284daa1be3aba398a356a":{"kind":"REWORK","_number":18,"created":"2025-12-29 14:14:40.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/18"}}},"commit":{"parents":[{"commit":"ec066451766f43873081b6f1120791ddfe05ef09","subject":"Revert \"[release] Use OpenStack 2025.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec066451766f43873081b6f1120791ddfe05ef09"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-12-29 14:14:31.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa6a29a54330896aaab284daa1be3aba398a356a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa6a29a54330896aaab284daa1be3aba398a356a"}]},"branch":"refs/heads/master"},"64440fa93df3f7c7ab99af94a0d8cc667cb0a115":{"kind":"REWORK","_number":19,"created":"2026-01-15 13:32:42.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/19"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-15 13:32:40.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64440fa93df3f7c7ab99af94a0d8cc667cb0a115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64440fa93df3f7c7ab99af94a0d8cc667cb0a115"}]},"branch":"refs/heads/master"},"d4fc5e3704ba65c2b01c5c667fcc088edaf622d6":{"kind":"REWORK","_number":20,"created":"2026-01-21 13:30:39.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/20"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-21 13:30:32.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4fc5e3704ba65c2b01c5c667fcc088edaf622d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4fc5e3704ba65c2b01c5c667fcc088edaf622d6"}]},"branch":"refs/heads/master"},"d8562cd630e0b4906ae9a644d0e6b782dc2796f0":{"kind":"REWORK","_number":21,"created":"2026-01-22 16:03:28.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/21"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-22 16:03:20.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8562cd630e0b4906ae9a644d0e6b782dc2796f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8562cd630e0b4906ae9a644d0e6b782dc2796f0"}]},"branch":"refs/heads/master"},"b99702732f5810df51474f42aa4ec573a4ef1d62":{"kind":"REWORK","_number":22,"created":"2026-01-26 13:30:38.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/22"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-26 13:30:32.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b99702732f5810df51474f42aa4ec573a4ef1d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b99702732f5810df51474f42aa4ec573a4ef1d62"}]},"branch":"refs/heads/master"},"5f3fc21668ea5400649408ecad84703c55824478":{"kind":"REWORK","_number":23,"created":"2026-01-27 15:09:34.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/23"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-27 15:08:59.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f3fc21668ea5400649408ecad84703c55824478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f3fc21668ea5400649408ecad84703c55824478"}]},"branch":"refs/heads/master"},"0aceb06f6e718efb1a329d924c2c15394eec669d":{"kind":"REWORK","_number":24,"created":"2026-02-02 13:25:55.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/24"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 13:25:43.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0aceb06f6e718efb1a329d924c2c15394eec669d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0aceb06f6e718efb1a329d924c2c15394eec669d"}]},"branch":"refs/heads/master"},"315492a199b78e40bf05cc198797d88ac0e718ea":{"kind":"REWORK","_number":25,"created":"2026-03-03 11:33:42.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/25"}}},"commit":{"parents":[{"commit":"aa230f9d05eea695f2aaa09a348f56a2079e6eda","subject":"CI: Disable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa230f9d05eea695f2aaa09a348f56a2079e6eda"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-03 11:33:35.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/315492a199b78e40bf05cc198797d88ac0e718ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/315492a199b78e40bf05cc198797d88ac0e718ea"}]},"branch":"refs/heads/master"},"fb1767ad7f42235f576ec5ae3fd6f623f207e512":{"kind":"REWORK","_number":26,"created":"2026-03-16 11:14:14.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/69/935669/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/935669/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/935669/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/935669/26"}}},"commit":{"parents":[{"commit":"aa230f9d05eea695f2aaa09a348f56a2079e6eda","subject":"CI: Disable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa230f9d05eea695f2aaa09a348f56a2079e6eda"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-11-19 16:09:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-16 11:14:08.000000000","tz":0},"subject":"Deprecate kolla-tags and kolla-limit","message":"Deprecate kolla-tags and kolla-limit\n\nDeprecates the options ``--kolla-tags``, ``--kolla-skip-tags``, and\n``kolla-limit``. Regular ``--tags``, ``--skip-tags``, and ``--limit``\nwill now be passed directly to the Kolla-Ansible invocations. The\n``kayobe-generate-config`` tag is added to ``kolla-ansible.yml`` and\n``kolla-openstack.yml``. It is always called, to allow for limiting to\nOpenStack services with just one tag, e.g. ``kayobe overcloud service\ndeploy -t nova`. You can still skip this with\n``--skip-tags kayobe-generate-config``.\n\nAlso adds the ``bifrost`` tag to ``kolla-bifrost.yml``, so that we can\neasily limit to bifrost in the seed service deploy.\n\nAs there is no clean way to handle some of Ansible\u0027s \"special\" tags\nlike ``none``, using both regular and kolla tags/limits together is no\nlonger allowed.\n\nChange-Id: I6f466305d49031da4d048f8fa7d2625b261a6fa0\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
