)]}'
{"id":"openstack%2Fkayobe~936416","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~Id292915fc4c4fefd86a6da431b35daf5faf45523","project":"openstack/kayobe","branch":"stable/2024.2","hashtags":[],"change_id":"Id292915fc4c4fefd86a6da431b35daf5faf45523","subject":"CI: use quay.io/podman/hello due too dockerhub rate limits","status":"MERGED","created":"2024-11-27 21:52:08.000000000","updated":"2024-11-28 11:52:56.000000000","submitted":"2024-11-28 11:51:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936416","cherry_pick_of_change":936130,"cherry_pick_of_patch_set":3,"meta_rev_id":"cf5292848b4ed0dc745a20bf9e616dc7b35fbd1a","_number":936416,"virtual_id_number":936416,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-28 11:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2024-11-28 09:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2024-11-28 09:33:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-27 23:58:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"862f48f880958e02444bc50c83f9aba0fc2e699a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-27 21:52:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2ccb9398c1935ffd33cd72c74e3935c73eac4488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 23:58:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0700ed40a9774446aaf4d339ed33f14a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a68417233fa4c42984d732e30ac9e4b : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db4db81c2cf4832a335c30d2a864d8f : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65a7b4d2cd934bb69906531e78e81453 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4ddceffef94447dbf11d04a416efe82 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ddadc3e3f84ff4bf79d8bd09c0878f : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ab6b88a4fb744a686fa3b5527ede36a : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ae6318aafcc4110b85fed5e030726cf : SUCCESS in 6m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97bd61f6f0e8411681c1a90851481be6 : SUCCESS in 5m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/03cc9fae781940c79ab9b0e6fee35dd5 : SUCCESS in 34m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/281a5623215b40e1a9efb8756b2547dd : SUCCESS in 1h 02m 44s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03ffac197fc841b8b08070219b21ae9b : SUCCESS in 57m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/74f2d8dcf9da46f7aaf3ec993c573d9d : SUCCESS in 1h 14m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f13392ca820142bea0bc2ab820d2788d : SUCCESS in 20m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/16f0f98d58a84910bba7220e459a4479 : SUCCESS in 17m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fe7cdf3a96e46da8a58e4e7850b981c : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/466e1746cd53406cab76f6a2acac1b89 : SUCCESS in 2h 00m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d8ce187036f49a1abdd8045d61d50b2 : SUCCESS in 1h 33m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bf23ecbf41414767a0f1b425860a9650 : SUCCESS in 19m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/370947344cd849d297e6c793c63d5531 : SUCCESS in 17m 04s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d77e15db59d44dd687eed93083d585f9 : SUCCESS in 43m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0ade1d3156a1434d90143314aa7f3ac1 : SUCCESS in 36m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a921753b758d44d3b5c9da01cc741539 : SUCCESS in 26m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/35bdb5038f7040c8a95eb7dc4be5bc8a : SUCCESS in 14m 55s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/2c3492cdb68843789154026a3ecb9072 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dea707d84d2a4627b4c54a2c3485fee5 : SUCCESS in 11m 45s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/385ffa37d75b447d863b26928fb77fed : SUCCESS in 11m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/428c795a1b4d4a1584c23a01131d2e16 : SUCCESS in 16m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99ffea34dbbe4da5a74e4ad38a506846 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a0a5fbe216153d12d4f114ac46061746c90da0aa","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-28 09:33:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"00b2425de391ba05c4bd9979013d1ffc1fa75e75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 09:34:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7b6f3dbbf4a50d5a05abc0e4fc9afa1b6bbcbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 11:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3249e849815d4462abea2be3498f2113\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/509f0a3d0801424c93d70577a4a8b422 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9b842d406624460b2fdae33511960f1 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b9cb1caeb74f84926f67d676761311 : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca17a74dce56484189d0c97ffdb44b24 : SUCCESS in 3m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c37e7db22084e30b9bb1d785ed669bf : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a790a92168bb4f4d9aaf104226687087 : SUCCESS in 8m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5a0c87dc67104853a22f831e24fff41d : SUCCESS in 1h 01m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03a183b779e34a37a6aef069e95bf25f : SUCCESS in 1h 05m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6fffe3c3eb824025998e83cf413ac77b : SUCCESS in 1h 14m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/65060e42a37b498fb3bf935ad712d82c : SUCCESS in 18m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb716121ca1e4436b90c8f0ababc7903 : SUCCESS in 12m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/faabd9445867455398ba8d9049e01062 : SUCCESS in 1h 38m 31s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cf9d922222341e0a11414b408eaa71a : SUCCESS in 1h 41m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/412a22182b744055b2dc39200320ec40 : SUCCESS in 18m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ff9a0c399314bd49342b1c71617068f : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/767d78ae9c154feab98a19815bd449ad : SUCCESS in 35m 37s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e1b7d7cd8594fc08375924653011ee4 : SUCCESS in 23m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/642c994050374293b99b238a973296ea : SUCCESS in 16m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c536b8939f344636966770a9d44721e3 : SUCCESS in 7m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/16ea055e2a0b453384daf1708e7b74a6 : SUCCESS in 19m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a574e96d37440a9aaf185df2a7aa220 : SUCCESS in 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"97fe7f1d696408131cc494928272f481d8e7681c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 11:51:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf5292848b4ed0dc745a20bf9e616dc7b35fbd1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 11:52:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcce996fcf224bf1931b56e2d94b6efe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b33b9982586d48a1b452fa3b28a5a2c7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c42631e7e2628b8436394975cd6057868d71231","revisions":{"6c42631e7e2628b8436394975cd6057868d71231":{"kind":"REWORK","_number":1,"created":"2024-11-27 21:52:08.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/16/936416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/936416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/936416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/936416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/936416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/936416/1"}}},"commit":{"parents":[{"commit":"b4a79437eda7862d9d06620473d55886d2031982","subject":"Ensure ironic-python-agent matches ironic-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4a79437eda7862d9d06620473d55886d2031982"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-11-25 10:42:00.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-11-27 21:52:08.000000000","tz":0},"subject":"CI: use quay.io/podman/hello due too dockerhub rate limits","message":"CI: use quay.io/podman/hello due too dockerhub rate limits\n\nChange-Id: Id292915fc4c4fefd86a6da431b35daf5faf45523\n(cherry picked from commit eacbe8a08d313d281a818c1e57e08aa8c5a64ebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c42631e7e2628b8436394975cd6057868d71231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c42631e7e2628b8436394975cd6057868d71231"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
