)]}'
{"id":"openstack%2Fkayobe~936417","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Id292915fc4c4fefd86a6da431b35daf5faf45523","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-11-28 16:06:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id292915fc4c4fefd86a6da431b35daf5faf45523","subject":"CI: use quay.io/podman/hello due too dockerhub rate limits","status":"MERGED","created":"2024-11-27 21:52:32.000000000","updated":"2024-11-28 16:08:43.000000000","submitted":"2024-11-28 16:06:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936417","cherry_pick_of_change":936130,"cherry_pick_of_patch_set":3,"meta_rev_id":"bb4f1f50ef075162ef117b104654db2fd00d3bc8","_number":936417,"virtual_id_number":936417,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-28 16:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-11-28 15:03:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-28 09:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-11-28 09:33:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-28 00:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-28 15:03:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"a65299334ce579df60458e614ac59ebfeed54fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-27 21:52:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8e1fd65d6cb0498cc3a4f31fcb1b18ca33d32f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 00:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c27c9788ae9416b9b04965892516822\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73517da1474b4c3a86efd8d66b4068ea : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa7ecd5deacf45f994f1d8a49f5e0843 : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe2c79024b7947fa97f2db3f96ed296e : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/010855fb5cf3486294d282aca54a6463 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52bf017900a7469f8e644275da2e1c97 : SUCCESS in 9m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1d2534b4755f4cddba1dda39bd142c60 : SUCCESS in 58m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6dba6335262b47d99fe1fdca0f90e4cc : SUCCESS in 59m 10s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d2f33b9857748839b6464a4d3843e3f : SUCCESS in 49m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0645ea033eb94c0fb118d361516c6547 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5ebbee4fe2794484929b97f7947c3bcc : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/656a16ab2ac849488b7e247191215f59 : SUCCESS in 16m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38aad6d40da84706b57995e29d5b9f5b : SUCCESS in 12m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6fa85dcca7a7408ca66ecd68a7109987 : SUCCESS in 1h 48m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b1cbcebba2d4797ad8efc011a1c984e : SUCCESS in 1h 15m 44s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/de1d260b07a64d1a8f3ccc7c617b9c04 : SUCCESS in 2h 25m 19s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69f71931224a49a9ae80cf36042b8a54 : SUCCESS in 1h 05m 29s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c7ec292bf340476abe2b3a5f79c3e0f0 : SUCCESS in 16m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d57efe2d5894348a3cc1542369cc7c9 : SUCCESS in 7m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9e747e1277ae485c99f50ad2eb668ea3 : SUCCESS in 22m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/907b591894084ada9666b8ffedc49339 : SUCCESS in 18m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3dc0a258c7a4119a28ed54709d5f62f : SUCCESS in 21m 23s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e3aeae1812534ed8bf9b7170bc80bafe : SUCCESS in 28m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c806521d217c4383ba7b7d42d8e64e02 : SUCCESS in 21m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82339c9e808848278255fb91b4b61029 : SUCCESS in 13m 06s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d609940ebfc4419f955a55bbb03c16df : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9ddd8b5355b497ebddf30a5d237492a : SUCCESS in 9m 04s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/20eede57cb704019affeb27c6f1e1d73 : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0ff73ddbe6c14a0992b0135984e6385d : SUCCESS in 22m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93193670ee784fb1aa10c6596cf16fff : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8a7391c91d34246b16481d0f0a1c425 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c14d2f04a3e144edbf055fc749d8f29b : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f112b67d6bed4a90b767af8593cc2fba : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91c1db0b9e88480b8db4b0a1f39b8177 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"57de1c6001ef01c4d1c8d7ca767a98c9ec5849a4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-28 09:33:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"943fc7603774c0e092ca1c94d552133fe0b0b73d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 09:34:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a34c352dc6205f92fb94a6277c4ac04a00c8983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 11:33:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/429500284aa0462faa508aaba494b7e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82b1d14b18b4d809c6bf17d08dea8c8 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/208c21f4b80347679ec239fde809dabe : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e171bdbdc2e04aeba6624c2c889bc484 : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e90afad3d999459f99dd0aa72a960fe6 : SUCCESS in 8m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a0d3968a0ef944368ae0ce40ac701a8f : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9071bb863824e4e93b122fc6d30f086 : SUCCESS in 47m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/11c6197ed82344d0829fb5aa1f7b225e : SUCCESS in 1h 04m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ba66ddb233b14732926637ce38f4fbc5 : SUCCESS in 18m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd0b86e8a2364b73be19f305afae0fcf : SUCCESS in 13m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/01c78c1ee3f14ad0bed992a1a18bbdef : SUCCESS in 1h 52m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e02201680b349c9a515d2dfcdfa6cdf : FAILURE in 22m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d26ded6100fd4928a1140c1f0fd673b0 : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d974686956224d4f944a87c25c6dec89 : SUCCESS in 13m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08344c690a6b47e8a95043c7e0540ecb : SUCCESS in 20m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efef4679bf9e429c8407c5f7f1fb18c3 : SUCCESS in 34m 07s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2def0c1b2d3c475984b81311702c6a54 : SUCCESS in 22m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d448d11daddf43c2b76576b4d3daab49 : SUCCESS in 15m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/18634570939841f8992359b1ccd8e97d : SUCCESS in 9m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b51e37453f6a42b1a20e80ab974fd49c : SUCCESS in 16m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/014736f0c39848a79b8c5e64b99f0208 : SUCCESS in 17m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f45d925ed5114dcf8c09920cc9d792a2 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4c4a92818b33093eee05cfd92df034b3b8794412","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-28 11:55:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e6647706455fd3702b80daa51aefacc18178f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 14:13:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdacb84b2e4644a58aa1832f3b09f8a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86982280552f4762b6e25b25a5c5b36a : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322e4c73d4a34fd0b508ab36d65ed108 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/567040e36c8d45a9a736616c641bdfde : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6cc141deed4542d9ac35f21b2acd1e15 : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bc72611b22a493eb58262f73bf22454 : SUCCESS in 5m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/16d214642fec44d0a9f178fc029632e3 : SUCCESS in 59m 42s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/22a6a4debae746f2903e69d95009e4c4 : SUCCESS in 57m 21s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99e137ae994441fca0ed151ca22d8420 : SUCCESS in 48m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/375cf1893c4b451f8efb1516377f5496 : SUCCESS in 58m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/02b63f4aa4e6427cb883cd28251f5f7b : SUCCESS in 22m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c3d3f1ecb689477da45be070c5921858 : SUCCESS in 16m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e0518620f4a455c9e212ce820130654 : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/51daf8cfcbe34605be2a8ec232844dd2 : SUCCESS in 2h 04m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/278fffe436774dfea81347b29c679d16 : SUCCESS in 55m 12s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2e5eb7b012fa47719f5d3a16714b87aa : SUCCESS in 2h 15m 12s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50e0487e6129418f8607ecaaea596744 : SUCCESS in 1h 54m 06s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b358e62b38ac464e86b353a22e8b151e : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d8ca41b5a494e2ba56aaa12c4242b78 : SUCCESS in 13m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a8d1fe0dd7b24c10a040a748d047c33b : SUCCESS in 36m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a8ed315ab205451b8d1a5eed1afc9b4c : SUCCESS in 29m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05415b4e2856441a8f7c6c3a9134e28f : SUCCESS in 25m 13s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/26a7adebda48419fb45b6e4bdfa46133 : SUCCESS in 32m 34s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc183b9f9eef4724bcbc013e3d139b2e : SUCCESS in 25m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/327a517f2fc04e73a28b79cb2232713a : SUCCESS in 13m 53s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3a0554ba99bb4a8c84ade84d6b3954ef : SUCCESS in 14m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb25e33ba3d449c7a0af31b4e38da2cb : SUCCESS in 9m 02s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/778dca4b7e6e46bf9adb9f080f7daf2a : SUCCESS in 9m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82b5d762e831467b8d3d166ac93c7547 : SUCCESS in 20m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d9ec9252fe141a7a41ea827f8e36386 : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f26c645a100041108763b48e95e9b558 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4263a82415d493c93fe7d29d0fc3b71 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2454237a6a4f4369ba5858fe7ff3a102 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf0739fa123f4fd9a871c6682264e795 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"55c6031a25bdd6cedbe4c0a45be62b9cf52f94cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 14:19:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1045ef0c4355c95f62ae41d4540c4eb86df23782","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-28 15:03:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"87b78563e137228721066e3f637e63ff703eabdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 16:06:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eaf946778f6435aab3af6e350f2c863\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c84b7ec1992240d08dbbd38e0f0a9e8c : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cfc865ee58ff44b0a83b6279b01db313 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bde1c5a3c25c4368b461f6a36b16368c : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca37da547839498380cd78958ee5c452 : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/021e3588a05640088a60deb438cb391e : SUCCESS in 1h 00m 37s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1064daa2ef0b441e80a78a0e896ccfa5 : SUCCESS in 26m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/61351efd5d9247a8bb85fd66371776ad : SUCCESS in 1h 06m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bcf252ee2d764f95923c2132c357d59b : SUCCESS in 13m 01s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3030fa130d6c46a1a80b739f1436aeb0 : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/19b9dfc8cd8a44a78e16d369d4d9f5c1 : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7195a2ff5fb145648fa45143bb698348 : SUCCESS in 1h 38m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/61acdfec79d34478a959995e7dfba472 : SUCCESS in 17m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2896fc3c5c7742f9b7853f15bd87ba99 : SUCCESS in 7m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6c04fc7cf5ab42b0bea5251bf718f406 : SUCCESS in 26m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5207c9617ed849e086fabb34870d0adc : SUCCESS in 24m 39s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/14a7f68520f44127b6b00fae991514a7 : SUCCESS in 30m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/607dc7e67bfe4a15a796400fd03d93ac : SUCCESS in 25m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7bf0fb030cb2437c8745a10c1967b3d0 : SUCCESS in 15m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b082ad50c304664b53595ee4e0f6019 : SUCCESS in 10m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c50bde0f9fea438997c1676ebeb05a8c : SUCCESS in 14m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f62340cba86a4a09b6398b1924b1378e : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"07bac8fda0c20930b07c1cf547efd8362655bdf1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 16:06:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bb4f1f50ef075162ef117b104654db2fd00d3bc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 16:08:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c293d5e6d394fc0ac73dfb489f573eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba396f98133f49e9b85befeda18a4d2a : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ed7724db0c97d0c5fd696bbd9c4c2f8763cd7d6","revisions":{"1ed7724db0c97d0c5fd696bbd9c4c2f8763cd7d6":{"kind":"REWORK","_number":1,"created":"2024-11-27 21:52:32.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/17/936417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/936417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/936417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/936417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/936417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/936417/1"}}},"commit":{"parents":[{"commit":"4e07e042b74ba6c97ea921987021cf9030aac0ba","subject":"Add internal VIP address to no_proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e07e042b74ba6c97ea921987021cf9030aac0ba"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-11-25 10:42:00.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-11-27 21:52:32.000000000","tz":0},"subject":"CI: use quay.io/podman/hello due too dockerhub rate limits","message":"CI: use quay.io/podman/hello due too dockerhub rate limits\n\nChange-Id: Id292915fc4c4fefd86a6da431b35daf5faf45523\n(cherry picked from commit eacbe8a08d313d281a818c1e57e08aa8c5a64ebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ed7724db0c97d0c5fd696bbd9c4c2f8763cd7d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ed7724db0c97d0c5fd696bbd9c4c2f8763cd7d6"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
