)]}'
{"id":"openstack%2Fkayobe~936418","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~Id292915fc4c4fefd86a6da431b35daf5faf45523","project":"openstack/kayobe","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-11-28 08:02:05.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Id292915fc4c4fefd86a6da431b35daf5faf45523","subject":"CI: use quay.io/podman/hello due too dockerhub rate limits","status":"MERGED","created":"2024-11-27 21:53:05.000000000","updated":"2024-11-28 14:05:03.000000000","submitted":"2024-11-28 14:04:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936418","cherry_pick_of_change":936130,"cherry_pick_of_patch_set":3,"meta_rev_id":"4c85bc439db42e4cc62d3e880886a92d078e77d2","_number":936418,"virtual_id_number":936418,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-28 14:04:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2024-11-28 11:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2024-11-28 11:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-27 23:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a615c7bce1dcd9b2be4defb8835069bb93ad32c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-27 21:53:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f79a2156545f6b17daf1ea05c427a1679826039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 23:38:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72c611cd96aa44e985e8520c36889d9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9ce36e1186947da8ac2179dfa9bf38c : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd99ca511e9a4b3da274636aea157bb1 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e962e38d4e54a42b25a0c7f84a26078 : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/70b3d898e80c471db278c0a48690bebf : SUCCESS in 12m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36be3b392af748b28e20c259554f6b32 : SUCCESS in 10m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0dcfec97d10f44adbaf5e5d9fd866a2f : SUCCESS in 57m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/eecb7ad80c194380bef761c0310f75dd : FAILURE in 17m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26f0680596aa4363afed71a623d6f75d : SUCCESS in 53m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bd52397ba70e435bad94f4f6e2516484 : SUCCESS in 1h 01m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/916b39af6d1943dcb97ecbd42c64f680 : SUCCESS in 12m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0f21edff50a74dd295fdb839b4902f62 : SUCCESS in 16m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1853d107e3464b358bb107916b02dac9 : SUCCESS in 13m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2d4c20e96d844afd849c9c87c8686869 : SUCCESS in 1h 44m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be02a18140d243dda9c66e9fc056f30c : FAILURE in 30m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0df3df0e66b64c5eb83bd6a6b2862a17 : SUCCESS in 17m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae6455f0521742c09ecfd27d9e7b40e1 : SUCCESS in 7m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ab8d9b9ce0ed476099d51d23e34d2224 : SUCCESS in 32m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5419f7ad0cba4b62bec559ac466034f8 : SUCCESS in 20m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/964e892e3afd4e15bbfb223454708953 : SUCCESS in 15m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/740099c71972492babe4eefc3471d1cb : SUCCESS in 13m 40s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/12968c20b19046e18be447706b694ff8 : SUCCESS in 14m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c18bd7292214f1286588855d84f9c89 : SUCCESS in 9m 07s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/189b8bff72be43b8b9cc32eeef111770 : SUCCESS in 8m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/56b2bb5519e34c3187ddcb142dfc5322 : SUCCESS in 22m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f811c3a90b84c4eb52dddce38c3db14 : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c507a1ceccc47abbbf182fac308e179 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf6f55542eee4eb3ba7bd94dc9a5d6d4 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a1fc079f24d43049d7b7063ab32a01c : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8df82abaa0fe40889542f0f2fd8cf7b7 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ac72d59892c4b42db98723f201c1aa1ddbc68ab3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-28 08:02:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e536c6bf2c9728c1c6d3807879c67f05b012506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 10:06:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e7c5ca4644841aaba44daec065ffa07\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f7c8f00d80540299a4a8b753dd5b6b0 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/231b109ea0b24d87984b5754db21a6b9 : SUCCESS in 5m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d07cf1e58dd344eab54aba10b92536a9 : SUCCESS in 3m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e4390d71b3c84c069c570ff30ca12745 : SUCCESS in 13m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3bc94a9da5304ad5876473d423987527 : SUCCESS in 9m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6bc457510a004b359e441ffd83039d4b : SUCCESS in 34m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e78a0b57407d49aeb047915be679821f : SUCCESS in 46m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd23223b8d5b4f99acc31c5d392b0234 : SUCCESS in 54m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9c9728fd6653446ea49b4ec0ef92773f : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1034dd1faaab41b9b2c353daf2c5d8e7 : SUCCESS in 11m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/20b31617ea624030aa31910b1c99b0e6 : SUCCESS in 13m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcbd5f9ffa8449a1ba0db8452587fb86 : SUCCESS in 10m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d61e5c024aaf4944bad14bc887ce07f5 : SUCCESS in 2h 03m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb005724ec5e4e959623743e77aea74f : SUCCESS in 1h 08m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/50e25b120b6b40dbb4b32412e1f8f37a : SUCCESS in 20m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dc1ec93141f463ba7748927d5f7f0e0 : SUCCESS in 10m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/142409117d8b4465981a883ae98fb96a : SUCCESS in 42m 25s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ab60519fc870422ea96dbc4c236ca4d9 : SUCCESS in 33m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d291643fe64405790a41cbc184ee79f : SUCCESS in 30m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5bf4870b050d4a6d9c0ee64c1031911a : SUCCESS in 15m 54s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e30ec445571e4d15bf1f16487d142bef : SUCCESS in 13m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fba25de751774e429e373fc62204a5a9 : SUCCESS in 8m 41s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/d630d23df6c944a29d80e0c9aeaf072a : SUCCESS in 12m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d3529f98d8f04376abd8d946f83d3ebf : SUCCESS in 16m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/813d885cc9d84529bf02f8cba4b55c84 : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2464cf51c8c94517a36cb07d5ad5db24 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bb0c80e1ac84df6becb7cae19d18813 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6d0ee2b954c4b3ab12d2575a6ed7c40 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9de70165dcbd408f8ed4adfe29330fdf : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"af0db5c1df4d317c3c1788291fbaf5237c9d6a26","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-28 11:55:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9fc10a46ceda5234fe7d2aecd7942fe6c39a90a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 11:55:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccede8da3aeb2f39895dfebc4172f141be56952c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 14:04:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a940ae78ad1147c5a82326b8aa33c259\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c0d2ffd9e74b38a0e3caca9b656557 : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/711d4b464c9943d9bc5d27f949e51a2a : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ad4ff9e0fb545eba37738f859ed4db3 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5cf3bb02561f4a948173bd65046c92e2 : SUCCESS in 9m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/28968c3046ff41e3b15794f98c514bf7 : SUCCESS in 55m 39s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07cf035a8741497b82f4bbb2c8f444c9 : SUCCESS in 55m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e66d2d69d98f4156bba7c8409a5e8cad : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/347c2b9f09ab411ab410535566d9f2df : SUCCESS in 21m 30s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ca7b18c7fe94a5eac1d3d3a4a0176bd : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d1f38866872140649e510b2b419ddca8 : SUCCESS in 1h 57m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bed424dadbb48b3aa9d86101a90daff : SUCCESS in 2h 07m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/009abb15c7da4896b263d3d0e120688b : SUCCESS in 19m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98bd6fec25e64fa2b1d777f16a5d82c3 : SUCCESS in 16m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/973dfbec2591403384c04c02c69065f4 : SUCCESS in 33m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9473945a4bbe40eb964d9d6a82a413e8 : SUCCESS in 15m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cc0ea736734e40ca9966300f02fdfb48 : SUCCESS in 15m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cba4b7331504fd6ac5251c0974c5827 : SUCCESS in 10m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fac484715fa84da1b27ba167df63c6f8 : SUCCESS in 19m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33a6a23ba0c540d883244cdd5aaa7d87 : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0c0de8eb328a2be4907aeea9988c207009ef1337","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 14:04:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c85bc439db42e4cc62d3e880886a92d078e77d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 14:05:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c2c8d2aa3e847a5bb0790cc97f59c36\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa318439079345d1a316333d2946df95 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40779aa4eb90b4f6dcbe2cb06f41326111e3858d","revisions":{"40779aa4eb90b4f6dcbe2cb06f41326111e3858d":{"kind":"REWORK","_number":1,"created":"2024-11-27 21:53:05.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/18/936418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/936418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/936418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/936418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/936418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/936418/1"}}},"commit":{"parents":[{"commit":"715b4b93f87cc5ea733868b5036852513a638dcf","subject":"Add internal VIP address to no_proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/715b4b93f87cc5ea733868b5036852513a638dcf"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-11-25 10:42:00.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-11-27 21:53:05.000000000","tz":0},"subject":"CI: use quay.io/podman/hello due too dockerhub rate limits","message":"CI: use quay.io/podman/hello due too dockerhub rate limits\n\nChange-Id: Id292915fc4c4fefd86a6da431b35daf5faf45523\n(cherry picked from commit eacbe8a08d313d281a818c1e57e08aa8c5a64ebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40779aa4eb90b4f6dcbe2cb06f41326111e3858d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40779aa4eb90b4f6dcbe2cb06f41326111e3858d"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
