)]}'
{"id":"openstack%2Fkayobe~936746","triplet_id":"openstack%2Fkayobe~master~I8b6cd359af2d831af590c830feb2f49fb1065416","project":"openstack/kayobe","branch":"master","attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2026-03-09 14:10:02.000000000","reason":"A robot voted negatively on a label"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-03-18 08:06:33.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-18 08:06:33.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-08 07:40:18.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8b6cd359af2d831af590c830feb2f49fb1065416","subject":"Support Prometheus integration for Docker","status":"NEW","created":"2024-11-29 13:43:55.000000000","updated":"2026-06-08 07:40:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4d4206a494717022dda93d3fc5617967627aaa45","_number":936746,"virtual_id_number":936746,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-06 12:47:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-29 14:45:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-30 17:40:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-02 11:10:25.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-03-18 08:06:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"699e2cec486923a93f25fd039e7f49fc8c99485e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2024-11-29 13:43:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bef9a35d74aae322b0f8088421670271c8525f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:45:52.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/936742, https://review.opendev.org/936717 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"f26d7db5b7669eb300876391f7903f5ddbcbd8d2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-30 17:40:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"861a51c1509e9a1cfe425c76d3f3a53fd8e969fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 19:20:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d096a3dd900b45d1ac6aa8c2bb0ec70a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56f1499b05904bd596a18b8160d09372 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361ad8e333cc4ce1a732773025104941 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9b4626487d14fc8b03d86fa96d5a8e7 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8710876d0cb44b98a6b77e8aa4b9844 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1980d9d1fdaf497a961d34cdd4194131 : SUCCESS in 2m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/453556448b0a4cbc9b31893a08a59927 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cb4dc10e1ed4aaeaf809370d82f42e3 : FAILURE in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e81613aad7449078d24c7c3c35199b5 : FAILURE in 5m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/aca1f7c314f04587a1365f524bdc968b : SUCCESS in 1h 14m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/12a4550bb0cf4464a55570cbffe496be : SUCCESS in 58m 56s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afaa517a8cd945dab49b9db12fd337bf : SUCCESS in 1h 06m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f3a41df492ae4f70a862299dff6342bf : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/21a230d4e173441cbea20683b51335d2 : SUCCESS in 23m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cb735c55a34e4f8bb95fe5f499d10b1d : SUCCESS in 17m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b2af558aba74471825c4f6851a747ee : SUCCESS in 9m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3b8b8547f6d845a8836559ca064309a5 : SUCCESS in 1h 38m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77886a8f8d9744c7b27d9a6cd97b8764 : SUCCESS in 1h 33m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a94e3557a00d46f98130131b87ea1a48 : SUCCESS in 12m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/375d03e78d944658b4aa937ba6e91afb : SUCCESS in 8m 49s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ff61aa5aba9a493e8f275ae9b5c680cc : SUCCESS in 22m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/64e6e792ce6948da895ac27b52875fda : SUCCESS in 21m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaec0ab26ba94b3fafac0bd684935add : SUCCESS in 26m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d28a89a8c0fa4900b3a35b584da9b3db : SUCCESS in 15m 25s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4252e30c68aa4392aecfb6efd9edc891 : SUCCESS in 15m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf0c898a821d400d8b390877ca9879de : SUCCESS in 9m 20s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e825ba34a72b40ecbe6611eee1ed512a : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c9659a9da44b4ad5b4db033208153a71 : SUCCESS in 33m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5f31171b9cb4aa39cf7bb4d13475559 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1dc5935e1319cef1571778c512169e430425cd37","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-12-02 11:10:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9e8391b58af0b1405483fdaba75cc21417986c73","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2024-12-13 17:32:54.000000000","message":"Abandoned\n\nThe metrics endpoint in docker is poorly supported and doesn\u0027t appear to work properly. The metrics don\u0027t all work, so this change isn\u0027t worth the added support cost","accounts_in_message":[],"_revision_number":1},{"id":"02cc755cc86817822a4bb080d34c56a0c09be3b9","tag":"autogenerated:gerrit:restore","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-03-09 14:09:33.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"7b80ffb92e9302ef5994c2ff47f9cf46b736b3ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:10:02.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/ansible-collection-kolla for 936717,2","accounts_in_message":[],"_revision_number":1},{"id":"69352d056ed0bd7605e986dd205bd8e57f0e3a02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-18 08:06:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1383fb1e4e994c54d5d0d99cd5ea059211df0105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 10:12:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/717b75443974487fa0869298ae5c5346\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b480ea5d5ad44c3fb6c206ad6f1d4253 : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/352b9fc5b0c243fc999b17a9e62c05e5 : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c692e2274d14ef4bb96ff13e90b096f : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/893b3e845c49418686fdea58c9b11d2c : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e38c7d17ea948afb98494feb28d7c09 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a39987efcb747acac1b698168f061c4 : SUCCESS in 6m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/078dfad61e4b43c7b95b55b991d831cf : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82bf532416b04f95be780ec9baf5cbfb : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/120b18baf04a4d4b9c515f93d90194a3 : FAILURE in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17bf0fbb01ba4e8e9770b416231e362c : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5e098b31d3cf48018741a9a455712d53 : SUCCESS in 13m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/025444be852040e9967ce78632757848 : SUCCESS in 6m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/35275f58146b411a9d037cf744f62062 : FAILURE in 10m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb5eb5685ddd4746ad1a6fd3504b328b : FAILURE in 8m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8eddd433be88463dbbdf1cc2c62095de : FAILURE in 16m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b20f6db527584555a69b81ce50f0d4c0 : SUCCESS in 1h 16m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/af897ff1a30d4aeeb5de348473709942 : FAILURE in 16m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b553e68112e04170835e42b91790c266 : FAILURE in 10m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/91af8ae3bffc455c81e5cca97f56cb05 : SUCCESS in 1h 19m 08s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/60f0e636e34940b5959d2d6b5bf6fdbd : SUCCESS in 1h 44m 29s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/97cf5db18c804bad874da984d38e31ec : SUCCESS in 1h 55m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15f36e6e16714f1585a416e1e0305a11 : SUCCESS in 1h 53m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e9f3b2272504f93857651a5c89f9d5d : SUCCESS in 2h 03m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/29ef0297c02f4d44bafb7e35e86052bf : FAILURE in 6m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/febf3fbfc5aa43b3afb9f2b84fb52b54 : SUCCESS in 10m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dc5ebde9fd94cd8ba4fcd119f17d9e4 : FAILURE in 11m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3aad6fb0f2d74d198940fe036105a1a5 : SUCCESS in 15m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/38c9f90dfed04b1b90deed17edb29e5c : SUCCESS in 34m 35s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5fb93d796ee545639247deac06f216d9 : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2939cdfd301f43fab05cbc3774e6d7be : SUCCESS in 30m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f856015bc6441838705581f77c12609 : SUCCESS in 32m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dda835013acd49fbab2d3fb4cdae7f88 : SUCCESS in 8m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40e1482a678743dcace499e4a800a4e9 : SUCCESS in 16m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b1d7c278c7b9419b039097d90dbc26a9c89b8d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-05 20:58:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8c51c012faab7b62e5d0a2760d9e52c2c6312a35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-05 22:02:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c74ad3f0ecb7bf8bb0e87ba0e4ef3924338bed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 23:45:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65a6b9afde63401ea0ca93babb9396cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4145204ea4d4cb3bd48ebc72df6f8d1 : SUCCESS in 4m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9a69b004ab574c8e8c1314bc3f346e6a : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c17f6a579bca4882b03bcb4dca0e3dce : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a21db0a75f645828ed120f0d1882e3c : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/363114545f4948149bb3efae8360bc85 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5fec6228e4c344428b7e8cba60b4bd15 : SUCCESS in 3m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf5ff1c0d2ba44c0b78c78df64f7a4d2 : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc44404eea2948188cce690a98620b9a : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bbcc042526294fcfbb47cb0dc048a437 : FAILURE in 6m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7be2cb537a24ef3ade360cf39265f19 : SUCCESS in 11m 41s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f288fe8435174d599dc4ba3cf1fdee50 : SUCCESS in 11m 53s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3b21871f6bf24a889cbc3a57dee35927 : SUCCESS in 12m 51s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb4c1a2517144501bc860a1a4a11d82d : SUCCESS in 14m 49s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e6ffe8f3c974db5835c7b5010dfe8a1 : SUCCESS in 21m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1676a7dacc6c46889ce7fda52b7f5d94 : SUCCESS in 12m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d7a48ba7e2b43e6824e118334555a56 : SUCCESS in 21m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6997bd8ae52742dab8f87eeb111841ec : SUCCESS in 23m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2b4efb80cdb495684d2cdd66ee1433d : SUCCESS in 16m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f024474b94ce4e259bad3d0ca061ce1e : SUCCESS in 59m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0f125017d74b4d4e9373867a1b4dbd90 : SUCCESS in 59m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d0f11181144d4bd7b4b772ab0b112176 : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1175c6a58c7f45a486617f5956f2f87f : SUCCESS in 57m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f23d776ec2ec4be7b023527946164e1f : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cf61ab752a1e48c29a6391ab0f0e343e : SUCCESS in 1h 41m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b11cb9c604dc4796ae53db620291ef4a : SUCCESS in 1h 39m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/92d7cd208d534bf4af0b4dc7c9dbe30c : SUCCESS in 25m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/eb425955a332444d8ad1700ba090fcf5 : SUCCESS in 28m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c65cb83a1476459b87fd38bbb4f349d3 : SUCCESS in 22m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ffce3c5691342b9a0bd5259ecde8aa9 : SUCCESS in 21m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/925576f78856489ea4c160cf62496865 : SUCCESS in 41m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04bf335b22e14a6b935704ecb5e09ca5 : SUCCESS in 36m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5257214a607f42f69db5eb898a19a18d : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5721865295084cee9d8226e4ef4a8f77 : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":4},{"id":"932df86ba3cde96de17117892c4917da0ee985e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-06 11:04:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"95e31c90efcae776a8085601935ea6f7b8bdfa00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 12:47:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebd0fd77acae41a59a6e6cc0c73dd5e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7ec4b566ca4488801295e0d96f3ef2 : SUCCESS in 9m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6977148f9164d7c9886642b693501bf : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/020e9dbf02e147a0b057b1d14c04e2bf : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d14c1c06a384c8e8b7fed418d46b068 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdf4f40afe4f421da1f14d0b161d4187 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4d7c227747b847c588be69f04df935f5 : SUCCESS in 4m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b5b8ebec0294f1e856049e4f70a71be : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0abd03c3f62c4c299397633c188660d1 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81f852b0baad4125a25bd21c197d403c : SUCCESS in 6m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e3e398d62bda43b99c9ce2a9f90a81cf : SUCCESS in 6m 59s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d97e06e9edc64d73ae2705604077ac4c : SUCCESS in 11m 57s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3835f1c621d94f309379b840ff140485 : SUCCESS in 14m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a00691af58a84f119ee969075d45ef58 : SUCCESS in 9m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a42740453923438aae67217161e21403 : SUCCESS in 13m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f1c5789d7fac47ee8812b6430851a62b : SUCCESS in 22m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abc02bee6e94405d93062e5a6bce14ca : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5c3be0eb56c14b20af8766f7155aba81 : SUCCESS in 15m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c80c993b9ac41febb37e29cb127ee76 : SUCCESS in 13m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/df8252f6b7664fe2a7440a6b4342c687 : SUCCESS in 58m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/80a520a3e2e44001b94be2ebf1c11161 : SUCCESS in 58m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e1a88985830543e395ac74dfb58d0bf4 : SUCCESS in 59m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b0123f004e44a4697fedd1837ae9933 : FAILURE in 29m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de7ae7c234504b03b6f3e0cc5533a9d0 : FAILURE in 28m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b3526d2e5974e64ad9f792d1cc5ec7e : SUCCESS in 1h 41m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5362adb651124c4b92cfbcdb1c97c817 : FAILURE in 29m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e31d6364a37d4ae7bad4b17f57178be2 : SUCCESS in 26m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/791afd5f3e8a4d60a0d5692c5496ec73 : SUCCESS in 24m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a6986d5b16e48938ccb964546f9e2c6 : SUCCESS in 13m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31e43f3d53b34c4e991e684cf13b320c : SUCCESS in 22m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4943181b66f84386a021f0383f284a8a : SUCCESS in 36m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70eb7334ed554a2ba0cb28e7b3434e70 : SUCCESS in 34m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3503edd12dd64ee0bca23975516b90e4 : SUCCESS in 19m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8f333b79cb9476bb89eeb899acf8a01 : SUCCESS in 21m 38s","accounts_in_message":[],"_revision_number":5},{"id":"c1b4e2f52579181d49b93109c7c098b5a08ddaba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-08 07:10:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4d4206a494717022dda93d3fc5617967627aaa45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 07:40:18.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/738045d132e54e74ba0f6860dde83598\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/092e9b305aa04df984d4db5e1b71ce50 : SUCCESS in 5m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7ef21c7da8f543eca4cd13677dcc63fe : FAILURE in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d3e62fc464748a4bbcd9af5d89862f6 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f79b6fc5b6647d780a1c9644685278d : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/82848aa94f674f958a81b7a5b2706086 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/394bb71070a94b0b9cd439af9b93d549 : SUCCESS in 4m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdf46358db984e54a6cbf94cb7af3334 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc357d8469ca4c2c91c50359fb6ff301 : FAILURE in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71afebd61e38451fbd77697fb3b02039 : FAILURE in 5m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c794e20c93314567ae51d33032447a45 : SUCCESS in 14m 31s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ce1876b6fad14dc683189b375d49e1f6 : FAILURE in 9m 05s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7170b6fb5de047bca633b89e2f277fe7 : SUCCESS in 14m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb7af144392c4c97908934c1eebfe9d9 : FAILURE in 9m 34s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d72e8b6c3b3b41e1a9f23cac54a57264 : FAILURE in 7m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7acd8d8ae79e4e42888ed22654b0aa08 : FAILURE in 6m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3db7734e22624cd4bc3336a4c9c72d5b : FAILURE in 8m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a9d03ff060b142f78354b26f8c3c4fd4 : FAILURE in 6m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd744783a66240fcb81079eff662184a : FAILURE in 7m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6a42f34e984343b9ad92114094e5616b : FAILURE in 7m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/42710e21a86347f3a428bd32d063144c : FAILURE in 8m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ad909a0f3e32454a9d253ac99bcdf8cc : FAILURE in 7m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30eb38d869c14c9fa6e010d296615ad4 : FAILURE in 8m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5f2a4a6e99eb4132a57b8aa6d51d0929 : FAILURE in 7m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d99d66448055486381e90d83468d5bfb : FAILURE in 6m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93ff4eefcd6e4d75a3f3e3c586f0ae2a : FAILURE in 6m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ba62fbbe08244fe8b0b3a33d96f2154a : FAILURE in 4m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/de86f2a34c034a7e9d7345126824375a : FAILURE in 5m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6a0bb8f5c15404e93cd59f5b769d516 : FAILURE in 7m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/11c1ccc10f6d4d1a99ddfe5ba952e132 : FAILURE in 5m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cc5447ecfdb843bebc334037e2cdc2a5 : FAILURE in 5m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4679a4150794472b9a68fb69830632c : FAILURE in 6m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3559f53d60be4f05ad259f539de03eee : FAILURE in 8m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ebae98829a94019be4436183fcc919d : SUCCESS in 23m 17s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c37cf1ed5af62b071d7ad3e580d18614682f336c","revisions":{"427d0f7aacb63ff913e6de2aefb6ebe73d462afe":{"kind":"REWORK","_number":1,"created":"2024-11-29 13:43:55.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/46/936746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/936746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/936746/1"}}},"commit":{"parents":[{"commit":"3c178846a4e40bfc1c5f6f3bed2f5f70bf226cf0","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c178846a4e40bfc1c5f6f3bed2f5f70bf226cf0"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2024-11-29 13:36:14.000000000","tz":0},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2024-11-29 13:41:51.000000000","tz":0},"subject":"Support prometheus integration for docker","message":"Support prometheus integration for docker\n\nThis change adds configuration options for enabling the builtin\nprometheus metrics endpoint in Docker\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/936742\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/936717\nChange-Id: I8b6cd359af2d831af590c830feb2f49fb1065416\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/427d0f7aacb63ff913e6de2aefb6ebe73d462afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/427d0f7aacb63ff913e6de2aefb6ebe73d462afe"}]},"branch":"refs/heads/master"},"33d745f784f650294e9ec9758709ba91a2e3c467":{"kind":"REWORK","_number":2,"created":"2026-03-18 08:06:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/936746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/936746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/936746/2"}}},"commit":{"parents":[{"commit":"cc33e0e2a38ce4030c82a08f8aa926fe8eb4886c","subject":"Merge \"[firewalld] Fix broken conditional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc33e0e2a38ce4030c82a08f8aa926fe8eb4886c"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2024-11-29 13:36:14.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-18 08:06:23.000000000","tz":60},"subject":"Support Prometheus integration for Docker","message":"Support Prometheus integration for Docker\n\nThis change adds configuration options for enabling the built-in\nPrometheus metrics endpoint in Docker.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/936717\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/936742\nChange-Id: I8b6cd359af2d831af590c830feb2f49fb1065416\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33d745f784f650294e9ec9758709ba91a2e3c467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33d745f784f650294e9ec9758709ba91a2e3c467"}]},"branch":"refs/heads/master"},"fcc8d1b33c9360ab28d2db26c296f2852c41322e":{"kind":"REWORK","_number":3,"created":"2026-06-05 20:58:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/936746/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/936746/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/936746/3"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2024-11-29 13:36:14.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-05 20:57:37.000000000","tz":120},"subject":"Support Prometheus integration for Docker","message":"Support Prometheus integration for Docker\n\nThis change adds configuration options for enabling the built-in\nPrometheus metrics endpoint in Docker.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/936717\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/936742\nChange-Id: I8b6cd359af2d831af590c830feb2f49fb1065416\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcc8d1b33c9360ab28d2db26c296f2852c41322e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcc8d1b33c9360ab28d2db26c296f2852c41322e"}]},"branch":"refs/heads/master"},"4b3e67561d928b77e47070f7b3b2b272b4f54f34":{"kind":"REWORK","_number":4,"created":"2026-06-05 22:02:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/936746/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/936746/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/936746/4"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2024-11-29 13:36:14.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-05 22:00:28.000000000","tz":120},"subject":"Support Prometheus integration for Docker","message":"Support Prometheus integration for Docker\n\nThis change adds configuration options for enabling the built-in\nPrometheus metrics endpoint in Docker.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/936717\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/936742\nChange-Id: I8b6cd359af2d831af590c830feb2f49fb1065416\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b3e67561d928b77e47070f7b3b2b272b4f54f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b3e67561d928b77e47070f7b3b2b272b4f54f34"}]},"branch":"refs/heads/master"},"c37cf1ed5af62b071d7ad3e580d18614682f336c":{"kind":"REWORK","_number":5,"created":"2026-06-06 11:04:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/936746/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/936746/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/936746/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/936746/5"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2024-11-29 13:36:14.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-06 11:04:11.000000000","tz":120},"subject":"Support Prometheus integration for Docker","message":"Support Prometheus integration for Docker\n\nThis change adds configuration options for enabling the built-in\nPrometheus metrics endpoint in Docker.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/936717\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/936742\nChange-Id: I8b6cd359af2d831af590c830feb2f49fb1065416\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c37cf1ed5af62b071d7ad3e580d18614682f336c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c37cf1ed5af62b071d7ad3e580d18614682f336c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
