)]}'
{"id":"openstack%2Fkayobe~937486","triplet_id":"openstack%2Fkayobe~master~Id04e7cfff80c42ce48e5ef2a98cd26c6af0220b9","project":"openstack/kayobe","branch":"master","topic":"fix-tox-ansible","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-11 08:57:45.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-12-11 12:49:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id04e7cfff80c42ce48e5ef2a98cd26c6af0220b9","subject":"CI: Bump size of Swift device","status":"MERGED","created":"2024-12-10 22:03:10.000000000","updated":"2024-12-11 12:52:03.000000000","submitted":"2024-12-11 12:49:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937486-fix-tox-ansible","meta_rev_id":"c9449ef230c692189f59ad682fc9289ae762a1e9","_number":937486,"virtual_id_number":937486,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-11 12:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"all":[{"value":1,"date":"2024-12-11 08:36:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":2,"date":"2024-12-11 08:37:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-11 08:57:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-11 08:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-11 08:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2024-12-11 08:34:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 00:44:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-11 07:36:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-11 08:36:42.000000000","updated_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"reviewer":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"state":"REVIEWER"},{"updated":"2024-12-11 08:37:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-12-11 08:38:20.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8e22ea6fc2e3561ffc840ad87aed6c82824896c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-10 22:03:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b2297d91e1d1a31b1920a9ea4216799bc17f939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-10 22:25:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"998c4d6e238f67ee588f8fa46081cd81164463ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-10 22:39:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c5ec7a2acc5af821758275035cac8583b93fa42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 00:44:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae686229fc464875a46969372241606d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40d4d4013805425097304c400318f9d6 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d114c023243456993695ed87abbd115 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdd674851d0e4384a5230b18ce0b95d8 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0315485bca4129abdf88a23e1713ec : SUCCESS in 7m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25463ed6307342caa90036e0fbd43d74 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33c98de6e4454cfe9bac004601e06c25 : SUCCESS in 6m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d5a251895c149b3913012945d19d021 : FAILURE in 6m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5cc52f36d8aa4048bcd3a2ff78213b3c : SUCCESS in 58m 47s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/50a9d9249ac44191bfe94e07f4a7a43d : SUCCESS in 1h 12m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d2864cac6cbd4c30bbdea708b46ef8bc : SUCCESS in 1h 08m 15s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25091ff8513549ec86347eb750b9240e : SUCCESS in 52m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/00c2d5704d534d9f997ee7612402f488 : FAILURE in 39m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f058541390694b47af324e647a52b47a : SUCCESS in 1h 11m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f6d3f6bfda3942929e6fcba3517c060b : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e7e9439b4ad34b1cbf68f96b2053ddfd : SUCCESS in 18m 05s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07d96e4df3e34bbd82a68e680a1a4533 : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e90092c2476404bbe9d027cd5c9de44 : SUCCESS in 2h 00m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07bd970895cf480aae39ad64519ed73a : SUCCESS in 57m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0de504d406774d228b68537e7c8a567c : SUCCESS in 21m 46s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/66592056e6774fe1b097a55d4042592b : SUCCESS in 19m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b974822ec3924957b1678fc405c08dac : SUCCESS in 13m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7b6d98066066424e89c8dba41683c1e0 : SUCCESS in 13m 20s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3c10600743db4124953cfe49a706aa5d : SUCCESS in 39m 50s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/3fa9742976604e6daca506477054353b : SUCCESS in 1h 02m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/93f139afb7a048a3a3f14f87fcb9ea3a : SUCCESS in 30m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f614940757ad484e930ee484d0991122 : SUCCESS in 25m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/437554300bcc4c9f975d97d1d1cdf0e6 : SUCCESS in 15m 42s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3b671ba666d74205bd77e9572145cae5 : SUCCESS in 10m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ae1664e657b4d9497423e4dae62b3c1 : SUCCESS in 10m 55s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/78a02b8ae7654ce899f2a853763ba86c : SUCCESS in 6m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a9cebf5c4a4246faa699653cbaa6e378 : SUCCESS in 21m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca13ebfd2c314ebea648be0676fcba75 : SUCCESS in 19m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f1e573785d96c973f297356ce2c5d7f930ca5fc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-11 08:33:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f971c021cbb66dc0d2ac024a1e31286b82ac2c01","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-11 08:34:12.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"feb89ac44c008ea254c427468842dae9854721e6","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2024-12-11 08:36:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7a091049fcce0fe0ac1815b6ea8d9be8ccd6b6bf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-11 08:37:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d80822c29e92f1fdf0429d0fb6a298b4b3a285d3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-12-11 08:38:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1960207fe9b1ae7b06bd588e9c7c4d157976903c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-11 08:57:45.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6f786f7366157438a58a05e7c113d81ce8aa33bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 10:31:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da4a67b533143ac86549b3f85c74e9a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fac9c5da7706411c9da2f7cb9cc4230e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83e2f025e9847b7b6ccf46781fb19e0 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/074aed4076744d50ab9a2d7a3fab6dfc : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a04e2459ab489884524c814846dbc5 : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fff2ffc593e9445693cb96956724b12b : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1aa0e44aa4f742bb8f8b9e8645a16ae5 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/158a8c585aab4114bab73c23d85a6e5f : FAILURE in 7m 53s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a46942ac3ef34689a626035f0173fc9a : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f002f043ec6f495eb4ce763e055e5076 : SUCCESS in 58m 25s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b09b92c008f84228ac859e019727989f : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0e013a3ac4cd48278c4e380312149aa4 : SUCCESS in 1h 10m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c2b19e6516eb42c280e65be0cd2843f8 : SUCCESS in 20m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5be09fdc2a0d434cad038c2e7c4d9ba5 : SUCCESS in 18m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96d6ac730eda4e21a857969d66794915 : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0030ab16ae4a40249daed1eae7f92df2 : SUCCESS in 1h 49m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/847fbb109a634f99b88e35c41f4611a6 : SUCCESS in 1h 38m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/821f820391de4dd1b16f2dc1756db6d4 : SUCCESS in 22m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6517f6fe6714a3d831cb678737890ec : SUCCESS in 16m 19s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3f0074d72da84100b04c6ae1b310c440 : SUCCESS in 45m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d8b2e9690b0842a8b708b73bd31984e2 : SUCCESS in 36m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20c1da72f96d4503b9c024356e7c9a31 : SUCCESS in 28m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f2e722359e8a4c108f466ec980e5dfa5 : SUCCESS in 19m 18s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/18a8e98795c74ab5848bdf26c718a606 : SUCCESS in 15m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/223a68ea82fe48859569065bf39fd1d7 : SUCCESS in 14m 26s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0c53dd28a7a64eee8c5fa56572f6bed8 : SUCCESS in 13m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e13b1cdacace4619a3455fd5017ea4cc : SUCCESS in 26m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d3b7e855229488386f479b781aa9f84 : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":4},{"id":"0ff30b19edc34ad6185401f7a7c504e6500b1717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 10:32:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6c6b4ea7e199c6c4fffa7ba65a52461db45aff32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 12:49:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e23392c90d6b4a16aa74c60054a01d99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75ff6e47c2214007a377313b544c6095 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9869cae4361a4f2c81c91ce9341dc2a1 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2c966279fe480a949a34357ae5c456 : SUCCESS in 7m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/047dcc89781a4750b923167b76cef0e2 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cbf2385307a845448d57f3f95b0e85cf : SUCCESS in 14m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d6a719386b0341009333c170fa6f2b1c : SUCCESS in 59m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/160aa562719a4aaf9cf00019af0ec82f : SUCCESS in 1h 11m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/12266eb5423f49f2bf3bda9ef65b8dad : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ee9a59352e9c4219b5a88cda415fb313 : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2651f96184294a6ebc9c6cb0c436e1c2 : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/654ef82b16da49358041e8de103535b5 : SUCCESS in 2h 09m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e213d15d3bbf4c249a30d8dcab524f9f : SUCCESS in 1h 43m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8b506ccc2c3545c294eaafbf1b529c54 : SUCCESS in 11m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c998970f5f04957979ee7bc7e660fec : SUCCESS in 16m 15s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1ccc45b169804e4c994b5f6463799f32 : SUCCESS in 35m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a3e7db5e6fc48c78f4fa6b0c3f744bf : SUCCESS in 30m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c1cc135cf904a14b8381de342a1f53e : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/860da5bf139b493590881bffef523205 : SUCCESS in 11m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/280028a245fa4cab86d8e1c4a47163d1 : SUCCESS in 18m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5634677eca284051b66091e0b0b4e56d : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":4},{"id":"02d2222023967358a81b1119f02fe8ba1eda84d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 12:49:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c9449ef230c692189f59ad682fc9289ae762a1e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 12:52:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cba9f57d841a42fca0f574c57a6a3031\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c4c0961c3c4ccba984a715448e4322 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","revisions":{"35ffd0a3ab119ace55514ce33adff78ded95df31":{"kind":"REWORK","_number":1,"created":"2024-12-10 22:03:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/937486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/937486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/937486/1"}}},"commit":{"parents":[{"commit":"c1c7656b52485e2ce2e702f1b7286774c8f20cab","subject":"Add podman support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c7656b52485e2ce2e702f1b7286774c8f20cab"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-10 22:02:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-10 22:02:21.000000000","tz":60},"subject":"CI: Bump size of Swift device","message":"CI: Bump size of Swift device\n\nThe kayobe-tox-ansible is failing in the swift-block-devices role while\nformatting the Swift block devices as XFS. The error comes from a change\nin xfsprogs [1]:\n\n    Filesystem must be larger than 300MB.\n\nBump size of backing file to work around this issue.\n\n[1] https://git.kernel.org/pub/scm/fs/xfs/xfsprogs-dev.git/commit/?id\u003d6e0ed3d19c54603f0f7d628ea04b550151d8a262\n\nChange-Id: Id04e7cfff80c42ce48e5ef2a98cd26c6af0220b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ffd0a3ab119ace55514ce33adff78ded95df31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ffd0a3ab119ace55514ce33adff78ded95df31"}]},"branch":"refs/heads/master"},"b4eebe7527937ee0969323f29398b897ae93aec9":{"kind":"REWORK","_number":2,"created":"2024-12-10 22:25:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/937486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/937486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/937486/2"}}},"commit":{"parents":[{"commit":"c1c7656b52485e2ce2e702f1b7286774c8f20cab","subject":"Add podman support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c7656b52485e2ce2e702f1b7286774c8f20cab"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-10 22:25:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-10 22:25:36.000000000","tz":60},"subject":"CI: Bump size of Swift device","message":"CI: Bump size of Swift device\n\nThe kayobe-tox-ansible is failing in the swift-block-devices role while\nformatting the Swift block devices as XFS. The error comes from a change\nin xfsprogs [1]:\n\n    Filesystem must be larger than 300MB.\n\nBump size of backing file to work around this issue.\n\n[1] https://git.kernel.org/pub/scm/fs/xfs/xfsprogs-dev.git/commit/?id\u003d6e0ed3d19c54603f0f7d628ea04b550151d8a262\n\nChange-Id: Id04e7cfff80c42ce48e5ef2a98cd26c6af0220b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4eebe7527937ee0969323f29398b897ae93aec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4eebe7527937ee0969323f29398b897ae93aec9"}]},"branch":"refs/heads/master"},"8a5ff20c334035b16afe78b858cd6d9e426a51ee":{"kind":"REWORK","_number":3,"created":"2024-12-10 22:39:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/937486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/937486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/937486/3"}}},"commit":{"parents":[{"commit":"c1c7656b52485e2ce2e702f1b7286774c8f20cab","subject":"Add podman support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c7656b52485e2ce2e702f1b7286774c8f20cab"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-10 22:35:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-10 22:35:45.000000000","tz":60},"subject":"CI: Bump size of Swift device","message":"CI: Bump size of Swift device\n\nThe kayobe-tox-ansible is failing in the swift-block-devices role while\nformatting the Swift block devices as XFS. The error comes from a change\nin xfsprogs [1]:\n\n    Filesystem must be larger than 300MB.\n\nBump size of backing file to work around this issue.\n\nAlso switch from pytest-molecule, which is unmaintained, to\npytest-ansible.\n\n[1] https://git.kernel.org/pub/scm/fs/xfs/xfsprogs-dev.git/commit/?id\u003d6e0ed3d19c54603f0f7d628ea04b550151d8a262\n\nChange-Id: Id04e7cfff80c42ce48e5ef2a98cd26c6af0220b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a5ff20c334035b16afe78b858cd6d9e426a51ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a5ff20c334035b16afe78b858cd6d9e426a51ee"}]},"branch":"refs/heads/master"},"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5":{"kind":"REWORK","_number":4,"created":"2024-12-11 08:33:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/937486/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/937486/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/937486/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/937486/4"}}},"commit":{"parents":[{"commit":"3c178846a4e40bfc1c5f6f3bed2f5f70bf226cf0","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c178846a4e40bfc1c5f6f3bed2f5f70bf226cf0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-11 08:32:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-11 08:32:54.000000000","tz":60},"subject":"CI: Bump size of Swift device","message":"CI: Bump size of Swift device\n\nThe kayobe-tox-ansible job is failing in the swift-block-devices role\nwhile formatting the Swift block devices as XFS. The error comes from a\nchange in xfsprogs [1]:\n\n    Filesystem must be larger than 300MB.\n\nBump size of backing file to fix the job.\n\nAlso make the kayobe-tox-molecule job non-voting as it is currently\nbroken.\n\n[1] https://git.kernel.org/pub/scm/fs/xfs/xfsprogs-dev.git/commit/?id\u003d6e0ed3d19c54603f0f7d628ea04b550151d8a262\n\nChange-Id: Id04e7cfff80c42ce48e5ef2a98cd26c6af0220b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
