)]}'
{"id":"openstack%2Fkayobe~937652","triplet_id":"openstack%2Fkayobe~master~Id3ebd6836ba8f3b2101b94637a89f97fb0eb59e1","project":"openstack/kayobe","branch":"master","topic":"overcloud-extras-service-absent","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-01-16 15:17:42.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-01-10 08:45:07.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-01-16 15:17:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3ebd6836ba8f3b2101b94637a89f97fb0eb59e1","subject":"Use kayobe_container only when service is enabled","status":"MERGED","created":"2024-12-12 21:06:28.000000000","updated":"2025-01-16 15:18:51.000000000","submitted":"2025-01-16 15:17:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937652-overcloud-extras-service-absent","meta_rev_id":"114f9caf158f8ab50c25a27ac1811ffe9d29cfae","_number":937652,"virtual_id_number":937652,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 15:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-01-16 11:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-01-16 11:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-16 11:13:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-01-16 11:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 22:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-13 08:27:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2024-12-13 08:52:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-06 14:17:21.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-01-16 11:03:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-01-16 11:03:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-01-16 11:03:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-01-16 11:03:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-01-16 11:03:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-01-16 11:03:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-01-16 11:03:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"644c241af6d271f79e6ebef85686fc668133398b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-12 21:06:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c7f080839b52d179352ace414437ec30f88e175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 22:41:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb00fb490a324092a29751b7b55e0531\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3211b92f8434a7cb18740427a126ba0 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2a8aa804ca4bf6af81804637adb91a : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f839fee81db4ae394667ad279dac9cb : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bef582a3a94f1cb9ce0a9945eb3fbf : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38933f09262f4a8b97554169efaffab0 : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aee48c694e474e83a4bf79b911748208 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8cb5d0f17da84cebb544d7992700a7db : SUCCESS in 10m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6995932f5444b068dd9ea7fceb2640f : FAILURE in 5m 07s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/42a484c5fee44911b1afd6057ea05b1f : SUCCESS in 57m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3fc64694733b47f881f0ca46647569d2 : SUCCESS in 56m 56s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e72df2624395471092829b9efd0df877 : SUCCESS in 57m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2c338b3bc0644056bc0624ef8482d446 : SUCCESS in 58m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/32017fb59f0e4972b448cf34b567dd87 : SUCCESS in 22m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/69ef92204a674e88a34511dd2ffcff79 : SUCCESS in 10m 05s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba8b69316dea4e79bb07da673a656e54 : SUCCESS in 11m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3af32dec4c6b4a80ad5e74bd0b71085a : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82038e16ac674a33a8ff1999cab8bbf0 : SUCCESS in 1h 34m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/acffa55ec1b3452589c82aeb6d3ddbc6 : SUCCESS in 12m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9646e74159384ef28ec9c052c7b10c23 : SUCCESS in 13m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5d96244ad6a349a0bd62299ecc77bb08 : SUCCESS in 36m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae6256d44d414cf4b6969846c7095965 : SUCCESS in 34m 12s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0ae72193b724aec91060b45d3671c24 : SUCCESS in 27m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c6a9592c20b455b8d7bbec6c9d92a26 : SUCCESS in 17m 09s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/535d6328c6d94cee96d918594bd7fac9 : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83ad6e0ef3e2442bbdef447803605c84 : SUCCESS in 14m 33s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/3ae61aeb3aab4dea9232fe24120eb866 : SUCCESS in 14m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/46c97fc10a4f4556ae8dd69ce91eef41 : SUCCESS in 24m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b024e1da61004468bc2969c54f166fd8 : SUCCESS in 16m 48s","accounts_in_message":[],"_revision_number":1},{"id":"734c1518f9334d81f6236179c6302b030d21bf44","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-13 08:27:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52e7d1edc82991a6f1999438ff26047a5f210e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-13 08:50:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"19e8e8ae07abab4d52ef07a998540b6d52ab50d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-13 08:50:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c8bd2130e5bdfdf19532bbd67052d76eb083dd8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-13 08:52:22.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c84b96aa8262ccbbb1e2c75818b4e6e23cc5a65f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 10:47:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce1b1f2e3ab4222850929077bc8773c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c96c5462494f44559c47d962bd6bdf54 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fe995d47b974fd2bc52f5dafeeb2244 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c40a5bb91929484099eba8ae762b65ef : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd12bc0d8e44db7be08524817f5fb8d : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd5229e331b14b00bb2a3ce15f55b464 : SUCCESS in 2m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c799fcccff947f580234f3df7dbd40c : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23bb287c367b4dba970cc07d88ee586a : SUCCESS in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/934966d180f744cb814145ae20d3fa0b : FAILURE in 5m 43s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e3fc2e01f3794be69388039229211222 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2919a6f8ca07437bb3c76a9b2f1b4f54 : SUCCESS in 1h 02m 23s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37a5f406293d4a92a141a0b5fc349cb3 : SUCCESS in 32m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/887569a26a3748f4b956319305a01058 : SUCCESS in 40m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/01ec6adcebc349b882980a8853e4edec : SUCCESS in 14m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e4bb370801f54eab832acc8c7128b0f8 : SUCCESS in 11m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5217e516c1a4428a8a3b1db87edcf27a : SUCCESS in 9m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fa96c483aeb54f4c9ad35dbb3f28276f : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1718b0de73e4089a4b4a80065141f67 : SUCCESS in 1h 48m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fa2bf2c3b2e54ffb9abc0397c9124634 : SUCCESS in 16m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd1420569c344cc193d11e05593705ed : SUCCESS in 13m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8428cb32c68e4cdab183f460b755547e : SUCCESS in 41m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9b57cb6fd18d49d4b3a4c786bb6ae96b : SUCCESS in 36m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b7a56815aa5441ea7ae0ea71484e51b : SUCCESS in 26m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e13296394d294f7e81a96db65e989381 : SUCCESS in 15m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/59b313d7c1814165abe21c178797ffc1 : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5146f850c324e5f8c620bfebc1ecf1f : SUCCESS in 12m 18s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a5fca6afa3914576985027e677edbd8c : SUCCESS in 10m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/31cd8107f06e41bbb6f251dede5a797a : SUCCESS in 23m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6bdeb45de994aa3a6b552387deb849f : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0d3240269f005e7b072207e78e32cded022794ee","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-06 14:17:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"852bd6b1feb17df26a5cc2d3363dc0e7ba16883f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-06 14:33:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e2a65f52e23e80f53bf73bd0f0d5e957609d936b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-06 14:34:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"30febfec19304f12c04897c859f23ab8499d7ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 16:45:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91974fdb4f864ca5bd3cabcee01a3dc3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21980fb8caf34469a2cd2d6ef20b948d : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70e4731701904e448e1a4744cc81e526 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5021fd0cadc84dcdb409337649d62429 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2ba0f34cab475089db5c21d0fb7f07 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc9a6af77b2d4678a73f5de09f7c4ae9 : SUCCESS in 2m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73a4f866c0e241b6b6f0b345a86ccaf4 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e873de2d5d344a6787bbb413916da937 : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01ef3b77d2964d8aaef6391ba9451f43 : SUCCESS in 9m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/928e04b527dc46aa9fa44875e6498b68 : SUCCESS in 37m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0f7775de8cfd43b092c49b097a520a35 : SUCCESS in 57m 48s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e310c427a4984867a87e44a8da96de67 : SUCCESS in 56m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/02c92ef4c23a4031871a1f4b16533a0a : SUCCESS in 1h 15m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/29059c1797d9413890d95c2a17f01cf0 : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c0e673e214514938ae091a78467bc028 : SUCCESS in 14m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aaff6c962f141a9b078c6f7ec2aced0 : SUCCESS in 8m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/794a474211c5439b9e2a8934d434f169 : SUCCESS in 2h 05m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b43eaabf24d64ede9369ea2a7b3da9af : SUCCESS in 1h 30m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7491a38e209b4b83a840dc597a8ce332 : SUCCESS in 19m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64f27b8507624d4eae97c7150accab18 : SUCCESS in 12m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fa1dedbce110474fa84ebf7975a8cf14 : SUCCESS in 41m 23s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5f12b496fadf4481b77a3642aae23bef : SUCCESS in 39m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59cfd4e7eeff4f79829584cb047c2f98 : SUCCESS in 25m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/60a5c614bb2e413194aea95b035a6e1c : SUCCESS in 16m 04s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/5b6a14c813fc4556ad07200b2a4ec1dd : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cc9106a5fe74377bf465749beb40717 : SUCCESS in 11m 31s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/2ae921ec33ab4161aee854d5d7018d1f : SUCCESS in 9m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f2e9395dec5c4c5080ea6f810b9e102f : SUCCESS in 22m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00e2898a9b794ea0a2cf8101f77c22c6 : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":3},{"id":"36edc4ecdae335702576adc4b6e967557c7a2b2a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-10 08:45:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"858d1e68b886f19e482f243930d8026bef37aeea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-16 11:03:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"244b258e159159d1a2e154bcf2cbb51a202583e6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-16 11:13:25.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"325ee0c45221d4de824c7cba4d5ecb3e498be6f7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-16 11:18:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b23a21828b619c51fa125fa59fef4e48ce75cb99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:06:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d30a2e88198140aaaa1505aebf936173\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba495c56baab4554916e50b373e3d913 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de06a9661d44c1f9c3fc297c2f20886 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb22cf29e3da4e14bbe98533869acd3a : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57178273e9b14dd0a41f4988847be9b5 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5901e9c69d9a4f9d9b0d7ceddbca5df6 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/998debbe27df4fd2bfac45d395813f3a : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d5352312bd14babbd097dd4756a2910 : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/72f6c7d08844483390f0d895267be400 : SUCCESS in 10m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/62bb296a9f5643a7b322ebe6009607a7 : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/dc15d517bede4315ad0f7003067c9c8a : SUCCESS in 58m 46s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5259345af7284262bdee21900b39b93e : SUCCESS in 32m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/52e20214264d4cf0959da3baf016d4db : SUCCESS in 1h 16m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e6efe15ead25461092889970b0ccddf6 : SUCCESS in 28m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/44c7f3267c3845c4a05e63f438e25056 : SUCCESS in 19m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8131fcb2d6a845adb7374bbf19970c03 : SUCCESS in 9m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/56718a829a6d4815841c68fe730356e4 : SUCCESS in 1h 55m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e12667740e74648b4ad92a3107bc698 : SUCCESS in 1h 44m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6c072688cea240909476f82de3d3ab30 : SUCCESS in 24m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bacff7026f443e299127fa47d0c1e4d : SUCCESS in 21m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e44994649e0141af90792359ccc59e69 : SUCCESS in 41m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4cdb96375f7c460ea1a3d23780b31976 : SUCCESS in 39m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ae04817caf14038936bff99566bc04f : SUCCESS in 27m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4995b482b82c4cfe95d9c36880f95c67 : SUCCESS in 15m 48s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/5f7e384ba0b048eea3e48edb66663526 : SUCCESS in 15m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e6172fe659a49768e0577a8dad15b91 : SUCCESS in 11m 19s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/bbe4a45e6aa943feaf6961c791468431 : SUCCESS in 7m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2322be8238154d2fb7e0fe2de7d54e04 : SUCCESS in 16m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b27fc095a4c84395a39eb2ee8831cec0 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":4},{"id":"35fcfbdfb901e51a3a3c0982c7e35b7b79bcf5a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:07:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a6b12bfc5557e31d464b902ca63843bbdef5d05b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:17:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4777a14c4d4842d2bca618383c6fd2bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/882c6b1a51c54bbfabfd3c6f57877322 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5ab06215cb143af832312d966b89796 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf4cab6170e43abb1034fb79debdb4e : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5621aeca95bf42d09172c65bd6f3b0f1 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e6a3acde9f83479683718e58f45db170 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3b6b685fc3e4f17932d6c6377e73f35 : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb9589a70ab2493f91a815aa3e373427 : SUCCESS in 8m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0f38f0c826ec4f46b88a49add5a875f4 : SUCCESS in 1h 01m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e72cb70d4b6c4f0da0bacf686864c196 : SUCCESS in 50m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/08d16e371c5b41c6a2d1f1ff80e6135e : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/74a457d87e024ce5887572f43a89eacb : SUCCESS in 20m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c747af7210134723b0f77f9e7d1e14f9 : SUCCESS in 10m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a4c3c738f4704aee8d7f2e9b20975829 : SUCCESS in 2h 04m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/deec2db24f5b451280310aa0378bf17b : SUCCESS in 1h 39m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/35eb23909b9f40baaef20f7a255c6aef : SUCCESS in 17m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/774655d12a644647b7c5c11bc6af8ab4 : SUCCESS in 14m 27s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5288511c0ca9437490470c7e7ba77de4 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eddc9b6b4f924c14b34652bcee06a523 : SUCCESS in 28m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e525304748804f02b7698f34b715c546 : SUCCESS in 16m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eda525fc22ea428c8ba3f177ef1c1ec7 : SUCCESS in 8m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f0a9dd037a1f45aba70b077cacbfe062 : SUCCESS in 15m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bd02bc6ea4f497e9c3611209a461838 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":4},{"id":"88412bfaff94b640312b316682825d08f89df59b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:17:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"114f9caf158f8ab50c25a27ac1811ffe9d29cfae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:18:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ec4a40fd674490fb27ed2d6c1d54d7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da4a0fc41ed64a73b516476370a7b459 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/076f71605b574180979532ab14b75bc9 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a6ba61f46282b12ff04d1dd5190595f1216c4693","revisions":{"5198026eedd990cf3ad614736f591c20466dd80e":{"kind":"REWORK","_number":1,"created":"2024-12-12 21:06:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/937652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/937652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/937652/1"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 21:06:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 21:06:18.000000000","tz":60},"subject":"Use kayobe_container only when service is enabled","message":"Use kayobe_container only when service is enabled\n\nThe overcloud-extras playbook called during overcloud service deploy\nwould always invoke the kayobe_container module, even when the service\n(inspection_store or opensm) was disabled.\n\nWith podman, this has the side effect of attempting to pull the image\nfrom the registry, even when `state` is `absent`.\n\nChange behaviour to invoke kayobe_container only when the service is\nenabled. This means that a manual container removal is necessary now to\ndisable these services.\n\nChange-Id: Id3ebd6836ba8f3b2101b94637a89f97fb0eb59e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5198026eedd990cf3ad614736f591c20466dd80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5198026eedd990cf3ad614736f591c20466dd80e"}]},"branch":"refs/heads/master"},"bc7dbe1a6d53f9e8508dd25ddad37fd6f2cd08fb":{"kind":"REWORK","_number":2,"created":"2024-12-13 08:50:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/937652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/937652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/937652/2"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-13 08:50:20.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-13 08:50:20.000000000","tz":60},"subject":"Use kayobe_container only when service is enabled","message":"Use kayobe_container only when service is enabled\n\nThe overcloud-extras playbook called during overcloud service deploy\nwould always invoke the kayobe_container module, even when the service\n(inspection_store or opensm) was disabled.\n\nWith podman, this has the side effect of attempting to pull the image\nfrom the registry, even when `state` is `absent`.\n\nChange behaviour to invoke kayobe_container only when the service is\nenabled. This means that a manual container removal is necessary now to\ndisable these services.\n\nChange-Id: Id3ebd6836ba8f3b2101b94637a89f97fb0eb59e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc7dbe1a6d53f9e8508dd25ddad37fd6f2cd08fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc7dbe1a6d53f9e8508dd25ddad37fd6f2cd08fb"}]},"branch":"refs/heads/master"},"185474c8a1a727cf634414d6c5b44a21956bbae9":{"kind":"REWORK","_number":3,"created":"2025-01-06 14:33:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/937652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/937652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/937652/3"}}},"commit":{"parents":[{"commit":"a6e7ae265befa05777632a14eef1b9ccf92d1c95","subject":"Merge \"Add missing space character to help message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6e7ae265befa05777632a14eef1b9ccf92d1c95"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-06 14:33:22.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-06 14:33:22.000000000","tz":60},"subject":"Use kayobe_container only when service is enabled","message":"Use kayobe_container only when service is enabled\n\nThe overcloud-extras playbook called during overcloud service deploy\nwould always invoke the kayobe_container module, even when the service\n(inspection_store or opensm) was disabled.\n\nWith podman, this has the side effect of attempting to pull the image\nfrom the registry, even when `state` is `absent`.\n\nChange behaviour to invoke kayobe_container only when the service is\nenabled. This means that a manual container removal is necessary now to\ndisable these services.\n\nChange-Id: Id3ebd6836ba8f3b2101b94637a89f97fb0eb59e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/185474c8a1a727cf634414d6c5b44a21956bbae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/185474c8a1a727cf634414d6c5b44a21956bbae9"}]},"branch":"refs/heads/master"},"a6ba61f46282b12ff04d1dd5190595f1216c4693":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-01-16 11:03:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/937652/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/937652/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/937652/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/937652/4"}}},"commit":{"parents":[{"commit":"70af189ae2c115b6d1d98cdfb54a1ebeceba32d4","subject":"CI: pin to ansible-compat\u003c25.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70af189ae2c115b6d1d98cdfb54a1ebeceba32d4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-06 14:33:22.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-16 11:03:17.000000000","tz":0},"subject":"Use kayobe_container only when service is enabled","message":"Use kayobe_container only when service is enabled\n\nThe overcloud-extras playbook called during overcloud service deploy\nwould always invoke the kayobe_container module, even when the service\n(inspection_store or opensm) was disabled.\n\nWith podman, this has the side effect of attempting to pull the image\nfrom the registry, even when `state` is `absent`.\n\nChange behaviour to invoke kayobe_container only when the service is\nenabled. This means that a manual container removal is necessary now to\ndisable these services.\n\nChange-Id: Id3ebd6836ba8f3b2101b94637a89f97fb0eb59e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6ba61f46282b12ff04d1dd5190595f1216c4693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6ba61f46282b12ff04d1dd5190595f1216c4693"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
