)]}'
{"id":"openstack%2Fkayobe~938038","triplet_id":"openstack%2Fkayobe~master~I956ed42a4a862a14d1ac9e40c40e0cdf62e36464","project":"openstack/kayobe","branch":"master","topic":"fix-molecule","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-12-20 11:22:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I956ed42a4a862a14d1ac9e40c40e0cdf62e36464","subject":"CI: Switch to pytest-ansible","status":"MERGED","created":"2024-12-19 08:31:34.000000000","updated":"2024-12-20 11:25:29.000000000","submitted":"2024-12-20 11:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938038-fix-molecule","meta_rev_id":"556ea9b7a4a883c717c2d7a8fbc128e11404a25e","_number":938038,"virtual_id_number":938038,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-20 11:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-12-20 09:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-12-20 09:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-12-20 09:17:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-20 08:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-19 08:50:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-20 09:17:54.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-12-20 09:25:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"d332bcde2cccfbe9308245a30e8adcf3cceffd83","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-19 08:31:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1a5de47ee302a8d23e850bafc3eb220052f8b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 08:50:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f8156f96f884a96a73602adc65ff122\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e7afc9237db49a0bbc774d70ab5d1fd : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c162d8635f64380a8f1b2ad2e84f68d : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d64cd25ca9854b40a1d6d8b01f9f0942 : FAILURE in 2m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7bdea07d71f54d708da1eaff4d8a832e : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6379194482154d4e9ee882b47518776b : SUCCESS in 5m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9493a867dba44d7bbca1f57e9edb2a9f : FAILURE in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c1efcf34fcfbc9e788292f9a2817817ed3200df8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-19 11:28:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a1aeb2247d6aec5cb8afed0e4df1cccb3fd38065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:44:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/43a6e33f03434bf5a7dc2c8ef40a5752\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0435ddd01584415bb063e3fd547b6af : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25047381c4234d678e7a3a76c281ba65 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf1961f3aec3427b812084a41ace1d8b : FAILURE in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89ed0b9cfb4e42ba9e03581795a0a6f9 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f7d459d5a940ed97fd5b1d54e0cd24 : SUCCESS in 7m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04ecb3c35f2543c9ba9effc124cfec5f : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":2},{"id":"18963e1704a92e2385dee0b5183f76baa4b49562","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-19 12:01:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e05199de7d90f818fb3324aebfc27414027140c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 13:37:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d85884fbe564469ea360f43f6916b890\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c19763f22044f18b34e3cb082c67077 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d5c703b70848ff8f524db21ce0d6e2 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc81acdc778946e39bc486386cb92bf4 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe7cd44fde64f2091a1de39637ac373 : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/973fca8f91104a9088c10825c0681cf0 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/266ab71450424f09b5e6817270296164 : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94f7b66169574f71a52e2f2efaaababf : SUCCESS in 10m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/789cbcbc9970431d9f63f0a7f0d4d08c : FAILURE in 26m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/eda303f2deea431e8d89f46cc85de81f : FAILURE in 19m 32s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be53d2db582b400ba99aca4a4d60bad4 : FAILURE in 18m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2bad0a16f435458994eacb1e2d6ca0e3 : FAILURE in 26m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8386c6ad9f6b4bc4aa59efc5f8f654ca : SUCCESS in 19m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/04706d4c2d9d4b36b64e65851be9a1ab : SUCCESS in 10m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/568e8bf49ca64beca5ac11f015b1b2b5 : SUCCESS in 15m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5f3e2ab848114d7bbc0883ed40af29f7 : FAILURE in 1h 25m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c68031781e24788bd841aa92abf85d2 : FAILURE in 1h 07m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c4c12044080e45cbb5374169db2949fa : SUCCESS in 20m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c726a41e7f64cffa37d4b6d9b1b0ff5 : SUCCESS in 14m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c94ad96c45df4ae0a403d33410b1e1c3 : SUCCESS in 43m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2b2c51d8150043e5ba167c1921d588a8 : SUCCESS in 33m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/515110d1db534e2095db36a311561469 : SUCCESS in 29m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/167454dda58f4c139d2acdc4cc2798ed : SUCCESS in 17m 23s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/bdcf3a9b023b41e992e9ab74ad46d38a : SUCCESS in 15m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9295d03e93b4019b99a4c20fa50f915 : SUCCESS in 10m 06s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/54489b8aaed44d5280cbc1b24ff74f0e : SUCCESS in 11m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d2c99fe55e7146928f0780534ae91e93 : SUCCESS in 22m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/245b3b94b67e4b17a1a743d6c5422cff : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":3},{"id":"2535c71a48c07edc454cc60c828fbe28bdee0c4c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-19 16:39:13.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0b5a42da8ece44a5bef1033ab98b98036eb3fb93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 18:34:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72f321fca6ce4d42bfb1e4a165afc482\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/046c599325e44942b71078da402895fa : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd734733db544c3844d7c1c63c650de : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2964cbcd071542f2a3bffed423b6d35c : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d693bd592164eadad19942529dda70b : SUCCESS in 7m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5cf6e0327bdb497d98de886eb36cfddc : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f48b180db0a49b1a2a7ed2653969535 : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/46afacfbf82f49f78646488f2344c5e3 : SUCCESS in 7m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3a7941f7055849e581b06030eaec62b0 : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9d988588b5404334be093c3da8474685 : SUCCESS in 1h 03m 22s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f24c002987ac449c882d78943c9bc134 : SUCCESS in 54m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a9f9e8eccd78493fb22bd40918f60215 : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/66a5f5e1ce8b488b8c250b13572ce73f : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d4eb90b092a84e0ba6db509ebd07e339 : SUCCESS in 10m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0faf7b33f52c46a1bf99d395327b8a8e : SUCCESS in 14m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/02eaf779dff543bfaf5e2df246c98212 : FAILURE in 52m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5f2275477f04cae8947aa1a0e6b551e : SUCCESS in 1h 54m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/307ce93317ad4a1baa2bb54de36fbb8c : SUCCESS in 17m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/664ab32864034180b6b5432b5a698eb2 : SUCCESS in 17m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6a2e30404e4e41d5919366bd5c09de5e : SUCCESS in 27m 52s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f51d796f395a4bf09020e4f8377efaf4 : SUCCESS in 31m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e2da4690ad74aca9ed3044111735b8c : SUCCESS in 28m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f66bd8a4c1d34486bed1111305aa6c56 : SUCCESS in 10m 43s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e5121e672537466aa606f0863cae91da : SUCCESS in 13m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92afec1515db41f3a6072e0a7e26e5e2 : SUCCESS in 11m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/4b317d99c160451d88382c23ea8aa1a0 : SUCCESS in 10m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/64ddfc7cbbb2437ca9686977b14f43e9 : SUCCESS in 20m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b356445bc0cd4826bd629a7776b841a1 : SUCCESS in 15m 39s","accounts_in_message":[],"_revision_number":4},{"id":"889d224d906ec908ba815196fb08b35d3340edd0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-19 21:46:19.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"bc87fe34892279fb6717f935867c9f852383f90d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 23:56:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12875dbdc7c34143b7113146a03a578d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c43260e4254f42e28197180731e558b4 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3438cff6f06c46f795cbde8af7586843 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/443c4f68561a458795c1e55bc4b9086f : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2b61dd3a1c4a52b3f3611c3d456fb0 : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0100fa2df57747068d05e5bd83a8494c : SUCCESS in 3m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae7723fc78234aed9ad3ebf688be31be : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2cc5015f4c63495c9fc07c54a9662409 : SUCCESS in 11m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/413a6fa411574cda9fdb0be716fd2292 : SUCCESS in 59m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/879466b3cf9c42c791f2cc94c87b1321 : SUCCESS in 57m 46s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/317c6ea9d8fc47e686ab7efe9fe2d4cd : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/205c822fafcf4c34abd2bc415ecaf9de : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f2fdb4c8d07247189588f894a96b8827 : SUCCESS in 18m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e457d3b3115a468c8b469cedf777db21 : SUCCESS in 18m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97ce082910bb4d0eb29ccd43cd10e71f : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3913b3c037264238a4d91ff4f809dea7 : SUCCESS in 1h 58m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f3941680b25494f92dde148386a0187 : SUCCESS in 1h 36m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c4d573642db64f6f9ed119c62803993f : SUCCESS in 21m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1341131dd3ed4fe390f9570761749358 : SUCCESS in 15m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/33618b55a40a4012bc9e6bb2a91b3575 : SUCCESS in 44m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e967432d78d44bd9a7bf61a76d29c9a5 : SUCCESS in 21m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4ab1c42adb34271ac339abb2e60a609 : SUCCESS in 25m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c6e65b1dc66247d8886082ed9779f3e3 : SUCCESS in 14m 17s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8d8baadbb86746e7941fe899b4113f52 : SUCCESS in 18m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9eb2f39d04de42c78eba93c2d4eea175 : SUCCESS in 11m 29s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/cf1aa868754c441583c69c3eb2128b1c : SUCCESS in 11m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7f2eefa591144a6f890178f07e0c2ee6 : SUCCESS in 24m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4775ed25cc6e465f8aa1e07385739b94 : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":4},{"id":"76a6c2c08bd1e14da68404417d07be44ccd9afec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-20 08:39:55.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"a902ca2f8d013e3555dd9dd0f5cadba04616e3be","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-12-20 09:17:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"72bfb72b80872d3fe1c7600033816ab39a80b975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 09:18:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2bf1ef86b125231919dc3336299098eb4346296d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-20 09:25:28.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9685f10596ce77eeff228df07dbf96080562b6cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 11:22:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b417043eb1e43d3b1916bb3f4a1d755\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bdcf71b681a4b3695071fb1a8d961a5 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52f29ed288ed4e4aaf2a0be4104bc7c6 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fce631bcf0d4b23a0fdc4ea80d6dec4 : SUCCESS in 8m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0a9625117f44013974eddb78f5ff03d : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c6e811aed624909bf6934146e3d62aa : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37437bfc33a44367a1b5d6003ee76400 : SUCCESS in 11m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f5d5917a684c4d488ca227029e0ddf92 : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa423a8866ed4760be69ddcd2b71df56 : SUCCESS in 57m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f28accd9592549aa9809a96c20f196ec : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b41f519c45ac457391556262304967cc : SUCCESS in 20m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88ed81c4eb974bfda5996d599d003584 : SUCCESS in 13m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3b8fcd945cc14fe8ac0843d0c87d744b : SUCCESS in 1h 59m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99b0706e9819496a8869bdecddecceb8 : SUCCESS in 1h 02m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3876f636028b458081c4dee4f1cedd2d : SUCCESS in 20m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77f9b192fe19452594ca7af9a748cd59 : SUCCESS in 10m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/401d56b774d543a9b1f2450c5f96fe64 : SUCCESS in 32m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3852e9945054c84a833443a80be058e : SUCCESS in 23m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a0ccbf19d13b42358042dd98b63d2249 : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/768cdf373d504df2826556bd781acf06 : SUCCESS in 9m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8223759d70a548eca84f5ca381f2dac1 : SUCCESS in 24m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62bb78f1f76b453c997d969d91853be6 : SUCCESS in 16m 37s","accounts_in_message":[],"_revision_number":4},{"id":"ca668ebb3888bfe6bc15610a009f38fb423c01ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 11:22:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"556ea9b7a4a883c717c2d7a8fbc128e11404a25e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 11:25:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee702f8972e44c49cf904422af82b4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6dcaa904f36408d815d69dfc4b9cfd9 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d3c0ff56a5e26561f689570a554c3b71970e96de","revisions":{"51dae95b006796cec595ccd8073a1431f37d8462":{"kind":"REWORK","_number":1,"created":"2024-12-19 08:31:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/938038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/938038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/938038/1"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 08:31:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 08:31:29.000000000","tz":60},"subject":"CI: Switch to pytest-ansible","message":"CI: Switch to pytest-ansible\n\nThe functionality provided by pytest-molecule has been merged into the\npytest-ansible plugin.\n\nChange-Id: I956ed42a4a862a14d1ac9e40c40e0cdf62e36464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51dae95b006796cec595ccd8073a1431f37d8462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51dae95b006796cec595ccd8073a1431f37d8462"}]},"branch":"refs/heads/master"},"c3119b81770df1eb369172848c14c654d751a4e8":{"kind":"REWORK","_number":2,"created":"2024-12-19 11:28:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/938038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/938038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/938038/2"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 11:27:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 11:27:44.000000000","tz":60},"subject":"CI: Switch to pytest-ansible","message":"CI: Switch to pytest-ansible\n\nThe functionality provided by pytest-molecule has been merged into the\npytest-ansible plugin.\n\nChange-Id: I956ed42a4a862a14d1ac9e40c40e0cdf62e36464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3119b81770df1eb369172848c14c654d751a4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3119b81770df1eb369172848c14c654d751a4e8"}]},"branch":"refs/heads/master"},"1fdd5c671b4ed3bd88d0a4392aa20928768ee4d1":{"kind":"REWORK","_number":3,"created":"2024-12-19 12:01:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/938038/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/938038/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/938038/3"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 12:01:01.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 12:01:01.000000000","tz":60},"subject":"CI: Switch to pytest-ansible","message":"CI: Switch to pytest-ansible\n\nThe functionality provided by pytest-molecule been merged into the\npytest-ansible plugin. The repository has been archived [1].\n\nSwitch to pytest-ansible for the molecule job. We need to disable\npytest-ansible when running testinfra because they define the same\n--ansible-inventory option [2].\n\nMake the kayobe-tox-molecule job voting again.\n\n[1] https://github.com/ansible-community/pytest-molecule\n[2] https://github.com/pytest-dev/pytest-testinfra/issues/58\n\nChange-Id: I956ed42a4a862a14d1ac9e40c40e0cdf62e36464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1fdd5c671b4ed3bd88d0a4392aa20928768ee4d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1fdd5c671b4ed3bd88d0a4392aa20928768ee4d1"}]},"branch":"refs/heads/master"},"d3c0ff56a5e26561f689570a554c3b71970e96de":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-12-19 16:39:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/938038/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/938038/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/938038/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/938038/4"}}},"commit":{"parents":[{"commit":"50ac65c58acdcd34025247c0ab29ff065cd6b37d","subject":"Add nova-metadata group to kolla-ansible inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50ac65c58acdcd34025247c0ab29ff065cd6b37d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 16:39:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 16:39:08.000000000","tz":60},"subject":"CI: Switch to pytest-ansible","message":"CI: Switch to pytest-ansible\n\nThe functionality provided by pytest-molecule been merged into the\npytest-ansible plugin. The repository has been archived [1].\n\nSwitch to pytest-ansible for the molecule job. We need to disable\npytest-ansible when running testinfra because they define the same\n--ansible-inventory option [2].\n\nMake the kayobe-tox-molecule job voting again.\n\n[1] https://github.com/ansible-community/pytest-molecule\n[2] https://github.com/pytest-dev/pytest-testinfra/issues/58\n\nChange-Id: I956ed42a4a862a14d1ac9e40c40e0cdf62e36464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3c0ff56a5e26561f689570a554c3b71970e96de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3c0ff56a5e26561f689570a554c3b71970e96de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
