)]}'
{"id":"openstack%2Fkayobe~938052","triplet_id":"openstack%2Fkayobe~master~Ida95972c95fad7587da5e8d9e80bd4d663b4de19","project":"openstack/kayobe","branch":"master","topic":"nova-metadata-group","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-12-19 16:14:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida95972c95fad7587da5e8d9e80bd4d663b4de19","subject":"Add nova-metadata group to kolla-ansible inventory","status":"MERGED","created":"2024-12-19 12:28:15.000000000","updated":"2024-12-19 16:16:01.000000000","submitted":"2024-12-19 16:14:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938052-nova-metadata-group","meta_rev_id":"7c5838c5e72996b8cd36fb3c83013921bd411f95","_number":938052,"virtual_id_number":938052,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 16:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-12-19 12:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-19 14:50:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-12-19 12:33:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-19 12:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-19 12:33:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-12-19 14:10:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-19 14:50:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b3bdf616b348536d42a53a9fed5aea392cf7fed0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-19 12:28:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce3ce96ce85d40ad9d56f171c8e1a5e56deeb439","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-19 12:28:47.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"05ad3f7fbee0db7e8dce5ebf53adce7533c1589e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-19 12:33:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4403166eca8ad54df2ea5a445a247503f67a47ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 14:10:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/193c8783cb0049b7b393235235392f5f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51dd7d7096ac4864b3ef10313fbb8625 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66ef1afd08d4bf883e09add8ba79f6c : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a166e191bcdf42a18c56147e93b5f8bb : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98726d4e54b444b087ea23e825fd6096 : SUCCESS in 5m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/638edd34bd524f56a37f3039ac24ddd0 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/776106c176864ca2a804beabb1801d3d : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f953f93490e44f5d8393704c8eba7407 : FAILURE in 4m 39s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/481c526c2a4b4030a3588d33199fd771 : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/64652fa0ae8846469975abb4a4fd5040 : SUCCESS in 35m 39s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/998d14673ade497a8822354ca709678e : SUCCESS in 59m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/aab2c85478d74ffab6e5566b81b834a1 : SUCCESS in 1h 19m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fb4977141fbc47eba15abe7a05b47623 : SUCCESS in 13m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0110b65a4d624f3e8bf5f867f5cf5a2c : SUCCESS in 17m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dc40ba4b49f46528582fac853f88ef4 : SUCCESS in 14m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a20dba4f88104d9ba5b7f72b70d06bb5 : SUCCESS in 1h 11m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d045896bd2ed4479b140a668d920d51a : SUCCESS in 1h 36m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f769f1f82e9b412494b100c6a764b906 : SUCCESS in 19m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26136f04323047a1a209f5bb8a0015dd : SUCCESS in 9m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8e3330babf454a858216c1920379934e : SUCCESS in 36m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9bc9a51d25f445fdb4a5b76ff73f2108 : SUCCESS in 34m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/302dac495c294e039778212ad2025b29 : SUCCESS in 16m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c9190016311340ec81364ec4136925b7 : SUCCESS in 15m 11s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8b23535404f847e6ad2b3d392c8d45e0 : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b54905844baa4a9390d45e5eb3bd8471 : SUCCESS in 7m 32s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c42c249f2f93491e884e1483a28ba2ff : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b5483dee879413fa788fa043e8b395b : SUCCESS in 22m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34fff9a433714a7dab9ce1c91331a044 : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cee0b50710235e5746b170a609ebd931418f4df3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 14:11:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d8df082101e6e114edbfecd257e355b6a3cf05b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-19 14:50:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a813da063d6b96a415de96e5eca30eed8a488abc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 16:14:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63e201951f1644b7a133afa0e7f48b3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727c2f205abd4283b42c17cab051c88a : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0056f3e240f41128bcb318944bd766a : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817fcf6169a2426dbd559009c0f5f926 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fcd186f80e684d6a98954da246503ce6 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8eab10eb4fa34409988b76842cd04816 : SUCCESS in 12m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b88cb59fb2bf40c0952cc1ba8babb06f : SUCCESS in 36m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/514278b446b24105bb98d557282dd0ae : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/affe0f0591de4decb25fe3a306ed1953 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/866bcfea17a04275bb0b22eb6ac83832 : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9ad32da3f014e868d4e6ac536d4773c : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ff9068f4361f40a4932df6249f08b18b : SUCCESS in 1h 58m 25s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a43ac9428524e879d592470f9a141bc : SUCCESS in 1h 36m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/83ef5b4abf56473d954755e3fa9c0521 : SUCCESS in 20m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/754f2ef68c084033b435eaa3b39ccde0 : SUCCESS in 16m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ff73714193144925b3226a06f7545fc2 : SUCCESS in 33m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e9484c49c374f44abb6f10176abcdda : SUCCESS in 28m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f4f42fe2cb2f43998adf844156d180aa : SUCCESS in 17m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0937de5efa8b40a6abc3683230c1dff2 : SUCCESS in 11m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1f98680bdfcb467aa9f6a3f895e6780c : SUCCESS in 19m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d22241880294037ac1936dfab921a55 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ec9c134020bf73c5b5596df86d8f139e299cfdbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 16:14:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c5838c5e72996b8cd36fb3c83013921bd411f95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 16:16:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d2e05d17cff462cbd7b3d0771a9a7d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906ddc90b23449bb9342aa563b054d7b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50ac65c58acdcd34025247c0ab29ff065cd6b37d","revisions":{"50ac65c58acdcd34025247c0ab29ff065cd6b37d":{"kind":"REWORK","_number":1,"created":"2024-12-19 12:28:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/938052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/938052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/938052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/938052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/938052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/938052/1"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 12:26:31.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-19 12:26:31.000000000","tz":60},"subject":"Add nova-metadata group to kolla-ansible inventory","message":"Add nova-metadata group to kolla-ansible inventory\n\nThe nova-metadata group was just introduced in kolla-ansible [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/936425\n\nChange-Id: Ida95972c95fad7587da5e8d9e80bd4d663b4de19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50ac65c58acdcd34025247c0ab29ff065cd6b37d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50ac65c58acdcd34025247c0ab29ff065cd6b37d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
