)]}'
{"id":"openstack%2Fkayobe~938549","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Id3d68b55e18dfe9c44805b6e2602e00eadecc76a","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-01-10 19:10:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3d68b55e18dfe9c44805b6e2602e00eadecc76a","subject":"Fix overcloud deprovision hosts list in prompt","status":"MERGED","created":"2025-01-07 11:18:43.000000000","updated":"2025-01-10 19:11:11.000000000","submitted":"2025-01-10 19:10:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938549","cherry_pick_of_change":937676,"cherry_pick_of_patch_set":2,"meta_rev_id":"274025157446b3eaac540060dd3d809bf288817d","_number":938549,"virtual_id_number":938549,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 19:10:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-01-10 17:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-01-10 09:09:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-01-10 17:13:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 13:15:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 16:32:09.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-01-10 08:10:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-10 17:13:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"1601dc0bcecaa4a090d7ceb1dd419d9857b430fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-07 11:18:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"625ecba5daaee31badec54a8ed96bc16002c457b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 13:15:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51ed43415c4245e7b532bed2caa1f7ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70316eabbdb54389acde1b9c04efd50d : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46eec4119ec4ef89997c0d32ca82cd1 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46e740aabdc949d4b0068b6ce3173f35 : SUCCESS in 2m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/861d923645ec4a39a9ae3d34f5b00700 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87fa1847bfba48bc8e637ac11802af64 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0819049d3484de8aa9a1fabf2210d51 : SUCCESS in 8m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e0cc03df90234ee68b94a4e0511ac23b : SUCCESS in 59m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e4bf022a95f64d368ed949fa837be269 : SUCCESS in 56m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d7406f2622f4164b3717ad947476b8f : SUCCESS in 44m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/39321e84b4ed4eb382e901132b33d8eb : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0c73f00753f949e7a969f56db89a9ef0 : SUCCESS in 18m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c22e23c814ab48bc826d9d57d7b006ed : SUCCESS in 9m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b3027ba7cb14683a7a15ee8f1cb337b : SUCCESS in 12m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/67fb95828b904b46971170956d62a779 : SUCCESS in 1h 55m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60299dcb58024182a96c73e19f00bff4 : SUCCESS in 58m 55s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5f60ff76214a4142ac6cd7fffd57f89c : RETRY_LIMIT in 5m 14s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46d3e4df62c64e7589b8912ad5e2add6 : RETRY_LIMIT in 3m 39s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1b09ec5bba7f4154a2f6da712abd7606 : SUCCESS in 17m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f640c0d333a94f0985bdf6c651fb9d3e : SUCCESS in 12m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/67bd197064b54bdcb1011547ad6ee037 : SUCCESS in 41m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/80122248b1004ebab086caaefb5c8a34 : SUCCESS in 33m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa4961f67166444e8221778cdd257e10 : SUCCESS in 23m 37s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b378ca0479334ec9a06406f984ac8814 : RETRY_LIMIT in 5m 59s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24d53c2841b84175bd6974b6f8183b5e : RETRY_LIMIT in 4m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15f2e935adb646738ae88592b98bc9ab : SUCCESS in 15m 38s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0fc85094e7b04b7a8e1ec59ab9aece10 : SUCCESS in 16m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c67faa2454ae4e51af84ac16f2b872d2 : SUCCESS in 8m 43s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/a9da801f90084be5969d2eda4f14fa2f : SUCCESS in 10m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c7b6ac1f79864440bf73e3ed4e82f7dc : SUCCESS in 20m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9f6d4711be2428990e040d88e8fbd29 : SUCCESS in 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bc5dd3e35b44ed83fb2f6913975ba4 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9968466ce7b147449a4422569e516029 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/037220ac71ae4ef59c23a2582e8cd2eb : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/610e1e115d404677b7a15592adea5d02 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"035b88eeb9a692b8ec1f142f672197f704098742","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 16:32:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"003543fa5d51dcfbd3eca3d23f3459714b09e4ce","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-10 08:10:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6dacb786df4281a4816cefc57b01dca2ca03d455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-10 09:09:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_35263\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"_revision_number":2},{"id":"dd375c71c3dc8e929215e3f9179ed4447ac64c45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 11:18:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35db1055ccbb4bd7a340d35da6c6584f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f33475a322444fcaa1c2202cd0fe13f2 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3755caf66af4a9080eeb1977c0e86fe : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe7dabfd04b74721b6277d7d5da615a3 : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8878ff847fe0430386f64d055212b480 : SUCCESS in 3m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f48dff8b495c4a8cbaeb7e781328478e : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db59fd7ac7e9412695f0871aff5c680a : SUCCESS in 10m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4f5dffa614514beaaf12389b244151fc : SUCCESS in 59m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/445f1fe805744ffdb5c16d33016949ab : SUCCESS in 1h 01m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9601e57c796244a787be18996b0aaf73 : SUCCESS in 52m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c52280656e524283993e10a34e373a28 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ea1d2e648ba5449b9b76adfa49003474 : SUCCESS in 18m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f1a78e4d5faf4ad1b0d0edc70c82c933 : SUCCESS in 10m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14307c4e1ea4446a81c6fd2857f3189d : SUCCESS in 13m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/259ff7dedd2f4d5d9c9b0019d615875e : SUCCESS in 1h 57m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3b56c4154f64cf2ac52a27eb4723bec : SUCCESS in 1h 59m 49s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fe21bddf4af7442daf6891e3edd561d0 : FAILURE in 1h 05m 48s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92f167f2b89140c3a1f34745e6cb4f89 : SUCCESS in 2h 01m 16s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0e0f0b907b4e47e4bdc884e9adbaa666 : SUCCESS in 17m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68b519f391bf4030b1b51f01ca3b3d23 : SUCCESS in 14m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/23548897f7054937a17efeea1f28c412 : SUCCESS in 40m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f1ccaa95b4d04d3d9904d0380982b1df : SUCCESS in 32m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c37e5406bae4b17acf9920ca9bfff0f : SUCCESS in 24m 13s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/67b20a2e84b64813906e71df138a5349 : SUCCESS in 34m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6daaf9f78679433dabc52c82c77b9ec8 : SUCCESS in 25m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5d358c4c60db469ab59a91388d8233cb : SUCCESS in 14m 53s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/24223c88e82b4d95b2172e0c268ff5a6 : SUCCESS in 15m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43e850e13b6d4fad8039c494e0531393 : SUCCESS in 10m 52s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/384a1146cde04c7299ec48fe230c2fd5 : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5c405a2d090549b2a221bd36367bdb6c : SUCCESS in 20m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c155cad4df724ef7913ed8236dbbb7d3 : SUCCESS in 16m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e1050351ed43f38aff618cddeca2b4 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c674f686f654a9e8d90b7d2621d0249 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/977ee83efc7c44aea24ed87dac63685b : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a20016ac69e84d6c8e3e21259aec96f2 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"af0ebce78ad3a82bdf150b2e7f6b3786640d14b2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-10 17:13:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a1af1b0a8b0e49fd79f41b4e5889d03cba5fdf2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:13:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a470c1a6b5e85fa3f4f62d07dd2234be1b97f349","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 19:10:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff422234f4846089e5321367418e56a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b4af5aa9af44488109b252c3cdf6a3 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d0704a9f8084bc28158cec4216aa31f : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10cc3c212afd48caa81d26341979ee9b : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/02ab7f2657994730814de17e00ed2475 : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/053ea0e3e9d74ffbac0d47f7725a41b2 : SUCCESS in 9m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9e1282c4a6ab47cabe634fa63a346dff : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c529aaa74574e7bbcade620848e30c7 : SUCCESS in 51m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f62f5c89485743989e40819b63145bf9 : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6931bccf492b4651b6bd3fe29e99b444 : SUCCESS in 12m 43s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02bd14c82c364f6db6f583d77c2ef56e : SUCCESS in 13m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/def876615e1749e9baf1b5b1bef1fcd9 : SUCCESS in 1h 51m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/869cf7073188438d8ed9e1c6809d9187 : SUCCESS in 56m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/47eb6683ca0f4a21b6940d8e843d0cba : SUCCESS in 18m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1f251472e3b45b09914d6a27ac527fd : SUCCESS in 13m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d5b170bb41454c10a7abbd39b3deb13d : SUCCESS in 19m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23ed3e7219b74ad4a7c7862423a1a46e : SUCCESS in 23m 13s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1d766f70d42345b69efe8e7e17d944e1 : SUCCESS in 20m 57s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efb728e3ffe0485b8f687b141ca19bd2 : SUCCESS in 24m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ca2a0f62f3b4e4993bc36a1fcdd9524 : SUCCESS in 16m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c4ae82fe2644d4c8d26b875705fe7e0 : SUCCESS in 6m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/398e3de698b4436a882d2bfe256a76ae : SUCCESS in 20m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4012991ef844217951b72e2647ca7e9 : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":2},{"id":"5378846fc1a0c43ca4cba936a61060a1901ea206","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 19:10:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"274025157446b3eaac540060dd3d809bf288817d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 19:11:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70344e59dfd4433b8a233e03ee950232\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d52f63eff24935aa19bd04ec37d84e : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97d314b535784f08ac20328ce29fb556 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"89097334567e4ceb7a0f64fac502d539a69e4b71","revisions":{"206b34cc7773c0e4bf13c960abf03cf2200329ff":{"kind":"REWORK","_number":1,"created":"2025-01-07 11:18:43.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/49/938549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/938549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/938549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/938549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/938549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/938549/1"}}},"commit":{"parents":[{"commit":"d20514b7e24505e891da7ec6ea731c43fb96d7d7","subject":"CI: test database backup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d20514b7e24505e891da7ec6ea731c43fb96d7d7"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-12-13 11:18:12.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-07 11:18:43.000000000","tz":0},"subject":"Fix overcloud deprovision hosts list in prompt","message":"Fix overcloud deprovision hosts list in prompt\n\nHosts that weren\u0027t part of the overcloud group would be included in the\nlist when this is run without a limit.\n\nCloses-Bug: #2091621\nChange-Id: Id3d68b55e18dfe9c44805b6e2602e00eadecc76a\n(cherry picked from commit f27755aeb2f1bd5a7bae1deabc0df37a6578dc24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/206b34cc7773c0e4bf13c960abf03cf2200329ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/206b34cc7773c0e4bf13c960abf03cf2200329ff"}]},"branch":"refs/heads/stable/2024.1"},"89097334567e4ceb7a0f64fac502d539a69e4b71":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-10 09:09:22.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/938549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/938549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/938549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/938549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/938549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/938549/2"}}},"commit":{"parents":[{"commit":"9da82a952ddfb585aeefc5d4d79933738591fe4b","subject":"Fix slurp upgrades due to 2023.1 branch name change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9da82a952ddfb585aeefc5d4d79933738591fe4b"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-12-13 11:18:12.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-10 09:09:22.000000000","tz":0},"subject":"Fix overcloud deprovision hosts list in prompt","message":"Fix overcloud deprovision hosts list in prompt\n\nHosts that weren\u0027t part of the overcloud group would be included in the\nlist when this is run without a limit.\n\nCloses-Bug: #2091621\nChange-Id: Id3d68b55e18dfe9c44805b6e2602e00eadecc76a\n(cherry picked from commit f27755aeb2f1bd5a7bae1deabc0df37a6578dc24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89097334567e4ceb7a0f64fac502d539a69e4b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89097334567e4ceb7a0f64fac502d539a69e4b71"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
